)]}'
{"id":"openstack%2Fneutron~853115","triplet_id":"openstack%2Fneutron~master~I26e3319232e05c0dccbc12588eed5c13a5e7a467","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-08-30 07:15:20.000000000","reason":"\u003cGERRIT_ACCOUNT_35058\u003e replied on the change","reason_account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-08-30 07:27:11.000000000","reason":"Change was abandoned"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-08-30 07:27:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I26e3319232e05c0dccbc12588eed5c13a5e7a467","subject":"Cleanup py27 support","status":"ABANDONED","created":"2022-08-15 04:38:58.000000000","updated":"2022-08-30 07:27:11.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2033d94ec4b8a5db4d7d104b9897ef132173ccf","_number":853115,"virtual_id_number":853115,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":-2,"date":"2022-08-30 07:10:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 04:53:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-15 05:17:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-15 06:19:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 07:41:32.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-08-15 11:57:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"16515d9260824a5c5c98b8a9f99cc896012dee3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-15 04:38:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5585ed547ee9ab275d6bafb7d14c2f97ee0fc781","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-15 04:53:20.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/853115/1 : FAILURE in 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"df947d6e06deddf64e7dd4e7052ef707800df7df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 05:17:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7858bce9744123b5d67d0a95db4a5d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63d19a6131e548318b65bea1d7af8b66 : SUCCESS in 34m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d886f31545c405b9f117397f20168bd : SUCCESS in 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70383be9695b0be8459257f09922f9627276e701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 06:19:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013f4729386e404091a70ac6fb5a3105\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fca22e482e4045e781f5c60762f9b3e3 : SUCCESS in 5m 53s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c741d8f630ae4956acc3413ac27017aa : SUCCESS in 1h 23m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a04f923d4d494147983e58dc9ac2b2c9 : SUCCESS in 1h 21m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9faec6c040034e499358b7da04d875b9 : SUCCESS in 1h 08m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bee60219df05477895bbead14939cefc : SUCCESS in 1h 15m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/298d96117cee499c960c0a2b03b078ee : SUCCESS in 48m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51521385b9a1463485a51ea3cf278b74 : SUCCESS in 30m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af20f07f70e44faaf69c2edcdc9a5de : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71dab1c3924a4557ba78a2d9fbf5c89f : SUCCESS in 33m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2109af060dcb4dba9069847fbee1a906 : SUCCESS in 24m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ea1cdbf20c14372a9d1b49cfc1366cb : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5945d1d9fae54242890b53655fcb429e : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38ea3b4c443b430297810a6ab6e40804 : SUCCESS in 1h 04m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce65091485ce4585bbe72129059d1224 : SUCCESS in 1h 38m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a3c87a97d5a348eb98c1b7eb190b3d16 : SUCCESS in 1h 00m 28s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ab7e96026c14f559a535ff9dd1bcc10 : SUCCESS in 51m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/473b2974ec3649f48c5c62ae26e5ad47 : SUCCESS in 56m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7037adc6069e4b64875ab46465d33f63 : SUCCESS in 1h 25m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/27f942bd01eb4678bcc8c52d702f30da : SUCCESS in 1h 12m 00s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ccb21c3e59e4154a9496b30a1a091c8 : SUCCESS in 1h 13m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3e176bcabc5b417fb41fc46194d267f1 : SUCCESS in 1h 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31464c51b330806357b171df23d1fccd4e42e631","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-15 07:41:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53258109524a289833bf8956a2632a324a64b44c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-15 11:57:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36e982b365e6dfae2c6dc64afd7d8b2622784b95","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-16 00:36:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"a7cba7d83ade8710c933d6cd12eaf99bae7f66c8","tag":"autogenerated:gerrit:restore","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-25 00:52:11.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"bbe3ae817e6364c8dc500484abc78f1141c04b6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 01:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d3bdaea9eec496fa8dc35dee255f205\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aee7638374d64881a00dfd0b6025856d : SUCCESS in 40m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4783ff4245024afe8220ee7edfcd0f41 : SUCCESS in 29m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26f420cbf79b15017e6fc2a34eb253203a5446a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 03:00:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9cfe420864f4d078683eeb0c877ff4e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e8257b49ada42a4b3a361e7a882af48 : SUCCESS in 2m 32s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/de2d823e846d42e38c95b5cf42272958 : SUCCESS in 1h 13m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0f60dce9452547d4b33fb70049864d4c : SUCCESS in 2h 02m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7756b8d379b345aab005a525f47fbf97 : SUCCESS in 1h 12m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6fbf76a7e503486bbf0b67ac9d58d63c : SUCCESS in 1h 42m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c4900571e10346c5b7d665ee51926518 : SUCCESS in 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d402028b06084b57bbf2c2f0979a8a78 : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a241842a05a41769f490a09a708d6be : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/957da253feb549fbbf3d3b56adcef773 : SUCCESS in 25m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/316e667a169c40aaad3ef506646daf9e : SUCCESS in 24m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ff91251d74c4a0bb5133aae50cf1db7 : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7440e59da5454ed4987ddd846f8675fc : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50840801e49545ef9a2a2907c7aab866 : SUCCESS in 1h 03m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f7463f595094fc190a96af7fa620853 : SUCCESS in 2h 06m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c5a43f75c481450f8d17451939bcd3b1 : SUCCESS in 1h 06m 34s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b33c3894b37415eb3b89805aba530b3 : SUCCESS in 1h 01m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6252d27b2acc4ad3b2ca1dfe66f0ee00 : SUCCESS in 59m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/482b444467424c13abecd033fae32710 : SUCCESS in 1h 17m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24d1eaea3a324413852f2f4926be2cd5 : SUCCESS in 1h 17m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/516d027112244a10b1525d34079a9e10 : SUCCESS in 1h 08m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7caf1858f29c43829bb62151f65e1f59 : SUCCESS in 1h 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"278cf2e15cfaf0baef87f43db8eb73b0d7af8ec4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-25 03:56:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/853115/1 : SUCCESS in 2h 04m 15s","accounts_in_message":[],"_revision_number":1},{"id":"14e843fdccd98b0333222959ad5fda97843c4e69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 06:16:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0094bb8e418d78047702dd238329878e92527b3a","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 06:16:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a834d9e012827697da187be053c01c6fda0f4bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 06:59:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79dfcca5f7c4546980cd8e0ca928050\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06dfbb1b27a4470e8c2026124c2a0f82 : SUCCESS in 36m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb2e80ce50b1477fbe4ed92618bf47e2 : SUCCESS in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc1fc2113142f9ac01e0cbcc9304deec97021695","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-30 07:10:04.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"090e5cb4925589164801a7215180215645bde897","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 07:15:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2033d94ec4b8a5db4d7d104b9897ef132173ccf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 07:27:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac1561248f56b063e2d3d8de2d9e5e8026545354","revisions":{"0a3db5b21418432d78167fc32aee1cb2b112438a":{"kind":"REWORK","_number":1,"created":"2022-08-15 04:38:58.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/15/853115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/853115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/853115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/853115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/853115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/853115/1"}}},"commit":{"parents":[{"commit":"129429c93f56ad85b66ad7f19c39f58c3c899bd3","subject":"Merge \"[ovn] Specify port type if it\u0027s a router port when updating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/129429c93f56ad85b66ad7f19c39f58c3c899bd3"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 04:37:43.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 04:37:48.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n\nChange-Id: I26e3319232e05c0dccbc12588eed5c13a5e7a467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a3db5b21418432d78167fc32aee1cb2b112438a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a3db5b21418432d78167fc32aee1cb2b112438a"}]},"branch":"refs/heads/master"},"ac1561248f56b063e2d3d8de2d9e5e8026545354":{"kind":"REWORK","_number":2,"created":"2022-08-30 06:16:15.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/15/853115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/853115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/853115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/853115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/853115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/853115/2"}}},"commit":{"parents":[{"commit":"129429c93f56ad85b66ad7f19c39f58c3c899bd3","subject":"Merge \"[ovn] Specify port type if it\u0027s a router port when updating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/129429c93f56ad85b66ad7f19c39f58c3c899bd3"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 04:37:43.000000000","tz":480},"committer":{"name":"Ke Niu","email":"niuke@inspur.com","date":"2022-08-30 06:16:15.000000000","tz":0},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n\nChange-Id: I26e3319232e05c0dccbc12588eed5c13a5e7a467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac1561248f56b063e2d3d8de2d9e5e8026545354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac1561248f56b063e2d3d8de2d9e5e8026545354"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
