)]}'
{"id":"openstack%2Fneutron~853433","triplet_id":"openstack%2Fneutron~stable%2Frocky~I8ab7995ebd28ab08f12e324c140a45bf53157ce4","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1986682-stable/rocky","attention_set":{},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-10-16 07:13:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8ab7995ebd28ab08f12e324c140a45bf53157ce4","subject":"[DNM] Test stable branches CI","status":"ABANDONED","created":"2022-08-17 12:05:22.000000000","updated":"2022-10-16 07:13:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":853473,"cherry_pick_of_patch_set":1,"meta_rev_id":"d092498178a5e5ae4e3588a27c115b40c1adde56","_number":853433,"virtual_id_number":853433,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-17 15:09:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 15:09:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-16 07:13:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"04096c4ce912b2d3fd78d617aa814f6db721adc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-08-17 12:05:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"3296d0d589c2e04306556fc40d36e1a5f173af5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 15:09:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ad2c2436c394492bd550fa0180b22bb\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/51b3fad0577742dcae5790f634b08cb3 : FAILURE in 30m 05s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/72d35d7157a9411194edffe1f979d6f1 : FAILURE in 21m 25s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/adf5642253854cd8bc68a760e4dd64b4 : FAILURE in 1h 02m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/2b8ca0572ade4ce390963dce82641c5f : FAILURE in 1h 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b09265c27ce64f4ea0a3f1efc5bbfa05 : SUCCESS in 10m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d364dc25c614e2f86d7fbb1477251a3 : SUCCESS in 27m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3cdbd1b8fdda4227ae8414ff3d5bc877 : SUCCESS in 27m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5272ab46d8e84fe999569b8278f483d5 : SUCCESS in 7m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e30466c6480f4d23be4f2fd427ae2d88 : SUCCESS in 1h 11m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c96000be9f554bf5832ca180a63b9ad5 : FAILURE in 2h 54m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71201a11bded4b61a1ae158e3d179e95 : SUCCESS in 1h 06m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35986b907262462cad2caf7b93fcc67e : FAILURE in 2h 05m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/96448d45e75648c794c4a93a3dd978de : FAILURE in 2h 55m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/083208ec59cf4129b841c1108cdb382e : FAILURE in 2h 55m 27s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87ee26f0c6ba4ae281cbf2fb989a7a67 : FAILURE in 2h 54m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7bfd6a86248435a9702b2c3e3a3314c : SUCCESS in 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d092498178a5e5ae4e3588a27c115b40c1adde56","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-16 07:13:38.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b5ab6f9af35bfa5d4be522ee89d0e572ad3889c","revisions":{"1b5ab6f9af35bfa5d4be522ee89d0e572ad3889c":{"kind":"REWORK","_number":1,"created":"2022-08-17 12:05:22.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/33/853433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/853433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/853433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/853433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/853433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/853433/1"}}},"commit":{"parents":[{"commit":"9c009ec60eaf1bdebb2885158b6090ad73f2d029","subject":"Wait longer before deleting DPDK vhu trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c009ec60eaf1bdebb2885158b6090ad73f2d029"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2022-08-17 12:00:38.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2022-08-17 12:05:22.000000000","tz":0},"subject":"[DNM] Test stable branches CI","message":"[DNM] Test stable branches CI\n\nPrevious backports show several failures in stein and older branches,\nbut a relevant fix should help for most failures:\nhttps://review.opendev.org/c/zuul/zuul/+/853073\n\nChange-Id: I8ab7995ebd28ab08f12e324c140a45bf53157ce4\nRelated-Bug: #1986682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b5ab6f9af35bfa5d4be522ee89d0e572ad3889c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b5ab6f9af35bfa5d4be522ee89d0e572ad3889c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
