)]}'
{"id":"openstack%2Fneutron~853633","triplet_id":"openstack%2Fneutron~stable%2Fstein~I984dad79505b85822ffdd9a088a04b6ed5a14f27","project":"openstack/neutron","branch":"stable/stein","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2022-08-23 00:38:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I984dad79505b85822ffdd9a088a04b6ed5a14f27","subject":"Mellanox_eth.img url expires, remove the mellanox_eth.img node","status":"MERGED","created":"2022-08-18 10:08:19.000000000","updated":"2022-08-23 00:39:49.000000000","submitted":"2022-08-23 00:38:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853633","cherry_pick_of_change":853631,"cherry_pick_of_patch_set":1,"meta_rev_id":"cf85ed274a4b4323537e81fce539c9d50d05b0f9","_number":853633,"virtual_id_number":853633,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 00:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-08-22 22:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-08-19 14:41:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-08-22 22:49:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 10:26:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 14:41:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-08-22 22:49:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"fb04384529421b52741a58b18e1088f9d0f08c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-08-18 10:08:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"bc31a5b375b2083823e11ff6973a8d862369597f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 10:26:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de9be8c1206b485fa17d839d1e3c429e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093ec695fcaf499ca904aa591007bed0 : SUCCESS in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5854b4767f454f5b93bf0fc9108a9112 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"92e66eb61966164b983c4792cc0d57c9305e16e3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-19 14:41:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4223084175086275c45331da6feb31c4f8307742","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-08-22 22:49:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9537f8ca7176dc40532e3aadf8c94d9b21504dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 22:49:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"439d5f621734307c12d7e3b20447824b05759e69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:38:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/785d90c0e5d24c79bf7f81a04b633c8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54eea71334a0466f8b4ca812b5fde859 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5cf77943814485b0b5f53f6eb6af44 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"59d2fa54a9f8418de63b856bc75a3eeb953b7d6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:38:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf85ed274a4b4323537e81fce539c9d50d05b0f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:39:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd3dc6c39444db685fa0b7c611eebc6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19e40afb36146b7bfad6c9e2630d640 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cea46ce515ba75aaff7c26a9b2e0777625839a7","revisions":{"2cea46ce515ba75aaff7c26a9b2e0777625839a7":{"kind":"REWORK","_number":1,"created":"2022-08-18 10:08:19.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/33/853633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/853633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/853633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/853633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/853633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/853633/1"}}},"commit":{"parents":[{"commit":"7d6bde20ce3252e090400909ff21d286f1c21459","subject":"Wait longer before deleting DPDK vhu trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d6bde20ce3252e090400909ff21d286f1c21459"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-06-29 05:26:52.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-08-18 10:08:19.000000000","tz":0},"subject":"Mellanox_eth.img url expires, remove the mellanox_eth.img node","message":"Mellanox_eth.img url expires, remove the mellanox_eth.img node\n\nMellanox_eth. Img download link failure,\ndelete the http://www.mellanox.com/repository/solutions/openstack/images/mellanox_eth.img node\n\nChange-Id: I984dad79505b85822ffdd9a088a04b6ed5a14f27\n(cherry picked from commit 1e5d43a2f324bcc8ebbe544de8a6cb6a08deb795)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cea46ce515ba75aaff7c26a9b2e0777625839a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cea46ce515ba75aaff7c26a9b2e0777625839a7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
