)]}'
{"id":"openstack%2Fneutron~853751","triplet_id":"openstack%2Fneutron~stable%2Ftrain~Id4ba8d32e6b2301f25c12f83a0708b62aa5b9d6d","project":"openstack/neutron","branch":"stable/train","topic":"bug/1903989","attention_set":{},"removed_from_attention_set":{"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2022-08-19 14:58:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id4ba8d32e6b2301f25c12f83a0708b62aa5b9d6d","subject":"Fix migration failed due TypeError of new_vlan_tag","status":"MERGED","created":"2022-08-19 05:18:23.000000000","updated":"2022-08-19 14:59:35.000000000","submitted":"2022-08-19 14:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853751-bug/1903989","cherry_pick_of_change":853750,"cherry_pick_of_patch_set":1,"meta_rev_id":"1008bfa625034b93d74a73e447e06c7037f657c9","_number":853751,"virtual_id_number":853751,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-19 14:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-08-19 13:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-08-19 09:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-08-19 13:01:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-19 05:18:23.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2022-08-19 07:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 09:33:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-08-19 13:01:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9ca59abf4b2198dd78cf7fc9dd5ed6e304ec1817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-19 05:18:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"96e4a7e3cb08befa82eee063a04fbd66651493c7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-19 05:22:18.000000000","message":"Topic bug/1903989-stable/victoria-stable/ussuri-stable/train removed","accounts_in_message":[],"_revision_number":1},{"id":"cd18c1438d350b4cc5fa310045418f72673a225f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-19 05:22:20.000000000","message":"Topic set to bug/1903989","accounts_in_message":[],"_revision_number":1},{"id":"d9ae6a81ad9701d2d9e3187bf9114653250b154f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 07:33:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/209ab1435c7941dabf616081285e3953\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/6a0c9407b6cd428f92a9d5c0a5d63ee5 : SUCCESS in 51m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/144c8245491a4a148d50376a761a9c5f : SUCCESS in 31m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00779d2555b54cc9bf0d498a99db6091 : SUCCESS in 12m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba999ad700af427cb69a44b47845be52 : SUCCESS in 25m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3fe47b163de46fa8dbf5ac869675c18 : SUCCESS in 22m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a26de2ab5e544439876f5fb4604735a1 : SUCCESS in 23m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196ce14f1b92473fa365940a5625951e : SUCCESS in 9m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/82b2d52e3fab4c9ab99551364a17ddb7 : SUCCESS in 43m 58s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/9238951659914c208a30773627cef858 : SUCCESS in 50m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/00d89fc0e1f34808a4d88d77c85fde39 : SUCCESS in 1h 12m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/71b23f540db04d8d8b8c7702d382f4fe : SUCCESS in 1h 25m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b1e1174668d14c939b08921555550abb : SUCCESS in 1h 23m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/25712e76a1574ecda41516fa6af8d44f : SUCCESS in 1h 09m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/064a1d8834eb4626816a0af3ff98e228 : SUCCESS in 1h 03m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f8c8b70dd41c4dd4910aaadaf5441b9d : SUCCESS in 1h 09m 06s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/23a0c32597e64b7eb1bd2d0a2b5fbef6 : SUCCESS in 1h 04m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aac73866a3ab40c3815550900a497358 : SUCCESS in 55m 11s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45598b32487c464d9eb681bb07396e54 : SUCCESS in 1h 31m 37s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7aab7480863a45ed86097cdfe2b28878 : SUCCESS in 2h 12m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d630dd71fe17493ea62a7de5ffcc0453 : SUCCESS in 1h 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a01121ec1700431a84583f360aaeab49 : SUCCESS in 59m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0abd9d0bf8da4ac7a621030a2625786c : SUCCESS in 1h 01m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ebb97d4853bc45ef9de7171321d1aa62 : SUCCESS in 1h 09m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e770ecb8baa41bf94a8c2aac757d4a1 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":1},{"id":"66855fcf761228dc3089659e173986f1f422f60c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-19 09:33:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edc873bd16151bb8ebd9c6267ce18af8dff607f8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-08-19 13:01:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bab26da0e26b129fc34e3970e839abd31289cd36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 13:02:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66a4d98716cf43cf01032867455a3d77bb3029a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 14:58:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db3fcb428db1480fa03831ceaa10babb\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ca3a6f0877da49ffbce4b8f630f6b83b : SUCCESS in 57m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4ffe3aaa1540d4a82abac4c09015a2 : SUCCESS in 12m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4ab16019124491db07d361efa424bae : SUCCESS in 22m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5438542cc72d4b0880f36af3ddb94995 : SUCCESS in 22m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77d33da233c14010a3e70a15e0d71c78 : SUCCESS in 24m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4228cde64f741ba9812f505459d4978 : SUCCESS in 9m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3c1ca5bd6451459490fa925322589373 : SUCCESS in 51m 34s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3c285e5791844138a9bde53f8b7456b4 : SUCCESS in 44m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/429d8e47aca44503bd9a31886e74144c : SUCCESS in 1h 23m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/88eeb001c2fa41f2894085d08f2cd532 : SUCCESS in 1h 33m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ab0f188ed9424b55ab713ba66341f95c : SUCCESS in 1h 15m 35s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd762de6d0d847bf8d9b8c1e033f56cb : SUCCESS in 1h 08m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/431f2584dde54fd2af256c66b2361c66 : SUCCESS in 1h 01m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/02eb8fb9696b44bea4c5f13f26b8c448 : SUCCESS in 1h 09m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/57507a2e19664efd9f802d11e5bca1ba : SUCCESS in 1h 02m 53s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f998080a24da4608bef5873dd25d15a5 : SUCCESS in 1h 16m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a6eb7d46967144a7b5c2f7b23c91f13d : SUCCESS in 1h 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46b8e4f4bdd640f997c080c49905d125 : SUCCESS in 1h 08m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/435040a15e744a47bced74c988170feb : SUCCESS in 1h 00m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fa430cc05b9147b9bb01b71175cc6a74 : SUCCESS in 1h 08m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34a09a0e58eb446a8417578abbe1b276 : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":1},{"id":"567cee791f52155fc6c70ddb38fffb2c43b422a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 14:58:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1008bfa625034b93d74a73e447e06c7037f657c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 14:59:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/920c04cf8a144a9e851b7e30ff9ac899\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6fcfc4ca4647c39cd07f4a8cb8570a : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"754c34148e8619675b379a2e4cb37f226b65b499","revisions":{"754c34148e8619675b379a2e4cb37f226b65b499":{"kind":"REWORK","_number":1,"created":"2022-08-19 05:18:23.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/51/853751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/853751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/853751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/853751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/853751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/853751/1"}}},"commit":{"parents":[{"commit":"530c1cbadab18ec9bf19171e0147cc2616140574","subject":"Clean up db residual record from dvr port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/530c1cbadab18ec9bf19171e0147cc2616140574"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-11-12 12:43:24.000000000","tz":60},"committer":{"name":"liuyulong","email":"i@liuyulong.me","date":"2022-08-19 05:18:23.000000000","tz":0},"subject":"Fix migration failed due TypeError of new_vlan_tag","message":"Fix migration failed due TypeError of new_vlan_tag\n\nThere could be a situation, when there are ports\nwith port_tags None created in OS infra.\n\nUnder normal circumstances, it would take some error\nto the log and go further.\nBut only because there is a \"new_vlan_tags must be string\"\nin that log, it is throwing out typeError and dropping\nthe execution of code and the migration is ending by code 500.\n\nAnd this is all done without a message from\nthe log, which is extremely bad.\n\nThis patch is fixing this behaviour, and flow\ngoes on.\n\nChange-Id: Id4ba8d32e6b2301f25c12f83a0708b62aa5b9d6d\nCloses-Bug: #1903989\n(cherry picked from commit 7cf52bb5325ac73b7afa3212cb84d29831e3724a)\n(cherry picked from commit 87c375bab79226a8df1e306f32fa29188d4d0ea5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/754c34148e8619675b379a2e4cb37f226b65b499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/754c34148e8619675b379a2e4cb37f226b65b499"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
