)]}'
{"id":"openstack%2Fneutron~853752","triplet_id":"openstack%2Fneutron~stable%2Fstein~Id4ba8d32e6b2301f25c12f83a0708b62aa5b9d6d","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1903989","attention_set":{},"removed_from_attention_set":{"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2022-09-22 03:24:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id4ba8d32e6b2301f25c12f83a0708b62aa5b9d6d","subject":"Fix migration failed due TypeError of new_vlan_tag","status":"ABANDONED","created":"2022-08-19 05:18:37.000000000","updated":"2022-09-22 03:24:07.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":853751,"cherry_pick_of_patch_set":1,"meta_rev_id":"357a53eec3b5dc22a61f312e02e3464af7f93a9e","_number":853752,"virtual_id_number":853752,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-02 14:25:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-19 05:18:37.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2022-08-19 08:46:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 15:13:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"74eff4fc073fdd0d500644b0aefa2a3591055c59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-19 05:18:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"572d6c0e3ff5b64be72aa9b15b40a30d483e0a47","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-19 05:22:46.000000000","message":"Topic bug/1903989-stable/victoria-stable/ussuri-stable/train-stable/stein removed","accounts_in_message":[],"_revision_number":1},{"id":"0b40b907469b64f94b57a1abfa471edc965b904e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-19 05:22:48.000000000","message":"Topic set to bug/1903989","accounts_in_message":[],"_revision_number":1},{"id":"42328a8293c140076158d757bedd2e1beaca4c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 08:46:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7525bd8a5974acca4c92e83229df0ce\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/d5898d659965437e955c67211c9e2cda : FAILURE in 32m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/5579fdc6cdfb43a48d411f08a31b2198 : SUCCESS in 1h 03m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c9e181f8de64acfa737b1204ecc6070 : SUCCESS in 31m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba3d6162736a4d75881c5fb4f2a6a7cc : SUCCESS in 12m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d3648cdde1148bdb0a2257b910607b3 : SUCCESS in 22m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f12c27ac2354055a3eac1cd7baa1ae6 : SUCCESS in 23m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d45b00ae6c542559ef09cf6d14a5a23 : SUCCESS in 23m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3173dedef4234b02a131ecca058c3dea : SUCCESS in 26m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/babdeef2afe24fbabe0816c69878ce64 : SUCCESS in 6m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d345e67abc384904aa778063c4329881 : SUCCESS in 1h 22m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f14722a0c9d3482a840d5649acbb9f6d : FAILURE in 2h 54m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29a66b8c39ef44c0820aec3ec07c4f98 : SUCCESS in 1h 12m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b0d090997c8c4e498358956bb21b001b : SUCCESS in 44m 38s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/294c3ddadbea496b9128c79890dd2c8e : FAILURE in 47m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/66d8dd946fee4a02b1339eed07bbc908 : SUCCESS in 1h 34m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/177e8fc434bc499a89fdd68df4345680 : SUCCESS in 1h 39m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/238da74454244fda8c5532fb7859deaa : SUCCESS in 1h 20m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8ce3afdd38b34e1bbfb22bd675206c3e : SUCCESS in 1h 19m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf8e7ecdd8e0477d87f53942e639e59f : FAILURE in 2h 55m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/633fb6c394c64948938cea0675c286bd : FAILURE in 2h 00m 55s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78afe423a54a414993b1a66b32d0f303 : SUCCESS in 1h 22m 27s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52c5592f45a249278b398412928bbd39 : SUCCESS in 1h 30m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/02ce5a3a2d05475eab1722b94ec3e361 : FAILURE in 2h 55m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8e6396ef6cae8acd3e23e750dbb31f68869addef","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-30 15:13:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b93a01c542655d8a20bee0c9e7d3c09c803a0d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:22:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9115e3d9bff4486bcc6c16fad539845\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/7f49e0f0ff634db1b4a2847c9c793192 : SUCCESS in 1h 01m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/4ff647e565f843d7b3bc32b614d8c587 : SUCCESS in 1h 25m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da152205c44040c3b26f7ee14347a42e : SUCCESS in 28m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d80c331bb22442748f865b042e18671d : SUCCESS in 11m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65cf5efe86e741048cd6892206a666bb : SUCCESS in 21m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2af0fabe2a245a7a57f720e05d6f371 : SUCCESS in 22m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e162c54f2b1741b6a33dd64c21642959 : SUCCESS in 19m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ae0749ecca0494a88ebd0c37fa1b4b3 : SUCCESS in 21m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c6d8cce4cd4d64a1477db1c0416c6f : SUCCESS in 4m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/328b58e8bf38451ca99a672367dbc61a : SUCCESS in 1h 26m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0259f3acb0564ab487adc85ea5cfa97c : SUCCESS in 1h 17m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cfcbc172ab3a44249ff7408487540424 : SUCCESS in 41m 35s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/080e3d0274c1418ca29efbcd1abb45dd : SUCCESS in 55m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/108a92f884684ab5913a6b80800a38d4 : SUCCESS in 1h 17m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b277c7f8492a41f2a6c4482036e75565 : FAILURE in 31m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/591e21421f7f4e6bb205f7f1310b1f84 : SUCCESS in 1h 17m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/026c6b2b463b434ba3758009d3b7bf12 : FAILURE in 33m 57s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ce426b33a0bd47de947a886277f05e0a : SUCCESS in 1h 18m 18s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e96f39c4aa14daeadd7563ed3276d34 : SUCCESS in 2h 01m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8f146470dbd0ef6c81170b4a67e52ab00bcc7e41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-09-02 12:33:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"44a363c467536f7a547b85361b1a30ee452bbf05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:25:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5610f453437450b82e187d816cb02e1\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/cfdea3e5132c409fbc11fbf9cd49b222 : SUCCESS in 50m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/7c517e24f2ea4c889636779d87cf5475 : SUCCESS in 1h 22m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc9216432e8e404ab89902d06ad7cd09 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f80c36b49d4411f978adb36549bbb68 : SUCCESS in 12m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1fcc8b479e5405ba8eb3eab0ef5e07b : SUCCESS in 21m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b6790a96e2b74020a45a5f14411080fa : SUCCESS in 25m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24ae1079d4974a87bb0c6b0e1069033d : SUCCESS in 22m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e52050acd4a740cb9c9376d220c0f242 : SUCCESS in 24m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec09d34f7654b4484cb207c697133f5 : SUCCESS in 6m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3c3209280d1543479596272a13dc60f3 : SUCCESS in 1h 40m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00c11125d54a41278161a2f631784cc7 : FAILURE in 1h 21m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/51f0498380b3454d9de7b7fe984d9d5e : FAILURE in 47m 05s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/f2e624f4230d48dabbc356ec2fed40c0 : FAILURE in 38m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/84fdd6f9991a49c69cb2d14c63a15505 : SUCCESS in 1h 19m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7b56fad4e41e4abeb5b2dd447392a50f : SUCCESS in 1h 35m 00s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bf1c172655b3413495d036fac7f37e3a : SUCCESS in 1h 18m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/61b6ac8fcf134b0c80af8ef93965d7d8 : SUCCESS in 1h 40m 52s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc4b2f63dc3b4d5991b576b4fa126fe0 : SUCCESS in 1h 42m 33s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/700791f339d1468587c0fab5f5852eca : SUCCESS in 1h 46m 41s","accounts_in_message":[],"_revision_number":2},{"id":"357a53eec3b5dc22a61f312e02e3464af7f93a9e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-09-22 03:24:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b63b28f0407cc4e1e56868025075ab79aa69100a","revisions":{"696268859aac8348eb55193b9be5176226468956":{"kind":"REWORK","_number":1,"created":"2022-08-19 05:18:37.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/52/853752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/853752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/853752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/853752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/853752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/853752/1"}}},"commit":{"parents":[{"commit":"7d6bde20ce3252e090400909ff21d286f1c21459","subject":"Wait longer before deleting DPDK vhu trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d6bde20ce3252e090400909ff21d286f1c21459"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-11-12 12:43:24.000000000","tz":60},"committer":{"name":"liuyulong","email":"i@liuyulong.me","date":"2022-08-19 05:18:37.000000000","tz":0},"subject":"Fix migration failed due TypeError of new_vlan_tag","message":"Fix migration failed due TypeError of new_vlan_tag\n\nThere could be a situation, when there are ports\nwith port_tags None created in OS infra.\n\nUnder normal circumstances, it would take some error\nto the log and go further.\nBut only because there is a \"new_vlan_tags must be string\"\nin that log, it is throwing out typeError and dropping\nthe execution of code and the migration is ending by code 500.\n\nAnd this is all done without a message from\nthe log, which is extremely bad.\n\nThis patch is fixing this behaviour, and flow\ngoes on.\n\nChange-Id: Id4ba8d32e6b2301f25c12f83a0708b62aa5b9d6d\nCloses-Bug: #1903989\n(cherry picked from commit 7cf52bb5325ac73b7afa3212cb84d29831e3724a)\n(cherry picked from commit 87c375bab79226a8df1e306f32fa29188d4d0ea5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/696268859aac8348eb55193b9be5176226468956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/696268859aac8348eb55193b9be5176226468956"}]},"branch":"refs/heads/stable/stein"},"b63b28f0407cc4e1e56868025075ab79aa69100a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-02 12:33:47.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/52/853752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/853752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/853752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/853752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/853752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/853752/2"}}},"commit":{"parents":[{"commit":"5b812fd45ec5cbf7a84b0442536ca2181248aac5","subject":"Merge \"Stein only: Make CI green again\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b812fd45ec5cbf7a84b0442536ca2181248aac5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-11-12 12:43:24.000000000","tz":60},"committer":{"name":"liuyulong","email":"i@liuyulong.me","date":"2022-09-02 12:33:47.000000000","tz":0},"subject":"Fix migration failed due TypeError of new_vlan_tag","message":"Fix migration failed due TypeError of new_vlan_tag\n\nThere could be a situation, when there are ports\nwith port_tags None created in OS infra.\n\nUnder normal circumstances, it would take some error\nto the log and go further.\nBut only because there is a \"new_vlan_tags must be string\"\nin that log, it is throwing out typeError and dropping\nthe execution of code and the migration is ending by code 500.\n\nAnd this is all done without a message from\nthe log, which is extremely bad.\n\nThis patch is fixing this behaviour, and flow\ngoes on.\n\nChange-Id: Id4ba8d32e6b2301f25c12f83a0708b62aa5b9d6d\nCloses-Bug: #1903989\n(cherry picked from commit 7cf52bb5325ac73b7afa3212cb84d29831e3724a)\n(cherry picked from commit 87c375bab79226a8df1e306f32fa29188d4d0ea5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b63b28f0407cc4e1e56868025075ab79aa69100a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b63b28f0407cc4e1e56868025075ab79aa69100a"}]},"branch":"refs/heads/stable/stein","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
