)]}'
{"id":"openstack%2Fneutron~854013","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Iaa7b0d7ceb244a38fddd7676066683bf2ca72341","project":"openstack/neutron","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-08-23 00:26:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa7b0d7ceb244a38fddd7676066683bf2ca72341","subject":"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata","status":"MERGED","created":"2022-08-22 15:38:03.000000000","updated":"2022-08-23 00:28:12.000000000","submitted":"2022-08-23 00:26:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854013","cherry_pick_of_change":854012,"cherry_pick_of_patch_set":1,"meta_rev_id":"08228a9d1dad0d4c8141a93a5f34d07c1ee8b773","_number":854013,"virtual_id_number":854013,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 00:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-08-22 15:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-08-22 22:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-08-22 22:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 15:40:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-22 16:02:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 16:28:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 17:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 17:32:43.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-22 22:47:21.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"13b87ec2d422073d89450ff82d377dc16dc100c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-08-22 15:38:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"de7c2e92d16b7913fbe366435ad0a593d0ccf479","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-22 15:40:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75878fb1265ea5f0f3f77b61aca1f642ab7ca69f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-08-22 16:02:29.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-wallaby-macvtap http://13.74.249.42/refs/changes/13/854013/1/openstack-sriov-neutron-wallaby-macvtap : FAILURE\n\n* openstack-sriov-neutron-wallaby-direct http://13.74.249.42/refs/changes/13/854013/1/openstack-sriov-neutron-wallaby-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"de0c6f42ce8587b18f483f56a1cabc3b38cf2948","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:28:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b831b7f1642446880188f4f2b55dd54\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b4b6a546ff848edb430f6c5c848cdc4 : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c85de06ac3914a2dbbea230d36cb4354 : SUCCESS in 39m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08b48e19f7aab1a0ea860da4d355e7dba3a9e3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:19:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53e6c25efa124871b1bfc341c5d089e5\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/9428c497ff9546caa371cda4553f7800 : SUCCESS in 1h 02m 29s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/ee33058b9f034c7f83f1e88bcd7a039a : SUCCESS in 1h 27m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9eb179b3c7614532899fd3aa0b31940e : SUCCESS in 51m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707ac2e379bf42d0a4ca39e4685a27cb : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c52c646452a94a828eeb43a5815de583 : SUCCESS in 32m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85a704d224b94efcab8c32d90b5d6e7e : SUCCESS in 40m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46c15c5e69124a648d25b623246dbc39 : SUCCESS in 40m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632f10d95e5248aabc21f71380c0e3bb : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/daa9b40fbdcf4b8080a8c372168ac396 : SUCCESS in 59m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a6b2837d6e984d8896b301fc7156ad54 : SUCCESS in 1h 21m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bab4a7fe7b9d419684a2b5353046cfa3 : SUCCESS in 46m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f5c5e745d97842e5bbe55c46e63aa43e : SUCCESS in 1h 34m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2cd9ba76bd67465e87b054dc912d9cef : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"58dfcded3f0d658a8c174549ef16f6b9d8ddd95c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-22 17:32:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/854013/1 : SUCCESS in 1h 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ac317773343684268559aadeddb850f369db09c7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-08-22 22:47:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46777af9390322a7fe42bb49e4acbe35b379f436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 22:47:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f15a14c6565ba288a64a64e2b821cc68f49527a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:26:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f189b343ce74804addf213059c6590d\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/61f48897b941484696d47f8d45cd1a91 : SUCCESS in 1h 22m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be5fbc70368429b8277ec90b111672e : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccdcdda48e014e1292d0f57261d3c651 : SUCCESS in 24m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97235d3fab98412a856f3831e7335267 : SUCCESS in 29m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a1d0f135ca4364ad34b659d5cc76d6 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a84df09328224b16aa79aa8bc88e6bb4 : SUCCESS in 1h 18m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/06a528aadfc2419c9726b3e2acf2dcf9 : SUCCESS in 1h 32m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d608f7b807c749518ea1c14ccb11e16c : SUCCESS in 53m 26s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2d7ee0f8f8104afa8b197b406e18eaf6 : SUCCESS in 1h 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e8fdbb9dec121743c32a51805d68df9fbc920b40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:26:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08228a9d1dad0d4c8141a93a5f34d07c1ee8b773","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:28:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0658a2051d3d42cab7cf6205c7d44dc5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ab72b2bf2b4ecda7277bea163e2ae2 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6432b3943611d2d2f5ff3a22230abc181edc1ac","revisions":{"a6432b3943611d2d2f5ff3a22230abc181edc1ac":{"kind":"REWORK","_number":1,"created":"2022-08-22 15:38:03.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/13/854013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/854013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/854013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/854013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/854013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/854013/1"}}},"commit":{"parents":[{"commit":"464e88f742914214b89f898cd58e688f52ac3e99","subject":"Merge \"[OVN] Fix updating network segmentation ID\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/464e88f742914214b89f898cd58e688f52ac3e99"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-08-19 20:16:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-08-22 15:38:03.000000000","tz":0},"subject":"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata","message":"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\n\nThere is a scenario where IP allocation pool is depleted but OVN\nmetadata port got removed its IP manually. The DB sync script will\nattempt to allocate a new IP address if DHCP is enabled in the subnet.\nSince the pool has no available IP addresses an exception is raised and\nthe whole db sync stops.\n\nThis patch simply catches the exception, logs and error and continues\nsyncing other resources.\n\nCloses-bug: #1987135\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nChange-Id: Iaa7b0d7ceb244a38fddd7676066683bf2ca72341\n(cherry picked from commit d237a2a83034fab809517a39dddee15995226131)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6432b3943611d2d2f5ff3a22230abc181edc1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6432b3943611d2d2f5ff3a22230abc181edc1ac"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
