)]}'
{"id":"openstack%2Fneutron~854041","triplet_id":"openstack%2Fneutron~master~Idfe5e51007b9a3eaa48779cd01edbca2f586eee5","project":"openstack/neutron","branch":"master","topic":"random_fully","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-25 14:35:59.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-08-26 08:42:03.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-08-26 08:42:03.000000000","reason":"Change was submitted"},"7130":{"account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"last_update":"2022-08-26 08:42:03.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-08-26 06:50:04.000000000","reason":"\u003cGERRIT_ACCOUNT_5948\u003e replied on the change","reason_account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}}},"hashtags":[],"change_id":"Idfe5e51007b9a3eaa48779cd01edbca2f586eee5","subject":"Allow operator to disable usage of random-fully","status":"MERGED","created":"2022-08-22 21:09:19.000000000","updated":"2022-08-26 08:43:23.000000000","submitted":"2022-08-26 08:42:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854041-random_fully","meta_rev_id":"4fe6add66d6ef68f409afc625f53e1fd103f415b","_number":854041,"virtual_id_number":854041,"owner":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-08-25 21:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 08:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-08-26 06:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-08-26 06:50:04.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-08-26 06:50:04.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 21:28:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 21:42:57.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 22:24:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 00:49:13.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 04:15:58.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-23 05:56:49.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-08-23 08:53:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-08-23 13:16:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-24 14:56:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-08-25 14:54:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"589a5a524b8a63303ecbba42ddca39592722bd63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-22 21:09:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c329e58156cf1eb2a4f63db8254bd7b8e81aaac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-22 21:15:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bdc619a7d9d3681f666cfa894e1468d875df7cfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 21:28:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a502d112e0a3407a828465a855b62430\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfa89645534e46cda7ea054831ddeb6c : FAILURE in 8m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5061d3c8c7643d8aab3116aa487b9d5 : FAILURE in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c90fccebf37959e5b9250d2871ff067805294d56","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-22 21:42:57.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"620c65920a6f5739455e53a1f0107ee0c4213081","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-22 22:16:46.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"fe9b8b27375de4cafe87e0ea8f79c8f50b5ef456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 22:24:50.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52cdf32100ac4e99a08f39fc5d6f0ded\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e231b4c9d8d54667944222859eac1afe : FAILURE in 23m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a7572286c81b473a910feeb3d0459628 : FAILURE in 23m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c2de4cdbd6f141b5a9a01d22766f0ede : FAILURE in 24m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ce0f3c3e3f164ece9591d764dabcbd8d : FAILURE in 20m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04ebed2e9bbb4749a8ffec32d6b6de71 : FAILURE in 26m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6e21e56d7394174b1a5682e13f49ad9 : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e237c24bd508416e815a106ba502caab : FAILURE in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38f85dfa39f840bc863dfed82e1f2342 : FAILURE in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3af55daf262431ba9a7013cb1b1ea2b : FAILURE in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1d71452587140709b621a612929bf2a : FAILURE in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c1ddc4eb8c4205aeb50cd7dcd1e756 : FAILURE in 6m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77cce07b469948328fb7d83c8d0ca3fe : FAILURE in 23m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d08fae70ebe4bd1aff5d772b2a76e2b : FAILURE in 12m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/911439f0d55348f99b474fd36e5d087e : FAILURE in 31m 45s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99f45c89085b48e0b640932d51ee453e : FAILURE in 52m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6f1a7f8f13d345229854b49cff02df3d : FAILURE in 1h 03m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a8b1074fdfa948118fbfedb34609101b : FAILURE in 36m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/51d521c2e88e4befb900db336a7dd040 : FAILURE in 22m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad5a343a363e48dbad36d979e5ffd816 : FAILURE in 26m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7e528d89b6314c2cb5d42668788b40d2 : FAILURE in 32m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3224e7ab533b679fbac46efbaabc87374eb6ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-22 23:26:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ad822542e12056f42d6c9705b6b0019504d2297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-22 23:32:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6698cb53d0f838e5ffa36e7285d8e2260ae2ecfd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-23 00:08:32.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"be2436d819e073fce29c1f6e39b712b4613f0a89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 00:12:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e9545bcd82f41149ad911f760815e51\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25921720d7db461a9497a9f3324ce42f : FAILURE in 32m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc8c582679eb4e19a2d05f79868de8d5 : FAILURE in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f530c58febd900824cafdc71949981d2610cf211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-23 00:46:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4f67ce180c9dca2e85d9330875e5865cb13abfd9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-23 00:49:13.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"c568950e53888bccdb3713b5896fe5ccaa93409c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 01:28:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cbf3c614ebe4559bea2fd44b3950be1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78aa874063b8425fb8fe745f85e8b357 : FAILURE in 40m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d974dd9ad1049d88ad859ef6107bb16 : FAILURE in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1346e6cbaff340f3f339cf59d6e97c36e78e3785","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-23 01:41:17.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"d29badf03ef5b1635bf8477436d2f7364b155921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 03:52:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60259afcaf6b436fae378c49ff8b4839\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/824a574bcbb54276950c30592ccd82a6 : TIMED_OUT in 2h 56m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b025ce84c0964385929100f179f9b27a : FAILURE in 2h 57m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7cac44db2a7f42ce80205950d7530ba2 : TIMED_OUT in 3h 04m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d71f66b7fd3145a896542bde34e38e1c : SUCCESS in 1h 38m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe4b1130d4cf460cb5889703f9fb51be : SUCCESS in 46m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78d01380c3794f4c99fbb40f3e1fd002 : FAILURE in 29m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8738fcc7c09e410ba794108a9fca43ba : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1d59b6d8ae244d2b61ea6df532d4b98 : FAILURE in 21m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a899670c97774cd6adf758f93cebb14a : FAILURE in 23m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b10cbef7251448f836f436a31a41d92 : FAILURE in 20m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08bc4bc9a8b34e1ba95aaedd30035fbb : SUCCESS in 8m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/414124d7e72543579f15b83e88e39f36 : FAILURE in 44m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/57202d243cca4fe695d5f433a7092684 : FAILURE in 1h 23m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/33eeb4e8581949f0824bbc2e1ee55019 : SUCCESS in 1h 01m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/32c28e510a734b0c8e9fdeda5d2329ab : FAILURE in 52m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6677ce531b940909994b927e5c7d6f2 : FAILURE in 54m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3a26412830cf4278a87d0c3dfecf2efd : FAILURE in 1h 08m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b39e1869bb44803bc8863eae4391191 : FAILURE in 1h 04m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54db946b7bef4b7db7e3cb46a76838bf : SUCCESS in 1h 02m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bfa959ead19f46bf954946a1492d5858 : SUCCESS in 1h 03m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d4718b330b9e627b359ccd930bf5935e2808e603","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-23 04:15:58.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/854041/5 : FAILURE in 3h 28m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3264a4ffbc4547db50507ea28da2885e6109866e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-23 05:56:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"20103b8b961d33ad079403c730b641fc5c84a970","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-23 08:53:53.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5c879b8be807fc0c71a0c612168d737e40f02f89","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-23 12:50:07.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0021a29b178183fe1cbb841b84d7804b09ee7514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-23 13:04:56.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"97c8e88071d84fec34bb69798357486162220a0b","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-23 13:06:49.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"988721370a9e2462bba8f01ce3035f8b20c85e5c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-23 13:16:30.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cc1ea2ac0915232db616c9684af0953d406fd665","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 13:47:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/199240bb05924ec4bd61ef0c93cd53b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e639431c104448afac610d118bcb0010 : FAILURE in 32m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7a90b8cf7ca41baa73537abf0663e1c : FAILURE in 41m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"767b9e5a64c82462ae5e16b1ab4677893b8328b4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-23 14:00:24.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4df3d46952181ca0115c6d365cb6ef99be84bf68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-23 15:56:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4efe45fd344ca3f2452421d5f5e742bfeb56442e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 16:46:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b5622708b7942bebae51f91465d515b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42324957e75c479e8e7f9304580e11be : FAILURE in 46m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c02380fe009243ee8fc392dfa3ad8578 : FAILURE in 46m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"225fa3984ae41913489fd3e8a1a7c19befafad42","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-23 16:48:42.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d2fa05b1bc961520bb7c51a1c66b32084f4d3748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 18:58:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06ed39833a8642f1a65c80ce38ab7a35\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4f14f6d525c54fd0b1fe46935a6fd518 : TIMED_OUT in 2h 56m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/67d06d73a5b44599837f086b5160ca88 : TIMED_OUT in 2h 56m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5c955cdcf3d4d6f90e08c2087e695f0 : TIMED_OUT in 2h 59m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6df728395599482eb338a2f67be8e83c : SUCCESS in 1h 09m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a6656542f454523b7ab75e4a67db3b2 : SUCCESS in 37m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e78cdbb45e84276a77389bff03ada20 : FAILURE in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b030baf79d4665b6523f57bbcf7282 : SUCCESS in 14m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62105f8ee1004f518d7c7f82916baf85 : FAILURE in 27m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81eaec0a8e6243bf8bdbdd15931a9aaa : FAILURE in 25m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17a4c4bac1d74254afff175770edd5aa : FAILURE in 29m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1175e5af53384941afd4871e02608b00 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c3cf1a6de4d49e79b89fb216da03492 : SUCCESS in 7m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d024c917e76e49b5a3a8ac8f57f438b8 : FAILURE in 48m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a20da6cdf0224c5c977adfc8c829d090 : FAILURE in 1h 28m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/26d432e1dd354d568640813384f3430c : SUCCESS in 1h 09m 41s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d3980394b0842b286dd66e5a2f2866d : FAILURE in 51m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8efaecb2d0e5401cb6d0e375e4ba34de : FAILURE in 53m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2489e3b77d1e4e988ad2cfb0f13ae48e : FAILURE in 1h 18m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/385b8e3eaf614b3eb161a33734ab733a : FAILURE in 1h 19m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c38ff0e20a314b77a1f826430791a659 : SUCCESS in 56m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c97e2fed3cc04d74b4e5a042a97dcfc6 : SUCCESS in 1h 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"276e4573f4e6ae31fe4d97fb56a75b792ee0c9a4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-24 00:27:31.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/854041/7 : FAILURE in 3h 10m 59s","accounts_in_message":[],"_revision_number":7},{"id":"f5c4dbd0faa9fa8a15c0304d0d5d461a668a241b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-24 14:56:42.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8e09792b497f54d181b0d9dfcbf6495677443a92","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-24 15:00:02.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"54ceb1d15fc725e23f90f7ddd2098b6d73984b89","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-25 12:15:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"db8e6b3bbc553d898a5f6689f851ac1e37949bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-25 12:38:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5ba1d242fd440e13831bccad5fa8002af1ff000a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-25 12:42:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3f3a47241d9751c3cc9105c6073116d659ba732e","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-25 12:43:25.000000000","message":"Patch Set 7:\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fe65c4b40fc0e1b7f6acfbfe9c0ac43e3093300f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 12:57:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/238d25fff5cb44dfb59d28d92227ef0b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5fa03bb6f0c4995924f75f1ec961d15 : FAILURE in 12m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89be29c5cf274184ac97f3831ebc6857 : FAILURE in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1dee4c5a4a600c7946051cbcfb30d8433a47a770","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-25 13:03:10.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"3dd285b5a106656c24e3b869ec5356774ac47f56","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-25 13:29:06.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"18aad8291759683d54e92cfe2379814dff37cbea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 14:04:48.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9597c5eab2c4dbdb8646260aeee1f44\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/77e5d3363ee44faf99b8701e0335f1ac : FAILURE in 40m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a45f6f28b901438380e332e6f45c17f3 : FAILURE in 26m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/27506c96fe5646da9bbf04f2926af34d : FAILURE in 29m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/89f178ac112e477e95262a54a4d56c04 : FAILURE in 25m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/918d28bafa794bc3a1c683c8bd5b8f77 : FAILURE in 29m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbf0380833454373915d4e7fb9725572 : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d9fd00a9bf42fb963450655956af68 : FAILURE in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f2231585f514d29b4c33417005268dc : FAILURE in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efff305995e347ea83d343afcd3f2070 : FAILURE in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/181bb2ce63364baaa21ed940f46736a5 : FAILURE in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efae85a27a1c4d9b9588bde390553167 : FAILURE in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e80d9e46c9e492aae80dcc42dd70d20 : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4585ce8b1ff4756be6bd95c7f9638a5 : FAILURE in 23m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/602823482172488c895ef2a130551613 : FAILURE in 21m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/575070b97a1b4449ad49e0752d88c81d : FAILURE in 32m 14s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/419c1407a3fe459883e6a68700f03b46 : FAILURE in 55m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5eba1341201d445bb9248d142bc84cb9 : FAILURE in 1h 15m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8ed21c5b06324138b69e3e12f2ec7f1a : FAILURE in 29m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4ee474b2de3c4793842f6ba3a5152afd : FAILURE in 25m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6643c5c1fda8443a805243e33e130127 : FAILURE in 22m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d3c146b5a8a42ed8c7ca05f85163958 : FAILURE in 23m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4481b16a1b0d56ab60a458bf66976881161d695a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-25 14:18:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6e9da52a5e02f4b8a67bbd428adab5217c9987cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 14:34:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76084d6fc32446097892d1b8f1a6982\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35af56673b15455cbb72344fc16d0dd9 : FAILURE in 7m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d752230600124d38b0264ac86e271539 : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"54a86bf4d09478434692a9edf7e73b00c76b243d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-25 14:35:59.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7de5a4380ce96422117d21da568110e18886f712","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-25 14:54:13.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"632c60855304460fdaffe546b9fa656077c04d2d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-25 15:36:43.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"bf833d325007e9736ebf166684ef3378885720af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 15:42:13.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e68bed57e824418181fbf679bdf2a2dc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/97bb56d9e6dd4de882c39d07ef8fbb6e : FAILURE in 26m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/98773ade379a487b9af6425645f06618 : FAILURE in 28m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28a8c49da25943768df2d5938c1e3e59 : FAILURE in 25m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a1e5373e93a547bbb248bb655f47a1cb : FAILURE in 22m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18a800503c1c498799e5141b2741148c : FAILURE in 30m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e0843e28b224792831d532575259f06 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2342914f3154996ab92fc6f6dcea034 : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ea88b5549c3456590f6817b862fcc96 : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b01492847b084cdda4a75e0cd4252065 : FAILURE in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7217ddc110746e4aa6078f9646b530f : FAILURE in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfbaca42b7d4e828be9f66c8e61d4b6 : FAILURE in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfdbbf3e91924b32b9dc63f741d9ca35 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b702e19f998431fabd1170e13d5c452 : FAILURE in 17m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc8ff8d687574dbda9b7c1c855a13010 : FAILURE in 16m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fa54d7a4013c4228af79055ea1748c6f : FAILURE in 21m 25s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a251ca6a15084499b946f8782ef63160 : FAILURE in 1h 05m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e316e8eff21e414ab98685919607f369 : FAILURE in 1h 05m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7ad6c1b69bc6455d8593688f7ea234cc : FAILURE in 25m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/79d5be179e4a4b1db30d576728f311f2 : FAILURE in 28m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f91e271ef48444cea7ec640f8d5138d2 : FAILURE in 32m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84788f9c6c224a2ca0feec5c55b5fe03 : FAILURE in 33m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"20efd02d0c2084652458d2e543ddbd11a7e47030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-25 17:49:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a30b02a700a65d93066893d94d6632f0470a682c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-25 18:14:32.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"01cd4403a2b4add5e6d3e8f8bf6c636f2b90deef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 18:43:54.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360c05868fa74e6cbdc9b44a8c8e9f2c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e4c0946885d487ba778b899eb80f96f : FAILURE in 39m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f7036342d9c4269ba707b0d3bfbac3b : FAILURE in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1beeb9c274eb5f580bcf7acce7409e67396782b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 18:58:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32a931ecbdae4d758ccd8e9445b5607a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/75f107db842642549101a53573b69178 : FAILURE in 24m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f9467b9d5d1a409c95d9325f3d9f0e6e : FAILURE in 24m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98108ec9eb3740ffa51fd05d291e4318 : FAILURE in 25m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/967a93085b0c4701b1d79b8c7d3b8219 : FAILURE in 28m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad8700168aa94b829f6567227cc361db : FAILURE in 28m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c0c1d8e1c68420699431b5f841bff8e : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01da00a5982049f984059ba3af6d4227 : SUCCESS in 13m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00cf8c3dd37a4fe7a4e4d067b77f135b : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d6149d8598466db20b60bb9fdb3dba : FAILURE in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bb52035f9f4426287316dfd9dc6ab9f : FAILURE in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46edb7923d504edea2251bde462af629 : FAILURE in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/861662f06dcd46eda8cb63d32201c50d : SUCCESS in 10m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1eee548f622f44dfa2088d13524f377b : FAILURE in 20m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/64344715f0d946bf8eec1ed7ce5f52d9 : FAILURE in 15m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f3d376b8026147fd9151acab6af782b4 : FAILURE in 25m 37s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a99e6c4fbc54b9a815126afea4b9b31 : FAILURE in 1h 06m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/88403caefed548b886e257dfe7047528 : FAILURE in 1h 07m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d301034b2c8840e0b3826476cd4c492b : FAILURE in 31m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/35caf69a2a0745898db36ff147d0f682 : FAILURE in 28m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6e6b4b14c2b4dbfa6aaa5b577bc9f9c : FAILURE in 25m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5c3a72b31fd422e90d3299715e4d50b : FAILURE in 27m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e7642840145faf9e88d61ad451bbd14364c42305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-25 20:48:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"14817b6229f0d0aa0f5c31bfd47836c3dd30576a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-25 21:25:34.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/41/854041/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"5027ebff4cb01e99b3cd0be823c4b34887a5eeb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 21:29:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123e8a3b758a4656a7b3278c62c1d313\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f47558fb52e49df8a040fee4fb72455 : SUCCESS in 33m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d28b9e0244a147d482a661d8d9c1f3bc : SUCCESS in 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f7e378a2c0aa38b805d44cad06442f8b26997ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 22:30:02.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a478dce79e7c40ce9d72773a822e9662\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/299d33f122e843a2851e1a52afbee0ac : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dd85ee74f90649688a9d029162459c0a : SUCCESS in 1h 33m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2ef0f1f7b444e1daa83bd4285b30fbe : SUCCESS in 1h 08m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c65cf47380bf4980b027f7a7df92d070 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dca05d1dfa144074950e1423043265a7 : SUCCESS in 34m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edb7da27117b48a3a4c8de350597bea9 : SUCCESS in 35m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eff95363ca54574a77e1b29d9d0204c : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8ef5979b93544c39d1561111796ac69 : SUCCESS in 26m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c58eeb7f82c24bb992c96a59344ec9ea : SUCCESS in 27m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e975b92be0614d44bee4b0422c6b50c8 : SUCCESS in 32m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa298cdfbb334992b04a320dca0ad15c : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9c2fbc64f9d41bbbb9afab4289e7f5a : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b51c815f1a5c4cffa08ec5a0452c7ce2 : FAILURE in 1h 09m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/54abfaecc35a49039394a2130319ec2a : SUCCESS in 1h 34m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a78eb65a77e343fea1081b72c79ac9e6 : SUCCESS in 1h 24m 37s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f34c8fe90904487bcb5c0737107d7df : SUCCESS in 1h 11m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0e47f057b8740f09867e131bcaa5b63 : SUCCESS in 1h 10m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/733a21893bea432abb7e234a00b3d50c : SUCCESS in 1h 22m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7122470e6c914fab96b9c9b204093ac5 : SUCCESS in 1h 30m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8191881680c44a3da8c7ac0de018b8d1 : SUCCESS in 1h 10m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e135af7609d42ad837e1aa0f8814edb : SUCCESS in 1h 05m 45s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e644743aa6f2534a1aa42b48467897b96845ce89","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-26 02:55:36.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d2c5a9be50573ee7e9388ffea053841ed86d0d22","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-08-26 03:00:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9259029e7ba8d775cd8dcb80eaa3e24feb1c892d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:41:59.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6908fefe20544c8ca410d03c799441d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/564a7fde060644f9bbd20b526ca71ced : SUCCESS in 37m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/033b3e20682b4b519d1fd1082f8250d6 : SUCCESS in 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2434ad5d0b8f2722e0f21a3a971aa730b4baaea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 04:57:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc819b8716ff4227b31beef672f47368\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/86db571021544781a347f27248b36772 : SUCCESS in 1h 45m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/630e27e98f6f4f52ac7cc555388e1766 : SUCCESS in 1h 58m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f552730740647128c6b8dfdca88193f : SUCCESS in 1h 43m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1d98d570aadf49cf94a10bf03951088c : SUCCESS in 1h 55m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/21172123f7d64c0a84c3a3fac9a9e9eb : SUCCESS in 53m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11d369b124c04b7c840912f5009be6a8 : SUCCESS in 33m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71eafe02a94d4ec2ab20a79db01b1af5 : SUCCESS in 12m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8349850df2d5405ca8eb5e22f463ddac : SUCCESS in 32m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79ebda59777b4fec97d9f79a88d76d7b : SUCCESS in 31m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/074e36c69b614f8e9e8a637bbe468c13 : SUCCESS in 31m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3355a9b18db4f1fb6ae5280ec57f9d3 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d9c997de8e4d1da4bac93f8077144c : SUCCESS in 10m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa0b79ef420144748c9cdcede7bd2aa1 : SUCCESS in 58m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/171a2fa6b20e4ba59fca8ac201481112 : SUCCESS in 1h 47m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bc168258e7e34541bc3ac45724a91827 : SUCCESS in 1h 06m 07s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e03de3acb81b49cb9421f6fbdd3c4344 : SUCCESS in 1h 16m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06a4ab7f8b6f4905ac67c4d1295bdcb8 : SUCCESS in 1h 08m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c5536f231a83410b9c3f08a7f57908b1 : SUCCESS in 1h 20m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/05d60628485d41dcb4d6a92ebc84c108 : SUCCESS in 1h 41m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5131fa660054c32b56d91f182bb50bb : SUCCESS in 1h 22m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e887962733104b79a6e84e654d4a7153 : SUCCESS in 1h 08m 41s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"138df776364940772038ffb6bc828a4ff819e0b1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-26 06:35:25.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"69b6c2232185026680c6f0010b023f55a0f5f8ed","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-26 06:50:04.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"3497cc2e58f1b75e2d0ce7863a4234711cedd559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 06:50:36.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"6f482d4f6f510ee945518eedb7ad84317ad54e74","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-26 07:07:35.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/854041/12 : SUCCESS in 1h 52m 12s","accounts_in_message":[],"_revision_number":12},{"id":"5fbaba744df759b3e56dbf16f3cd6847f22cbd27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:42:03.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beed7dec86c54dc9804883d3fcef4680\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1747305b689a4f7d8f63ca0a73f90138 : SUCCESS in 1h 35m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8043440e74124adc91fefd6d718aaf5f : SUCCESS in 1h 49m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6fb490bd09a04f72bec45c8177a19803 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/24a8a1699df94f22a3f868bb0f356aad : SUCCESS in 1h 44m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab73a28698344d8c838d98748ca43a47 : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c167d34815f744bdb66081953de0b5bc : SUCCESS in 22m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba65746b66424bb2a3bb74d4d65c16fe : SUCCESS in 22m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e258ed519a48bdab056d62cdb20702 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e30162241cb64651bc8d41d0832f26f5 : SUCCESS in 8m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bf9e622517c43c789004ff690914ad4 : SUCCESS in 1h 11m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/de743cf3f0c34c92b8b79f78cb65c20c : SUCCESS in 1h 34m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e1ee49c82d5f492e8ca660f3584ed866 : SUCCESS in 1h 14m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d16f90e692d4b24bd4a5ccd0d1d5cb7 : SUCCESS in 51m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/45fa62958f78427292a04f801feabd78 : SUCCESS in 1h 11m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8d4a43a638b4746b15744ed47abb63e : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":12},{"id":"b21a83d42fb946b7054a1f14904de7a7bde383f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:42:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"4fe6add66d6ef68f409afc625f53e1fd103f415b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:43:23.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f13365c9cfb14db6bba7127eeeaf3706\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164d4743f3144cdea943a54f64291a09 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0f08f7846a14f8f80571422214c37e9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"bbefe5285e7ab799422fab81488f57c9c22769b6","revisions":{"494f9f5c1e6ee5305a40a96b861e2ec7e1285217":{"kind":"REWORK","_number":1,"created":"2022-08-22 21:09:19.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/1"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/494f9f5c1e6ee5305a40a96b861e2ec7e1285217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/494f9f5c1e6ee5305a40a96b861e2ec7e1285217"}]},"branch":"refs/heads/master"},"73bb9e038a8fcb7ddc92ea05626f54aa37f10f54":{"kind":"REWORK","_number":2,"created":"2022-08-22 21:15:09.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/2"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:15:04.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73bb9e038a8fcb7ddc92ea05626f54aa37f10f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73bb9e038a8fcb7ddc92ea05626f54aa37f10f54"}]},"branch":"refs/heads/master"},"ce907d1de598f588ad4a24a3c3365b9ebe6b794e":{"kind":"REWORK","_number":3,"created":"2022-08-22 23:26:59.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/3"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 23:26:54.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce907d1de598f588ad4a24a3c3365b9ebe6b794e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce907d1de598f588ad4a24a3c3365b9ebe6b794e"}]},"branch":"refs/heads/master"},"0dc99c5f659282562ba4b94a7652bf8a74372a4c":{"kind":"REWORK","_number":4,"created":"2022-08-22 23:32:22.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/4"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 23:32:17.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dc99c5f659282562ba4b94a7652bf8a74372a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dc99c5f659282562ba4b94a7652bf8a74372a4c"}]},"branch":"refs/heads/master"},"22900163b35112ceeb475e3e1a97c929dba3a6ec":{"kind":"REWORK","_number":5,"created":"2022-08-23 00:46:19.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/5"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-23 00:46:13.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22900163b35112ceeb475e3e1a97c929dba3a6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22900163b35112ceeb475e3e1a97c929dba3a6ec"}]},"branch":"refs/heads/master"},"948b241f0cacce1460848748c1fef83b91b169ef":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-08-23 13:04:56.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/6"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-23 13:04:35.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\nCloses-bug: #1987396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/948b241f0cacce1460848748c1fef83b91b169ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/948b241f0cacce1460848748c1fef83b91b169ef"}]},"branch":"refs/heads/master"},"f39df22210e4c3c30a8e13d1f68b364c99c5dd0f":{"kind":"REWORK","_number":7,"created":"2022-08-23 15:56:55.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/7"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-23 15:56:51.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\nCloses-bug: #1987396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f39df22210e4c3c30a8e13d1f68b364c99c5dd0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f39df22210e4c3c30a8e13d1f68b364c99c5dd0f"}]},"branch":"refs/heads/master"},"2745ece9e740f661032d26a3540a9c69b70a971c":{"kind":"REWORK","_number":8,"created":"2022-08-25 12:38:32.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/8"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-25 12:37:44.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\nCloses-bug: #1987396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2745ece9e740f661032d26a3540a9c69b70a971c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2745ece9e740f661032d26a3540a9c69b70a971c"}]},"branch":"refs/heads/master"},"1c920b5021f9469de07a5ea637a74fbd045d149c":{"kind":"REWORK","_number":9,"created":"2022-08-25 12:42:50.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/9"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-25 12:42:45.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\nCloses-bug: #1987396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c920b5021f9469de07a5ea637a74fbd045d149c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c920b5021f9469de07a5ea637a74fbd045d149c"}]},"branch":"refs/heads/master"},"23df1a334914279cc1a4dee71d00fae941a20913":{"kind":"REWORK","_number":10,"created":"2022-08-25 14:18:19.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/10"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-25 14:18:14.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\nCloses-bug: #1987396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23df1a334914279cc1a4dee71d00fae941a20913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23df1a334914279cc1a4dee71d00fae941a20913"}]},"branch":"refs/heads/master"},"21f14ee3a89a48c6877080e523511da6e0f8713a":{"kind":"REWORK","_number":11,"created":"2022-08-25 17:49:13.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/11"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-25 17:49:03.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\nCloses-bug: #1987396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21f14ee3a89a48c6877080e523511da6e0f8713a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21f14ee3a89a48c6877080e523511da6e0f8713a"}]},"branch":"refs/heads/master"},"bbefe5285e7ab799422fab81488f57c9c22769b6":{"kind":"REWORK","_number":12,"created":"2022-08-25 20:48:51.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/41/854041/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/854041/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/854041/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/854041/12"}}},"commit":{"parents":[{"commit":"7a743ad1d93ceaa9eeb13c65bd07b55783348915","subject":"Merge \"ovn: Don\u0027t fail db sync if new IP allocation fails for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a743ad1d93ceaa9eeb13c65bd07b55783348915"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-22 21:03:49.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-08-25 20:48:45.000000000","tz":-240},"subject":"Allow operator to disable usage of random-fully","message":"Allow operator to disable usage of random-fully\n\nIn some specific use case, the cloud operator expects the source port\nof a packet to stay the same across all masquerading layer up to the\ndestination host.   With the implementation of the random-fully code,\nthis behavior was changed as source_port is always rewritten no matter\nwhich type of architecture / network CIDRs is being used in the backend.\nThis setting allows a user to fallback to the original behavior of the\nmasquerading process which is to keep the source_port consistent across\nall layers.  The initial random-fully fix  prevents packet drops when\nduplicate tuples are generated from two different namespace when the\nsource_ip:source_port goes toward the same destination so enabling this\nsetting would allow this issue to show again.   Perhaps a right approach\nhere would be to fix this \"racey\" situation in the kernel by perhaps\nusing the mac address as a seed to the tuple ...\n\nChange-Id: Idfe5e51007b9a3eaa48779cd01edbca2f586eee5\nCloses-bug: #1987396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbefe5285e7ab799422fab81488f57c9c22769b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbefe5285e7ab799422fab81488f57c9c22769b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
