)]}'
{"id":"openstack%2Fneutron~854553","triplet_id":"openstack%2Fneutron~master~I94536a3bd7d761752f693ddfb5165273cab16250","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2022-09-09 07:20:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I94536a3bd7d761752f693ddfb5165273cab16250","subject":"Add new novaplug mechanism driver","status":"ABANDONED","created":"2022-08-25 07:42:53.000000000","updated":"2022-09-09 07:20:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a9dedb960d2e395ae24d2caa522c2cb173f7ce71","_number":854553,"virtual_id_number":854553,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-08-25 08:56:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-25 10:05:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 08:24:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-25 08:56:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-25 10:01:41.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-25 10:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"731ecdc4e0b190ee58d1966515a455d747f111a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-25 07:42:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70a6fdcd15f392f19c61286f2bc6d0165bdebdc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 08:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d138658a9d40c0bddee1945fd5d523\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2282c7c94434d978a7074adda0ce93f : SUCCESS in 32m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/518d543b1ad34c29bde386df73efc853 : SUCCESS in 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bc81b5d3432c1666a9e789e426694f6d911c5ff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-25 08:56:17.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/53/854553/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"96ef681a143b35c28fa30e7d2fa748ba759ce1ce","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-25 10:01:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/854553/1 : SUCCESS in 2h 18m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7e3438916ab6c65d07f49e798ca0142278006bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 10:05:53.000000000","message":"Patch Set 1: Verified-1\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75a59ba7b2694f338ccac7bbaaeea15d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6da8c3245f14ae1a3aa236bd5fa5792 : SUCCESS in 2m 54s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d6ef5f6b8089417ab01e942b81868a1b : SUCCESS in 1h 09m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/23e063f55fb54313be7ecb26f30980fe : SUCCESS in 1h 09m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08e1707278a94f989fa19fdfff47d905 : SUCCESS in 1h 09m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/22f4e2b8291e409c87c17276b9e44aff : SUCCESS in 1h 06m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a72ee6bbb03d40c48ee2da30f85e5dd3 : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2355523a3cc4e21a6c82bea1c71cd5f : FAILURE in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82768fae5c8d4623b8ca49d0c4cef4a1 : SUCCESS in 27m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86d1e3c718ae4b9392438f9b4245564b : SUCCESS in 25m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/629ea962d26a447dafca51eac1245cfc : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1581ec45de8423985eb893771442bbb : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/396a658ddc5c451f81e947928f03f27b : SUCCESS in 1h 24m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8250887a8f5740949d3ad63db97112f2 : SUCCESS in 2h 21m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e8269e81de654593ae3d4a1124585181 : SUCCESS in 1h 16m 55s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dc9e5295f944adc9dbf4cd510a3db92 : SUCCESS in 51m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bcebdf16fe374b80b7ed164c846eaae8 : SUCCESS in 1h 09m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4738340f8cf24c30a4b2309a9f81763a : SUCCESS in 1h 29m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a155d69de9a4ec6b06c6b990f5b1b63 : SUCCESS in 1h 29m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eecc671d8f37484fafcc28b7f29a3537 : SUCCESS in 1h 15m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/82071ef2707444e7a07b0451ce846afa : SUCCESS in 1h 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9dedb960d2e395ae24d2caa522c2cb173f7ce71","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-09-09 07:20:53.000000000","message":"Abandoned\n\nwe will not do that, we will do a policy change instead","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee9a8901ab15d28d2bd8f982796027ac045eeb71","revisions":{"ee9a8901ab15d28d2bd8f982796027ac045eeb71":{"kind":"REWORK","_number":1,"created":"2022-08-25 07:42:53.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/53/854553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/854553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/854553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/854553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/854553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/854553/1"}}},"commit":{"parents":[{"commit":"d4790238cd6af16fbaabc035097de3d061f6ad62","subject":"Merge \"Allow shared net to be added on router\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4790238cd6af16fbaabc035097de3d061f6ad62"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-25 07:42:05.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-25 07:42:24.000000000","tz":120},"subject":"Add new novaplug mechanism driver","message":"Add new novaplug mechanism driver\n\nThis new mechanism driver will do a call to nova client in order to\nattach the port to a an instance when device_id is given on port\ncreation.\n\nRelated-bug: 1986969\n\nChange-Id: I94536a3bd7d761752f693ddfb5165273cab16250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee9a8901ab15d28d2bd8f982796027ac045eeb71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee9a8901ab15d28d2bd8f982796027ac045eeb71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
