)]}'
{"id":"openstack%2Fneutron~854986","triplet_id":"openstack%2Fneutron~master~I4ca66d86705a55b2b63083c229629c16b6136283","project":"openstack/neutron","branch":"master","topic":"bug/1988037","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-30 17:03:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ca66d86705a55b2b63083c229629c16b6136283","subject":"Format the protocol number to be passed to pyroute2","status":"MERGED","created":"2022-08-29 10:47:33.000000000","updated":"2022-08-30 17:05:11.000000000","submitted":"2022-08-30 17:03:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854986-bug/1988037","meta_rev_id":"332ed90efad41fc52798226a7eaa422d8e973e4d","_number":854986,"virtual_id_number":854986,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-08-29 11:18:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-30 17:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-30 15:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-08-30 11:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-08-29 14:41:34.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-30 15:11:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 11:18:42.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 12:26:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 12:48:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 14:41:34.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-08-29 19:55:27.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-30 11:26:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-08-30 15:11:43.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"1723c951806a59dcbb250398fcb51fb8c5162558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-29 10:47:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7d0da94827d6eab178bdc9a0e1996f3c8c37618","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-29 11:18:42.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/86/854986/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"90f690ee4c5aece259a6985d80f3906bd88baa4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 12:26:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0af53a9adae446993de60b5f5a5291f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1dcaeef34b44371b2c337ab42257054 : SUCCESS in 39m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efaa874e31474ab7a7353eaeef6c0499 : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fa5a96200982a047864cf8724264835537d41cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 12:48:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b0c33ec0bbc4a49a7b99375321dc9bb\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/99b603959df64e38a83c99d3063fd265 : SUCCESS in 1h 19m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ff6fea6c224a450b917ef79e51c41ee2 : SUCCESS in 1h 18m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6679844ce9a4d3f958071e6627b9c34 : SUCCESS in 1h 13m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b33a7d8322d4134b732491cd1204e1c : SUCCESS in 1h 16m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c8bb45a716e430096bd0844cb65aa82 : SUCCESS in 34m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be321c5207b74f14beebd70ed3031ae4 : SUCCESS in 14m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a29dcf889df475c85652ce533ac7aca : SUCCESS in 24m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f26a1c1cece14055bf6237aa313c05c7 : SUCCESS in 28m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78fb539c8c6344e9b5e8af7531c38c8e : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394b7a88306243dc96ed1ec3f9f4d936 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b66d8e48260f4560b69091aa7d44ec13 : POST_FAILURE in 1h 20m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e01e0fa3d034766a5b6832f6bce19f8 : SUCCESS in 1h 41m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/94b542885cfc4220ac94aa98ec439c11 : SUCCESS in 1h 15m 35s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9350177e8aed4e949b0c774cbf1c8292 : SUCCESS in 1h 17m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a95012e164494f4a9c7330de7e4bc459 : SUCCESS in 1h 08m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/203629b6268c45549757df1213f509ff : SUCCESS in 1h 24m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dee79befbdd84168bfb0f59332781089 : SUCCESS in 1h 08m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ce951ed224c4042b34fa923d2b27158 : SUCCESS in 1h 04m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe5a84f36f174da59ea85bbf258ce3f2 : SUCCESS in 1h 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b476db886061deb1564512cd6d7dd892a3d08f39","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-29 14:41:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e296451a3c423a9489089826cb4758b024a6533","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-29 19:55:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/854986/1 : SUCCESS in 1h 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"654a44f67ad5a3c4445bf254f195a0c2a5caf7b8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-30 11:26:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85eca893ea1ebd099ccd7ec5fcc0a498767c8adc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-30 12:11:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06698f43b6847a16a47866169a5403afb45c7467","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 12:17:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed13ecedeac14cd4a97dc07767848b32\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0764c20d986d49a098570364e10213b5 : SUCCESS in 49m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29d8e6479bcc434ab2a7e4275312fddd : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"024ddf61ad6dc1cbd20f3cb02331c5d7d06efcc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 13:13:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff31190dfe484658aff0bd0f298df5d9\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a74d65e6f5e34740a97868ec49c53e68 : SUCCESS in 1h 37m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8016cd5574c34b91953053fe1e345f0b : SUCCESS in 1h 19m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f2f3a9fba2eb4aeeb343b26f5662beac : SUCCESS in 1h 10m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/efd16082fefe49efafd4549047e78f74 : SUCCESS in 1h 18m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c531c3e82f9648d0be5c05d32e5e808c : SUCCESS in 37m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6786794581324d878e5ebbabc2c090d6 : SUCCESS in 12m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c98dcd619a8e4b7dbff4c114e6371b4c : SUCCESS in 28m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4490fa661354a66b38668d306e7bafc : SUCCESS in 27m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/122dacfccc06419694637e45acf23efd : SUCCESS in 25m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e083a30c0d4e4ba396c04c12d15d74 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce7629cfdb574164a1f66553161ae2f7 : SUCCESS in 1h 04m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e410be2ab5ac430395a564195a161b4e : SUCCESS in 1h 37m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/55886185db624ffa95d2a8193ff4c71f : SUCCESS in 1h 12m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/86ad714fa2214faba61c0ff248c0e938 : SUCCESS in 1h 04m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ebd0062a7ed140feaf01c944e966e137 : SUCCESS in 1h 01m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8c54ab6ad44c43e8ab2ebf06cc4530f6 : SUCCESS in 1h 41m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9747547035b842b8b3523b86ec6a232a : SUCCESS in 1h 14m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1beb620df09749c096d4d9242b738766 : SUCCESS in 1h 05m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/adff2d0fef6341b29f119aa3d746f22e : SUCCESS in 1h 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3362983c50e85e09b42cea1343b175e8c1a2da87","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-30 14:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/854986/1 : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d9fa9c9c04e0b34433294c86f1e1f79893002489","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-30 15:11:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dda5240b35ba24c3550223ae07a999c0542afdb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 15:12:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a3b4a40eeeb60eaf7a83c36022430089095f6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:03:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efe687d55c0f42649ba86e6451b9dcbb\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fd9310717d724fad8bdbf3ac1207556b : SUCCESS in 1h 49m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8e7dd5479f4d46c28aedfcebb6d32e26 : SUCCESS in 1h 25m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1a51af80e8024c9d8996430f82f3fe48 : SUCCESS in 1h 37m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc61f501a6ba4a78bd545b378cbe6287 : SUCCESS in 1h 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad10243dc2840f69c3f41c70df8bc66 : SUCCESS in 12m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43ebed954cde431aa5d1faed7041d446 : SUCCESS in 23m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab4259159af34a26a80c1dccf411b306 : SUCCESS in 26m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582b64eedd2a4b659e133d9366271c65 : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/31ef8284b3a3441c91c5a134d7eeba16 : SUCCESS in 1h 04m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a18af8ae08ff4ff3a225e34459e4f19f : SUCCESS in 1h 42m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc6caca05a614edba7555609e7c0fcaa : SUCCESS in 1h 20m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5a6d49c68cd42fba4e8741f67384b32 : SUCCESS in 1h 10m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c96137533340431fb335f0524a3b8ca9 : SUCCESS in 54m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3424cf550b54fb49fd2b08f40ab3425 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dccf176a7f0417cedcf12a5357cae54456389097","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:03:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"332ed90efad41fc52798226a7eaa422d8e973e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:05:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2b4dc3aff7948fb8619c6231cc9faac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbf7e9c8f23402ab71379dcc7aa49ed : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b8d794a359641f33928490e3a288e1ab49a61f7","revisions":{"1b8d794a359641f33928490e3a288e1ab49a61f7":{"kind":"REWORK","_number":1,"created":"2022-08-29 10:47:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/854986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/854986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/854986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/854986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/854986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/854986/1"}}},"commit":{"parents":[{"commit":"7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f","subject":"Merge \"Update NDP proxy documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-29 10:44:14.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-29 10:44:14.000000000","tz":120},"subject":"Format the protocol number to be passed to pyroute2","message":"Format the protocol number to be passed to pyroute2\n\nWhen a new IP route is created, before passing the route protocol,\nfind if it is a string and if this string is on the pyroute2 defined\nprotocols. In this case, pass the protocol number.\n\nIn the same way, when the IP route is returned, if the protocol is a\nnumber, convert it to the corresponding protocol string.\n\nCloses-Bug: #1988037\nChange-Id: I4ca66d86705a55b2b63083c229629c16b6136283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b8d794a359641f33928490e3a288e1ab49a61f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b8d794a359641f33928490e3a288e1ab49a61f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
