)]}'
{"id":"openstack%2Fneutron~855257","triplet_id":"openstack%2Fneutron~master~I50091c84e67c172c94ce9140f23235421599185c","project":"openstack/neutron","branch":"master","topic":"bug/1988199","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-09-13 21:47:51.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-09-12 14:55:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-09-13 21:47:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50091c84e67c172c94ce9140f23235421599185c","subject":"Add an active wait during the port provisioning event","status":"MERGED","created":"2022-08-30 16:19:38.000000000","updated":"2022-09-14 07:39:30.000000000","submitted":"2022-09-13 21:47:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855257-bug/1988199","meta_rev_id":"ea068e37a675610b8832594811618b18ae6b9b7d","_number":855257,"virtual_id_number":855257,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-13 21:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-13 20:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-09-12 14:55:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-09-13 09:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2022-09-13 16:18:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-13 20:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 16:54:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 17:05:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-30 18:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 18:19:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-31 08:53:24.000000000","updated_by":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"reviewer":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"state":"CC"},{"updated":"2022-09-01 06:01:29.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-09-02 14:34:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2022-09-12 14:55:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-13 09:39:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-09-13 16:18:11.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"},{"updated":"2022-09-13 16:18:22.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2022-09-13 20:07:05.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"5a79b5cb56fdf8f8b171ae9d17228882b35c5e0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-30 16:19:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"274c6aa9a232e99f5f3a72b4ae9d8568339a043f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-30 16:54:53.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/57/855257/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e0f927fcf4fdf26240e5244677f69e3d5105b712","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:05:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56f7f473593740148be384810efc663f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/664c68ef1b484957bc3fda57a6e03e81 : SUCCESS in 43m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a929898b9c1f48ebaf5c962bd57dff85 : SUCCESS in 38m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2480a8d2405e59cb308d7b40b0480ab6c7136d4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 18:14:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/641e6a7b1866497e91ecfb041448d741\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e1545f317db7461da68bfefcb7cc5c3b : SUCCESS in 1h 36m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa7d4d90345f466f96ecc55825e59eb6 : SUCCESS in 1h 17m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e74d74659cc4da8b8cd34c94e3ab81c : SUCCESS in 1h 45m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5bceb199c32400eba5042de8f182431 : SUCCESS in 1h 27m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6e10980bba748cb93b506decd44c1f3 : SUCCESS in 37m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/666ae89ba28b46bfacacc3dfc18c68c4 : SUCCESS in 46m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cffeb99db9e847b2a5e2f491b7da437f : SUCCESS in 16m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01f3aeaea7484c7d9cdfd3e10922770c : SUCCESS in 30m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/343f25d06e3c4616a3fad80731e09ec3 : SUCCESS in 29m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/288455e68aa64305a8112145b1b3d4dc : SUCCESS in 34m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c3d3bb42ed4f8f8b08ae11694a02a4 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f801a74d0b543bcad6921776d44e2ed : SUCCESS in 54m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb1bf22ae8e04ae3afd975b871a1e9ba : SUCCESS in 1h 48m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e71f36117b654e6589c130797d3c6b77 : SUCCESS in 1h 06m 49s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2cefbe19211432ab9757564f9a36c40 : SUCCESS in 1h 04m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f79a5cc7bf4844949e5a9f06ccbb0f88 : SUCCESS in 1h 08m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/37b11778619d42668b4f3cd493d0b6db : SUCCESS in 1h 23m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cabaa88820b24eb582e58188a0521812 : SUCCESS in 1h 18m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a14db9fb078489b9b12bec885f76621 : SUCCESS in 1h 03m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/95e6d778e7be4afd8938d3aaa9440457 : SUCCESS in 1h 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c23d3fb1d383483b525138b48625f60b848282e8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-30 18:19:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855257/1 : SUCCESS in 1h 58m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f0fc13294e5a3a4722c2f91b435b205f87fb8963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-31 14:46:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b22208ac9d1af589e0701d1f892f08c918fa3540","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-31 15:39:18.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/57/855257/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0d59a58691b95b294602da76473d7a1df442f083","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 15:41:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17277e279a9941078360eea6c318d7dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ca3403a0beb49a4902b00d546b91cd3 : FAILURE in 51m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ffc004091f04027af86bf37438c5c4e : FAILURE in 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e695cc04d9a96e250d9f50c48333853ac9a317ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-31 16:36:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af523ec00488512622384d091eb044a37a2d1faf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-31 17:12:20.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/57/855257/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f4f1b147b61c1c8b0097a42b20eb5f535c846547","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:24:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/640b5187c90c418c95acc74152f289b6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a09667592384f719f55dff419fe22fd : SUCCESS in 41m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c043f58e4eb840f5a9b87aba9fa9d91f : SUCCESS in 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ebf954e0f703a34ef81b66b3ba52e80fa6208b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 18:21:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27cdde6cdd2b434dbecf18d978f85674\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/cd4274ca0b6d4449a0090ef477f92d98 : SUCCESS in 1h 15m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d5e72d3fc66a4476bc58268d434b520e : SUCCESS in 1h 20m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/73c89d61404040acae54dc1bf6ddd841 : SUCCESS in 1h 04m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a83ae6be160428a9b6b5c7dc940390a : SUCCESS in 1h 32m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da7dc098b3754ce187fc6977f10b8138 : SUCCESS in 37m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72460c993f484eafb2898bb60701bb16 : SUCCESS in 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d00bfdd87842a288d02b854bf39c99 : SUCCESS in 13m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec15425f81b64ed3a463ac9fb63d4706 : SUCCESS in 27m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8954a85330814e2996bb31f886e5c0e2 : SUCCESS in 28m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84d08bd4c38a41f3966622a01811a7f4 : SUCCESS in 26m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c08b32ef884bd0b26cc041761531cc : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c029dae158cb40bfa9dc44dbb6f838fa : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f204b991ed964914b2bd3f32fc7b60f2 : SUCCESS in 1h 09m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e16e489d1594313a8ec730604e76741 : SUCCESS in 1h 38m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f669d8ebc4194bc8bef5d8759e604546 : SUCCESS in 1h 20m 29s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8913f973c76d44bf8503c355ed5d47d6 : SUCCESS in 1h 06m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca1bdbeafabd45bf80a06b81416f7771 : SUCCESS in 1h 08m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3db14c401f114ee3bc8fb98695dd3a32 : SUCCESS in 1h 43m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b0611989d5443eab32c6851764d6723 : SUCCESS in 1h 26m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f545ba000f34a899c3089c9b17d2024 : SUCCESS in 1h 10m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/31f189b371d446d1bc89040f8f2a1707 : SUCCESS in 1h 20m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4869506017fa23366c3ad00d8baa9fed18ee23b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-31 20:13:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855257/3 : SUCCESS in 1h 48m 50s","accounts_in_message":[],"_revision_number":3},{"id":"d4dc1a7ca1cdc8ea1b764544e1f8898483e5afff","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-09-01 06:01:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dddb8b4cd34458afb0c0e5f1311d1c3c942cafad","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-02 14:34:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dd7936916f820430b61a0b0e03f6c2699bbcd83e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-12 11:04:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"314e8fd27960f81f04dbbea4629348cc57fe6dc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-12 11:39:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec4b26aad7a8661f04cccff5219724887fc3168c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-12 11:50:56.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/57/855257/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"e0dcf1a4ebcee1e4395766288982cd8a9a6dd488","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 12:24:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e009ef239c14a7182b96ca3be255938\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9efd73ec1d8c4db5bdef66385a807f50 : SUCCESS in 42m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50916172f219406f93f73b5e09646e8f : SUCCESS in 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"006a2ab65f74fe910249fc4b3125b7d3fa5166c2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-12 13:19:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855257/4 : SUCCESS in 1h 39m 40s","accounts_in_message":[],"_revision_number":4},{"id":"bcae1e3ea0bcc12a8d9fece518e11dc52ae230b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 13:24:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef066bda19342df9c54daa6cae2ceff\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/946e6f8cd6f648019d06f1368b66041d : SUCCESS in 1h 35m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c39840d44e754ef6aa2c13718cd71e73 : SUCCESS in 1h 38m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e26976957ae4d428c88ab3cd81a88c1 : SUCCESS in 1h 33m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87fc5d740b7f46ab82ec7a2029991d31 : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e0801d8b0b840bc9569a3a4458fea98 : SUCCESS in 48m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f972225eb9d4733ad86716cf4cf6a0f : SUCCESS in 42m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c5fb55fbf94da795cb791e33b17a6e : SUCCESS in 13m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d399cfbfbd48485b8c88f8269466fe53 : SUCCESS in 31m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d50733b4b6e4e009b1478a210a9215b : SUCCESS in 29m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/013d480a23504f00b62e219f8a35ab9f : SUCCESS in 28m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505c8f6a9d0a443a88522fd65f801973 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b8905aabf414574bc584119f3f14d23 : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ea0cb4ca7344e009ee7367b18b21518 : SUCCESS in 56m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bab85f85bd0a461194a696f69735afa7 : SUCCESS in 1h 39m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c00681dd981450597d5346fd4313552 : SUCCESS in 1h 16m 02s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c9e87699452443b839207cdb05cd862 : SUCCESS in 1h 07m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7f0dff7979f451381600548360c0ef7 : SUCCESS in 1h 14m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/970e0302578e494bbdad7669c82225fa : SUCCESS in 1h 33m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f8254ecb81af49d8bb948083251a1320 : SUCCESS in 1h 25m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13533a7a8bcd4054973c0ddca305d089 : SUCCESS in 1h 01m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7bcc62b5326b471ead6d622b3090b7be : SUCCESS in 1h 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2cd825316d6d488e427bad75b89d03c7317561b0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-12 14:55:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7797dab198703b57204c28c918ad252607d9185a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-13 09:39:48.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c6fa0df64cd0804c396482b4c233c39fdf53d91","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2022-09-13 16:18:11.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"490538009d102712c92072f78b5ff1cc8f08bdeb","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2022-09-13 16:18:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7e2b27ae030a19c4c68e696d9ff10d4c200c8104","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-09-13 20:07:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0e39dee78357ee572dd50f93f986462f7d733f83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 20:07:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f47a7b92eee4f6506d467135d85a0cb0920d038a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 21:47:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5927fbf783a74e458222e32ac28369fd\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/18d85e6552794664957b05ffbf2ff8c9 : SUCCESS in 1h 34m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef600294d0e14e41b69e5895fe64ab8a : SUCCESS in 1h 32m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a4be2c6f2414eabb406bd95e65a7431 : SUCCESS in 1h 20m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3542512f0d148cb9f2c8723531bc8b9 : SUCCESS in 1h 08m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9eaf1b78b15462e9dbb970160e9e75b : SUCCESS in 16m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2948974e054b4ec1b62e0066d023f2a2 : SUCCESS in 29m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aca3bfd809147499f6c182f458d21e8 : SUCCESS in 28m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78dede8cd0114e7594687ade82dfb212 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/078ec904e4fc43a688fb888c8ed51c00 : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/43c0bbec120346e88161196433f358ae : SUCCESS in 42m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebd29ac220fd4a099c6c0a9a8b707495 : SUCCESS in 1h 20m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/db4a61688b964d2396cd9637b685c4db : SUCCESS in 1h 05m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/faceb876648240dc967e5fecec02bad6 : SUCCESS in 49m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9068396b17284ccd80a7bd923d39af79 : SUCCESS in 52m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/823720083166498491d10a5caf15edf6 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":4},{"id":"225da1124f3194d046f594dfd3ae30df06b4dbab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 21:47:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"87d87ceb6babe4e3e4b76d18cea92bf96708a8d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 21:49:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e8eed84868142b881812242e8b57478\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a73bf15a2e4cf494b97dcbd8e9c137 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f092541d9124bc5942cc81ad5f88cf2 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4},{"id":"ea068e37a675610b8832594811618b18ae6b9b7d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-14 07:39:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"91f0864dc0ccf0f67be7162f011706dbc6383cb3","revisions":{"32e652c77ab3cd66864e1b4c8a6391e810326a5d":{"kind":"REWORK","_number":1,"created":"2022-08-30 16:19:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/57/855257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/57/855257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/57/855257/1"}}},"commit":{"parents":[{"commit":"7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f","subject":"Merge \"Update NDP proxy documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-30 16:09:34.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-30 16:19:21.000000000","tz":120},"subject":"[WIP] Add an active wait during the port provisioning event","message":"[WIP] Add an active wait during the port provisioning event\n\nTODO: testing\n\nIn ML2/OVN, during a live-migration process, it could\nhappend that the port provisioning event is received before\nthe port binding has been updated. That means the port has\nbeen created in the destination host and the event received\n(this event will remove any pending provisioning block). But\nthe Nova port binding request has not arrived yet, updating\nthe port binding registers. Because the port is considered\n\"not bound\" (yet), the port provisioning doesn\u0027t set the port\nstatus to ACTIVE.\n\nThis patch creates an active wait during the port provisioning\nevent method. If the port binding is still \"unbound\", the method\nretries the port retrieval several times, giving some time to the\nport binding request from Nova to arrive.\n\nCloses-Bug: #1988199\nChange-Id: I50091c84e67c172c94ce9140f23235421599185c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32e652c77ab3cd66864e1b4c8a6391e810326a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32e652c77ab3cd66864e1b4c8a6391e810326a5d"}]},"branch":"refs/heads/master"},"5f70543c58a9a8f3eb673d8872aae6a1665355ad":{"kind":"REWORK","_number":2,"created":"2022-08-31 14:46:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/57/855257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/57/855257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/57/855257/2"}}},"commit":{"parents":[{"commit":"7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f","subject":"Merge \"Update NDP proxy documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-30 16:09:34.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-31 14:45:41.000000000","tz":120},"subject":"Add an active wait during the port provisioning event","message":"Add an active wait during the port provisioning event\n\nIn ML2/OVN, during a live-migration process, it could\nhappend that the port provisioning event is received before\nthe port binding has been updated. That means the port has\nbeen created in the destination host and the event received\n(this event will remove any pending provisioning block). But\nthe Nova port binding request has not arrived yet, updating\nthe port binding registers. Because the port is considered\n\"not bound\" (yet), the port provisioning doesn\u0027t set the port\nstatus to ACTIVE.\n\nThis patch creates an active wait during the port provisioning\nevent method. If the port binding is still \"unbound\", the method\nretries the port retrieval several times, giving some time to the\nport binding request from Nova to arrive.\n\nCloses-Bug: #1988199\nChange-Id: I50091c84e67c172c94ce9140f23235421599185c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f70543c58a9a8f3eb673d8872aae6a1665355ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f70543c58a9a8f3eb673d8872aae6a1665355ad"}]},"branch":"refs/heads/master"},"6c3435489c750a64a31ade32c2e5af9abcc2c395":{"kind":"REWORK","_number":3,"created":"2022-08-31 16:36:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/57/855257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/57/855257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/57/855257/3"}}},"commit":{"parents":[{"commit":"7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f","subject":"Merge \"Update NDP proxy documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfe41ab8f9ecf6266c7a51c0223ff8f8822c16f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-30 16:09:34.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-31 16:36:02.000000000","tz":120},"subject":"Add an active wait during the port provisioning event","message":"Add an active wait during the port provisioning event\n\nIn ML2/OVN, during a live-migration process, it could\nhappend that the port provisioning event is received before\nthe port binding has been updated. That means the port has\nbeen created in the destination host and the event received\n(this event will remove any pending provisioning block). But\nthe Nova port binding request has not arrived yet, updating\nthe port binding registers. Because the port is considered\n\"not bound\" (yet), the port provisioning doesn\u0027t set the port\nstatus to ACTIVE.\n\nThis patch creates an active wait during the port provisioning\nevent method. If the port binding is still \"unbound\", the method\nretries the port retrieval several times, giving some time to the\nport binding request from Nova to arrive.\n\nCloses-Bug: #1988199\nChange-Id: I50091c84e67c172c94ce9140f23235421599185c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c3435489c750a64a31ade32c2e5af9abcc2c395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c3435489c750a64a31ade32c2e5af9abcc2c395"}]},"branch":"refs/heads/master"},"91f0864dc0ccf0f67be7162f011706dbc6383cb3":{"kind":"REWORK","_number":4,"created":"2022-09-12 11:39:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/57/855257/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/57/855257/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/855257/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/57/855257/4"}}},"commit":{"parents":[{"commit":"653949808d2d5078102e2f6e2a70643071006b5b","subject":"Fix a wrong assertion method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/653949808d2d5078102e2f6e2a70643071006b5b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-30 16:09:34.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-31 21:20:37.000000000","tz":120},"subject":"Add an active wait during the port provisioning event","message":"Add an active wait during the port provisioning event\n\nIn ML2/OVN, during a live-migration process, it could\nhappend that the port provisioning event is received before\nthe port binding has been updated. That means the port has\nbeen created in the destination host and the event received\n(this event will remove any pending provisioning block). But\nthe Nova port binding request has not arrived yet, updating\nthe port binding registers. Because the port is considered\n\"not bound\" (yet), the port provisioning doesn\u0027t set the port\nstatus to ACTIVE.\n\nThis patch creates an active wait during the port provisioning\nevent method. If the port binding is still \"unbound\", the method\nretries the port retrieval several times, giving some time to the\nport binding request from Nova to arrive.\n\nCloses-Bug: #1988199\nChange-Id: I50091c84e67c172c94ce9140f23235421599185c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91f0864dc0ccf0f67be7162f011706dbc6383cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91f0864dc0ccf0f67be7162f011706dbc6383cb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
