)]}'
{"id":"openstack%2Fneutron~855335","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I296003a936db16dd3a7d184ec44908fb3f261876","project":"openstack/neutron","branch":"stable/yoga","topic":"change-821401-stable/yoga","attention_set":{},"removed_from_attention_set":{"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2022-08-31 16:40:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I296003a936db16dd3a7d184ec44908fb3f261876","subject":"[ovn] Specify port type if it\u0027s a router port when updating","status":"MERGED","created":"2022-08-31 09:14:28.000000000","updated":"2022-08-31 16:42:40.000000000","submitted":"2022-08-31 16:40:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855335-change-821401-stable/yoga","cherry_pick_of_change":821401,"cherry_pick_of_patch_set":9,"meta_rev_id":"05ae33fc0d399df39a7e5f836781c76e231e7b48","_number":855335,"virtual_id_number":855335,"owner":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 16:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-08-31 14:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-08-31 09:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2022-08-31 14:45:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-31 09:28:32.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-08-31 10:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-31 11:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 11:44:53.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-31 14:45:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2c49f8c172a220c96dfdd59f9c40ff338ae7541f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2022-08-31 09:14:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c7267f2f526fa272e8d003c778ab5cbd15aff09f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2022-08-31 09:22:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"78b1780386605b376769a549b5be6450e3a288a4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-31 09:28:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bfad321a79fcb769093670387726a8d2a9b9263","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-31 09:31:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fe2604d9fbab4e67660ba36da28f89128f09c85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 10:01:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f8659f62368446d8e93019efea9285a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/625ba22a423445bca0d1dd8e027b3c7d : SUCCESS in 32m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f570c5ddbb94b1f8d5f7a592f5ac24c : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4eb24aed9ed0c58eb34c9f2b5609d0ca3ae8a454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 11:09:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c43333fd3e54a65990d37b219e5ab96\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/525b38a190ba4c80b8567ff86d48eddf : SUCCESS in 1h 01m 57s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/212ded6cfde84f749ea61998c5a12bde : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/2450e278d91547658d412720e6ac5eef : SUCCESS in 48m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f8de49ef1674d40a335c8e78792706a : SUCCESS in 39m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57a95cde3fb242feaf429ce5f9ccd51e : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60c37e2a066a4c27917d6a32e0b56ccc : SUCCESS in 26m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5395862da364b7dac31f03eb07604e6 : SUCCESS in 24m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b9455e71e743d59e88f4ab40317f96 : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b7366d9b56c40768f4b350fc5014794 : SUCCESS in 54m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7bbe419dbf7b48f7b591c34abadbf5de : SUCCESS in 1h 43m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2a5ad669d944301ac53f35fa6d6875c : SUCCESS in 1h 06m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f01e7726b00b47fdb705fdbb0188c6f5 : SUCCESS in 1h 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c31e3da687464c37bc2669048e6d1e3c6f9f02da","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-31 11:44:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855335/2 : SUCCESS in 2h 21m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cfd62ee77620e4eb4c2608dc731db914ed0ec003","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-31 14:45:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8d7f358a6af3b82e12a525424f066c81a3ed005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 14:46:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63b715bfcf27a5b37145496440a3901f7528660a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:40:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d574c4ebc264c6b909514a5b6508da0\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/d295f06ba2c24ef19c301c8dc6fb0a74 : SUCCESS in 50m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b162b44235472d8ffa6f65df97be6a : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c173665a144742979dfa2d3eea3d52f9 : SUCCESS in 28m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d211d54f40614582b0dc849391334c0c : SUCCESS in 33m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ebc22e9581440f6ba71e715b24fee60 : SUCCESS in 16m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1d096fcf0cb45d29656cc0909082b0c : SUCCESS in 1h 18m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0278c5c3a1364c80b1d5ab7e4b894ff2 : SUCCESS in 1h 44m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c1b4335b65f4eae90c602dbbcfeb761 : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":2},{"id":"49c457fa1f44b6d3f8ae253c98ca35c74d381528","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:40:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"05ae33fc0d399df39a7e5f836781c76e231e7b48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:42:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/093a6ab893994ece9bd0f681b6940663\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd768b603ea54063893fc5a92f0f9c47 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d036d59fe53c1654fa7749b3ed7a17203f9f903","revisions":{"813cb7f7002c47546b3f39934186dd221ce798d2":{"kind":"REWORK","_number":1,"created":"2022-08-31 09:14:28.000000000","uploader":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"ref":"refs/changes/35/855335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/855335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/855335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/855335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/855335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/855335/1"}}},"commit":{"parents":[{"commit":"3b3b8eb5832090ee64eea41930cc082ad9d317b5","subject":"Merge \"[OVN] Remove ACLs with remote SG during deletion of SG\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b3b8eb5832090ee64eea41930cc082ad9d317b5"}]}],"author":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2021-12-16 20:53:27.000000000","tz":60},"committer":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2022-08-31 09:14:28.000000000","tz":0},"subject":"[ovn] Specify port type if it\u0027s a router port when updating","message":"[ovn] Specify port type if it\u0027s a router port when updating\n\nIn order to avoid multiple LogicalSwitchPortUpdateUpEvent and\nLogicalSwitchPortUpdateDownEvent is mandatory to specify the\nrouter port type when udpating the port.\n\nIf the port is not specified when updating the port, the\ntransactions will trigger a modification on the ovnnb db\nthat will set the port status to down[0]. Triggering an unnecessary\nDownEvent followed by another UpEvent. Those unnecessary event\nmost likely will trigger a revision conflict.\n\n[0] - https://github.com/ovn-org/ovn/blob/\n4f93381d7d38aa21f56fb3ff4ec00490fca12614/northd/northd.c#L15604\n\nCloses-Bug: #1955578\nChange-Id: I296003a936db16dd3a7d184ec44908fb3f261876\n(cherry picked from commit 8c482b83f2cf6f5495f4df2e5698595db704798d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/813cb7f7002c47546b3f39934186dd221ce798d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/813cb7f7002c47546b3f39934186dd221ce798d2"}]},"branch":"refs/heads/stable/yoga"},"3d036d59fe53c1654fa7749b3ed7a17203f9f903":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-31 09:22:08.000000000","uploader":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"ref":"refs/changes/35/855335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/855335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/855335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/855335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/855335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/855335/2"}}},"commit":{"parents":[{"commit":"3b3b8eb5832090ee64eea41930cc082ad9d317b5","subject":"Merge \"[OVN] Remove ACLs with remote SG during deletion of SG\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b3b8eb5832090ee64eea41930cc082ad9d317b5"}]}],"author":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2021-12-16 20:53:27.000000000","tz":60},"committer":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2022-08-31 09:21:25.000000000","tz":120},"subject":"[ovn] Specify port type if it\u0027s a router port when updating","message":"[ovn] Specify port type if it\u0027s a router port when updating\n\nIn order to avoid multiple LogicalSwitchPortUpdateUpEvent and\nLogicalSwitchPortUpdateDownEvent is mandatory to specify the\nrouter port type when udpating the port.\n\nIf the port is not specified when updating the port, the\ntransactions will trigger a modification on the ovnnb db\nthat will set the port status to down[0]. Triggering an unnecessary\nDownEvent followed by another UpEvent. Those unnecessary event\nmost likely will trigger a revision conflict.\n\n[0] - https://github.com/ovn-org/ovn/blob/\n4f93381d7d38aa21f56fb3ff4ec00490fca12614/northd/northd.c#L15604\n\nConflicts:\n  neutron/common/ovn/constants.py\n\nCloses-Bug: #1955578\nChange-Id: I296003a936db16dd3a7d184ec44908fb3f261876\n(cherry picked from commit 8c482b83f2cf6f5495f4df2e5698595db704798d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d036d59fe53c1654fa7749b3ed7a17203f9f903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d036d59fe53c1654fa7749b3ed7a17203f9f903"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
