)]}'
{"id":"openstack%2Fneutron~855456","triplet_id":"openstack%2Fneutron~master~I313289f25f9eb57b4fd974a874d3c696a4c996eb","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"35238":{"account":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"last_update":"2022-09-07 16:04:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I313289f25f9eb57b4fd974a874d3c696a4c996eb","subject":"Add validation to ip_version field","status":"MERGED","created":"2022-09-01 12:08:42.000000000","updated":"2022-09-07 16:07:08.000000000","submitted":"2022-09-07 16:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855456","meta_rev_id":"3dbe458b89d31ea961cc8a47c7cbef7cdfa25391","_number":855456,"virtual_id_number":855456,"owner":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-09-02 17:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-07 16:04:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-07 13:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-09-07 13:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-09-07 13:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-01 12:48:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-01 12:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-01 14:07:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-01 14:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-07 13:12:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-07 13:54:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"be3334a6a1a588054f0069a09f371d40531f518a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"date":"2022-09-01 12:08:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18f1a3e5d256d264c55cf58c50a79acdb834d7af","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"date":"2022-09-01 12:12:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"af94791a588a3711aad76b2f9c223f5d0f9f362d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-01 12:48:52.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/56/855456/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d10f3458c015b86391d78532db075f73d640a0a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 12:52:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f5ca22b1a64994a28679a8932b5d66\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/54dcf56b66d5493d9a566a4a8e26369d : FAILURE in 33m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13a6e7b23e36430a8bf9788a125ea899 : FAILURE in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1cd73465a4ecff739721c01e7cf7f3beabe3593","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-01 14:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855456/2 : SUCCESS in 1h 54m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3b3cb5c204ab9c380b08e0330a2f89073d25a1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 14:10:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dd9f5554bbd43e6b6165ad9cbf1aff7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/49069d2d57844ab28a562682876be9c2 : SUCCESS in 1h 18m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/913e52816b6647e7b4793259dda0178a : SUCCESS in 1h 14m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac8ac792fffe49819d8f22514af86bde : SUCCESS in 1h 09m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f7e06e68d5946f4a1961c0779028e96 : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/694766a85a3349d3be3b1020c68b2878 : SUCCESS in 35m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d64328a13504e889de9c05d8048ed2c : FAILURE in 30m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a114471ad144038236c7f9564dcc0b : SUCCESS in 13m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1e5806d0641449ebcc4c16a9efa3590 : FAILURE in 26m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe86dd5929a64f8e886d040951464661 : FAILURE in 29m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a7d48884d344912bc80e947287bffa0 : FAILURE in 28m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5740b902d3244744a6853c1e1aa131a8 : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6bb5780b2dd46cfb1265b5d1ba4310c : SUCCESS in 1h 13m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/78afb743cdf84725a03be6a9e32d4bc4 : SUCCESS in 1h 51m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b706ad36a93e4e108a41d911e8376c2e : SUCCESS in 1h 28m 38s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a3741b211364eada38c62e5cd363e53 : SUCCESS in 1h 01m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6d4f95aa75a41e3a9c2ee703d5be5a4 : SUCCESS in 1h 08m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9cd8f956f10d42fd8447910a2786b6d1 : SUCCESS in 1h 18m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e7951f0fe58417c82c63dfe899ba1ec : SUCCESS in 1h 26m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd155efedfc44b80a902f6af42d8b0e0 : SUCCESS in 1h 01m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75a60a36ad114bed93869b49c41d52fe : SUCCESS in 1h 09m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee275286bcc6bba2f146877e2a041b3dfe1412c6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"date":"2022-09-01 14:45:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"70d7e1bd83cb1f54aaac7cf2eb6b0c61e4ce5d35","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-01 15:21:53.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/56/855456/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"510b066b1688c32f327d86946a829c434aeb2f4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 15:34:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ae27da43cf45afbd439e26f95e5f78\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05312e7bd824435c91a0fb5e49d3f979 : FAILURE in 47m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca8b4b2020774b4589499d74df46a469 : FAILURE in 45m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba6a513cb52412960e0a3227ce0c9fe6729b254b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 16:30:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b68621a2b5b644f38a8239154892aa26\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c64c4a49aafa4970983506b9bdd40b74 : SUCCESS in 1h 34m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0cc4e9e4dce34c49a8032381d936d258 : SUCCESS in 1h 41m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/81b1e31509194ac39db25fe3cc7a4ee7 : SUCCESS in 1h 22m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/854101b5503645168528345f38b523fd : SUCCESS in 1h 35m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e55afb45cc3473cab4c3f2ef06739eb : SUCCESS in 43m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cdad5afbf184f21800fbefd228eff4c : FAILURE in 35m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f72941a26f432a9cd7e5e40f28d6a6 : SUCCESS in 14m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/931b24c5db2345b3b8b592096fa741cc : FAILURE in 29m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28620bf37ef644fc8f8784d6432c72b7 : FAILURE in 29m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2455abed10b942ebb29521fd3915e16e : FAILURE in 26m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7641884625f14064a4f38ae434e6a9ff : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aede058f501f426799ae94d081a63ada : FAILURE in 55m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0afe153f99c24f1584567dc189ebe9be : SUCCESS in 1h 37m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b08898ed9d5146f6bcbe8eede191404a : FAILURE in 1h 17m 52s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dbad2041979944c7ba8f468e2f7b13b7 : SUCCESS in 59m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01cb6a2bc1864a9fa5a27ac0e5fc86bd : SUCCESS in 1h 00m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/77353cc59de0452484dff531cad6ba59 : SUCCESS in 1h 17m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/71bb9aba8608411ebc9c8ab7f3b8ece2 : SUCCESS in 1h 20m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92d9578172464c188047d63e1fef6e0b : SUCCESS in 1h 04m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b4681abe2561455b9f99c413c6d7a737 : SUCCESS in 1h 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b2d23bb2effa11277cb847f7423570a50e4f7bc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-01 16:46:24.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/855456/3 : FAILURE in 2h 00m 24s","accounts_in_message":[],"_revision_number":3},{"id":"e3d72625edd6f93a50bab111d6f3e8756b6ac47d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"date":"2022-09-02 16:29:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"4803b364b556ccdcfa5d212eb803fac9bac9172c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 17:08:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9da2f24b9fd48a7928ad222d6ffbc79\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db2766ac200e4e50bd91948b03a9175f : SUCCESS in 34m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35c566a994614beda545d8d2c9caeb47 : SUCCESS in 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c78ac34774854b032903e880bcb79d9f734a317f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-02 17:38:11.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/56/855456/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"e2fd04d85287c39c1508495478357a31c6afac48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 18:21:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cbbeb853cd24ea49edddb65e3c00009\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b3ab5ecaac3b404a809a61a62309d825 : SUCCESS in 1h 26m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2389414337ad49b3a1137a93a1a6bcbf : SUCCESS in 1h 16m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b1ff9b7c891b4c85b536eccd68aba956 : SUCCESS in 1h 13m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/637199f211ca48b39ed566b583e434b8 : SUCCESS in 1h 13m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ba0003651884873a87c9191d4ec6eec : SUCCESS in 32m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e009493a54d54347b9c4022a88e253cc : SUCCESS in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23bcbd8b06a54392a65af814ff85c9f2 : SUCCESS in 13m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f98078925b5546ccb793ed25ca1d1244 : SUCCESS in 23m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a2503fedf83448b81b2170c33e67883 : SUCCESS in 26m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b80b135094a48e0b075f3f5434bc3d2 : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce70d01db7524eba8dfe839d0a4fba6d : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d31de5b19b7545ed9c2dbe0286cc1c27 : SUCCESS in 1h 02m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/19c7e7a451244e499293470f349baa5b : SUCCESS in 1h 50m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/15ede5f07274443cb7c4d69e10c2ae9c : SUCCESS in 1h 19m 19s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e82f169bd2f4e72bac77df211d30780 : SUCCESS in 1h 00m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/352e2843cac54beea50750e9340c1fdf : SUCCESS in 1h 03m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/159a61366b7c4f769a96af5d4c3faa6d : SUCCESS in 1h 27m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dc2752ea7a8346d6bcbbf493a5410a52 : SUCCESS in 1h 13m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e62507942194b1bb4be0553096d61a2 : SUCCESS in 58m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/322eae89ae3b459980fb4fecd2078e3c : SUCCESS in 1h 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2137aa948829ace4bb34b1aa0241184c49d4af38","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-02 22:05:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855456/4 : SUCCESS in 1h 59m 30s","accounts_in_message":[],"_revision_number":4},{"id":"109cc4eee24a8a923dc1519e219ac2042d4e26a0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"date":"2022-09-05 07:06:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"83ac0ca211e17f3adfedb0ab386adcff448e4f09","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-07 13:12:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cd58613732e9b0cf063db1ff05f9df3754591f2c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-09-07 13:54:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d03cc925fcd486a929fef4f31e810b8ab9cb80c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 13:55:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"05ff337694a7ac8d18f0de6f9c4d2334cede090a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 16:04:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7837b349b463491c9dcf4b859f3f0046\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e55b63d5fadf400d87e4576c4fb8c5c5 : SUCCESS in 1h 11m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e198ab9a3f744cc99508e383d36b91bf : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2225860ea77346f8ac76d00a197dd149 : SUCCESS in 1h 36m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/69c24e994ad8465ab34bc97a35c241b4 : SUCCESS in 1h 29m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61813d0cb1f4cc5bb9122d750d4c371 : SUCCESS in 12m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13de5eb0b3254dd892a24d3e48762279 : SUCCESS in 32m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5d9644ec0884f63be2eb670381f33c4 : SUCCESS in 30m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a19fecefa74959a5e13c82c783389a : SUCCESS in 10m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2845c758c3c34181a8d0f4c37acdb534 : SUCCESS in 1h 01m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbe70914629a4277b36ad4e2db001128 : SUCCESS in 1h 35m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/babfaf0fdba4490d85b6fa9d3d40cd82 : SUCCESS in 1h 22m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2bf6b2424524a5a998e048a5e28d80d : SUCCESS in 1h 04m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/967bcd189366409c889ef6d61c0bab99 : SUCCESS in 57m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1bc2bcb47cc4ffda4dd762b4ed4919d : SUCCESS in 1h 21m 05s","accounts_in_message":[],"_revision_number":4},{"id":"4d0018bd7379442c3f597c0399c31d827ce3a67c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 16:04:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3dbe458b89d31ea961cc8a47c7cbef7cdfa25391","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 16:07:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb5c0914ae04f94a8d5c6e48af80403\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad096f1945a4aafab0945e2a137f60a : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"116bb13651da249cf13534c7a8c0961d81ea3d5d","revisions":{"a684eabfbcde9ffa11e7fce1e1c75da9640ca4cf":{"kind":"REWORK","_number":1,"created":"2022-09-01 12:08:42.000000000","uploader":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"ref":"refs/changes/56/855456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/855456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/855456/1"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-01 12:08:42.000000000","tz":0},"committer":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-01 12:08:42.000000000","tz":0},"subject":"Add validation to ip_version field","message":"Add validation to ip_version field\n\nThis patch adds validation to value of ip_version field in ExtraDhcpOpt object\n\nChange-Id: I313289f25f9eb57b4fd974a874d3c696a4c996eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a684eabfbcde9ffa11e7fce1e1c75da9640ca4cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a684eabfbcde9ffa11e7fce1e1c75da9640ca4cf"}]},"branch":"refs/heads/master"},"d3b5836c8a408e1833805fb598394c1b1647336d":{"kind":"REWORK","_number":2,"created":"2022-09-01 12:12:26.000000000","uploader":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"ref":"refs/changes/56/855456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/855456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/855456/2"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-01 12:08:42.000000000","tz":0},"committer":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-01 12:10:28.000000000","tz":0},"subject":"Add validation to ip_version field","message":"Add validation to ip_version field\n\nThis patch adds validation to value of ip_version field in ExtraDhcpOpt object\n\nChange-Id: I313289f25f9eb57b4fd974a874d3c696a4c996eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3b5836c8a408e1833805fb598394c1b1647336d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3b5836c8a408e1833805fb598394c1b1647336d"}]},"branch":"refs/heads/master"},"d09f470305f736b335fe1d9203f63f274dbbe056":{"kind":"REWORK","_number":3,"created":"2022-09-01 14:45:21.000000000","uploader":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"ref":"refs/changes/56/855456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/855456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/855456/3"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-01 12:08:42.000000000","tz":0},"committer":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-01 14:44:30.000000000","tz":0},"subject":"Add validation to ip_version field","message":"Add validation to ip_version field\n\nThis patch adds validation to value of ip_version field in ExtraDhcpOpt object\n\nChange-Id: I313289f25f9eb57b4fd974a874d3c696a4c996eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d09f470305f736b335fe1d9203f63f274dbbe056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d09f470305f736b335fe1d9203f63f274dbbe056"}]},"branch":"refs/heads/master"},"116bb13651da249cf13534c7a8c0961d81ea3d5d":{"kind":"REWORK","_number":4,"created":"2022-09-02 16:29:24.000000000","uploader":{"_account_id":35238,"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","username":"everzakov"},"ref":"refs/changes/56/855456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/855456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/855456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/855456/4"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-01 12:08:42.000000000","tz":0},"committer":{"name":"Efim Verzakov","email":"verzakov.efim@huawei-partners.com","date":"2022-09-02 16:29:13.000000000","tz":0},"subject":"Add validation to ip_version field","message":"Add validation to ip_version field\n\nThis patch adds validation to value of ip_version field in ExtraDhcpOpt object\n\nChange-Id: I313289f25f9eb57b4fd974a874d3c696a4c996eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/116bb13651da249cf13534c7a8c0961d81ea3d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/116bb13651da249cf13534c7a8c0961d81ea3d5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
