)]}'
{"id":"openstack%2Fneutron~855539","triplet_id":"openstack%2Fneutron~master~I7aef4a6dde8291f33e32ed42bae4a274ea45553e","project":"openstack/neutron","branch":"master","topic":"fix_dns_any","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-09-01 20:44:12.000000000","reason":"Change was abandoned"},"7130":{"account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"last_update":"2022-09-01 20:44:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7aef4a6dde8291f33e32ed42bae4a274ea45553e","subject":"neutron l3 db notifications send 0.0.0.0 dns","status":"ABANDONED","created":"2022-09-01 17:31:55.000000000","updated":"2022-09-02 07:52:41.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"818a1e646647812ccaf21dba29806bf9ce9298bd","_number":855539,"virtual_id_number":855539,"owner":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-01 17:38:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-09-01 18:15:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-01 19:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-01 19:53:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-01 23:13:43.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"1c2a756ebd52c2f89930baa79e622eeeed7cf022","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-09-01 17:31:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5a080b254feae1622b94ae3de51ef9e48f2ca3e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-01 17:38:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b49ef4d6ee8c9dc000fb9cd6a58e256cf361a70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 18:15:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216bdac74cb44fabbfc958e56f2eb16d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/adad6f42baa5424e909e90c9fe97317b : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c470d7c3ec04426387f38f0a9d65b21e : SUCCESS in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb1ca2996f92f201280b4e41331017b4a968f042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 19:23:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaaebe80f8b04a0bb5fc0c227a55b783\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d50284a3a27c4745a7578fe4542dc920 : SUCCESS in 1h 17m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03dab809469047de8a7f243237699153 : SUCCESS in 35m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f8798d35cd8464ea5024458f102670a : SUCCESS in 44m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5057be42343542d6bb44ce005d5c89f2 : FAILURE in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15f6b1a98c584a66b834b29523fe0f59 : SUCCESS in 29m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3368cbe4128e4bc284ff5d5b5a2d4a63 : SUCCESS in 30m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2e75dc0cb2b4ec9992c634ecd76efd4 : SUCCESS in 26m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92962e95730549d68ed911f20620c0b6 : SUCCESS in 8m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/10104be7e0e14933856b5542f8c24b91 : SUCCESS in 57m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/de9acbe843a94f5f89d06fb2e0cc41be : SUCCESS in 1h 45m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ecf86ea636949ada5e2d3da62df9c73 : SUCCESS in 1h 04m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc090e3d2a8b448ea855216f3c5307cc : SUCCESS in 1h 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28b854c53f8ba67f54c66700342c4a14b1040e8e","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-09-01 19:45:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d1daad87385fba96c38d81344b64b88b20e24b8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-01 19:53:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855539/1 : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":1},{"id":"11737b73c11a5ddab8549b58eba59d0f8844774f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-09-01 20:18:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea0c85d3a1aa9e7786387a934a2bac2545f36fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-09-01 20:22:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32e9d337f8cf5f0518543249a76d5971136a37ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 20:33:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e21d9bbb164cc19de33bbf465409a7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8085e0eb984248c7ac57a12ce70eab21 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af7041ab5b154698b65a7fe0840f9eb2 : FAILURE in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8649db577d22381dfad02ce49eafa33b20877d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-09-01 20:39:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b411cb840f4e7f71eb58e508bec73e8461ded819","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2022-09-01 20:44:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"9bf3167dbaf1e946eef67d77ee5a55e8009fc646","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-01 23:13:43.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/39/855539/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6715ace542491b52fba9ac9985da192034f49b58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-01 23:37:22.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/39/855539/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"d7d036f036085c27e363daf3a805b193a6848a09","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-02 00:11:43.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/39/855539/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"818a1e646647812ccaf21dba29806bf9ce9298bd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-02 07:52:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"18f40d6b38f5bf36a87a2ce6749641aaac6b67fc","revisions":{"ce394be7c3e51362e55c3368c9246dab532ac364":{"kind":"REWORK","_number":1,"created":"2022-09-01 17:31:55.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/39/855539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/855539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/855539/1"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 17:30:13.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 17:30:13.000000000","tz":-240},"subject":"Return no DNS if --no-dns-nameservers is used","message":"Return no DNS if --no-dns-nameservers is used\n\nReturn no DNS if --no-dns-nameservers is used as the current behavior\nappears to be to return one of the available DNS servers from the\nconfiguration when the user requested to provide none.\n\nChange-Id: I7aef4a6dde8291f33e32ed42bae4a274ea45553e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce394be7c3e51362e55c3368c9246dab532ac364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce394be7c3e51362e55c3368c9246dab532ac364"}]},"branch":"refs/heads/master"},"00d5f22c72d171947f8f1a847af3301991e18965":{"kind":"REWORK","_number":2,"created":"2022-09-01 20:18:20.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/39/855539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/855539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/855539/2"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 17:30:13.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 20:16:34.000000000","tz":-240},"subject":"neutron l3 db notifications send 0.0.0.0 dns","message":"neutron l3 db notifications send 0.0.0.0 dns\n\nneutron l3 db notifications send 0.0.0.0 dns but it is a special\ncase where it should send no nameservers through the DHPC service.\n\nChange-Id: I7aef4a6dde8291f33e32ed42bae4a274ea45553e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00d5f22c72d171947f8f1a847af3301991e18965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00d5f22c72d171947f8f1a847af3301991e18965"}]},"branch":"refs/heads/master"},"92166cfd3449897744cac9151e55fdf502362328":{"kind":"REWORK","_number":3,"created":"2022-09-01 20:22:44.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/39/855539/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/855539/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/855539/3"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 17:30:13.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 20:22:40.000000000","tz":-240},"subject":"neutron l3 db notifications send 0.0.0.0 dns","message":"neutron l3 db notifications send 0.0.0.0 dns\n\nneutron l3 db notifications send 0.0.0.0 dns but it is a special\ncase where it should send no nameservers through the DHPC service.\n\nChange-Id: I7aef4a6dde8291f33e32ed42bae4a274ea45553e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92166cfd3449897744cac9151e55fdf502362328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92166cfd3449897744cac9151e55fdf502362328"}]},"branch":"refs/heads/master"},"18f40d6b38f5bf36a87a2ce6749641aaac6b67fc":{"kind":"REWORK","_number":4,"created":"2022-09-01 20:39:58.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/39/855539/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/855539/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/855539/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/855539/4"}}},"commit":{"parents":[{"commit":"4b83bf462d14824caf5f07569377de4276ed99c0","subject":"Merge \"Migration revert plan\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b83bf462d14824caf5f07569377de4276ed99c0"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 17:30:13.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2022-09-01 20:39:54.000000000","tz":-240},"subject":"neutron l3 db notifications send 0.0.0.0 dns","message":"neutron l3 db notifications send 0.0.0.0 dns\n\nneutron l3 db notifications send 0.0.0.0 dns but it is a special\ncase where it should send no nameservers through the DHPC service.\n\nChange-Id: I7aef4a6dde8291f33e32ed42bae4a274ea45553e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18f40d6b38f5bf36a87a2ce6749641aaac6b67fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18f40d6b38f5bf36a87a2ce6749641aaac6b67fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
