)]}'
{"id":"openstack%2Fneutron~855867","triplet_id":"openstack%2Fneutron~master~Id527e7b5240240f5299a6f163bc124002c64fa18","project":"openstack/neutron","branch":"master","topic":"improve-neutron-ci","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-09-07 08:22:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id527e7b5240240f5299a6f163bc124002c64fa18","subject":"Remove \"fetch_journal_log\" role as it\u0027s not needed","status":"MERGED","created":"2022-09-05 11:02:27.000000000","updated":"2022-09-07 08:23:39.000000000","submitted":"2022-09-07 08:22:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855867-improve-neutron-ci","meta_rev_id":"03929b597f7a29cf4eaf1180a4c9b6f6746f2f03","_number":855867,"virtual_id_number":855867,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-07 08:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-09-06 15:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-09-07 06:42:07.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-09-07 06:42:07.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-05 12:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-05 12:41:15.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-06 15:39:54.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-09-07 06:42:07.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"d90e58991b6954c349db217e93d5c8c2ff30370e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-05 11:02:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff209f006867dc3a059c6438f19cc6e6bb1accca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 12:39:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ee2042e06d410fa2af688a99d4086c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/83c4704e0fa64685b87b24c26d6fb854 : SUCCESS in 1h 17m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3ecf394390a246d794dd6b05554cbc4b : SUCCESS in 1h 22m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/171446df78e54ec088059b5d9af80a9a : SUCCESS in 1h 13m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a3a2859166142cc8501471e5679e180 : SUCCESS in 1h 08m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f65c11c7b10447a87d37c07e3944ece : SUCCESS in 35m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d77111ed2b465ab3eb4b1e110ceaf2 : SUCCESS in 13m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794627de5ce44b4fb952483ef91b8c4f : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9e89c502b63462fae5d579c390e1398 : SUCCESS in 1h 11m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/01b632ae3e6f4363926ab675c9ef460f : SUCCESS in 1h 24m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/68d46b65498341d28e9192efab6b7495 : SUCCESS in 1h 19m 27s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f8af65951ff643e39de5b4f9bb34d6cc : SUCCESS in 1h 14m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f349989f9f6a4a65ba4f23304f99f39f : SUCCESS in 52m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8fded102ef23468f968075b030815181 : SUCCESS in 1h 29m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e776d2bf86e425a8d028e39b8bedc81 : SUCCESS in 1h 14m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f16158651e04c33a9071824daf60e8f : SUCCESS in 52m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5ecf115dea684ea38e949ed515a7f816 : SUCCESS in 1h 22m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e0e85cce70b46e4479ccf96710a131a8e20baf0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-05 12:41:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/855867/1 : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e7df15ad15c619ad8de44c91a5d8b9dd124d0c95","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-06 15:39:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3d865b19195a53920559b9f9b809e34d4d166d1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-09-07 06:42:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a47801685b723e24e326643ae589bc1ad875bcb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 06:42:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a81d8548dacd51c109d5e383e843a344af8bcac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 08:22:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61de08741a5840e885cbb6e0e98fd283\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c57b6fe24c364dc29b810ebfb1448a5c : SUCCESS in 1h 18m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9bb3ab4db6b84dc2b1662a848afc84b8 : SUCCESS in 1h 26m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b67fc70720cd467ca3601fd90b90b622 : SUCCESS in 1h 19m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ebd0ea5224c4a10ad45ab9072cf40ed : SUCCESS in 1h 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1660cae25834ddd8a3d331cde256960 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df735a6a2d104f359d2062e0a2a27a87 : SUCCESS in 11m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e608c51835b94cc6aff30701f1354042 : SUCCESS in 1h 02m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/29137d8efe014d5a88bbbd1789bd1a58 : SUCCESS in 1h 38m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dd78a2e1ee3a40a2bb516fbdec6aab62 : SUCCESS in 1h 23m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0278aeb171de494fa6400843cc001566 : SUCCESS in 59m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0cfe7ce5b2c442a8a13052548982ee66 : SUCCESS in 1h 04m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abf6d4b040d34011ababa6b9da60e3d4 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cb5b3668dd8825d50175c11870241d8bb0838acb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 08:22:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"03929b597f7a29cf4eaf1180a4c9b6f6746f2f03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 08:23:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4cd8b4d048b472d9dfc57b0d3a3d149\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3646a6467d6446bb5a4eb151b56d542 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58f80b4ec7c5053266292c201636b3ae3f03e9d5","revisions":{"58f80b4ec7c5053266292c201636b3ae3f03e9d5":{"kind":"REWORK","_number":1,"created":"2022-09-05 11:02:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/67/855867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/855867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/855867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/855867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/855867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/855867/1"}}},"commit":{"parents":[{"commit":"6db791ac0fde866a61d8cfbf9679dbd02d327635","subject":"ml2/mech: bind port on segment that subnet belongs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6db791ac0fde866a61d8cfbf9679dbd02d327635"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-09-05 10:56:30.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-09-05 10:56:30.000000000","tz":120},"subject":"Remove \"fetch_journal_log\" role as it\u0027s not needed","message":"Remove \"fetch_journal_log\" role as it\u0027s not needed\n\nWe don\u0027t need to run that ansible role in the functional and fullstack\njobs as devstack already stores archive with journal log from the test\nnode.\nIt is available in the \"devstack.journal.gz\" file in logs.\n\nChange-Id: Id527e7b5240240f5299a6f163bc124002c64fa18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58f80b4ec7c5053266292c201636b3ae3f03e9d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58f80b4ec7c5053266292c201636b3ae3f03e9d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
