)]}'
{"id":"openstack%2Fneutron~856034","triplet_id":"openstack%2Fneutron~master~I0bec5ceb0dac5d233ca371865e99f93cbe417b68","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"12825":{"account":{"_account_id":12825,"name":"Szymon Wróblewski","email":"szymon.wroblewski@ovhcloud.com","username":"bluex"},"last_update":"2022-09-08 06:36:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0bec5ceb0dac5d233ca371865e99f93cbe417b68","subject":"Fix test_nova_send_events_* tests","status":"MERGED","created":"2022-09-06 09:40:59.000000000","updated":"2022-09-08 06:37:32.000000000","submitted":"2022-09-08 06:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856034","meta_rev_id":"80f7ca63949e240b01fbd5b077ebe72233bb593d","_number":856034,"virtual_id_number":856034,"owner":{"_account_id":12825,"name":"Szymon Wróblewski","email":"szymon.wroblewski@ovhcloud.com","username":"bluex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-09-06 10:16:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-08 06:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-08 06:03:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-09-07 13:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-08 06:03:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 10:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 10:16:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 11:29:03.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-07 13:02:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-08 06:03:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"d7dee19a2599e96a972ac03d9c73bc24cea5e91c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12825,"name":"Szymon Wróblewski","email":"szymon.wroblewski@ovhcloud.com","username":"bluex"},"date":"2022-09-06 09:40:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ac3271b40d70274b8617b1d74c02404c9b940e1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12825,"name":"Szymon Wróblewski","email":"szymon.wroblewski@ovhcloud.com","username":"bluex"},"date":"2022-09-06 09:41:16.000000000","message":"Topic retry removed","accounts_in_message":[],"_revision_number":1},{"id":"b17a56caa8fb9f362e0095d0f8e136df0e9b0b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 10:14:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5731237b0a59427ba1546cf4c304a9ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e2e5da5aa4477e862d5b00e54a8c15 : SUCCESS in 32m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbbe4e429ede41ea8ff470366c5594af : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b843f866f87458f90c4c4ab326c2779 : SUCCESS in 32m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ba736eb3bbd4036b4686a2954c4dcea : SUCCESS in 22m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd6ae31c9b4e4608af11185eb81bdddc : SUCCESS in 23m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff59781c2803496fb0c7f6cfe7301286 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9b1cd6cda2c140930c444d62296724b2b6bf33af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-06 10:16:21.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/34/856034/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b03f7e810a6fcd163437fc75dfbe3f75e2de859a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 10:29:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca5d6980aae465792379b74f359be4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21b7cc573d984a1482ddb3822736bacd : SUCCESS in 46m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcc624f961464656aebaab7640b4eb8e : SUCCESS in 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f46f111228d5bf93dee2c5040ef41bf04f838803","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-06 11:29:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/856034/1 : SUCCESS in 1h 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8bb4e7a2faeadd3aef9a6620a4c5d3e5ac8a5787","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-07 13:02:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f5895db7874fa70113c548bcd656393d2f71b1f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-08 06:03:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7bfa79c3010feae94cccd185125e18cf10c4dd46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:03:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4885770ec8a6647273c981e8ebf27f76f9289570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:36:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fae50f8f56db40bc95bab512ff95f7a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff5a96587764e52a20f07adf934ad65 : SUCCESS in 13m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cd684d366b64efe81c98f7bee832d57 : SUCCESS in 26m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f9956ec177f4717b5839a26984156c8 : SUCCESS in 25m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a21a88c1014d7190c35259d429022a : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"94816e4465813f55b1130f62d31b93d8cf2c9bf6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:36:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80f7ca63949e240b01fbd5b077ebe72233bb593d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:37:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/550a428aa897488e95fdbe2215001fc5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7ff28bb9924229a25665aa4aca3117 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9081ce369e6c2cb4c87bd8b16d866ed861e83864","revisions":{"9081ce369e6c2cb4c87bd8b16d866ed861e83864":{"kind":"REWORK","_number":1,"created":"2022-09-06 09:40:59.000000000","uploader":{"_account_id":12825,"name":"Szymon Wróblewski","email":"szymon.wroblewski@ovhcloud.com","username":"bluex"},"ref":"refs/changes/34/856034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/856034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/856034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/856034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/856034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/856034/1"}}},"commit":{"parents":[{"commit":"cd475f9af898b81d98b3e0d3f55b94ea653c193c","subject":"Retry connections to Nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd475f9af898b81d98b3e0d3f55b94ea653c193c"}]}],"author":{"name":"Szymon Wroblewski","email":"szymon.wroblewski@ovhcloud.com","date":"2022-09-06 09:40:10.000000000","tz":120},"committer":{"name":"Szymon Wroblewski","email":"szymon.wroblewski@ovhcloud.com","date":"2022-09-06 09:40:10.000000000","tz":120},"subject":"Fix test_nova_send_events_* tests","message":"Fix test_nova_send_events_* tests\n\nmock().foo.bar and mock.foo.bar generate different mocks.\nFix mocks to reflect actual use of nova client and\nadd assert to verify create function was actually called.\n\nChange-Id: I0bec5ceb0dac5d233ca371865e99f93cbe417b68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9081ce369e6c2cb4c87bd8b16d866ed861e83864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9081ce369e6c2cb4c87bd8b16d866ed861e83864"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
