)]}'
{"id":"openstack%2Fneutron~856039","triplet_id":"openstack%2Fneutron~master~Ie167e42ec876b77a38030f4255d8669a8cf32796","project":"openstack/neutron","branch":"master","topic":"improve-neutron-ci","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-09-07 08:38:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie167e42ec876b77a38030f4255d8669a8cf32796","subject":"DNM Just run small subset of the functional jobs to test new role","status":"ABANDONED","created":"2022-09-06 10:57:40.000000000","updated":"2022-09-07 08:38:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33f5ce198437905babc062788398316b8e8e81a0","_number":856039,"virtual_id_number":856039,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-06 12:29:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 11:39:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-06 12:29:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 12:57:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"9404b16a60d545abfdce1496c68fddf728a83bd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-06 10:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a767aa80177526f3a810ee05c99f364a58aeea81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 11:39:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d732882552294f7c9844d21bb1e8f867\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1954755bae64f32ac5b558749de9efe : SUCCESS in 40m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43b4bf3fca7b4c8c812cf0943e3827ba : SUCCESS in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"332483e87a8ca1551492fa96135cdd05e4115987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 12:29:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/358d70f8ec2a43e4a0568911debf4be1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cacb3b455c244c629293e4a15bf3626a : SUCCESS in 44m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ebee939d1a942e9b1a03773c4fa3745 : SUCCESS in 13m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a8a39f616374319a3829d8a57424379 : SUCCESS in 28m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c35bea0cc55f4bd686489a65fd21453b : SUCCESS in 25m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fbff46895324c949f5b87d72e99c63b : SUCCESS in 28m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b9a57d6c33460999f7edfc18df028b : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2f2951b934546078c5a9f9b4f821359 : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3b6f4f43ca646daa207782a2953aeac : POST_FAILURE in 21m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/58d02ac5d3f14c20849c4403933eb109 : POST_FAILURE in 1h 29m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e4a6a979462888e0a4d895146d854649420c7730","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-06 12:57:01.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/856039/1 : FAILURE in 1h 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"33f5ce198437905babc062788398316b8e8e81a0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-07 08:38:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5152f7b09b2da076e3869400d1a50a7de020201b","revisions":{"5152f7b09b2da076e3869400d1a50a7de020201b":{"kind":"REWORK","_number":1,"created":"2022-09-06 10:57:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/856039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/856039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/856039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/856039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/856039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/856039/1"}}},"commit":{"parents":[{"commit":"17bb87cdc433676b56fd51efa7665cb238b61736","subject":"Add new role \"prepare_functional_tests_logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17bb87cdc433676b56fd51efa7665cb238b61736"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-09-06 10:57:13.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-09-06 10:57:13.000000000","tz":120},"subject":"DNM Just run small subset of the functional jobs to test new role","message":"DNM Just run small subset of the functional jobs to test new role\n\nChange-Id: Ie167e42ec876b77a38030f4255d8669a8cf32796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5152f7b09b2da076e3869400d1a50a7de020201b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5152f7b09b2da076e3869400d1a50a7de020201b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
