)]}'
{"id":"openstack%2Fneutron~857773","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I9d8db215ba19ea88bc615b7b30084113a3c953d4","project":"openstack/neutron","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-09-19 17:30:45.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-09-19 17:30:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d8db215ba19ea88bc615b7b30084113a3c953d4","subject":"Fix indentation issue in wait_for_change override","status":"MERGED","created":"2022-09-15 03:20:08.000000000","updated":"2022-09-19 17:34:11.000000000","submitted":"2022-09-19 17:30:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"857773","cherry_pick_of_change":857802,"cherry_pick_of_patch_set":1,"meta_rev_id":"47cfe44fc10adeaad07bcad47afccf449cb365b8","_number":857773,"virtual_id_number":857773,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 17:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-09-19 11:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-09-19 16:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-09-19 16:11:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-15 04:24:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-15 04:29:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-15 05:16:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-15 12:57:48.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-09-15 15:06:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-09-15 17:44:53.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2022-09-19 10:53:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-19 11:06:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-09-19 16:11:52.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"6f937dec7a7175645d8ba7470bd79d57c07ce62e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-09-15 03:20:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ee1c2f0e06837b3f08d3af27ca0604a4a9bf52b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 04:24:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b091be6d56413f9313657c54ad3c90\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/849c1eacac5b48da85d177bdc2ded8e6 : SUCCESS in 34m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a41b67b349484d21bd338030686067a4 : SUCCESS in 1h 02m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a4909d9e62e985116d8dd2e7f32c7eb54533795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 04:29:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c221c61c8a6c416280e45e2457281ede\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/578f00d9c3f4459aa3ebc8aa2b916280 : SUCCESS in 41m 14s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/d9e0ecf5a96544a99ff9da50031b4870 : SUCCESS in 56m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2014a6fbf85d4bacaa7f2b20af72c077 : SUCCESS in 31m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b756d7a7d5c44e395f57dc56897d200 : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c72b352e425642d6bdb426d16b74b85d : SUCCESS in 22m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf729bfff4194c2e9d38a837b8c271e3 : SUCCESS in 23m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79188bb84aa64fd4bcc416fddc8751ee : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6b42fd0c336460b8da78de766dd6e44 : SUCCESS in 47m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b76983ad112143bca9e1f19fe0f1f060 : SUCCESS in 55m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54af3dd1fb984839a8cb7603adc5ee16 : SUCCESS in 1h 08m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85594f951e7f21ad8106f1155ffc305d2cc7e04d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-15 05:16:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/857773/1 : SUCCESS in 1h 56m 08s","accounts_in_message":[],"_revision_number":1},{"id":"434e91bba718dd4414c8d6327b9ee427304ae60c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-09-15 12:57:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f64515aaceebf6570a1f077d3c7ce39718a4384","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-15 15:06:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0cc9f1384a8960c050217ae5dd2f7ccb6b00a76d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-09-15 17:44:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e299583635bb3701b5635a0c8d4beba9236ac6ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-19 10:53:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80f61d4efc4de83cdc2205c63c40a4e9b3f78201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-19 11:02:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f92b0c84a9f7298b24e49139fe47f9c7ac2c0905","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-19 11:06:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3796fb2defe9b5c4e4d38a1dbd3457c6a025e836","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 11:45:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98bd5aa846334f0d8bfb74239d95dc5e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35734efcfa5b4ed2ac619142f4a5865f : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/229a0fcf225f4c1795a4615f53d6f868 : SUCCESS in 37m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dde4b9a2dea47e007c82aee80b33d44e28346c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 12:27:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1da44dc46bc45259eab6f4fe2cbe430\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/1021bd743e5d44d89471414efd1ef466 : SUCCESS in 40m 58s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/d8448cce143840bfab2d0f452ea2a0e5 : SUCCESS in 58m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded75f57817346bf968b55c7e8c002a5 : SUCCESS in 26m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34e3a674c8b447f5b1656ca55fccce8d : SUCCESS in 10m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26702c492a7b477da470c4caa137fd23 : SUCCESS in 22m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b058771c47434389b0600539e113d6d8 : SUCCESS in 26m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992ce5caed0342c2bdfb9ad7b7fe77d2 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/232f0f3671bc470d88d61b61e625a304 : SUCCESS in 48m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/794c63978cad431fbc3444d296fc27fe : SUCCESS in 1h 02m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5286444f110443f3bad35ec87e9ef475 : SUCCESS in 1h 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e5be61848931dd2070601d00b84ae7b17565ca65","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-19 13:08:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/857773/2 : SUCCESS in 2h 04m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3186fe0e9222ecd7755b0164df8ff14d76872be1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-09-19 16:11:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"579b77f1bf80019304ce8b31c7ce01de34372281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:12:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04b047d2256418ec1d1d5a7271a29d62a6764808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 17:30:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bc79b9fe7034b2c94c85bda51e0f376\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/7310df787869427c97425c3a9886f54d : SUCCESS in 38m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62a31b3b9504e7db184064c194e5e4f : SUCCESS in 13m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2563aa44034043aca3e46a7e45b79136 : SUCCESS in 23m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04fb16310adc44f79e9abdc849c7e2ca : SUCCESS in 22m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf3ea9f32864950889534059872c893 : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/155c58f98ae94e459652a61b18146d57 : SUCCESS in 49m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad695f3918cc41fb9e5ddf97aeacc94f : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c74ea8edb0b51cc3307b39d4c0040c1d905df79f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 17:30:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"47cfe44fc10adeaad07bcad47afccf449cb365b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 17:34:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f4b03a8bec54fe9aac2b7aa8914c346\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b7a58f4bb7477fa1094290c5dc2b6a : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15150ed9c7e8de21af504c087b4472542080058e","revisions":{"6f255845f742341f1733295f0f45c5ce6bb7b755":{"kind":"REWORK","_number":1,"created":"2022-09-15 03:20:08.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/73/857773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/857773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/857773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/857773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/857773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/857773/1"}}},"commit":{"parents":[{"commit":"b11fb8802f75bad438833f0e7db22c9f69a4b504","subject":"Merge \"Bump revision number of objects when description is changed\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b11fb8802f75bad438833f0e7db22c9f69a4b504"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-09-15 03:13:03.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-09-15 03:20:08.000000000","tz":0},"subject":"Fix indentation issue in wait_for_change override","message":"Fix indentation issue in wait_for_change override\n\nThe ovsdbapp.backend.ovs_idl.idlutils.wait_for_change method was\noverriden in Ia9e23113fdeebf0b99085da200c3d61b71567d36. Unfortunately,\nand indentation error was introduced which rendered the method as a\nNoOp if a seqno was passed to it.\n\nCloses-Bug: #1989646\nChange-Id: I9d8db215ba19ea88bc615b7b30084113a3c953d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f255845f742341f1733295f0f45c5ce6bb7b755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f255845f742341f1733295f0f45c5ce6bb7b755"}]},"branch":"refs/heads/stable/yoga"},"15150ed9c7e8de21af504c087b4472542080058e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-19 11:02:58.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/73/857773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/857773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/857773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/857773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/857773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/857773/2"}}},"commit":{"parents":[{"commit":"b11fb8802f75bad438833f0e7db22c9f69a4b504","subject":"Merge \"Bump revision number of objects when description is changed\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b11fb8802f75bad438833f0e7db22c9f69a4b504"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-09-15 03:13:03.000000000","tz":-300},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-09-19 11:02:58.000000000","tz":0},"subject":"Fix indentation issue in wait_for_change override","message":"Fix indentation issue in wait_for_change override\n\nThe ovsdbapp.backend.ovs_idl.idlutils.wait_for_change method was\noverriden in Ia9e23113fdeebf0b99085da200c3d61b71567d36. Unfortunately,\nand indentation error was introduced which rendered the method as a\nNoOp if a seqno was passed to it.\n\nCloses-Bug: #1989646\nChange-Id: I9d8db215ba19ea88bc615b7b30084113a3c953d4\n(cherry picked from commit fb7d937adae140520dae94a47d696ce202479ab5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15150ed9c7e8de21af504c087b4472542080058e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15150ed9c7e8de21af504c087b4472542080058e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
