)]}'
{"id":"openstack%2Fneutron~857774","triplet_id":"openstack%2Fneutron~stable%2Fxena~I9d8db215ba19ea88bc615b7b30084113a3c953d4","project":"openstack/neutron","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-09-19 17:42:05.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-09-19 17:42:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d8db215ba19ea88bc615b7b30084113a3c953d4","subject":"Fix indentation issue in wait_for_change override","status":"MERGED","created":"2022-09-15 03:20:26.000000000","updated":"2022-09-19 17:43:39.000000000","submitted":"2022-09-19 17:42:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857774","cherry_pick_of_change":857773,"cherry_pick_of_patch_set":2,"meta_rev_id":"a90955916910a4e46a43a25537b07e003b13b00b","_number":857774,"virtual_id_number":857774,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 17:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-09-19 11:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-09-19 16:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-09-19 16:11:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-15 04:01:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-15 04:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-15 04:58:50.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-15 08:47:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2022-09-15 12:57:52.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-09-15 15:07:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-09-15 17:44:50.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2022-09-19 11:07:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-09-19 16:11:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"3ab6dbf3a533d0f927c4dd8a06732e78056280ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-09-15 03:20:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d3163058549e363a3fa07062646453100a1b704f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 04:01:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4d183aa798f4820aa32c06bee139c41\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/976cb2678a49487a9393dde17d4502eb : SUCCESS in 32m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9269fed6ae3407d84d6895e4e8e7257 : SUCCESS in 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45060939b02e12786a8c0e172af8868b3151d3fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 04:49:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2667b8b60df4b8a93c12951ab68c521\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/4cb6b94c71ff4d9a9acf5de4379f74a2 : SUCCESS in 53m 20s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/50c085146bd24caab639b4dde9919e35 : SUCCESS in 49m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d5984ee1c2a4e029b540de5c3ff2710 : SUCCESS in 44m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a935df58d04854b02a92f53f74ad63 : SUCCESS in 15m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd4ade27b86a4545999d6c414bf19603 : SUCCESS in 27m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6658211ffbfe436fb82898596986337a : SUCCESS in 24m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd0b018a6d6b4b37a7eadf3157c93993 : SUCCESS in 29m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f07f1008a4d4783920f593581be68fe : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1bc20b3d4f874acf9819d07174e3cea9 : FAILURE in 53m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29e8c048f5724631b43bb865d5b8c429 : SUCCESS in 1h 06m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61431b620b3c4adc997c9a48687fc2dd : SUCCESS in 1h 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db8064fcc97c0e2b24f560da83b71421ccf982c7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-15 04:58:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/857774/1 : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9e1885e6ea6f7e350f8f77fe3a8f69e410d1037a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-15 08:47:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a52c9a0c3baa5f6e562b7de520a2108bca0f0266","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 09:32:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1b538f5af37432cab14e0924489460d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea021bf311b140bb99091601a0d80fdf : SUCCESS in 33m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e21bb564047a4b85b1b085a496592ae0 : SUCCESS in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59536db1955b3e6e278524e8d3f463cfdfddc5fb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-15 10:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/857774/1 : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"61152f1f71dbf22d7c826fd9b5e72b45694c83ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 11:44:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d00f939360cf4133b06c8f239f7a2a56\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/cf790e3bc2624700b0cadfe08cf95db7 : SUCCESS in 43m 35s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/7afd2131ff7f41e98e89322d7ee2df3d : SUCCESS in 56m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f87c9ffc04a4e93bfef848d56f93978 : SUCCESS in 40m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cdc6bcfd94e4be987aced23c32ff779 : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8e17a6085fa4693ae1b2c16c70cb0e8 : SUCCESS in 30m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/807f6a7e204e47ac8a40eb614ac2d579 : SUCCESS in 26m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7923dc2ebc142d391b28caf41131bc2 : SUCCESS in 32m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99bf69f3903e44c7b40e1045b25929bc : SUCCESS in 8m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/97719b9d07684c2085eaac008e96a171 : SUCCESS in 47m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47dd401d1c8a472ea3531733ed867659 : SUCCESS in 1h 05m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/036c28d5aa304ddfa5c22d4108ca01e6 : SUCCESS in 1h 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae9731c9b929d9ac01228f32414593bbae921f6c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-09-15 12:57:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37e1436b889ec4f7a27dd24e6c0aa0eb1906159d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-15 15:07:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a45b31baefebf429a6ff765d4c9f38937cb6569","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-09-15 17:44:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cc1a78ea4d6bd807b09f8a3b08686435310cfd9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-19 11:03:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74ac7c906d8e819f80e044cad9d5a6b247c8646e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-19 11:05:25.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4e88ca700c6ff895d04de238977fb2912ae7c18b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-19 11:07:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a792fbf932bd962453a87200679fafe11a87b37c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 11:51:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a6a073384d84b779ed48a1e87a42f1b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d32207776f1e41c681a70ca06998a721 : SUCCESS in 41m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c440db74b2c4bd2aaf7ac98246530fe : SUCCESS in 41m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"646e4b4635956bd1a2f3e6f140caceae8a2bf39e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 12:27:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cff2a5f129640e8a1428fa9a46059a3\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/41f04757bcdd485d8120cbebe19c4a11 : SUCCESS in 57m 42s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/7d9f1caee467415ea63d1d8d8004e2d4 : SUCCESS in 56m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/727154e3a5284643b32b84b497fce27c : SUCCESS in 43m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a3f4bf5b454d7bad931fe6c74c7ec2 : SUCCESS in 16m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e61bd16becb7465890120f1b253057e4 : SUCCESS in 29m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43a30da7af774ca888cd7e751b4bda5b : SUCCESS in 27m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3cee945bda4479db9a712d5d9b205c6 : SUCCESS in 28m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bda9a70734a409190d5b00d40d51b18 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f5496fbd9024b22942ff3517ed7a3ae : SUCCESS in 57m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1b3481871434b7caa9d5dd49e3ef9ac : SUCCESS in 1h 01m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f41c4dae0b7d48aeac1d988025a5fe8e : SUCCESS in 1h 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c49fa06cf0a6d367d7af3196f0a967baaa1ae1ed","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-19 13:04:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/857774/3 : SUCCESS in 1h 58m 05s","accounts_in_message":[],"_revision_number":3},{"id":"26ef009948017c547183e50984e0c6aa0a3bacc0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-09-19 16:11:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0c0bc8cee8f7a61d5ecb57fcb99103294685c8f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:12:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1a37735d6a55f1e4f6e7622ab4f1dfec8a390532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 17:42:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf6f163d03c243cf93fa0134a8c4ea7b\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/bdb25925fb104039a7ba51ffeb391c98 : SUCCESS in 1h 02m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2907c0e78a5448e8a0e746bc99f9c521 : SUCCESS in 19m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4209423122942e19bd9958dbc05f12c : SUCCESS in 24m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11bed1345d554474b2207de61ad73a4d : SUCCESS in 22m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749e3265bee5492587bbd14518537575 : SUCCESS in 10m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba3014ee0aca4210ad841f1a6d7f39dd : SUCCESS in 47m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/122a902e1059453680232958176cbb36 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":3},{"id":"7da5d2a1b71cb97fa998a27aa4112d998e33a573","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 17:42:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a90955916910a4e46a43a25537b07e003b13b00b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 17:43:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2db4ea9f101c417082f75a522d9315bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2817d7a03eec4480bb91a484c732f2a1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3f146db5a70fe9625269b750c2e824d271421f0c","revisions":{"931dfd12ef63efaa39077564ac79a088af336213":{"kind":"REWORK","_number":1,"created":"2022-09-15 03:20:26.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/74/857774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/857774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/857774/1"}}},"commit":{"parents":[{"commit":"7bc419a263b6c59d6b8e3098162bb08a122f09e0","subject":"Merge \"[ovn] Specify port type if it\u0027s a router port when updating\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bc419a263b6c59d6b8e3098162bb08a122f09e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-09-15 03:13:03.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-09-15 03:20:26.000000000","tz":0},"subject":"Fix indentation issue in wait_for_change override","message":"Fix indentation issue in wait_for_change override\n\nThe ovsdbapp.backend.ovs_idl.idlutils.wait_for_change method was\noverriden in Ia9e23113fdeebf0b99085da200c3d61b71567d36. Unfortunately,\nand indentation error was introduced which rendered the method as a\nNoOp if a seqno was passed to it.\n\nCloses-Bug: #1989646\nChange-Id: I9d8db215ba19ea88bc615b7b30084113a3c953d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/931dfd12ef63efaa39077564ac79a088af336213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/931dfd12ef63efaa39077564ac79a088af336213"}]},"branch":"refs/heads/stable/xena"},"54a1f283824a6d1a809233cca67197d963ebcecb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-19 11:03:24.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/74/857774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/857774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/857774/2"}}},"commit":{"parents":[{"commit":"7bc419a263b6c59d6b8e3098162bb08a122f09e0","subject":"Merge \"[ovn] Specify port type if it\u0027s a router port when updating\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bc419a263b6c59d6b8e3098162bb08a122f09e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-09-15 03:13:03.000000000","tz":-300},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-09-19 11:03:24.000000000","tz":0},"subject":"Fix indentation issue in wait_for_change override","message":"Fix indentation issue in wait_for_change override\n\nThe ovsdbapp.backend.ovs_idl.idlutils.wait_for_change method was\noverriden in Ia9e23113fdeebf0b99085da200c3d61b71567d36. Unfortunately,\nand indentation error was introduced which rendered the method as a\nNoOp if a seqno was passed to it.\n\nCloses-Bug: #1989646\nChange-Id: I9d8db215ba19ea88bc615b7b30084113a3c953d4\n(cherry picked from commit fb7d937adae140520dae94a47d696ce202479ab5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54a1f283824a6d1a809233cca67197d963ebcecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54a1f283824a6d1a809233cca67197d963ebcecb"}]},"branch":"refs/heads/stable/xena"},"3f146db5a70fe9625269b750c2e824d271421f0c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-19 11:05:25.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/74/857774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/857774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/857774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/857774/3"}}},"commit":{"parents":[{"commit":"7bc419a263b6c59d6b8e3098162bb08a122f09e0","subject":"Merge \"[ovn] Specify port type if it\u0027s a router port when updating\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bc419a263b6c59d6b8e3098162bb08a122f09e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-09-15 03:13:03.000000000","tz":-300},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-09-19 11:05:25.000000000","tz":0},"subject":"Fix indentation issue in wait_for_change override","message":"Fix indentation issue in wait_for_change override\n\nThe ovsdbapp.backend.ovs_idl.idlutils.wait_for_change method was\noverriden in Ia9e23113fdeebf0b99085da200c3d61b71567d36. Unfortunately,\nand indentation error was introduced which rendered the method as a\nNoOp if a seqno was passed to it.\n\nCloses-Bug: #1989646\nChange-Id: I9d8db215ba19ea88bc615b7b30084113a3c953d4\n(cherry picked from commit fb7d937adae140520dae94a47d696ce202479ab5)\n(cherry picked from commit 15150ed9c7e8de21af504c087b4472542080058e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f146db5a70fe9625269b750c2e824d271421f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f146db5a70fe9625269b750c2e824d271421f0c"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
