)]}'
{"id":"openstack%2Fneutron~858466","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I7f4366ff5a26f73013433bfbfb299fd06294f359","project":"openstack/neutron","branch":"stable/train","topic":"bug/1873091-stable/train","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2022-09-29 16:42:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7f4366ff5a26f73013433bfbfb299fd06294f359","subject":"Sync the dns-assignment with the actual designate dns-domain","status":"MERGED","created":"2022-09-20 17:38:29.000000000","updated":"2022-09-29 16:43:43.000000000","submitted":"2022-09-29 16:42:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858466-bug/1873091-stable/train","cherry_pick_of_change":731624,"cherry_pick_of_patch_set":5,"meta_rev_id":"abd0eef1ab3d9b8bb0e4bfceb7857098308b66c2","_number":858466,"virtual_id_number":858466,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 16:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-09-27 05:11:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-09-29 15:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-09-26 15:08:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-09-27 09:30:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-09-29 15:05:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 17:38:29.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"state":"CC"},{"updated":"2022-09-20 19:32:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-26 15:08:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-27 05:11:47.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-09-27 09:30:26.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-09-29 15:05:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"aafcb32b57f772ce05785bb7c0aa6df1561e6f22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-09-20 17:38:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eba4b1a4c520af0283bb3449c491fe504a3ef0a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 19:32:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8e645bd20c64884934e2deef96c5359\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/5fbae98bdbe547bc9ccc54b4babce169 : SUCCESS in 45m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f6c7a5d71b149c4b445f1277d56f99b : SUCCESS in 30m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701edebaf04649bd8e02fd786ff468ec : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18866ad2074945a28e1e9854ef1f4fee : SUCCESS in 23m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aff9f320bd2409c8702151ff223ca56 : SUCCESS in 20m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0619121d33c4d0b93c2f4932c4ec582 : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a785ed6e102422bb1c70b1aa3fdd534 : SUCCESS in 17m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b38cf6aeff974763afe8fd4fc869f95d : SUCCESS in 11m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c25ff693abba42c08139147a13eb075a : SUCCESS in 51m 05s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/0d1ccfb38c484621af0a4e2d6bb2ff11 : SUCCESS in 41m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f2f954208a2d4bea9875137fb58a8be2 : FAILURE in 1h 15m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6caf49116cbb4c698b2b5a0b3335f2a0 : SUCCESS in 1h 19m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1b0c11ead0a54f4e8d0c4c24cfc2abd0 : SUCCESS in 56m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7ae501fc4b3c4b779569f240d827ab3d : SUCCESS in 1h 03m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa06c1247f834c359daa26c445b6b7e4 : SUCCESS in 1h 06m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1716978c75b48dc841c4848c480e320 : SUCCESS in 50m 50s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8747bff54f50467bb720c48b63bf56ee : SUCCESS in 1h 19m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7bfbdab5b3b945388eff1a71a97c20fd : SUCCESS in 44m 12s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b302d61d4fc8463f9ae241c16a3ffa1d : SUCCESS in 1h 37m 21s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/91e70d1e49874128944f0f48c686daac : SUCCESS in 1h 45m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/39eae96817d74f37a8324f59cda3830d : SUCCESS in 1h 03m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05bc627c80a54db69445dc1a833bb2da : SUCCESS in 55m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/ec9086b5918441658a77de13c904bae0 : SUCCESS in 50m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c7493223b3f24384b1744e661d5dc0b1 : SUCCESS in 1h 10m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04939bb5d2564dfe9a3761721dd8eedc : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dbe1b4f5ae770ee86c6a3ea81a1e025547391d00","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-26 15:08:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"463875339da28fbde9c6f2ee5e64000b5dfb5f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 17:21:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e50aff718a54046bb73ac75a6081861\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/fd64df80ed27476a8b8a72e4671fc84f : SUCCESS in 55m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6cb91324fb14f8d9a0492fbb669b73a : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21fa755061ee41dd94dce8ff89b1b248 : SUCCESS in 13m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0d629424b1a4758bc10f24a4b00bf88 : SUCCESS in 23m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73ee13ae1c22448a9cf9b29c6463a608 : SUCCESS in 21m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7f36939bff242f4a2a2289621bc64e9 : SUCCESS in 23m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c46e900436c424b949b91ddd36b7d4e : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4e1507d29344416add52517b016e3c3 : SUCCESS in 12m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fe5aa3e686454f42b69ccf66053276d2 : SUCCESS in 50m 57s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/5c961e2892b3434091dfeaa7744a14a1 : SUCCESS in 52m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/475e857cbeff403fb37a2a76d5e157b6 : SUCCESS in 1h 24m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7de2c21a285d49799e4cbc6dd94992de : SUCCESS in 1h 36m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/12e767c8927948feb8561fa325772078 : SUCCESS in 1h 15m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1e6bf7935c2a4adebab7d3b922f8f935 : SUCCESS in 1h 10m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54678b44820f43b981d6ce2b6e8e0712 : SUCCESS in 1h 02m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4989dd6ac7ed40e680164b671c5a4b04 : SUCCESS in 1h 12m 10s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/93721e867ce945e9aef8357d49a36f83 : SUCCESS in 1h 10m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bf159e22285b4e2096f7beec744e9b27 : SUCCESS in 1h 03m 49s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a44b443c2374f6693a007f062f2ef02 : SUCCESS in 1h 37m 34s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/018e87cacd874989a9b5a9efbd23824d : SUCCESS in 2h 10m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9af45151501543dfbeace5de5b2b1c54 : SUCCESS in 1h 24m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91ade9ae5e6d46709ccf660bd35ac9b1 : SUCCESS in 1h 07m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/b5477a61aa13438897ea87fb8cc66e90 : SUCCESS in 1h 03m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/057cae7c0693446cb4de75b0f1932dc6 : SUCCESS in 1h 10m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ada39b661c3c4555b2056a9cd485e98f : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b608a42e063b4ff9ec0323b2b32efffd4afd1d19","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-27 05:11:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c9a319ee9260263031a6343be1c251e996008db","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-09-27 09:30:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"937047928322a4b62c45d11048dd5cec5b5f0b85","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-09-29 15:05:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0d9dc83fb37764ef15741e5f4c9482ae436bad2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:07:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3316f43396c10ba347284731b309f2cc2afa30b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:42:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09e97bce032c4cc5b1625c47ff4981b1\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/8b27faae591d463ca68317a7cafdd398 : SUCCESS in 57m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf7cfee6a3cf439a9135e87300ed4e78 : SUCCESS in 13m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b965a17aa00c4d0cb37f339a4e220e3e : SUCCESS in 23m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a9ad23beed84b87859713d6ede8a3a0 : SUCCESS in 23m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71fea1f5fdd647df876a91f8d3a8e2a7 : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11572d46e6d43ab97d4cbe9a6a709f3 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccc39e78ecad4e84b008f0039d64d726 : SUCCESS in 12m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a078c300a4984f448a2ebd5199b05154 : SUCCESS in 47m 18s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/4e5ed93dca2c46d89103b6284a7e94f1 : SUCCESS in 50m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ed78fb6578e444a085fcb47f893e84ad : SUCCESS in 1h 15m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/26a012eb780a443492d6b96bfe4bbd0b : SUCCESS in 1h 26m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3ec217e2ac0d481eb9f4070cdd938225 : SUCCESS in 1h 02m 49s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9bc376cc3b024597b63823802dc4fc70 : SUCCESS in 1h 05m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/906f3bdc1ce0489e8104bb25fb27fd69 : SUCCESS in 1h 04m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/442d96cefb4d46538d54417568b008c2 : SUCCESS in 49m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d671b5aa462f4c1b9a075df63c68512f : SUCCESS in 55m 07s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a1c148d234341beae061bdf555c95b5 : SUCCESS in 1h 31m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d62361a25f5461bbba9f01c336d8306 : SUCCESS in 1h 15m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ea9b4580283416ab8f1ed1cdb358c8e : SUCCESS in 52m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/93e22919607b4a1b90169c032685db92 : SUCCESS in 59m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/90c1c0da9d1e4662a754fccafd27237a : SUCCESS in 59m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51f8962fe0344f49a91ba950c4d87dc8 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":1},{"id":"abe499ac5e9968374c61f6218887f439cf69bf00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:42:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"abd0eef1ab3d9b8bb0e4bfceb7857098308b66c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:43:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c232385547a140509d847c7b5274dc5f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56666d63eae64e32b48b1842dd4efbcf : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4a7e9b71c114e8e855a77289f8559ff : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f1e45da6af4879725fb2d766f1989c8a10af474","revisions":{"1f1e45da6af4879725fb2d766f1989c8a10af474":{"kind":"REWORK","_number":1,"created":"2022-09-20 17:38:29.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/66/858466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/858466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/858466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/858466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/858466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/858466/1"}}},"commit":{"parents":[{"commit":"c81fb5b394995622614bb9e54ff139835b91363d","subject":"Merge \"Bump revision number of objects when description is changed\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c81fb5b394995622614bb9e54ff139835b91363d"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-05-28 23:17:28.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2022-09-20 17:38:29.000000000","tz":0},"subject":"Sync the dns-assignment with the actual designate dns-domain","message":"Sync the dns-assignment with the actual designate dns-domain\n\nWhen a port is created the dns-assignment (dns-domain part)\nwas always taken form Neutron config dns_domain which is not\nalways true, since it could be Neutron network dns_domain or\nthe dns_domain sent when creating the port\n\nChange-Id: I7f4366ff5a26f73013433bfbfb299fd06294f359\nCloses-Bug:1873091\n(cherry picked from commit ea13f2e83f8c2de3def69b6c883a5c161c3a6180)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f1e45da6af4879725fb2d766f1989c8a10af474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f1e45da6af4879725fb2d766f1989c8a10af474"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
