)]}'
{"id":"openstack%2Fneutron~858467","triplet_id":"openstack%2Fneutron~stable%2Fstein~I7f4366ff5a26f73013433bfbfb299fd06294f359","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1873091-stable/stein","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2022-09-29 16:43:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7f4366ff5a26f73013433bfbfb299fd06294f359","subject":"Sync the dns-assignment with the actual designate dns-domain","status":"MERGED","created":"2022-09-20 17:39:44.000000000","updated":"2022-09-29 16:44:51.000000000","submitted":"2022-09-29 16:43:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858467-bug/1873091-stable/stein","cherry_pick_of_change":731624,"cherry_pick_of_patch_set":5,"meta_rev_id":"cb4647fcf4fbf5a34a9ad13e452fe54bff724238","_number":858467,"virtual_id_number":858467,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 16:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-09-29 15:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-09-26 15:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-09-29 15:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 17:39:44.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"state":"CC"},{"updated":"2022-09-20 19:17:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-26 15:07:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-29 15:06:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"102d5ff872a022747a4fc1418f3b635f6c1e549c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-09-20 17:39:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b14e11a69f306d2689a95931352d940a83e9fb0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 19:17:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cec6f5f6e44b368b07908e3e688d42\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/54f55b9f0db84b91b5bccd5418a9b08f : SUCCESS in 56m 57s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/0534af8b5e42404fa48dcfcf8e0b405d : SUCCESS in 1h 13m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/057d818a41d343989c810e4e9f028463 : SUCCESS in 1h 17m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f449b8f72877400dbd781db74b5222a5 : SUCCESS in 22m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407d8a47749b4d0bb798cf2aadd901ab : SUCCESS in 9m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ba8667abf11405f917d0b01754448c9 : SUCCESS in 18m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c36b668373de4b67b19bdfd6f41aad7a : SUCCESS in 24m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20b24118da5348709d9647ad66133819 : SUCCESS in 19m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49232ed3b5f9490c9fc2d19598f9c9fe : SUCCESS in 25m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6a9318cdf644a5b85e5f5e5cda8fef : SUCCESS in 6m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4410a12b4ca84ecd9784b7e596df952b : SUCCESS in 1h 29m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c88cc1c5b924fb0b4bc9e77c25af033 : SUCCESS in 1h 15m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66a05cbad6f04bd3b838cf02abae7b67 : SUCCESS in 12m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c76b43f3e4ff4e05a657eb4397d813fc : SUCCESS in 35m 50s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/299f42aa1c024aa7b58d19507bda9223 : SUCCESS in 46m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c318f59d24724940bc97386111dcd677 : SUCCESS in 1h 08m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a49cbf7826494e56a4e192b3daec21bc : SUCCESS in 1h 31m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cb71ee3f9aee40e3a8ce37b6d781adf4 : SUCCESS in 1h 31m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f0064a944a294568ad9a43fbd97d076c : SUCCESS in 1h 11m 30s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f8e583d5f7b4ad78c1c487617fc644f : SUCCESS in 1h 27m 31s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d948a5013e4c4d79ac51b99a55a51e64 : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3093ba7d0ac36bc995cf16fea07f141a20be9087","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-26 15:07:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ad530e30badb3d8d8af047401787cdd83ea91e4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-09-29 15:06:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b87cacc6a63ef55b44de14603e2108a1c2448ab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:07:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15808adbf15474905472dce0e006a56e0fbb8a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:43:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8026da8b934444a9873a0ac2eac33ac\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/8f21583d941a42f994386ffce68acdbd : SUCCESS in 58m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482b4897b3ca4826abaca43191107ebd : SUCCESS in 11m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c9226c2a11e432da2e5e0ab5b6e913d : SUCCESS in 35m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4fbf984e4ded47f195b45a32e24bdd0d : SUCCESS in 30m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31cd5cbe11ae44e98ac828809a75c2b1 : SUCCESS in 18m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38ef7de91cdc4978be23217965aacf25 : SUCCESS in 36m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d1f53a7f2e4ea185d0925e678b12e4 : SUCCESS in 6m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d292bb553454440db293665d466ba4ef : SUCCESS in 1h 19m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f3e4385cf814737a9efaee9183ac69d : SUCCESS in 1h 32m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5ac64465dc146e1b45608d978c9c112 : SUCCESS in 14m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/93e3f21243714f16980f7a78d75abc32 : SUCCESS in 43m 30s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e5a25eaa427d4406a2a134b7c4b8ca1e : SUCCESS in 38m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bc40869e01ab4cb1af12e076c84fa7de : SUCCESS in 1h 31m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fc9867a122e947128c6fdc55163a20d7 : SUCCESS in 1h 10m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/59ec4b9ccc7f4bd18f7b68f4de89ae66 : SUCCESS in 1h 15m 36s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/99b66f23c7f84f05a1d922dd6d036920 : SUCCESS in 1h 34m 27s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0fa78bf6a8fa4b63bc6bc14667608cbb : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c4870d6f87b40071b6754c5d44830ca3ea31c273","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:43:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cb4647fcf4fbf5a34a9ad13e452fe54bff724238","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:44:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77efa49a16f342b9b7b0e6add105a235\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f708da966a430189b259329d8d3d23 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f34f3fe8c68c426a96049d114fa2cc65 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26afffc3a7b18d0db1aa0a81c961bd6a4642749d","revisions":{"26afffc3a7b18d0db1aa0a81c961bd6a4642749d":{"kind":"REWORK","_number":1,"created":"2022-09-20 17:39:44.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/67/858467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/858467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/858467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/858467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/858467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/858467/1"}}},"commit":{"parents":[{"commit":"5b812fd45ec5cbf7a84b0442536ca2181248aac5","subject":"Merge \"Stein only: Make CI green again\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b812fd45ec5cbf7a84b0442536ca2181248aac5"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-05-28 23:17:28.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2022-09-20 17:39:44.000000000","tz":0},"subject":"Sync the dns-assignment with the actual designate dns-domain","message":"Sync the dns-assignment with the actual designate dns-domain\n\nWhen a port is created the dns-assignment (dns-domain part)\nwas always taken form Neutron config dns_domain which is not\nalways true, since it could be Neutron network dns_domain or\nthe dns_domain sent when creating the port\n\nChange-Id: I7f4366ff5a26f73013433bfbfb299fd06294f359\nCloses-Bug:1873091\n(cherry picked from commit ea13f2e83f8c2de3def69b6c883a5c161c3a6180)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26afffc3a7b18d0db1aa0a81c961bd6a4642749d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26afffc3a7b18d0db1aa0a81c961bd6a4642749d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
