)]}'
{"id":"openstack%2Fneutron~858504","triplet_id":"openstack%2Fneutron~master~Ic8fc6d8701fe17ebd870f57df83f21c110b73849","project":"openstack/neutron","branch":"master","topic":"bug/1988604","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-09-21 03:51:22.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-09-20 10:32:00.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-09-21 03:51:22.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-09-21 03:51:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8fc6d8701fe17ebd870f57df83f21c110b73849","subject":"[sqlalchemy-20] Add py39 oslo master to experimental","status":"MERGED","created":"2022-09-20 09:04:47.000000000","updated":"2022-09-21 03:55:46.000000000","submitted":"2022-09-21 03:51:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858504-bug/1988604","meta_rev_id":"b5272e0e6262773b235da2d432cf0ed9b5775fef","_number":858504,"virtual_id_number":858504,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 03:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-09-20 10:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-09-20 23:08:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-09-20 23:08:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 09:08:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2022-09-20 09:10:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-20 09:22:47.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-20 09:22:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-20 09:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 09:28:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-09-20 10:32:00.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-09-20 11:50:21.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-20 23:08:18.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"d10fbc0dfa5dc7a4a036c5237e05dc6082899ebd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-20 09:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c42edcba01bca3511d1803962adec8a6bf0f71c8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-20 09:06:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7ace6bd929ee717c9bd1ac2c7c6770e464ce98a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-20 09:08:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6dbac4ad370ec2238c43de66ef6183fd2a42ead8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-20 09:10:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85f71f15c876c0b75072978bd6b2667d867fe9e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-20 09:21:27.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5cfbf4bf7021ccf1961d283c99829dd58a8b23e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-20 09:22:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7e2c0c95431dd72017bfcc05da8c83c7fe22f3fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-20 09:22:47.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 101, column 20","accounts_in_message":[],"_revision_number":2},{"id":"80c6414399ea6af5a67ddf910730e2e87e8df59e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:22:50.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 101, column 20","accounts_in_message":[],"_revision_number":2},{"id":"46980cbbdc135a6ac22cd80b94749f68140e45d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:22:55.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 101, column 20","accounts_in_message":[],"_revision_number":2},{"id":"80139d41c70c655cd00a0d422ae42973cc992e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-20 09:23:15.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b90cc93484e25992a1b04fa8c19d037fa321b18f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-20 09:23:37.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 101, column 20","accounts_in_message":[],"_revision_number":3},{"id":"5c3153cc417209eec1c6adb4a632a08086fad206","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:23:39.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 101, column 20","accounts_in_message":[],"_revision_number":3},{"id":"b07da0e13e2752ff8401d32bb2ef2dba5533dcc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:23:46.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 101, column 20","accounts_in_message":[],"_revision_number":3},{"id":"7c38db831c78a1d374425f3aa0521ed19d2625c2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-20 09:28:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f067cc3723481d698aa86433e932d0d67d63945","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-20 09:56:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07e69d4c8713b1c61f440dd86977670903d4d42e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-20 09:56:32.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"68b70c379261d703158b21252222a915708fc1d9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-20 10:32:00.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c9c750d0c9f9b5dd8c892541de35949cc88cde2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 10:36:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78676844d5d843539037edeb769c205a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0b00b6a1bca4755b678e93dcfa5bc63 : SUCCESS in 33m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db2f685e16c7460ba9f7882e47209a76 : SUCCESS in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"669d808f4d076263e0f62375910529be11d6fb8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 10:43:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57b2b367d6f4452bb0c2981ba44b8a32\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bca2967cb3944851ab791cc94a67b0a7 : SUCCESS in 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92b66314ffa34708b6d93864dcab5888 : SUCCESS in 14m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ad9d159cd6646238a965e5126444500 : SUCCESS in 30m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0a6674202b14673b65db85b052c2526 : SUCCESS in 26m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ac51e68b5834b27825add8f7622d7db : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a167a61df14b64b1d30b521e0f4e4a : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":4},{"id":"583e69332272ea5c65cc283f92ece3c60799976b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 11:42:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d891e7b78e4b7ebf4bc98724c87ade\n\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/66a889a89c8a494e8dd3916ded2cf09d : SUCCESS in 55m 01s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/9cda1cb9fc114657a47b431110484f6d : SUCCESS in 41m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/26ccd49ce5f04a43adc9a0789892bb2c : SUCCESS in 49m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/23bc4e9eb3c54ed7a96c7aac2b18d0e0 : SUCCESS in 1h 08m 20s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ea96bb48758445cdb9cf3a056cdb954b : SUCCESS in 45m 34s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/10b3d180b8dc48c1a10dd47be9dcdfcc : SUCCESS in 23m 24s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/d3f3fa74fc0b4f6f99c294ee057f23ac : SUCCESS in 25m 43s","accounts_in_message":[],"_revision_number":4},{"id":"e95af002a13edfd8b8f738c1755e38348b995776","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-20 11:50:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/858504/4 : SUCCESS in 1h 53m 09s","accounts_in_message":[],"_revision_number":4},{"id":"dc19475bbc48c553d602ac1dcba341c4a10de45b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-09-20 23:08:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"50f50368dd243e8d7f386932f1c77ed706409ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 23:08:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"404bb6f8679f1f877fe9b8c18b50066fce788bee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 03:51:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88d92c7afa5a4087ad601b64a630389b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80332f916b3542f488692c7f948fb7ee : SUCCESS in 14m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e5b4683fdde4e28808f9e942bd7c114 : SUCCESS in 29m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13cf5af56df0445ba0694500dfaff288 : SUCCESS in 27m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab4a6f41f814f1a946ad711d07f7e8c : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":4},{"id":"46113c53fa2cd7e89d3ab735a2c3a22d5cc138ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 03:51:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b5272e0e6262773b235da2d432cf0ed9b5775fef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 03:55:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43dd61f755da48caaa069b8a2c7ec113\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c560a59c6da74e8d916d81f49b711e08 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a765b7acbb3633245bca8d6e00c64c03b2384661","revisions":{"9694bd5094bc02bf76420633b5826ca0fac201d8":{"kind":"REWORK","_number":1,"created":"2022-09-20 09:04:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/858504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/858504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/858504/1"}}},"commit":{"parents":[{"commit":"9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-15 12:12:12.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-15 12:15:10.000000000","tz":120},"subject":"[sqlalchemy-20] Add py39 oslo master to periodic and experimental","message":"[sqlalchemy-20] Add py39 oslo master to periodic and experimental\n\nIn order to quickly detect new changes or errors with the latest oslo.db\nlibrary, this patch adds the \"openstack-tox-py39-with-oslo-master\" CI\njob to the \"neutron-periodic-jobs\" and \"neutron-experimental-jobs\"\nqueues.\n\nRelated-Bug: #1988604\n\nChange-Id: Ic8fc6d8701fe17ebd870f57df83f21c110b73849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9694bd5094bc02bf76420633b5826ca0fac201d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9694bd5094bc02bf76420633b5826ca0fac201d8"}]},"branch":"refs/heads/master"},"a30c7e1a5458c98c378a78c043ef2f054a4f6092":{"kind":"REWORK","_number":2,"created":"2022-09-20 09:22:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/858504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/858504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/858504/2"}}},"commit":{"parents":[{"commit":"9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-15 12:12:12.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-09-20 09:22:24.000000000","tz":0},"subject":"[sqlalchemy-20] Add py39 oslo master to periodic and experimental","message":"[sqlalchemy-20] Add py39 oslo master to periodic and experimental\n\nIn order to quickly detect new changes or errors with the latest oslo.db\nlibrary, this patch adds the \"openstack-tox-py39-with-oslo-master\" CI\njob to the \"neutron-periodic-jobs\" and \"neutron-experimental-jobs\"\nqueues.\n\nRelated-Bug: #1988604\n\nChange-Id: Ic8fc6d8701fe17ebd870f57df83f21c110b73849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a30c7e1a5458c98c378a78c043ef2f054a4f6092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a30c7e1a5458c98c378a78c043ef2f054a4f6092"}]},"branch":"refs/heads/master"},"55ce4955a38e8db48d947cbbe9621147551b676c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-20 09:23:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/858504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/858504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/858504/3"}}},"commit":{"parents":[{"commit":"9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-15 12:12:12.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-09-20 09:23:15.000000000","tz":0},"subject":"[sqlalchemy-20] Add py39 oslo master to experimental","message":"[sqlalchemy-20] Add py39 oslo master to experimental\n\nIn order to quickly detect new changes or errors with the latest oslo.db\nlibrary, this patch adds the \"openstack-tox-py39-with-oslo-master\" CI\njob to the \"neutron-experimental-jobs\" queue.\n\nRelated-Bug: #1988604\n\nChange-Id: Ic8fc6d8701fe17ebd870f57df83f21c110b73849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55ce4955a38e8db48d947cbbe9621147551b676c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55ce4955a38e8db48d947cbbe9621147551b676c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a765b7acbb3633245bca8d6e00c64c03b2384661":{"kind":"REWORK","_number":4,"created":"2022-09-20 09:56:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/858504/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/858504/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/858504/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/858504/4"}}},"commit":{"parents":[{"commit":"9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1b5d1e51e5ab2f8ad905c7749a82d1d21f01f2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-15 12:12:12.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-09-20 09:56:31.000000000","tz":0},"subject":"[sqlalchemy-20] Add py39 oslo master to experimental","message":"[sqlalchemy-20] Add py39 oslo master to experimental\n\nIn order to quickly detect new changes or errors with the latest oslo.db\nlibrary, this patch adds the \"openstack-tox-py39-with-oslo-master\" CI\njob to the \"neutron-experimental-jobs\" queue.\n\nRelated-Bug: #1988604\n\nChange-Id: Ic8fc6d8701fe17ebd870f57df83f21c110b73849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a765b7acbb3633245bca8d6e00c64c03b2384661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a765b7acbb3633245bca8d6e00c64c03b2384661"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
