)]}'
{"id":"openstack%2Fneutron~858728","triplet_id":"openstack%2Fneutron~stable%2Fzed~I4f62b723083215483a2277cfcb798506671e1a2d","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-09-23 15:32:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f62b723083215483a2277cfcb798506671e1a2d","subject":"[OVN] Rate limit the \"Disallow caching\" log from hash ring","status":"MERGED","created":"2022-09-21 14:21:12.000000000","updated":"2022-09-23 15:34:06.000000000","submitted":"2022-09-23 15:32:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858728","meta_rev_id":"2363122e3b5e49fd1a630b6424c7d4d58085bcc4","_number":858728,"virtual_id_number":858728,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-23 15:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-09-23 11:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-09-22 21:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-09-23 11:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 15:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 21:37:52.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-09-23 11:00:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ef39aaffe9d995f411ae728b81856511a7bb79fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-09-21 14:21:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfd2a8b19a8daed66e171b8d95408cb8829e7c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 15:06:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b84a6329d5cd45d8b29b917a41d46fb3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae8ad9e02eeb4a13ae7ee208bcd84b30 : SUCCESS in 39m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e0612d3e484118b49d5f63b2e34785 : SUCCESS in 13m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/212f9b574a264997abcbe3e9da245c20 : SUCCESS in 28m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/986057dc8a924417b0e342cbeffc778a : SUCCESS in 26m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/721e5acd87144522b193df7245eda2ce : SUCCESS in 26m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f0266372d2490da68a62be875edb25 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ef93ff1206842ca9ac2ad9caabe257b51ed20b52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 15:11:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cc773c90bc44c5293dd70b97a989a59\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/754c3318571e405c85a02ca057cc6590 : SUCCESS in 47m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff144275affb4b57893e7e72baaf976c : SUCCESS in 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2557c1fb280692ca2395c74885b8678f06373dcb","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-09-22 21:37:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1dc165885ddf8c12d54d98ba8c85096fe8396569","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-23 11:00:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b1b61d96f375836ba308455f3f1ae8f2a86d638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 11:00:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ee4a02fd0caff5d6a093e6ec7960d90a0d02174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:32:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/864e32132e1749658229e057bd9d4730\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbd983f0e754d4d92edbcabac244ad7 : SUCCESS in 15m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a56c930b9da447918070612012f58729 : SUCCESS in 32m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a2ffe779f804db4a933bc36a787139b : SUCCESS in 29m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30fa851ed0f94f54b3bb2a8ba8e7eb7b : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0bc030d002e94c2876bfb3f98af678fa1dec1dbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:32:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2363122e3b5e49fd1a630b6424c7d4d58085bcc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:34:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fbba2743aeb4ced92b21bd9a417f32e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ffea385d824993ba28e1bf5cd82239 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f36ab53b49c1f83e6df7737c4e92b41e9658660","revisions":{"9f36ab53b49c1f83e6df7737c4e92b41e9658660":{"kind":"REWORK","_number":1,"created":"2022-09-21 14:21:12.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/28/858728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/858728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/858728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/858728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/858728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/858728/1"}}},"commit":{"parents":[{"commit":"a1bdcc32148b0622fca803e43c0c6b01e5c92aba","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1bdcc32148b0622fca803e43c0c6b01e5c92aba"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-09-13 15:11:37.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-09-21 14:21:09.000000000","tz":60},"subject":"[OVN] Rate limit the \"Disallow caching\" log from hash ring","message":"[OVN] Rate limit the \"Disallow caching\" log from hash ring\n\nThis patch adds a condition prior to logging the \"Disallow caching\" log\nmessage from the hash ring.\n\nPrior to this patch, this message was logged when the number of nodes\nconnected to the hash ring was different from the number of API workers\nat Neutron\u0027s startup. This is because the hash ring waits until all API\nworkers are connected to build the hash ring cache.\n\nWith this patch, we will only log the message once (per worker) until\nthe number of connected nodes changes. When nodes connect and the cache\nis built the _wait_startup_before_caching() is no longer used until the\nservice is restarted again.\n\nChange-Id: I4f62b723083215483a2277cfcb798506671e1a2d\nCloses-Bug: 1989480\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 9655466763282f47a06862a47f7f31b48130277e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f36ab53b49c1f83e6df7737c4e92b41e9658660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f36ab53b49c1f83e6df7737c4e92b41e9658660"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
