)]}'
{"id":"openstack%2Fneutron~859093","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Ieef6c8a02ca0cf85bc72b9f7d09b46dbd0e073be","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1990374","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-09-26 16:40:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieef6c8a02ca0cf85bc72b9f7d09b46dbd0e073be","subject":"Use \"OVNMechDriver\" instance in \"TestOvn[Nb|Sb]IdlNotifyHandler\"","status":"MERGED","created":"2022-09-23 13:56:42.000000000","updated":"2022-09-26 16:42:09.000000000","submitted":"2022-09-26 16:40:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859093-bug/1990374","meta_rev_id":"73ad445ee78f7a41af667017357ba00c73b3f38a","_number":859093,"virtual_id_number":859093,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-26 16:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-09-26 15:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-09-26 15:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-09-26 15:05:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-23 14:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-23 16:12:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-26 15:00:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-09-26 15:05:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"68bb53c1307c4b32e42fd4a34706b1273045f861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-23 13:56:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6055940bf6b888ef054991ebf6f424783900570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 14:30:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5bdf4bb57314138861a630fc5bd3f66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/005f4e5ec5734c299ab18fe6032b2310 : SUCCESS in 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5c2f9817f940d7b136a80fd83bd56a : SUCCESS in 11m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd4da86e1f4343bda4e3b148c90b3a99 : SUCCESS in 23m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95ff1203b5514bc19f947d2c860e091b : SUCCESS in 22m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25981a88dfd8407fbb1a43ffba746d28 : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e4bf8e775d43bba32a6438f301fd01 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"11d1afd9e0f82800d90368127b94bfabd9ec604f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 14:46:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7274f1de4f9443b99927c42670e8d1f2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad445096fb3642278906688ffd126110 : SUCCESS in 44m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c949629587d44da89bd16f57cc63bb8e : SUCCESS in 47m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb2c1a5ac1b0bb4d4cd949b1f6f3fc682f5eae5e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-23 16:12:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/859093/1 : SUCCESS in 2h 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"658e4bea3e0325d8f50b23c070efd1b5f4358e86","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-09-26 15:00:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"788394f80082a71b3d2f8fb0f9ce845a2ea6a4dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-26 15:05:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ae4da79af3e909f4d01409bf04de9ca21941452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 15:05:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85904e1277a473e1e82bcc1177b70fc0f89eabf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:40:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce77e3c142fb4a93a4de920d668b64f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a441f905a7c4201ad5699be83a13ef2 : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/689ba8e25d174cc18c3a25f3ff7ee779 : SUCCESS in 31m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe00de46cf2e41cf80efaae16675253a : SUCCESS in 26m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967e1657e6534d82bdd98a38e59693b5 : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e7e5a3527d723f5fbe4619cbcb42b944a94066c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:40:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73ad445ee78f7a41af667017357ba00c73b3f38a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e9d0f36a96340dfb688ba2836b892ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3ccac9f20f41e680293adcd5fabefc : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00efb17f00a71c75c3ea9286adebf28d49b7a3ab","revisions":{"00efb17f00a71c75c3ea9286adebf28d49b7a3ab":{"kind":"REWORK","_number":1,"created":"2022-09-23 13:56:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/93/859093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/859093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/859093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/859093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/859093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/859093/1"}}},"commit":{"parents":[{"commit":"54947f974603f77f920ba7f68013cd170161d610","subject":"Merge \"[OVN] Rate limit the \"Disallow caching\" log from hash ring\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54947f974603f77f920ba7f68013cd170161d610"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-21 10:08:13.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-22 21:33:46.000000000","tz":120},"subject":"Use \"OVNMechDriver\" instance in \"TestOvn[Nb|Sb]IdlNotifyHandler\"","message":"Use \"OVNMechDriver\" instance in \"TestOvn[Nb|Sb]IdlNotifyHandler\"\n\nThe test classes \"TestOvn[Nb|Sb]IdlNotifyHandler\" now use the\n\"OVNMechDriver\" instance to create the \"OvnNbIdl\" instance, that\nis stored in \"self.mech_driver\", instead of \"self.driver\".\n\nCloses-Bug: #1990374\nChange-Id: Ieef6c8a02ca0cf85bc72b9f7d09b46dbd0e073be\n(cherry picked from commit e9b5a27001614836fcf8297dd8510d144816bd02)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00efb17f00a71c75c3ea9286adebf28d49b7a3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00efb17f00a71c75c3ea9286adebf28d49b7a3ab"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
