)]}'
{"id":"openstack%2Fneutron~859645","triplet_id":"openstack%2Fneutron~master~Ic8891e2deef4bb5e72cf7d7f37b043e936adbc00","project":"openstack/neutron","branch":"master","topic":"bug/1991092","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-09-30 11:46:36.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-09-29 17:48:37.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ic8891e2deef4bb5e72cf7d7f37b043e936adbc00","subject":"Port provisioning should retry only for VM ports","status":"MERGED","created":"2022-09-28 16:05:28.000000000","updated":"2022-09-30 11:48:53.000000000","submitted":"2022-09-30 11:46:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"859645-bug/1991092","meta_rev_id":"26e9e4885e1b405bd10693ad8816dbc70c2c2059","_number":859645,"virtual_id_number":859645,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-09-29 16:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-30 11:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-30 09:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-09-29 17:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-30 09:56:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-28 16:51:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-28 17:04:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-28 17:50:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-09-28 17:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-28 17:55:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2022-09-29 17:48:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2022-09-30 09:56:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"8105db6d12e07f804938b1f4b04d855f7820b702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-28 16:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2e6e556f8b91ab5725b659c4743d2db020f4daf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 16:51:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b3cdcf01e9429d9fb9f99027092e2a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac32a52f45d04244b769bc042b3a14f1 : FAILURE in 38m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf7c2eaf00f44317b0ca0f914f35e482 : FAILURE in 38m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d035a36f4608804c05b05c55e69d7184088d828c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-28 17:04:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/45/859645/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0a1cddc4e6cd3105e6bef658fb9210a4ded5b360","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-28 17:50:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/859645/1 : SUCCESS in 1h 44m 02s","accounts_in_message":[],"_revision_number":1},{"id":"91bcb43313a1631875bbf45b8d69699b311c1fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 17:53:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ecde32fbc5647518f37985d25c70f3f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/86d044b1292e47258b10937d83d083d2 : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a1502469436f4eeca78a9e36705cfb36 : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01d26fdab5d94e30a6c886560476ba58 : SUCCESS in 1h 25m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/855072e5dc8b400087fb6992a584e86f : SUCCESS in 1h 29m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96767aa4a28b4bda9586ebdcfe774c79 : SUCCESS in 35m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d52d50c7d36946e4be5300d90a78b261 : FAILURE in 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/678bf6ba707f4d338247c5b60bcdb57f : SUCCESS in 14m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/925bddec6ca0421bb6870d895987d17d : FAILURE in 24m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efc30124aa2d48f795a2bed95208f2dc : FAILURE in 30m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da7f465d66b24ee4b37bb7a8a0cbaf40 : FAILURE in 21m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1e5667aa0b41e58b5493bf3601719e : SUCCESS in 9m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7017844668da4edb97dcbff9184e7f63 : SUCCESS in 49m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/da65e4bf4afb476194a848d28a2941f3 : SUCCESS in 1h 42m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/417f74b0c8334c2a9eb91906c26afcaa : SUCCESS in 1h 17m 38s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99c23d92016743d99d011e7e2967f9e4 : SUCCESS in 49m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5fb357742fac453083e42ef788f45d99 : SUCCESS in 1h 12m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2af70c8e3c4343cdb5643f75b29f90c6 : SUCCESS in 1h 12m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/47e1056de37c4f5d96f65444c731deda : SUCCESS in 1h 06m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c7462458ec342f2a25a8709654fbe22 : SUCCESS in 56m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b88910ff187343bcbd5303a756231a2b : SUCCESS in 1h 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cebaaa8c9b8316d7efa08b589a1e16e20dfe107d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-09-28 17:55:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f3558293a36376d1430fcdc32e9b654f20eae4b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-29 10:52:38.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c296765e937cedf95f1c7fb1583cc47ecd8b092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-29 11:13:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"705a1b004d800e75427a5640c314b70a646b98d4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-29 11:27:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/859645/2 : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ac6cf687aa0824cbcb11fec34b6c57c8343483c5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-29 11:49:43.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/45/859645/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"331b81a5783c173e33aecbf949b66e12a722727e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 12:00:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59c7748ef8e54e3183c54be4ddff6b38\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6807135a44c544e5a11119387ca95813 : SUCCESS in 33m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ace5b02daf044e3add8783255f3daa1 : SUCCESS in 44m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f34e19e03866b2eed060b22397c9436803494be7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 13:17:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abebf810c0754d08861a26cecfb81562\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2c7aaea1396445b2a5948df09a0e062e : SUCCESS in 1h 40m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4961407bacb84e72b3cb3ca73c95a535 : SUCCESS in 1h 59m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04697e4998104fea8471984f272bf409 : SUCCESS in 1h 20m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac248a65106546b18242fecc5042e605 : SUCCESS in 1h 48m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b214a35fc6f74cd7be04c7348b702617 : SUCCESS in 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb1d9ad0d0594ee3a6f662d4f319af89 : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50f3ecb20c34498c9f4d9abab6ff3b41 : SUCCESS in 10m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c09b3dfb33a24c029a7027b8ca6014ee : SUCCESS in 27m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5086fd65dfe846eda17d1e29069fe34c : SUCCESS in 27m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca5f697bca324b8f9b1572f4378ccedd : SUCCESS in 23m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99bfe1d8aed64b7eaa2d5e791b846732 : SUCCESS in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6adb8a42c82f453689529a8e75bb058e : SUCCESS in 53m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0afd848038ae4be99f7a4a7740acac86 : SUCCESS in 1h 34m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f458217cd1847559376b24d48b4631b : SUCCESS in 1h 03m 23s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e11f54b92dd41968819cb6459a539fe : SUCCESS in 56m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/18be0dfc17004d178a448fff2837fb85 : SUCCESS in 57m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/50eba5f053b94a5487cef257cb3f554e : SUCCESS in 1h 17m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/196f3a699591442c8131234cb2bf7c34 : SUCCESS in 1h 17m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f66e128b0c9849579eb54fb14f81e348 : SUCCESS in 53m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1e0fbc3dcfa44829e17734b3c6cd86d : SUCCESS in 1h 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b0ddd17a3e75945b04ed9fc0e83f463a734353e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-09-29 15:04:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a03ae1deb31a7c3a75585152b4516023984ccbfc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-29 15:15:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f64444bb29d147e40bf7a3de1ca37cb16d81d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-29 15:40:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e28273fb71489e0f3046de1d6c7b88f2008da99","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-29 15:54:51.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/859645/3 : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":3},{"id":"212e06ec48f002cb3a26579c77de73901eb3d7ce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-29 16:23:38.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/45/859645/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"1a3b0fb3ee07b6730efe99d72f3eb564ff467566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:41:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b2a1fc1af5945e3a8134a3edc60c02a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee3fd26fb89f490e916e44528d12b5f5 : SUCCESS in 47m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9524b815d95a429997ee27cb6442590a : SUCCESS in 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3f509607027b9173f5cecfae5d1fcc183958d141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 17:44:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbde351076f14130b7b6b88652608d1d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5df3de3d3d874a22b31e3ca2d2cd2d86 : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/786b5b46b5f24cf7b2379db09df95bfb : SUCCESS in 1h 49m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50e5ebb513cb4ceabb2e69b1164c13e3 : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7a75a262e8cb4d34862b96dde0dbcdbd : SUCCESS in 2h 02m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3cbbc11d75b4640b74e4df8d05832e9 : SUCCESS in 1h 02m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ece3377386a34d4ab2b368f01a5d73a2 : SUCCESS in 38m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acbd16795b634482a76473492e0850ac : SUCCESS in 14m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f035724a2f64cbc96cc359b24953afa : SUCCESS in 27m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1c0054cfb44438187ce0bf6f6c665a6 : SUCCESS in 31m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4de5186b99604d53bb5b336e43e678d2 : SUCCESS in 27m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a7a608635b4c46b7cd1359ccfd9e6b : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a988f7fe4bec4a6e9a842075a09ba546 : SUCCESS in 48m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/658202f6f4ec4218a4796fe3432402dd : SUCCESS in 1h 40m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/509fe431340d495f884913cd65318285 : SUCCESS in 51m 56s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccce6211eca445d8ab5a824550a66fdd : SUCCESS in 1h 05m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8c0647f7b194329bbbf33aba46895f2 : SUCCESS in 49m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a9d535b445284fcdac265a636e2b7c2d : SUCCESS in 1h 38m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/243d6a5b4fd046da809964b7ec5324c7 : SUCCESS in 1h 09m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7ca1acfccb34b9cba06757f52726dee : SUCCESS in 1h 01m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dfc8773120574dedae37765a1bec3341 : SUCCESS in 1h 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8df5b68e9cf4bbee26828a392db4a8a45f36f7de","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-09-29 17:48:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"01a39aad3944285d334983b442f3546e042077bd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-30 09:56:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b88de2a6d5fee1dd29a9e94dbed4c2132bc30e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 09:57:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bc0a8a2a965c3dec59a3da76f068f656d7aa0c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:46:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83563dea22474302b08697ca8214ba60\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/41843ad0346a45c69b1ab4507a4a836c : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd13af3a1df7488199d6490f85fc1077 : SUCCESS in 1h 30m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/da95b136d8a34085abff8b1b92b0fcdc : SUCCESS in 1h 35m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ca80b113894b4ec6814fe2b310d6da91 : SUCCESS in 1h 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f1984a8f394b3ea1eaaee07d8d4466 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165f354fc98d475c9fdd692d3cdf5b5f : SUCCESS in 26m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bfffb3844104455ba2905fd42469767 : SUCCESS in 27m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4135861107a6458187f145b03a97d650 : SUCCESS in 9m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5aa31746c7d4214a4df267acd209c46 : SUCCESS in 46m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f6a2e989f254f7b9834edddb452cb51 : SUCCESS in 1h 42m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6225446274b2445481a19939ef1ac9c4 : SUCCESS in 1h 07m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d02ff8c08b04d38a5a8683fff2c07ab : SUCCESS in 1h 03m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a79558d098634ddeaeb8b6f87e29d68b : SUCCESS in 52m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f1f9f2c89914b5ab26094b2bb3ac164 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":3},{"id":"40744abde20c531c48a63aa8c35131bb17dee7c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:46:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"26e9e4885e1b405bd10693ad8816dbc70c2c2059","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:48:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c98e54ac92e426aa452869e46b9a758\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7cac98c5174e2e8f9492a32631a2f7 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"21491efd9f8ce6df98cb58c26da7896d75cb4a8b","revisions":{"0796d421ce66d1f7ea9127d0091727dc317477a2":{"kind":"REWORK","_number":1,"created":"2022-09-28 16:05:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/859645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/859645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/859645/1"}}},"commit":{"parents":[{"commit":"994e2aa8f6b12b0a7f5dd7b2a0fd0b3e63652bba","subject":"Merge \"Add missing comma in .pylintrc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/994e2aa8f6b12b0a7f5dd7b2a0fd0b3e63652bba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-28 16:00:25.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-28 16:00:25.000000000","tz":120},"subject":"Port provisioning should retry only for VM ports","message":"Port provisioning should retry only for VM ports\n\nThe port provisioning method ``Ml2Plugin._port_provisioned`` creates\nan active wait to provision a port if the port is unbound since [1].\nBut this active wait should consider only VM ports in the case of\nlive migration, as described in the LP bug [2]. This wait should\nnot consider auxiliary Neutron ports or baremetal ports (we don\u0027t\nlive-migrate then).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/855257\n[2]https://bugs.launchpad.net/neutron/+bug/1988199\n\nCloses-Bug: #1991092\nChange-Id: Ic8891e2deef4bb5e72cf7d7f37b043e936adbc00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0796d421ce66d1f7ea9127d0091727dc317477a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0796d421ce66d1f7ea9127d0091727dc317477a2"}]},"branch":"refs/heads/master"},"53b8dbde54200d78791dd96eda8372f1ff5b7d4c":{"kind":"REWORK","_number":2,"created":"2022-09-29 11:13:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/859645/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/859645/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/859645/2"}}},"commit":{"parents":[{"commit":"994e2aa8f6b12b0a7f5dd7b2a0fd0b3e63652bba","subject":"Merge \"Add missing comma in .pylintrc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/994e2aa8f6b12b0a7f5dd7b2a0fd0b3e63652bba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-28 16:00:25.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-28 18:13:07.000000000","tz":120},"subject":"Port provisioning should retry only for VM ports","message":"Port provisioning should retry only for VM ports\n\nThe port provisioning method ``Ml2Plugin._port_provisioned`` creates\nan active wait to provision a port if the port is unbound since [1].\nBut this active wait should consider only VM ports in the case of\nlive migration, as described in the LP bug [2]. This wait should\nnot consider auxiliary Neutron ports or baremetal ports (we don\u0027t\nlive-migrate then).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/855257\n[2]https://bugs.launchpad.net/neutron/+bug/1988199\n\nCloses-Bug: #1991092\nChange-Id: Ic8891e2deef4bb5e72cf7d7f37b043e936adbc00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53b8dbde54200d78791dd96eda8372f1ff5b7d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53b8dbde54200d78791dd96eda8372f1ff5b7d4c"}]},"branch":"refs/heads/master"},"21491efd9f8ce6df98cb58c26da7896d75cb4a8b":{"kind":"REWORK","_number":3,"created":"2022-09-29 15:40:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/859645/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/859645/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/859645/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/859645/3"}}},"commit":{"parents":[{"commit":"994e2aa8f6b12b0a7f5dd7b2a0fd0b3e63652bba","subject":"Merge \"Add missing comma in .pylintrc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/994e2aa8f6b12b0a7f5dd7b2a0fd0b3e63652bba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-28 16:00:25.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-28 22:41:01.000000000","tz":120},"subject":"Port provisioning should retry only for VM ports","message":"Port provisioning should retry only for VM ports\n\nThe port provisioning method ``Ml2Plugin._port_provisioned`` creates\nan active wait to provision a port if the port is unbound since [1].\nBut this active wait should consider only VM ports in the case of\nlive migration, as described in the LP bug [2]. This wait should\nnot consider auxiliary Neutron ports or baremetal ports (we don\u0027t\nlive-migrate then).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/855257\n[2]https://bugs.launchpad.net/neutron/+bug/1988199\n\nCloses-Bug: #1991092\nChange-Id: Ic8891e2deef4bb5e72cf7d7f37b043e936adbc00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21491efd9f8ce6df98cb58c26da7896d75cb4a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21491efd9f8ce6df98cb58c26da7896d75cb4a8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
