)]}'
{"id":"openstack%2Fneutron~859997","triplet_id":"openstack%2Fneutron~stable%2Fxena~I0de5fbb70eb852f82bd311616557985d1ce89bbf","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1979072","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-10-04 14:40:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0de5fbb70eb852f82bd311616557985d1ce89bbf","subject":"Script to remove duplicated port bindings","status":"MERGED","created":"2022-09-30 15:24:08.000000000","updated":"2022-10-04 14:41:53.000000000","submitted":"2022-10-04 14:40:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859997-bug/1979072","meta_rev_id":"cf70d0f1f20bc0aa0ae0f3c077813be98619c3d3","_number":859997,"virtual_id_number":859997,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-04 14:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-10-04 08:25:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-10-04 12:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-10-04 12:05:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 16:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-30 16:58:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-30 17:14:54.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-10-04 08:25:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-10-04 12:05:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7b2ef846d66c53d2c0953837d9ba16f0301e16af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-30 15:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66b4d5fabf5b459ab60853bfbb9db2008669c4a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:06:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/410ae55cdce44e438394e95d06867db9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2db19e7905c747898b74ff4238b39438 : SUCCESS in 29m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40a94c7c70bd4acaa53688aff76dd279 : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd8b7b7cd787eede89403c327923ab434446b666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:58:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c54fff268674c9799c99b8f62daf41a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/17cccbcb9002480cb71329746dd4bced : SUCCESS in 2m 31s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/e85f2d6217fa403780376883daf50350 : SUCCESS in 52m 45s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/d5d30c65df6c48759da46fb1463c0a14 : SUCCESS in 1h 07m 51s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/ddc3928477fb49ceaa745d5f1b2b2799 : SUCCESS in 55m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/4637305186ae4f00a2727c4a6e3716ee : SUCCESS in 47m 37s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/e75778359ef74ea2b4517bb2e5e76ffb : SUCCESS in 1h 00m 55s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/bb8f7196845942769b2288a4553a68d3 : SUCCESS in 32m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ffa4337aa3b4c46ac236f3214d5e395 : SUCCESS in 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26751bd52a8643a7b707b71f91622c39 : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/728f1bb2e44e4f2cb5b2b76351c9bd46 : SUCCESS in 26m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d3e755d2cc04e92b27ee6b3ea265593 : SUCCESS in 26m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83dc9977a12d4906a32f4d3d94f4b2b1 : SUCCESS in 22m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c838413f1f478d93087bfb9a1d7dab : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13f6ac09bb4846a99b99a11668ec15db : SUCCESS in 13m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3dac58cb5693456ab0ba9cd6ba6afa78 : SUCCESS in 54m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/667d372e67154ba3b62e61083b7847ef : SUCCESS in 1h 27m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/78ed503cee6a4166bd65a19c1e1d6a42 : SUCCESS in 1h 04m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/15d9054d4a614e868e446954300381ed : SUCCESS in 55m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eac6f18f1c894b9b98d56ad71ecd71f3 : SUCCESS in 59m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/831fa159590b43fd9d703d47498e6fad : SUCCESS in 1h 16m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c49fb2948d904d25a8d0b20206b2bf74 : SUCCESS in 1h 22m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b45862318824cd89e588977e7f462f0 : SUCCESS in 1h 03m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c36bb88d813b47089feb979c0e6292ef : SUCCESS in 1h 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"750265001f8f2a923346e87effc396e1d44fa285","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-30 17:14:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/859997/1 : SUCCESS in 1h 50m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2f2783f2ee8743c66bce15f75e9a54dcec4b9b6c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-10-04 08:25:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12dbeed7676db863df41d4331528083b67356516","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-04 12:05:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a90ff55abfd700ceb09f9e7586d9e39f140295fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 12:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c0e9d16d23bf8e692ecec20cbe991429481f6c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 14:40:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82203a23b22441a39aa346a384e60ba5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97fb8929bfa2467fadaa156459036281 : SUCCESS in 2m 30s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/b940f27ea54d4fd0a94fea5d76b777c5 : SUCCESS in 55m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a7d28e39434917a05856fd449ff736 : SUCCESS in 20m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ffb110bc9914a4a8205ce871b764f1c : SUCCESS in 31m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e4b176e4b9e42369435c877ecb68878 : SUCCESS in 26m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ed85342d1042f58617a251d50f9ea4 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/595bea3dcff540a488f0a75c1120ff35 : SUCCESS in 11m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b066a6b3f853417f8f118c0225973ed7 : SUCCESS in 41m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/efe261fee1fd4dfa99201d121812aef5 : SUCCESS in 1h 12m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e2d856dee17042769bcb14e55db15d39 : SUCCESS in 1h 08m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc702e5135d44ffe9600385294bbc9da : SUCCESS in 56m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73a69654120c48dd84c8ae651b05bdf0 : SUCCESS in 49m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bdbd9662504b91afa0809d98f710fb492cdd934d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 14:40:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf70d0f1f20bc0aa0ae0f3c077813be98619c3d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 14:41:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a6ccfad48c453b8d9c9175d46a13b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1563560a5114f059affc046f990c513 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65d17fe0a7544abd9eef4549b7f9ac1f : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d033ab6eb6dadeb1770eb259d4d59b469d9e3bc0","revisions":{"d033ab6eb6dadeb1770eb259d4d59b469d9e3bc0":{"kind":"REWORK","_number":1,"created":"2022-09-30 15:24:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/859997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/859997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/859997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/859997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/859997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/859997/1"}}},"commit":{"parents":[{"commit":"d847b52c4f36e17c9c360320d160a0d05330a71c","subject":"Merge \"Use \"OVNMechDriver\" instance in \"TestOvn[Nb|Sb]IdlNotifyHandler\"\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d847b52c4f36e17c9c360320d160a0d05330a71c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-07 06:31:22.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-29 22:24:55.000000000","tz":120},"subject":"Script to remove duplicated port bindings","message":"Script to remove duplicated port bindings\n\nA new script to remove the duplicated port bindings was added. This\nscript will list all ``ml2_port_bindings`` records in the database,\nfinding those ones with the same port ID. Then the script removes\nthose ones with status\u003dINACTIVE. This script is useful to remove\nthose leftovers that remain in the database after a failed live\nmigration.\n\n\"dry_run\" mode is possible if selected in \"[cli_script] dry_run\"\nboolean config option. The duplicated port bindings are printed in\nthe shell but not deleted.\n\nRelated-Bug: #1979072\n\nConflicts:\n    neutron/conf/common.py\n\nChange-Id: I0de5fbb70eb852f82bd311616557985d1ce89bbf\n(cherry picked from commit c5b76a8393a21adb87447c925da2ede4a75dd11a)\n(cherry picked from commit 3d307ef8f8c31ebc92a621aac75c62b83a7b3787)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d033ab6eb6dadeb1770eb259d4d59b469d9e3bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d033ab6eb6dadeb1770eb259d4d59b469d9e3bc0"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
