)]}'
{"id":"openstack%2Fneutron~860639","triplet_id":"openstack%2Fneutron~master~I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9","project":"openstack/neutron","branch":"master","topic":"bug/1991965","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-01-16 15:02:54.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-01-16 18:59:59.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-01-11 22:43:52.000000000","reason":"\u003cGERRIT_ACCOUNT_4694\u003e replied on the change","reason_account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-01-02 11:20:47.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-01-13 10:46:55.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2022-12-19 13:01:12.000000000","reason":"\u003cGERRIT_ACCOUNT_32586\u003e replied on the change","reason_account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-01-16 18:59:59.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-01-16 18:59:59.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-01-10 15:27:47.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9","subject":"Support for minimum bandwidth rules in tunnelled networks","status":"MERGED","created":"2022-10-07 08:14:47.000000000","updated":"2023-01-16 19:06:44.000000000","submitted":"2023-01-16 18:59:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":53,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860639-bug/1991965","meta_rev_id":"01c334121467a0cf63b6b9ce6a93f72731ed5ba1","_number":860639,"virtual_id_number":860639,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 18:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-05 14:05:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-01-11 22:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-01-02 11:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-01-13 10:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-01-10 17:55:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-10 15:27:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-01-11 22:43:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 08:58:14.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-07 09:00:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-07 09:47:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-11 15:20:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-10-13 15:52:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-11-11 14:23:25.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2022-12-05 11:24:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-12-05 11:24:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-12-05 11:24:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2022-12-07 11:53:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-12-19 16:13:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-12-23 17:23:18.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-01-04 17:43:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-04 17:43:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-01-05 14:05:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b356ca3090a4c81903c2016a5b2d4b9662101914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-07 08:14:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"456e7d3827e44bba9ad9a2b2dd71f2dd9d68de74","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-07 08:16:25.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"51daa4ab159cbd018b422cd43a23cae29a90d314","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-07 08:58:14.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/39/860639/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d5e2d535336940a54cfccfd325df5a2c71493202","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 09:00:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c301089267b94dfab459bdae2931e4ed\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcad6b9bd52d409f85628423394ab85d : FAILURE in 39m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f164c3980d64d28ba26f4cae1180e00 : FAILURE in 39m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cd1de11061c23dfe919b0a4e861450acc36ed7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 09:47:13.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df9fbe53f6d94a9684a519c0aa4d5f5d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/64a4ec1941df48e78b88a7a4eb2a0b50 : SUCCESS in 1h 22m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f9cef20c5e75464bb290cac082b1241c : SUCCESS in 1h 29m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e1fe40361f04894adf0be3620fead48 : SUCCESS in 1h 16m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4895b943c11f41b0b1a546548354d91a : SUCCESS in 1h 29m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebabb1b7e722434897271038e73563a1 : FAILURE in 34m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f7f8b6008049c6add5392124d7d5d5 : FAILURE in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b8fb318bb6448e7a55932142850f8de : FAILURE in 22m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc91735685b7427ab02cecaffce0c651 : FAILURE in 27m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89a7e5f1fa454eab8541f6335e39d7a5 : FAILURE in 24m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da28be43fb454e0db3b0a48846ff90f2 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/02e262fbf86d4170846a7a525aa8e923 : SUCCESS in 46m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92bf61463d384a63b9c017d365afe4a4 : SUCCESS in 1h 27m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bda86ecc660c47b9be592ae21446fc69 : SUCCESS in 59m 01s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa65406f267e46ae81ee6347717fdbad : SUCCESS in 48m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/752cccce105f4cf98c5234dadac00cad : SUCCESS in 54m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2e21d7cb07c243628d08116d15a5fe99 : SUCCESS in 1h 08m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c39af41d45044f4ea94bb14cea57310f : FAILURE in 1h 01m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bbb789780d2402e9c4d5ac0c5e0578d : SUCCESS in 49m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a7245ddc5ca4fe79d9bf7adde9eb613 : SUCCESS in 1h 06m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fdf240472118f866d5ac07604163257da0e1328","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-11 15:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0395ace9f7b2a69933d4436b132d3de3c8c9fa25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-13 13:56:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ef7c46e46eed722bb6ee802bcc3a1d9c3af8784","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-13 14:08:54.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"9e82e8bfadd97bffff4bd0011af14332268f5bfe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-13 14:32:34.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/39/860639/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e3a3b06cb8eb0edea779e683811c6adc69d60000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 14:38:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f6cd73cf8144c5a7b07c1be9e13c10\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35593545df554f5c8871f93170540f9e : SUCCESS in 35m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47689dc93f8d48a2b1a631ee94da9afb : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c07158bfd954b339cf177161ff71a94 : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c8f8e4c3ff7b86f8bc90acd8c2b996747f85b5bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 15:39:38.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efc22b85778640a49a75334ab3737e64\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/662ac43ad2c44905a3a4d03bb1e03744 : SUCCESS in 1h 41m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/53a605f730724240a3ea6b6b72ca5278 : SUCCESS in 1h 32m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f17badee7ea4270956c6d4dc2909ea8 : SUCCESS in 1h 08m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7f9285d36154ae9b2ebe576791358c5 : SUCCESS in 1h 19m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4790b4c99ee24c4fbb98079356705af5 : SUCCESS in 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc081bd885724a42b36005543c8a0e2e : FAILURE in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91fb7b76615e4e44b25e142855ae9b05 : SUCCESS in 26m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d46253ca6234b93bc1136793ae7b7d0 : SUCCESS in 29m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7edb9ba3ba054e1b916fa33cfcc71e4a : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be89b9623f9b4f31b693b0b180e4a962 : SUCCESS in 28m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/105db602434547bd961c26784232c254 : SUCCESS in 50m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d6616e9a69149b59313f78581de9432 : SUCCESS in 1h 31m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36f2e5d01c2b4f31a71931dcda010796 : SUCCESS in 56m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3a202a36276a40abaf463e2090d410a5 : SUCCESS in 52m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6068cef4257f4a7c8090ee9ca5be8215 : SUCCESS in 1h 12m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ed39d87772f64c8eb3001bc1ca57e99b : SUCCESS in 1h 16m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c855120c0f194c5f9acb45a4fc0f793c : SUCCESS in 59m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/227b39b6285940bca5bd8aa80919ff8d : SUCCESS in 1h 03m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ac3ffb4b112549bd9dc93b88cea21f2d : SUCCESS in 1h 04m 43s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c6b2c68a3f94c3c952d7e6512b56b60 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a20330f5a8ad6cfc2887c5e95c50aad9602d42ac","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-13 15:52:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860639/2 : SUCCESS in 1h 55m 20s","accounts_in_message":[],"_revision_number":2},{"id":"652df65874e0d3cb6c2a659a7430e0f4f14a9f33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-13 16:29:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1349ca70d81603aab6c81cbac94d8b5883f7378f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-13 17:02:42.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/39/860639/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"51edadb2bd894815c329883bd8d4a2c39974e543","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 17:13:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/760bfb7a8a534fd19d2165c259022ad6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b008fb9525654ef0b2bd6cf7f64b1cc4 : SUCCESS in 42m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cad69d9cf5134e61b8221b1b7b3d28c6 : SUCCESS in 39m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1f2a23aab0d46288c580bfc404e2be6 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3779775a13f5db5de95a9e8c20a6247ed7089330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 18:04:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99aac569d14144469439ceb85198eb96\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9258c479f3f644f18f97e5f7c3a10656 : SUCCESS in 59m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15c55cb38ec74911b76d77f08d321aa7 : SUCCESS in 1h 14m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/298d3ab644954fb6b85026ebe3ab4e48 : SUCCESS in 58m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/26553749632e4817b0473777678ac3a5 : SUCCESS in 1h 03m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0578615101064a84a8cd3121898cd99a : SUCCESS in 41m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb5e22d7cba34f7582ec2ed7ed18d251 : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1500e28f37e4b20945a4b2d2863f7d8 : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a67484d967c4434be56a92fc79bbbe3 : SUCCESS in 21m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51a152474a814bdf94bb22d924cc95cf : SUCCESS in 20m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9734bdd58c457ab9a3976f3a2fe281 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a36e7e64246451bb295d55698aed9aa : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b8b0381d8624b5ba30bdd2818570249 : SUCCESS in 23m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/92134cd09f4e45ef89b265ded9a9fcc6 : SUCCESS in 45m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3812d6af736743bcab2c93003f778d49 : SUCCESS in 1h 33m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4528133fefb54072a8d5a76f6c5ba405 : SUCCESS in 48m 33s\n- neutron-ovs-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/295aa9b5b20e4f499aa9afbb370474e3 : SUCCESS in 53m 20s (non-voting)\n- neutron-ovn-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/0cb7f08e7768490a9e67b449cfb8e29c : SUCCESS in 1h 02m 25s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/efe2151783bf4565b7ccef401c93a644 : SUCCESS in 49m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3c123b4af43c44328f86b3f7b7c26a19 : SUCCESS in 1h 09m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8eda598a41044b18aefd294e68940d46 : SUCCESS in 1h 28m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd22bcb7da1048f09e462d7325adc668 : SUCCESS in 50m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2f420cdec4e54ccaa0abc44c8008a76b : SUCCESS in 55m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/563d3320c1cc4fce90b695487331d809 : SUCCESS in 55m 33s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/afbb5345376749dbad1c66e501fccbac : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":3},{"id":"626f760e55d65040a3eeebfbe7b5e779604cc9b5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-13 18:22:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860639/3 : SUCCESS in 1h 52m 46s","accounts_in_message":[],"_revision_number":3},{"id":"2a36b40c1c290d2fc487d2ef741fff99301a1e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-14 16:17:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"30fd4a3b8e11c2fdd80d125827fb335bb184dc89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 17:12:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fd672f4251d476e8dfa1f87483c7fb5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2c3d69ef9e542dfb2761f9660a61efc : SUCCESS in 48m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81d5503b2c3e4616b65e167b41353025 : SUCCESS in 48m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96b76393f1df437a86b005e52e9ab853 : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":4},{"id":"d7a5743ef91e8f172c76544fce66e88523475ea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 17:47:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee4efcaf33d04245b8024edd3a815429\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c0cd94d3fcfb4055943d041db1b96693 : SUCCESS in 1h 19m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9bcc7cdb35e448738afb1b48d4e8d653 : SUCCESS in 1h 17m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3ee82f881a9f4b3384ff13eeeff4834b : SUCCESS in 1h 24m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94efdc697e5c43f3bae0e2a781482da0 : SUCCESS in 1h 06m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ba1c470bd7f493d8aabd0ddf42f7d20 : SUCCESS in 31m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e02303f068e4669bae6fd595a413fef : SUCCESS in 27m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b82fe934ae4e7f8ddab522fa19ca77 : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dc81ccf02f24da4a28c2b7362a9bfbe : SUCCESS in 22m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc8cb1d6194b4dd5b38f9d077027f956 : SUCCESS in 23m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d3a83e6575433d9222f40e390c36f2 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c211f1367e54868a36359e6ecf35220 : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5938439d261f4a849c219dffe2b59435 : SUCCESS in 27m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dea61313d3af44398b95dc44ea5fddee : SUCCESS in 45m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8eaa788a166d4e70823927e89e9cdc96 : SUCCESS in 1h 17m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c251ba40365497e9e63b697edfa621a : SUCCESS in 50m 31s\n- neutron-ovs-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/48107d43b1ac404bbbcf7d57fa867a42 : SUCCESS in 43m 52s (non-voting)\n- neutron-ovn-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/8de85edcf443470b84f187ef71317d76 : SUCCESS in 1h 02m 49s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ee6f90be865042eca785917e9a534ed8 : SUCCESS in 43m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eac0b53bd5194903a830fa9ce3c4d2df : SUCCESS in 1h 20m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/475e91d52407431380d0b5630e15f56d : SUCCESS in 1h 07m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/24aa9b8a16f24c5eaecbf6885a8725e7 : SUCCESS in 51m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c38f8160ca4a42a3bef5569c2e2c0865 : SUCCESS in 1h 11m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9432a7bdbf9e4e3a9180cac86974ef82 : SUCCESS in 1h 00m 29s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6dda96a48bb429e816c792271578e1d : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":4},{"id":"18d0c0ba94d84353da221832a4315478befb110a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-14 18:14:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860639/4 : SUCCESS in 1h 55m 44s","accounts_in_message":[],"_revision_number":4},{"id":"158b080ac2c551149fe4a894ea9f453114043166","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-18 06:41:27.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/39/860639/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"9f6a764a95feb9f739d160e24529f15c8f9f406e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-28 10:49:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"46fce94ccff985b70647483508718623cfb0e82a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:47:47.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef5f451b652e41c38fe248b2654e6394\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0d2c71c0bc94d4d8563cf76d047ea0e : FAILURE in 55m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/384918e1df1e4eb6b7ff7d459325575e : FAILURE in 55m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c72a13d803c4b4ba571ad1027c1ec33 : FAILURE in 55m 28s","accounts_in_message":[],"_revision_number":5},{"id":"759fccb93a821db8ad42f9ea9e4dcf2562499b3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:33:04.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f681b1beaa740fc8b3fe82baa2a0ddb\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/cfe4e82b8a8b4dff8f77dc6a1aa5d57b : SUCCESS in 1h 12m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/df657710f9d14de9a86b3a10e401398e : SUCCESS in 1h 07m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8ba24715b01845c8a6833e122dffb143 : SUCCESS in 58m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d6af225cd2cf42eb91f6278f411d8611 : SUCCESS in 1h 13m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c0904f095b3442286f9cd23498e9d47 : SUCCESS in 28m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9094827f5cf4a5e92cdaf30a7366ff0 : FAILURE in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25010d19f6c493b8715c5c2cf062895 : FAILURE in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b4a86380a8d4865b8521b35fd1db538 : FAILURE in 28m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9710c7170ac40c5be6d247f89445924 : FAILURE in 28m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90c6c8d1b2846f59ecd61d830225c5a : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94c34baffb5043a0aedcf4e283a69c5e : SUCCESS in 12m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b289d8fada7480da342a83847e69aa3 : FAILURE in 26m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ed6ad10954143b9b8dcc87f14c5a137 : FAILURE in 42m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6995bd7e6df546a793b0d47c28fda0b3 : SUCCESS in 1h 36m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f77bc2975ce04272b99b2f77eaafe91f : SUCCESS in 1h 02m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f9586348a66147048c83f002e890c49e : SUCCESS in 47m 55s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0f82c79dcd27470097db5209c98e507a : SUCCESS in 58m 33s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/77e7256cbf7b4a878629d97959210749 : SUCCESS in 1h 04m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fdc27cf39c0a45f88510681b4c0c3097 : SUCCESS in 1h 03m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2dab6244998445679e697579e7a5d2e7 : SUCCESS in 56m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d48b568f3a6d4f51a28eea755806773c : SUCCESS in 54m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/adcc4bcca7994fbca7eea889fea0392f : SUCCESS in 1h 01m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c8b76f395994f538c1ad5708111e166 : SUCCESS in 1h 02m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/07009cf515d94bc395ec96b0e4ac5765 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":5},{"id":"7bb8026ddc444d3b9238b31278ed9e45c2fe6085","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-28 14:03:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860639/5 : SUCCESS in 1h 58m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2320caf2c10f475e19bed556f377d8b60b975c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-28 16:38:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6207c025bec584bcd56faf0c567eb1cc932f372d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 17:29:58.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e89a4168cc4e4aedbd52dd65ee9d3ff6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47fbbf6dc078401796fc2df3c4d4ebaa : FAILURE in 37m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e3da43ac3124cbebbb3da05d1b184b7 : FAILURE in 40m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e577df0e0ba4bf7835c3bce8153f710 : FAILURE in 42m 46s","accounts_in_message":[],"_revision_number":6},{"id":"ec9cdd8bf202149b0767719f7db8cc2da8660e05","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-28 18:15:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860639/6 : SUCCESS in 1h 35m 45s","accounts_in_message":[],"_revision_number":6},{"id":"bfb049e93a82160fd405ed296bf78a663114f222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 18:26:28.000000000","message":"Patch Set 6: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98438f5d3f244b5ab328c8e4d736ed7e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d36dd1183e014f8ea04a6df2444c26c7 : SUCCESS in 1h 19m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f71d9a71e76c4d1cad9ba83be702298a : FAILURE in 1h 29m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b609e4395af4919a23193774b614ac9 : SUCCESS in 1h 19m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f34dd46939444402a67ba19104338700 : SUCCESS in 1h 22m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b458b86c6a28499f871fa791f6fe3253 : SUCCESS in 29m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29accfb61c1f4ac4841bd6a293a9814b : FAILURE in 30m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e2ce62a493f4a689a6d0ef8370d9a9b : FAILURE in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e0b95bcab0141288d95cfd8438e5b46 : FAILURE in 28m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7c15d1b9de1443c8a8b36076f54405e : FAILURE in 27m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b62daaccc60a49feb222f7ee9a01aea1 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55fb0d6bf3db4abea7956143d8dffc91 : SUCCESS in 14m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0423f8b8949b4b8dbda2010ff3f75005 : FAILURE in 27m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88eb6b02e3f94259b728e6863d350b4e : FAILURE in 50m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/00691b9e7b344387ba9047f0cea14f15 : FAILURE in 1h 23m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/23280dcde27a4fa1ac98792427d3d30c : SUCCESS in 1h 02m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ef11e332286d468794422dd0a83b6288 : SUCCESS in 1h 13m 59s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/adbd4cddad8f4c2080726cd53949404f : SUCCESS in 1h 19m 41s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/20fc79f68fca401e8206b4c60e76c315 : SUCCESS in 51m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/468d9cca9ea548818d249a405e8bb67a : SUCCESS in 1h 09m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8bb6e8d8387b4441897c7d1b1f85d4fb : SUCCESS in 1h 39m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1e58b94900cc4605a393bd6fe1a42f44 : SUCCESS in 52m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e43d7fc2b5964e3c8700476050ababd3 : SUCCESS in 1h 02m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/83de4438e0bf4eb5b3beb1e225a8e94e : SUCCESS in 1h 17m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/96d8253bfe3a4d5696ae9e6fbe804200 : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":6},{"id":"822a59e8a95cde8481d4a282a78f22e98bdf6ef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-02 10:38:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"16abb962cdeba5cbf04b0688f2baf00df7095222","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 11:37:57.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3ef3c47ac5845d4899a1fab35f761e1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a66e1c5c3d4c4c589a1a47ef65bd4130 : FAILURE in 48m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c7e25737bdf450bb228f031e0e56c5b : FAILURE in 46m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe8e95e2c5c9464392a9b1e0c3b9409d : FAILURE in 48m 26s","accounts_in_message":[],"_revision_number":7},{"id":"0d79ab23addb7c3c87966b56cbab7729b7dbc1e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 12:08:43.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a92206fc2e3d43479848999c3916e622\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5e733d16d45942fda42630a65a578917 : SUCCESS in 1h 23m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/85c6d4fd2ce44deb8ad7cb3617b1b23e : FAILURE in 1h 27m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d691310f062b4211a9c18fdba2597929 : SUCCESS in 1h 12m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/006d2df629bd4fb0a91c9de7b7dc6f89 : SUCCESS in 1h 25m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0954d7657cdc4a8e8184b0d54c700a4f : SUCCESS in 33m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a16083c517864a228b51528027883493 : FAILURE in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55be3f4d8114edfa4026242fc720524 : FAILURE in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4751b1a6b5a4690a1ed449659c2039c : FAILURE in 24m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e750357faaf94c2c9ace1ea3a6988bad : FAILURE in 24m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d9bb17cb6d4636817ee9952ba9d779 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cec987f471a48629f4a295102bdb825 : SUCCESS in 12m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45867349c62a4c79b2cf405894363733 : FAILURE in 23m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/40e3d45708d34d96a2f12907930cc9ac : FAILURE in 40m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf83447c77f640f08062d1b9c1640e19 : FAILURE in 1h 23m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa49f650eaf046c78f22e14478a717ac : SUCCESS in 52m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4f1e244a0c5d4267bdd680622de77940 : SUCCESS in 1h 01m 40s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/103b855f2fea4e1191fb8040a81be089 : SUCCESS in 1h 00m 15s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9f4a084fe73462b87669618138f9d96 : SUCCESS in 1h 01m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ef80b2ba29cf4515be32163129f3da55 : SUCCESS in 1h 11m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/70d445377ef443099040031ba39496d7 : SUCCESS in 1h 10m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aee4de4f02304f5682ce1860bf7ec2dd : SUCCESS in 1h 04m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d06adc3a649a41179ecfefc2a96d339b : SUCCESS in 59m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b1758cc2e734b118ae3e22a07799da2 : SUCCESS in 1h 08m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/afa8386070274061bf5e1d58e34ed274 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":7},{"id":"8dc708ac21ecb93b88b84f26aaaa3309cc4dddf9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-11-02 12:58:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860639/7 : SUCCESS in 2h 01m 53s","accounts_in_message":[],"_revision_number":7},{"id":"8afe2cc8e72832e22b5d15dfe49c3163f8d2259d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-02 17:40:16.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"6be009b93e5bfeacd4eb7a10d104871ef7b24b5a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-02 18:06:46.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"dbbf126e49739cdd5883440e759b9fe4bf33c4f6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-02 18:30:48.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"99bbb192264df71720e3f4fa818a666d765e4484","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-03 08:18:06.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"dd0858b7a938c98200b92bc2635009c66aba5cc6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-03 09:34:18.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"827e283c32af86bdf3a02f09fefe9b7c5db506be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-03 10:49:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"28ca0ed7652f91f24a2a05a52657739bd55ed88e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-03 11:13:40.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"28f1d7ee3c017f127b45d843c35787298ffb968b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:41:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4418af52dbf48d0a3a054b56160681a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04e5e32793e648bc9890b3d022d7fbed : SUCCESS in 47m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b111a342dccc4957b7675f1b0ae53d53 : SUCCESS in 49m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/601046c0ae0142e0a91edfa72fd56db6 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":8},{"id":"ce358b2e45a6450d137603b0ac780f06b5cf351b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 12:33:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fec006c18c0433c993eef2549dc6f11\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/74a4b9f0c78548ac872e5e690e4f1264 : SUCCESS in 1h 39m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b94f09db567a4ecf84749ea10da7edad : FAILURE in 1h 37m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a3684e5edf434512a67f21120c6b66ae : SUCCESS in 1h 18m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3933c74fa4234b0b9b78f6e5c82b653c : SUCCESS in 1h 28m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/021d27467a2c402cb92710a6d5191646 : SUCCESS in 42m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ba84f3eba1442ed9cb3dcf7515f86e0 : SUCCESS in 36m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e6b2006fbf4dac9a32e79496ec5a9d : SUCCESS in 15m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edc073efa3a749ed9cc4e3161b83062e : SUCCESS in 32m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f7edffa86744233a8d7ed3aa1206fe0 : SUCCESS in 31m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de98c71b40f4dcfb8e67f06ecbe6ba4 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac675ba8746948c08258b1ddf4bf3de3 : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bbb31a428d744299c1e8aac40775e41 : SUCCESS in 32m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae51859a02c04e5d9937de05ba86a3aa : SUCCESS in 42m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/55bcc2bacbcc43308526ad41f9dcc181 : FAILURE in 1h 07m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7171106a638d41c1a59cd12238a98692 : SUCCESS in 53m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4e3c001bf8dc4590ac1045d183e5be86 : SUCCESS in 56m 07s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/059d9b16b2b24a4180e38f2ebade53dd : SUCCESS in 1h 02m 59s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0e0b5dc3cb1c4cab87faee3fd6acbbca : SUCCESS in 43m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a851883aba444c9db64db085da15b8b5 : SUCCESS in 54m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b563f279602742f2868b8cdfbd6547b0 : SUCCESS in 57m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2a695caf29b74d5f85fb687155a0efd8 : SUCCESS in 43m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aefbbc9c69ce4d86bea6cd1b352f0649 : SUCCESS in 1h 15m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/738ba2a92c784b04817625127d7962d2 : SUCCESS in 55m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d6629b138a5f4fbd8667441dc8a5f128 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":8},{"id":"ea2d9a38afff8e118c14643fc7cd623f3c4bb833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-04 11:19:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"09ccf856d41255078f88dffc6e28b073a175866b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-04 11:45:51.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"743ef20a7ae61d14fbb9fb57910a2aedc52f30b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 12:11:24.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/578624ff62464e9d81a4e6613fc41160\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9dc5c483252c4cd7a13de43c93a57181 : FAILURE in 45m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6ff7d97e90d43409d4fee994f52b19c : FAILURE in 44m 24s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bca6f272d8284691b221888e95759946 : FAILURE in 44m 44s","accounts_in_message":[],"_revision_number":9},{"id":"54e315e1f68cc432626ab1f43dc6b5403f97a4c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 12:46:48.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfb181cd18034ac59141bd854db19b03\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5fd955ebcc764941a33e82392131947e : SUCCESS in 1h 10m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9e8ccf80d5674790b9ff47cef0f11ac0 : FAILURE in 1h 14m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a67c462788a74b99b238a2796f1dd39b : SUCCESS in 54m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d4c43428c8614e54bb3f6102e077d86a : SUCCESS in 1h 02m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42f11aaf760c4e71a6bd2b0f21feedd5 : SUCCESS in 24m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d98c1ab70f2d45e8a580fdbf54964e28 : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b250e068e4b4997858b78625ffc44aa : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2fadc87483d44faab18dc751be67fe8 : SUCCESS in 21m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/857fafee79a74393b4f8670e9a0d5e3c : SUCCESS in 23m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4b3fe822a34ade881b2cbc0678c7b0 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99faf778fa904730b898b0b784ef2c18 : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c497234ae924bb682554049670d95e7 : SUCCESS in 22m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/780ae5859cd1482985493a20839b4a8c : SUCCESS in 46m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c440dd0d4d2a42d5acffc794a09a18d6 : FAILURE in 1h 25m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb119ba9b3b24efbb18d69332aac8529 : SUCCESS in 44m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c506bde0ca2f48f3baed1856e23a7879 : SUCCESS in 53m 06s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/df6b9ce668544b16ae0d4897b7acf58d : SUCCESS in 59m 40s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ea9290a3b624f49bbcb8c4f6c15c34e : SUCCESS in 46m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d9480cdaf92e4a308a87cf2cf3e3b688 : SUCCESS in 1h 15m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f29f8a2dfe0244c487fae54484162beb : SUCCESS in 1h 09m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5b847ca0fd24d3682e8e1dea89469e7 : SUCCESS in 51m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f2350985354044ef8c2cd12e6a9c2cd2 : SUCCESS in 57m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae5871e8133e46bcaee1903186696fb9 : SUCCESS in 1h 03m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/66224c193f754efd8f8cc25f55ffb8b6 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":9},{"id":"1cf0d7cc40642546f5f061578de84b59526fcefa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-07 14:16:04.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a6420e95060084864c9e7d0cd3eaf32903e91983","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 15:08:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cef27c034f74ca193a36d42ff09ba50\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ae1d60e299a4130a4d7165a84b7f472 : SUCCESS in 34m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13c05d06929e444a9b0e3abe6e1cc903 : SUCCESS in 35m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aaabc160e4ca416d91f52526568fd12c : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":10},{"id":"187356ae139b84f0d01f391fac7fdc69ea59d7b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 15:52:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb370fb9349347d0a0b944dc09d459c3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/821b138ffb474fbeb9dcdca78d1f25c5 : SUCCESS in 1h 10m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5554c9d7e0774fd2b4665c6c69c545ff : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f09df86de27147f8bb63ea180b1ac6bc : SUCCESS in 52m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8069ef162ed048bda21b9eaa93d0c271 : SUCCESS in 1h 00m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/006221c6a84349a08bfb341fe58e39fe : SUCCESS in 29m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e166d5b390674ef780ce50644ea94bf8 : SUCCESS in 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f09cbcca654cf695752396061f598a : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31928acd41dc4c379edbd0ea14c6a0e1 : SUCCESS in 27m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9aac22850278480ca89f6fd275f1efe4 : SUCCESS in 29m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c70bdd9aca54680805299d3599e04ea : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef38e67cea7f4a21a45ca53c5debf299 : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb1ec5c483ad40f897000fe7bb991df4 : SUCCESS in 27m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89b841b8ba2c44a882e26e5190593bc1 : SUCCESS in 43m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/339b22764130452a9d0366caf5388a44 : FAILURE in 1h 21m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b347e2ef2a50403c9e4d9bc6a7f9aae3 : SUCCESS in 53m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/28ae964cccc340248001615a0658437f : SUCCESS in 56m 52s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/970b0766810e47d6a9c561e71e6e3836 : SUCCESS in 1h 15m 49s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c07c753b1dea4429a9cbed154eec090b : SUCCESS in 1h 03m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ff915be627ee4c568e47dce0f58bdfcd : SUCCESS in 1h 07m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b12eeba28142410ab0cfa5b48ea7855e : SUCCESS in 1h 29m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eb7800da1609470e9ef2a01256976de9 : SUCCESS in 51m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/093acd7d426d4a43a1f762cd67318552 : SUCCESS in 1h 06m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79c67c58f8954c0481e23e4593698eb8 : SUCCESS in 1h 17m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/50ccf8218f3e4a9cb235de2cc5455ce4 : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":10},{"id":"e90a848ad89256274c3ff74afdbf38cde4f5edc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-07 16:06:16.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9bafa8c9c33b7253a56e3089a41c2a19a9cad984","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:50:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44e658f26056441aa105d92d747e10dc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5483112dd5844d4bb58b1b485efd28d4 : SUCCESS in 37m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6eab35ca760a4311ac84fadcf47f0006 : SUCCESS in 40m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d1780edb1f0441091c9640c2c0b0388 : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":11},{"id":"9722b50d7e6207fe225a0090266078bb30b8f52a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 17:46:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8fa345cbaa249a7acf3f9a5fb1a8b60\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0a4173b1f54043888566b9655894ab26 : SUCCESS in 1h 37m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/48882245ab7d4c12a42af83696c8a8c5 : FAILURE in 18m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b985706720e4cf9b81a32f3bec4a677 : SUCCESS in 1h 16m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/23c5544ad5a74abd8a7b1cd67ebf8de0 : SUCCESS in 1h 25m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/339bcb53045741e8993801a84eff4a47 : SUCCESS in 40m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea6c816ae9746e6b4afc14afe266a46 : SUCCESS in 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf001996fc6849a2a0841f7de7915d38 : SUCCESS in 12m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65b360cf53a642ac905370a44636e960 : SUCCESS in 30m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56531dd825e240e6a0873f8c6c2beba4 : SUCCESS in 29m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb8cf1101954f1b88c93d7908bf7e57 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04d81baac86847c9884f2f2bbaa3b17c : SUCCESS in 12m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2294b7780d94519807d872e8c1954e0 : SUCCESS in 31m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f627e6765024be19f99f175f23821f2 : SUCCESS in 56m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/da3c5faee45f46f1a98eb3f6a2d7d26b : FAILURE in 1h 23m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a4bfad8716a486fb797866543127526 : SUCCESS in 1h 01m 00s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f2618c44f7d54abcb200ea2c3dc5aabf : SUCCESS in 1h 06m 13s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/937aa19d5d5144b5abb12df1233aeede : SUCCESS in 1h 12m 06s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d086cf9e47b1443b8d3b4a7d1757fad9 : SUCCESS in 1h 06m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/52663c47bd5448eebae9b77a543f1ac8 : SUCCESS in 1h 22m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bd01a8b3e67a4ebeac585fb71268cd20 : SUCCESS in 1h 20m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/97bf59c65cea49f295c38c600d922403 : SUCCESS in 59m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1856406c3ead41828186249610a993b6 : SUCCESS in 1h 09m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bfc7711ed0524830ad253603ddb40586 : SUCCESS in 1h 24m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2db29918a964c0694a2fd423fa7bad9 : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":11},{"id":"28547290b92fcba5319bff940f872e2bf65b433c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-08 11:35:45.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0f59c9a345c0265e9abe183ffa632358f8e828ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 12:45:15.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a688e9bcd4478cb27c16beec0690d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c032b084235e4df7adc7ca5a25601546 : FAILURE in 59m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42fea26cda094a5587110f2b37780fc5 : TIMED_OUT in 1h 00m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13387037aa5944a2bb61d201d1def809 : FAILURE in 1h 01m 42s","accounts_in_message":[],"_revision_number":12},{"id":"89660ee6f29832b37f41711f739190789a792312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 13:05:09.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00ca241f59f94b1ca59f6b44644f6f5e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9a4b7a58694c4f58bf91503841fba8a9 : SUCCESS in 1h 17m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b4d8bb015b064e48a435cc6598bdc0c2 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/536da5d937b04e6aaa3fe834b5130609 : SUCCESS in 58m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/deead21e1228456d9134f6d5a4f57a2a : SUCCESS in 1h 05m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b27b12577aa94f2e8ec2c20c42c9bc95 : SUCCESS in 24m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95f8aff2cce244c0af619e2d640c09c3 : SUCCESS in 28m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4860091852749f1b683ced0d4b8a701 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ec74619afbc4ffba74fbd01e98302a8 : SUCCESS in 21m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52d66b2845574c77b7ac6b5d829686a7 : SUCCESS in 22m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9753bcd7943e42c09e773a98b2026bb5 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d94090234cd48ef99035d98a1b310eb : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e453c3965ef0424c9e8142a4433c64a2 : SUCCESS in 23m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42555ea315994e0bb41ac09c41ab3f90 : SUCCESS in 45m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bf880a44f434634891fdb3fe4503010 : FAILURE in 10m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/91e50bd222e34f24b681d538bac8d4c4 : SUCCESS in 51m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cb59d889c0a74d45b5b98b9799ee68e0 : SUCCESS in 50m 16s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fdfffa6d79354944bc6b2c723631093a : SUCCESS in 1h 00m 32s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/20cb423b91444b098a7bb4e018aadff6 : SUCCESS in 49m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6fe980ef5a32489c91e216c89e9e3b18 : SUCCESS in 1h 22m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55c0db081afb42219c21cd205707da06 : SUCCESS in 1h 05m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a2821fcac46486c8d6d08519ef5a14f : SUCCESS in 50m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/49f5c1e7c6a64a8a9c13a935cf9219c3 : SUCCESS in 1h 12m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fee81881c4304906b10479a9d3a2c583 : SUCCESS in 1h 06m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1844d3d747fd46d4b3b01cf2445843fb : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":12},{"id":"640b76b1dd3a2bd5b0a6a9aeb82500795eb4f231","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-09 08:42:35.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"eb43326ef357e3de6d49bc83cdd0e7bafd525dd0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-09 09:17:30.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"55cdc11f5af269db19316594cd1bf2840b50973a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-09 14:25:51.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"346769e07a173c806b198c997624c381ec597977","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-14 16:31:22.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"afa1474688b31a07c7a57128ec08e0512f5163aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-15 16:58:10.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9125659b79b26066e2b44822a9f7d3389769d317","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-15 16:58:31.000000000","message":"Patch Set 13: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ad3de5e7bcf429b39188458a5e0c328f4098947f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-15 17:32:39.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"0d56e4a3dc625a0ab6891684d61dc6837f6accd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 17:52:51.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841fbbe6db364c81921684123a1df931\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1be7d2dbdfb44bc5bb1f7a2b7bcf3bcc : SUCCESS in 41m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/096ef1d9e9974d2088e6f1336febe189 : SUCCESS in 42m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5cede60cd737478a8479adbedcd4fb63 : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":13},{"id":"3782aea7cbb1de304224aaf28f0fe34d0e4e6768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 18:29:27.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b981f494a8db4f9997a1949b85c32594\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/00c9c938f2d74199be3295e001d1e006 : SUCCESS in 1h 13m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5490fb9533ba409eb3ba6cd5809995fc : SUCCESS in 1h 06m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ecc37ecc3c294060b78aee3062241a0f : SUCCESS in 59m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4b57289a3d2c4355bcea2ff48da699ed : SUCCESS in 58m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a15932df318b4dc6ac484806f66285df : FAILURE in 32m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b69ff05255f4b5c9114b5e52782fde0 : SUCCESS in 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db9fadf3224444891b8009e468f8dc7 : SUCCESS in 10m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c21cdd5f75a471ab0445be0737328e7 : SUCCESS in 22m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1492a2487f0248e2bd20b83ff54e9e94 : SUCCESS in 21m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f2ebbc1e364e96a062e42255c8bb72 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95c4a4a2c9cd4fbebcc6ef0e3bae961e : SUCCESS in 11m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5441ed90b2544aa6a72b7d47a15bee06 : SUCCESS in 21m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89dfc2e0fa23440dbb93a7b8a0349ed0 : SUCCESS in 44m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10a2964d635147709ce0f1f79a62ac84 : SUCCESS in 1h 29m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c58a5ac224a24c049c3b2b1b0b502469 : SUCCESS in 43m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d8a07ac239474754bc43fb11c08dfa99 : SUCCESS in 50m 10s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/99bccb12b8f247af8d4e9d9635e515dd : SUCCESS in 1h 00m 17s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/243ec1f96fea49cd882998fd568191cd : SUCCESS in 43m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/97060b210ffb448da32e31734cccd197 : SUCCESS in 1h 17m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f77ece2bfdfc4ab6badd970b1c617f69 : SUCCESS in 1h 10m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/abbd03634a3e47198c6ff12dd4377dc2 : SUCCESS in 46m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/09d7af50080345888aba5c12c798538f : SUCCESS in 1h 02m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe878055c5ca479f9ef215556646eb9c : SUCCESS in 1h 00m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fbf79cd1de304c0eba2ea6c5809de40c : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":13},{"id":"1c7ba4146ef59219737032d6b9f4647bc552ef74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 11:27:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ffb0726981f813da660e379586755239aad1b18a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 11:51:35.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e6f7eadf51195eda6110efe8f9f20744148ee586","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 12:14:44.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a9055ffa5f45a68e465b53440ac848\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af7e97e9645a40e899b0e59e4db3c4ce : FAILURE in 30m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43bfcd8bab8145baab54a392439a77b8 : FAILURE in 43m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25ff8657af5d42a880852513d7dc4d51 : FAILURE in 44m 53s","accounts_in_message":[],"_revision_number":14},{"id":"e238806aaaa0d480b0695edf46cfdc4e06f01d19","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-25 12:41:13.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"dc4a677557212234b3d664a62977a99296721209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 13:35:58.000000000","message":"Patch Set 14: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5d993041089400586c535aced7d3a70\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/64fb3653ed78415bb541f0302cd5a2e6 : SUCCESS in 1h 16m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0afd18291a59479fa30a0e4abe435c53 : SUCCESS in 1h 25m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9e0e1d949674598ad1d9c91c4b4aa4e : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0cd4c59e11c14700a2315bd753c85f62 : SUCCESS in 1h 07m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca0963bfeb7e45388468be98f3d55018 : SUCCESS in 33m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d745290f8b4743f0a920a6d7da413ad9 : FAILURE in 36m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812ac82b45304f04aeb74ebf8e46aefd : FAILURE in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/906fb89c022944288c141de19847f1f3 : FAILURE in 29m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acc1de2aeeb04ada901d59814a72b3cb : FAILURE in 32m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ec626703c844c69b73c98a82b42e1e : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbbd360e145e4b4b908fb62685471f73 : SUCCESS in 13m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e43b418c27a4fa5987f3226beea3303 : FAILURE in 29m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f168ccbf334a42679e24d19061032a64 : SUCCESS in 41m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a310b5b92c584fee90388b36306728dd : SUCCESS in 1h 43m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe7ee90250bc491686999fdc5e23964a : SUCCESS in 1h 06m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/16c70a8fe86643b3b589398feb253ed4 : SUCCESS in 1h 06m 11s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d85cfc2a3df41cab72e7a549b915a5b : SUCCESS in 1h 22m 15s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c4f320a836d84a71b416b885c0bbfd8f : SUCCESS in 1h 08m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/371393cbd55d4061af62d2f7235506fe : SUCCESS in 1h 31m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e0fa97518c094f57b72a59244ab73512 : SUCCESS in 1h 14m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/33f7e0ffea984302a7333a45fa72403f : SUCCESS in 1h 14m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4495e3404bd646ce960260f04a783134 : SUCCESS in 1h 58m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/669e8c2aeec94ad2a23550368c2da148 : SUCCESS in 1h 56m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e6889da426b04606a2a6f504a0061ab4 : SUCCESS in 1h 19m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c68255d376f847b7af501da608f95431 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":14},{"id":"c5683e43d6e263ce16f10f982a55bc439f05bb3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 14:56:06.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"dcb264bf514bba693891e214ddbd368cd1d5e5c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-25 15:33:23.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"75ecd96baa55057ef2460181e1021a5e7ddd450b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:53:52.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7846273164434e99fb4293da640930\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c63ac357b3494c61a3397dde478f3115 : SUCCESS in 52m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/196ba7a337da4a9f9f5ced9b6d9bd133 : SUCCESS in 55m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d61d017a2ee4c50b24e0b9342312dd8 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":15},{"id":"e7aaaaeed8efc8f37e1e0d01ede3c356779ba5bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 16:42:03.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32013f59401a4367ab88c2d793513a91\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/81d714dbe156430790ecd5d9dac36197 : SUCCESS in 1h 39m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7614225992a7403fa2079270464d3289 : SUCCESS in 1h 35m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7964d9743b6a489997b30c039d28aaf8 : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/76d058e8dafd47209b99a732ef9eb5d3 : SUCCESS in 1h 29m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05519a43251c408f9e6014cdfa261d3f : SUCCESS in 37m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58d2ae1aa9484394a2186f513c807f91 : SUCCESS in 29m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490fe66ca50c4375a386b3b7a5a5bede : SUCCESS in 12m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/795382d2e4ce48a08f64919661b5c804 : SUCCESS in 23m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e67d34fbb17841e9939f09adc9d3a4f6 : SUCCESS in 23m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1bdcd1b0e74f3f9d2baf205058b59b : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92dd17e6b3ad4e7a841ededabd5b2373 : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e27598cac144d4fab93a2c0702807a1 : SUCCESS in 30m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3bcbcb34a8840ab8c4ae6baca500ad4 : SUCCESS in 45m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/309f353f96b044a29919d3873ca41c36 : SUCCESS in 1h 38m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/85e909d347104aeb8bd99e6c55408005 : SUCCESS in 51m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/94af944b754a4ed6be69186ec56deda0 : SUCCESS in 1h 02m 48s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2494637d622d4c01aea41945c20086b4 : SUCCESS in 51m 22s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e3e2444b9e3c46fcb82b115a3db70b20 : SUCCESS in 55m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d744352a3ff04dabac53d26755c8edea : SUCCESS in 1h 12m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/44a2b5f2313e4f28b84815bae8141ad6 : SUCCESS in 1h 27m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/637d563b476e4e148891d90356dbe79e : SUCCESS in 1h 02m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3d2059a1a2f64fdfaaa80644da36919b : SUCCESS in 1h 27m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07f817b6c66c4dd690d429b989a30fe6 : SUCCESS in 1h 36m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c94af6d152046ba9fee31fac9147da7 : SUCCESS in 1h 07m 04s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/7f95d006b1894c44afd4c8888f9996c2 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":15},{"id":"271f4b13dac72faf5ced30d2d9f700a9b5af1fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-29 10:12:59.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"b32879f45aa08bd8085548bf1e24c5b180e66a5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 11:01:25.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e95e6740154d5bab49ba86b2fb8720\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8013d9976cf849c5828b1d0f05c18a5f : SUCCESS in 46m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/09d0e1e7da29430db0cb81109abc8557 : SUCCESS in 45m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8ec99c870f14b2f8109ee84d8369c32 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":16},{"id":"8ffc634e2ef2607c562eea69c11d90975e851a4d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-29 11:23:52.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"448ad9eb61969538bfacaf1bb60d88e6877061bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 12:05:23.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/401aae0665fc40d3996d720276faf13d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2b23fe23b31842afa105a4f65222dba4 : SUCCESS in 1h 39m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c9b21b6263054f00826d9009e4aab5f3 : SUCCESS in 1h 21m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/44abafd56b9e400cb4902da5e822ecda : SUCCESS in 1h 20m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e8d47ded05dc4e0d9470d3b06742dc88 : SUCCESS in 58m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60ad1a05a6424002b537bfa3b08bd29b : SUCCESS in 36m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/096c8c9a63f74e7b9ed5f6bc1ccd87ca : SUCCESS in 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae13b7d7b7d84fc283e76e1dc79dd73c : FAILURE in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05060e5797c8412f8ced90a280698cfc : SUCCESS in 20m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f28fd4cb5c4c4c8b9783be0f550c29c3 : SUCCESS in 23m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e92187f2a548da87ec700afe0972c6 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec4a49e90c2c4e3faf33c5422f4eb056 : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d0ed43418bc484ba56024500f98e6cd : SUCCESS in 26m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/30650af6819141bf962e0e238a01b72d : SUCCESS in 43m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/128bd60aecf546379711f6aceb81cb59 : SUCCESS in 1h 31m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a04ab8a30d54aa1b4c56b1eb01bb4db : SUCCESS in 47m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/48bf202d5615466895962b63c01340a5 : SUCCESS in 56m 18s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e8b007c668114d49ab57bb30bcd597d4 : SUCCESS in 1h 11m 33s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/450fc2ef9e114e30ac4adf158e7c752f : SUCCESS in 53m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a2c4fb9f0c6b4b89ad7ea0c21e8fafa4 : SUCCESS in 1h 15m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ad5ec44401d74c4c8ae898f2fe9be43d : SUCCESS in 1h 06m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/54b2102300e34fabb9c8c444407e9d6c : SUCCESS in 56m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/983157880d2247908ca003f506d9785d : FAILURE in 15m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d351fc8e4e92406a9126c9c792c67c21 : SUCCESS in 1h 47m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/75134ec836e147ada78f00e00e4981c1 : SUCCESS in 1h 10m 26s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/56f83b9c324a4f89aaa6a00dc7f22ee4 : SUCCESS in 1h 11m 35s\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/extensions/test_placement.py","accounts_in_message":[],"_revision_number":16},{"id":"5c518e8705eb25b0fa0a1b3d7d5e7ffb08f9eb2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-29 16:16:13.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"ed46b2c6a2920a08d6a91c8ae529b613da61e286","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-29 17:43:48.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/860639/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"cf6258e74fc74ff5358d20819930f0e4daa04836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:17:04.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d3aaa357d24fd38ede734477914e00\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c29f0f9cd1bc49bd9b6c9e929ac45494 : SUCCESS in 1h 37m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/178ded49c6204c18bb6416f877cfd49b : SUCCESS in 1h 26m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd41be4e012d445cab04a7ba1723ce37 : SUCCESS in 1h 17m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5e10aa1061341bc8f1a9cbe096cf5c1 : SUCCESS in 1h 31m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de20afa2d1b94c4f98e061660cdc8f73 : SUCCESS in 27m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cff1444097e14133890f26a8b162af23 : SUCCESS in 32m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9e65f760ae45c399c79a516fb77c6c : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7867d452a7f74fb5a6ae668bc1496653 : SUCCESS in 23m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66054dc0e13845cab205fe186f814db0 : SUCCESS in 23m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d8646284874b4ca1537c0f8f276ecd : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc7e68ba65554b92b29f09f288d3672c : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c88e7d15a9848f09c89c44c824cc660 : SUCCESS in 22m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b99accfea1144f7ebc2bb5604dbb7a5e : SUCCESS in 45m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8904ced5102e4610a5ee748b3c1052e6 : SUCCESS in 1h 24m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9510aa528d3449dbab252d5486d1985a : SUCCESS in 1h 08m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0971bfff59354a0aa3e7eff084238572 : SUCCESS in 43m 32s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ddbb3b55a24e45209c0d73a9540d5aae : SUCCESS in 1h 17m 51s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e9c31df97ece49e8bb251f1b17247b36 : SUCCESS in 1h 11m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0f5cdd92489d404c8334efac861012f0 : SUCCESS in 1h 18m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a7caf04d95e640af9cb83b1008194cf7 : SUCCESS in 1h 22m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/490bd37732374e19b191f4115ce27899 : SUCCESS in 55m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a5313dc7a9c4e4bb1cc739822a31d23 : SUCCESS in 1h 55m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3283b32a7c834634946466e510551178 : SUCCESS in 1h 31m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/629ff79da8b045d993cf42e40198f528 : SUCCESS in 1h 15m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/29d20ae6cd824da49adc171cb5806539 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":17},{"id":"8ade81843dfc9dd292382d8ab85bc659d083c2e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:23:28.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b36e8f2b7b4c9bbc69fc607af9e78a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e65abb5be094759ac473b78a5cdf9c7 : SUCCESS in 50m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3007b4a7ee994b668c88720bce9922bd : TIMED_OUT in 1h 01m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c533f81922624bbe8a3c4a78de714b2c : FAILURE in 1h 07m 17s","accounts_in_message":[],"_revision_number":17},{"id":"7745044964fac9ee2799636a6c6f7c52140672c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-02 12:37:15.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"575c882485550f5f22dbbdf4594b0c6c219aae3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 13:30:50.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc31f2b5e75c4ee0a091e1bc4e3c8444\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5861574affaf44978fa4a811643f0690 : SUCCESS in 51m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bbb4e4487004e9a9786f630751f690f : SUCCESS in 50m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9892dc15a30d40ccbf92c81ba0ca268f : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":18},{"id":"77dfe106db2cfa92c6551dacf21c578723039997","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-02 13:45:26.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/39/860639/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"02cbf0bf2b1b746e3117edeaa8b9e94f3811b60b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:21:40.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1ef297d354141efb8973c1002cf81c5\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4aa343edee264e9bac93b3b313fe28f2 : SUCCESS in 1h 16m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0b9c60bc9d544cb3a1eb6a5514b3e69c : SUCCESS in 1h 10m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e0341b6647bf46679b43b5fdfdd2b804 : SUCCESS in 55m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c61212c0ec5048d986490d9fe111e40c : SUCCESS in 1h 20m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/228a2880865541cdb0909415c753f82b : SUCCESS in 26m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ca211cd07614ee38852424facc09a7c : SUCCESS in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e99a4c524c04251894ad344ca4b5df7 : SUCCESS in 14m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55bcbcf410e446a5a338bfe3bb80d526 : SUCCESS in 25m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5eecf2a29cf74c2db45dd3189501d509 : SUCCESS in 30m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed68ddaa450144c08757866a911d09c4 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b62c4233ed84cfd93d0974111bc4073 : SUCCESS in 14m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96acf3badfae44c1880492e8d22c97c5 : SUCCESS in 25m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6502affcf7284e2a81b6d897fef411bb : SUCCESS in 49m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12956e8a9dd2473b8f585b78edc62144 : FAILURE in 1h 34m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/430388bbae4d4cc1976924afa88b3bc8 : SUCCESS in 1h 05m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7197426f54c34d26b434067232dca818 : SUCCESS in 45m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9ceb16d2019143c385c34f5aea9d5606 : SUCCESS in 1h 24m 39s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fbbe4f45269749959ee9da05f66574f3 : SUCCESS in 1h 13m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e80748c27f9147bc8893ab1f74bda970 : SUCCESS in 1h 06m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5f801839ca64629b62bd323635237ec : SUCCESS in 1h 10m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ad0a55e4bb041479a983669fcef7f4f : FAILURE in 42m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/414ef5a80efe46939c463587ece2c933 : SUCCESS in 1h 35m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b2a20113ecb48f4b120fa402b86e9c2 : SUCCESS in 1h 39m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5ea8bc516b574316a3ad320e23e0f54b : SUCCESS in 1h 35m 54s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/5a6fbd4bc0244e2d89c373be36f1dc82 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":18},{"id":"f6d946684b559ecfc0e3abd9f3e86e3ebdbbc9db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-02 14:43:43.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"106b6d82f940c00177c274b9fab8f8efefdc9fe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 16:11:55.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93580153fe074c4fa6de999caaa3f1c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d828b2ca29d14ade82f6abb59ef263a1 : FAILURE in 1h 15m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c05df722101747e6b3344a9119b11ea8 : SUCCESS in 45m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f04e343735b146bfb42e5d3d12aae23d : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":19},{"id":"8a63d6a7c75abe7a2e92ec3c22fd32e23ab19d55","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-02 16:21:44.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/39/860639/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"6398ef2a40ba4fe23a3308f4b17b997b6afd20bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 16:36:43.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e43866f825b47729322113b916cf607\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/526c8d7c65ab4f14923378b7b35986f1 : SUCCESS in 1h 17m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0db12c1d6f5b4aaabe49b0ac9513f225 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/876b9c5c0af244e9b7c683c05da47397 : SUCCESS in 1h 18m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1c62c9c919464e87ad3c2cd3a5efd9cf : SUCCESS in 1h 22m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/525099addfe54f758133965fc39bc9cd : SUCCESS in 31m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2122fb01a2874df998b9c0be44397b02 : SUCCESS in 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a86292c1c9c4deba669e8a35055e5b0 : SUCCESS in 14m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4c5699c0bc54d548d4781cd4d2620e5 : SUCCESS in 22m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/494c9ac9f809488998371bb36844fd09 : SUCCESS in 28m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b770b141f62458580837b090cb0bb63 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab31c99e4cdb4ea2a5beb4a068afccd7 : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52cc1d2f51b14e95866b0c2154fd62ad : SUCCESS in 21m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebb2b5d3bf1d4e55b17319b6773fe22a : SUCCESS in 40m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9acd493bf1a48b1b8f230c0e27694e5 : SUCCESS in 1h 42m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7dcebb728e344386a0b56cedfd1ef925 : SUCCESS in 1h 02m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d3ab297d44a04eee9909797dbc725094 : SUCCESS in 1h 02m 07s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/953f484b24e044b9be095250ed24e1c4 : FAILURE in 1h 13m 12s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1461e7d058c94583bb3dd3419d7e66af : SUCCESS in 1h 04m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/127a606ee60e42dbbf19c1d3e6b73110 : FAILURE in 1h 22m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4161dc7fcc334fe1a25392fe72bb4bf4 : SUCCESS in 1h 31m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/29928d022eda453b86cb7597a11f40d0 : SUCCESS in 55m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/00c2719a06894b3a8527ae6d7a6d0fca : SUCCESS in 1h 41m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c959c32ee874aa59d2cae98252de0d3 : SUCCESS in 1h 44m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dc8bc34db87c46d38460ebb14b2b5627 : SUCCESS in 1h 45m 02s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/279c25f99bc24d61b8b7da2a8da8dce9 : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":19},{"id":"bb5a318a185e25dcefa0bce4524342cf02270c9f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-02 17:16:07.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"b14ca4599193279b123699a2277ce2922dd947ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 18:09:44.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45e7866a25d4414899771ac5c1f363d3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35563bf733c0481a8a6e0ffb2eb1b3df : SUCCESS in 43m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4dd326a0f0944918a441878c0c301947 : FAILURE in 43m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73f35cc722e64f52987f5d98ad78b83f : FAILURE in 41m 53s","accounts_in_message":[],"_revision_number":19},{"id":"2915c74493c01ffef7654501a140ad47946c989b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 19:20:26.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55541c356cc64096b52eca9f2d0aaaae\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ebdea2338af044b6bacc521fee3cce14 : SUCCESS in 1h 25m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e9140f3274b4f9887af82be4a5e5ec2 : SUCCESS in 1h 30m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8fa967b6bbe42b8a7f465eef4fbe3ea : SUCCESS in 1h 19m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/32aa59c336604f03bc7e259fa555ac69 : SUCCESS in 1h 18m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07d41dcf449f4b8fab9573a206b7669c : SUCCESS in 37m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c55d1fc85fb944549517269fc85b381e : SUCCESS in 43m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37241f0ee8bc4a41879130d9715447fe : SUCCESS in 12m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/939b8a501b97461dbde62a8f19b09104 : SUCCESS in 28m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04c020f0a72c4e81aff8cfabc7a54097 : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d38375ba6224032aea55ea0044b6b81 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06ab94dc57dd4e5788bdee3ae7a1c2f7 : SUCCESS in 8m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4074ae6722de48b099a3457f4818c6e7 : SUCCESS in 30m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bafa4f3f603b4a098d7526f9871f7947 : SUCCESS in 50m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0c08038b44d4e7f98a23e347b73eea3 : SUCCESS in 1h 48m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b70fe89d526943f08f08ae54ae43803c : SUCCESS in 1h 06m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/70b0557e34964a7faea7aa73818dcd0c : SUCCESS in 1h 09m 37s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c7569c649cb147c09d50341761622192 : SUCCESS in 1h 09m 40s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7308e9e6711b4aefb81c49bd0214a9a1 : SUCCESS in 1h 11m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3dd8937cb9f8451bbe3d438c5eb382f1 : SUCCESS in 1h 29m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4276952e931c4c69b7a9f5cfb7d2c6fd : SUCCESS in 1h 22m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/271d4a0036b347108f02e6a9ecd968d9 : SUCCESS in 1h 07m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e2125aa9db9b4ef4997ed02ba6e9932d : SUCCESS in 1h 55m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f5b3f2a715ef4046ad77063e02b563d5 : SUCCESS in 1h 56m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46ac672c93694b6fa2d47ef813eac807 : SUCCESS in 1h 11m 28s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/84126b55947f4f9c98f26fb20c76e44c : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":19},{"id":"58fc0cc2dd96fae25c380cdcbf00887369d5f0d8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-12-07 11:53:47.000000000","message":"Patch Set 19: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":19},{"id":"2c706944526437ff651a04c42150e8e9e30d48a4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-12 08:42:27.000000000","message":"Patch Set 19:\n\n(10 comments)","accounts_in_message":[],"_revision_number":19},{"id":"db38d94913585856d6145bb18aa6d0c734b2ea51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-12 08:49:43.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"848c0d3ef77523559589ee28a4b5f3546ac7202c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-12 09:28:24.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/39/860639/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"5308e2bf7e23ba8078acd8324567444fa475d206","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 09:36:34.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b955ecc14d0541ae87278c8df9820b48\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5a37dbbd11a43618906f93e6fcca51c : SUCCESS in 44m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e44c161a4b594ae0a28d90472330d546 : SUCCESS in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a312cdc1853a1416d1fb4ca4987867035e91743a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 10:53:51.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a90ac1889e054fa1969c0719fee0b67e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/36368bd690434d72898fd70599ef8640 : SUCCESS in 1h 53m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b60e00f0be0c431199cf580d03b1d667 : SUCCESS in 1h 47m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/950db2a0ec434d7191e6a4c4e05700c1 : SUCCESS in 1h 16m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/91b33c58b5b848709ec6e2f2232d3035 : SUCCESS in 1h 25m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/20cede89293949d0b085d7d1e78e3d41 : SUCCESS in 38m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d4cbf90da0140b3ae27368b75c43c0e : SUCCESS in 29m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3311e6fdcdd48d5bf5c4f526b150fbb : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a5392852e6469bb71f22d4260c95d9 : SUCCESS in 28m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d94fa37abc794c2da56b298771206511 : SUCCESS in 21m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c0c51501074907b571d3853a016542 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ad4a55b43474fa5af69040202d215fb : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b1520d5163941c0964a3878eab568a9 : SUCCESS in 45m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af5988a911cb452fb479fe0ac69abf6b : FAILURE in 1h 16m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8734128b35654a9cb6523de02b0fd2ed : SUCCESS in 47m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e7ec00382d4e4c0c90e50593737a8457 : SUCCESS in 55m 38s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/988087d5bdaa48078786851c7d784ac4 : SUCCESS in 1h 05m 32s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/57f3cd91234548e4a9625c6b27015b48 : SUCCESS in 51m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a0a25bf9e4ee4bd8a5bbba7fbdf39ff4 : SUCCESS in 1h 20m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5c3d29520a704932bee0af38c0f3723e : SUCCESS in 1h 10m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5f55e21678454d498150205a3a3e6151 : SUCCESS in 1h 00m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cf353dd9db834b43bceb549623b6f5ae : SUCCESS in 1h 35m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/482574e552734891a76a87aaa07631c8 : SUCCESS in 1h 38m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/36f42867f1734146975db7e85df73577 : SUCCESS in 1h 17m 28s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/7fee2ff0821842dbb3f3f460a6a03b08 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":20},{"id":"f7db25d77020e0bb144e48e6ff95af9e7c61fac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-12 11:05:44.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"e7567221b5b5bf5e243ee9dbb3d540e7ab12b8d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 11:50:10.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c2dcc1fdb98431caa0c9c8bc2906105\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/19ba3d4312d04c8d92d7829966e5403e : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9dd12bb4f4354451ba5ea0a552eb3766 : SUCCESS in 41m 35s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f9cc1b1ec364bf20ebce20002c0d63f15793e215","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-12 11:50:53.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/39/860639/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"0b30f8ef7aeaf734497f9cc00109b39d5095a508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 13:03:19.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e2d124c1d943cc82ad7719d53883c0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/59ae4cfe9c7846fc8d980dfcbc4740b3 : SUCCESS in 1h 31m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9d6c211b233c45dbaff7b3168ca20741 : SUCCESS in 1h 33m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2306fab8d3f74cf4a7f370903fa449dc : SUCCESS in 1h 29m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/04672e6b794f4e3ea269eca6d23e361b : SUCCESS in 1h 24m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a29cade6bc284baf9815b6ea2efa9c6b : SUCCESS in 37m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/631007e25eb543e9b25b5dd5587ecf95 : SUCCESS in 53m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05814903ffd4de6881995d106b1fd00 : SUCCESS in 13m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66fc89ddab96418ba58bc2f2ddb8a30a : SUCCESS in 29m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6b5833333574153ac450168cdd57fdd : SUCCESS in 28m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ffe90a0c4b4d10bc0c92a02546b231 : SUCCESS in 14m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7e1390f17043699043cb7b95e0a757 : SUCCESS in 12m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28889743a9a04bb898e77ac48c52a54f : SUCCESS in 1h 05m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/20a2bec6e3884bfd8cca6c50503a29fe : SUCCESS in 1h 51m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebe4bb7aef32490fab8bd81919f0ca78 : SUCCESS in 1h 05m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9d9356e5cf8b42a99a895bd88664efbd : SUCCESS in 1h 04m 44s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/87ff1d9a09054044a7996c8350f235bb : SUCCESS in 1h 02m 38s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/af99d671cd764f91b2c9856812983bb9 : SUCCESS in 55m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/44aa964e5cc343188389ea70e473f79c : SUCCESS in 1h 19m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5b02c0d8ab6b4472b14b9e09029f0c93 : SUCCESS in 1h 28m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/63e6be108683408ab619aaaee4ef43a2 : SUCCESS in 48m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b7d27018343644e0ab97fda7aa1a0a28 : SUCCESS in 1h 31m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1208af4184f44a69974e75f53c6d6563 : SUCCESS in 1h 37m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/22e6437a1a7f4e53919d1c6c9d031482 : SUCCESS in 1h 02m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a86900c21efd4881a40decc37857d3c8 : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":21},{"id":"2d1e12374d879593a080ce2f195026a788ffa5c9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-19 10:55:42.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"519258d09767cbd8c148417a27c75e171de788a7","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-12-19 13:01:12.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"c794eee05adf69225ed2bb52c7f245fa288f458d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-23 10:03:40.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d09135e1a9e8207b074463de80110731ca75adc4","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-23 17:23:18.000000000","message":"Patch Set 21: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"8a6fe84530e8a8dbc604c51d7dd42039e17fae90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-02 11:14:50.000000000","message":"Patch Set 22: Commit message was updated.","accounts_in_message":[],"_revision_number":22},{"id":"7e641fe72c2aac65bee370585df5b02017bcb7ae","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-02 11:20:47.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"fee35274345dfaa15139e5dda4fdb4f980f3f1b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-02 11:34:18.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"6f11182d21d4a7fcfeb05d6284643f9afaa7568f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 12:23:12.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e8d633be54d460cac9e13abcdceea2e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0516279de84493f9066817644051751 : FAILURE in 1h 06m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e7f760ff61b4434a621a3e5a453e4f4 : FAILURE in 1h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7a0808df1cf7ec21fa551a23ea3ad6f7d26c5377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:33:52.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edc998d0a92d496095f93f2682ee1428\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/437349fca16d4ada8bda14f8bebdbcee : SUCCESS in 1h 32m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef80ac31993f4d52b3c9ba43c8a5638e : SUCCESS in 1h 12m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67b75068548d4130adefa36503177bd4 : SUCCESS in 1h 19m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1cc9b4defa584b94a9b502e610c19270 : SUCCESS in 1h 19m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ec1e791ac2f4b6e893e49aab8c7d7cd : SUCCESS in 37m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e95f06bf7864f31aad68ed45481eab3 : SUCCESS in 45m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61044fe885854c579a813f09bf46dd32 : SUCCESS in 12m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bae1d42585e4adeb6c96efc78269ae7 : SUCCESS in 25m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32ba9224c5aa43cb8cdf911d21a5842f : SUCCESS in 31m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528522d4374a4935849edd7bc66538d1 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95ec0de369994132bc1ea2630cd36be6 : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/531bc32d3c3b4839a19af2ee29eebff4 : SUCCESS in 41m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3f5aff9e72444a4abf472773dbe529e : TIMED_OUT in 2h 12m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb66427e3d96431cb9e6af57e8082e54 : SUCCESS in 1h 04m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5ee1befa90ab41aeb67a9922d95bb775 : SUCCESS in 57m 56s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f2ccd9b414c54adcb6fe74536360099a : SUCCESS in 57m 08s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2a9b4de2266646eeb27e2ac8cd1f3e27 : SUCCESS in 50m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/773141b0be5b4b9c8a15f63ac0e9df39 : SUCCESS in 1h 49m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3097ded565644ea0977dbd6521d81d02 : SUCCESS in 1h 25m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a1eb2eb84887480aa314b114b7f063df : SUCCESS in 56m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0d9001c7cc0f468a9986dc1430cc266d : SUCCESS in 1h 34m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3642765fcca14dd086aeac73beaf40c4 : SUCCESS in 1h 39m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/265a5ed8928543d1a08150f60245297d : SUCCESS in 1h 19m 30s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4888e09984e947da8b587dbb5fe5274e : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":22},{"id":"a12274f307de9235caaefe1ecd37f23277a4cda4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-02 16:47:04.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"3beeeac0a58cee366db5100fbae0715b77b3e814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 17:44:41.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7e916a533a549fba7a099c37eb4dc43\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/248ee8f2dc61474a921871ae2b47a1b3 : SUCCESS in 55m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e32ebadb282649c08ce4b8e5bdda4aa7 : SUCCESS in 53m 56s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d07facc4b174eac93eff88cc648097f4010c1fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 18:41:28.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/554bfcb710074bee9fa811e7a3f8b67a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/06c3e4258fa641be89db5f5581deed1a : SUCCESS in 1h 24m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/53e5c5668c284e7a9b1115db2410d14e : SUCCESS in 1h 20m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04221633a465405f8df9d1863fd26b21 : SUCCESS in 1h 25m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fd2e68b441754422b44133de5d89648e : SUCCESS in 1h 25m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b02f4fa53580438b92cbb556d66088b8 : SUCCESS in 29m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09098066affb463594dc15d54767d8b7 : SUCCESS in 35m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d392b8fa564db180614303f3b0bf74 : SUCCESS in 13m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54d6299dea7848558f7078e95e0f2ecd : SUCCESS in 24m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fafc60b441d84ce9be9bc6ea0d43e641 : SUCCESS in 27m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c441dd18cf4ff592346b0075bbd1ed : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84b57a3b1b3c47d3b9a78461e369e52a : SUCCESS in 14m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3713201490f04e20bd91bf60e517a7ca : SUCCESS in 1h 01m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/856e03529bd84402b959c2854f87ad78 : SUCCESS in 1h 48m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfb92f85c38d46a3b6572f40c134cbd8 : SUCCESS in 1h 07m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4204d0fd73194e5fa17e50a344d32be3 : SUCCESS in 1h 08m 42s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9d0f1542e03e4ef68b00b99332bbaa4b : SUCCESS in 1h 11m 08s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ce4a978342f4b95913a1702daa06da9 : SUCCESS in 1h 06m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f99b4531743e4bc1b4b8a592c1df36ec : SUCCESS in 1h 23m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99b301588dd54186b1feeed089ba9a9a : SUCCESS in 1h 18m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4b5a60cf490e4de0a5cef929c26692be : SUCCESS in 1h 03m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1062fe650f6c45e79a7a57e47f6a038b : SUCCESS in 1h 36m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/89dacdfbf09940348572dedf772a7284 : SUCCESS in 1h 43m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7a32edf5c95c49a4a57f4dd256341a90 : SUCCESS in 1h 12m 48s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/eb3b024fd41c449db5f85815d586e7f1 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":22},{"id":"c8f5331f44c5b75afa5b443cb7081787af5a66c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-04 17:43:15.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"5dd164dfd23a4be8fc8cd6c20b957c05579cca85","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-05 11:32:37.000000000","message":"Patch Set 22:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/39/860639/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"c163305c879eb4f39e16d85e03e2f9f78db49e21","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-05 14:05:08.000000000","message":"Patch Set 22: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a58acbfb2e5f2463056bcec6f8be87c5d62eca3c","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-01-10 15:27:47.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"5bbf9f8295d8ba788956f8883e1751f7ec5240e5","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-01-10 17:55:53.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"21dccee90f7cbf828889773fd9a6517694274e51","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-01-11 22:43:52.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"2fe7a484ebe60a586f7567104e3f1c38685f3ccb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 22:44:27.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"68cd4bd095b64689f29cfc8d341ab65cd0128e27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 00:46:14.000000000","message":"Patch Set 22: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0282a9976fc74cb8bed82a342bcbd79a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/45299cbe89b344b09660f33d4069f67e : SUCCESS in 1h 11m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/64f1e581937b42f1bc7044ebf0b6a355 : SUCCESS in 1h 09m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2b41efd5c189421ab5f5039ec4f518b3 : SUCCESS in 1h 05m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a6ce85df1104c979064a8b7d2d000fd : SUCCESS in 1h 02m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908b1ffab5db4cbda028caf45ff63836 : SUCCESS in 15m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/269f03b530c943b4ae0ba7ec4e53aa22 : SUCCESS in 29m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/091df21aed0b47ee837b583bdf938c8c : SUCCESS in 29m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58d1de6a80c47d8b0e52ebe2d3f69e9 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b6d9980aa1342129d39f48404addcd0 : SUCCESS in 12m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/86bf40cae7aa435d9499a2320fce312a : SUCCESS in 58m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/829263d4f65a45b98d708803a02110ca : SUCCESS in 1h 56m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/72b693888bf54a7fbb413fbde6a09ea5 : FAILURE in 1h 18m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d26cfaa2ce8e4a50b6abc06103a9f693 : SUCCESS in 1h 18m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/65d7fb558a7d436194c1f9c6952299b9 : SUCCESS in 43m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8150cc3fa3eb4d179e0262d9398ad698 : FAILURE in 1h 16m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bbb23c62b63942dcabceb558becf9d24 : FAILURE in 1h 21m 42s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b59fa8f2f98841ccbd543145d9867d24 : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":22},{"id":"625c6f972704cad38aaaf3dbdc7417f7f6192ed3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-12 10:23:54.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"97adc4c95e9a0cfe739737634dfded3ddc4bf7a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:34:37.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc52845cb645436395c120de1ea7aac4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff24cd84153e462abd71f18b5c4b8377 : SUCCESS in 46m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5c84cc4e54d45e58cbed15a530afb7d : SUCCESS in 47m 48s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"fb4eb4a89a56f617b2aa757805885967f2cb5c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:30:06.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bb451b813c2403e800df534ae8fb6ee\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6b555a40db5e4fc4810435f9dc9bc7fd : SUCCESS in 1h 28m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0904fe9358174e9b9d21e4e531f14c57 : SUCCESS in 1h 32m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6cf2986238384a7885814a62a41c84f2 : SUCCESS in 1h 39m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b4770e5d18345e1ad78678b6faf0761 : SUCCESS in 1h 29m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/43dc102c63d24a5481bad2407e3dcbc1 : SUCCESS in 34m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a66fd46a2be64ae7800abd6f50cdd165 : SUCCESS in 38m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09f69ba80e845728e10ddc11f3201e3 : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33dcd13e1d7415fb2c5b13a0e4b6d99 : SUCCESS in 37m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a13b51b3218543c9a7453f0239422928 : SUCCESS in 31m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff382804b48b4f0fb8f5febb7a1b8890 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d46e24c5bfa46d58128ab9235e6ad94 : SUCCESS in 12m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a159e7bdffbe43b08b32e37b111e6e91 : SUCCESS in 52m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c008ca565b7b4e9a8ad7c47b009ede62 : SUCCESS in 1h 46m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/68c91545b9ef4f5daaf13afabd3eac84 : SUCCESS in 58m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/73996679be10406a958f4737e4794e47 : SUCCESS in 1h 00m 07s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8b1034d83137422c98637e83e950e0c3 : SUCCESS in 1h 16m 49s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c536faafd9364d5597646102e4decd24 : SUCCESS in 1h 00m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cbf0d35d5d5d4c2e9a7d84bac77338e7 : SUCCESS in 1h 03m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a992ca055f704856af8135d31a03cfb1 : SUCCESS in 1h 15m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b38d91634fd846a3888cc12ad82e9680 : SUCCESS in 1h 15m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4392ab4f9ba247e19049131f99c81a2d : SUCCESS in 1h 52m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/290f34524d60403aabe8744ecd5179c9 : SUCCESS in 2h 00m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/41c0dfc4ebeb478a906b7e1d72ce71cc : SUCCESS in 1h 16m 46s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/cadebe6af9444f03831625fddb8b63b9 : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":22},{"id":"653ecb235943cdd7d30e04ed4da0467af64112e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:30:37.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"5339cd471db9ce2234989cdb2af39cfb63bfcc6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:10:34.000000000","message":"Patch Set 22: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b423e1699e5743229b37ad8139740cea\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0b8edb08b5d243c795bd2753057e3800 : SUCCESS in 1h 11m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/66f84291fc4e48d3bb4039dbf9393327 : SUCCESS in 1h 07m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0359f0b8a5b642e1a4f87c4adf42fb0b : SUCCESS in 1h 10m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/52507b356f3a4fb0a8f15b07c7c7c5ad : SUCCESS in 1h 07m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf1563778c54d2b988c2ed3ca3822de : SUCCESS in 12m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/988a2c241f5541248089076321b9600d : SUCCESS in 31m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe73f6a7ca574c66ada78036b894c6ac : SUCCESS in 28m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8eb86915d3949ebba3b0a5601886c39 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61fef6765db74fa7beff0baa9566e57a : SUCCESS in 13m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3aee35009e5e45d4bae825d5dc350aa4 : SUCCESS in 1h 04m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef9ae85243874184a6dd17bc2b7665b6 : FAILURE in 1h 35m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/71e952bf2be84ca0a9a7c5868bc4994e : SUCCESS in 1h 23m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8933effbee64d33b73fe8eddc404957 : SUCCESS in 1h 02m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d59de2134d5c4480897620e232fb8f51 : SUCCESS in 58m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ed0b526948954c2bade419cdfa8f54a0 : SUCCESS in 1h 05m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/717c0d89f2c04ceab16ff637de56ca67 : SUCCESS in 55m 44s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/77dfc098d8e24a329f59d99e2c11c726 : SUCCESS in 1h 10m 37s","accounts_in_message":[],"_revision_number":22},{"id":"93f4fe44d9ae9b6c64382e97c81b8a903290ce8a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-12 14:12:01.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"9e43068a1341e03c5210d28914ea2a61503ce5eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 15:49:59.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aebc4c6190648a2b860ede1d99edae2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/032a57a7f23a488dbb6a69b66bce334b : SUCCESS in 48m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/225a4321c9a147f893a5099c88bf8525 : SUCCESS in 50m 17s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"78504127ea7f16d8429b21f46112b9506fc8e28a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:26:13.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0aeda75aab34528a76f877998326a82\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a44d91b438624bccbec934b87b890822 : SUCCESS in 1h 28m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d024e543967243d78237be87c88e103c : SUCCESS in 1h 40m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f6f793a091334033859a72dd0e28fd47 : SUCCESS in 1h 25m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4834e0e487ce49ca953c47175baed482 : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87365227e35b461798e66d69fbcc4e20 : SUCCESS in 39m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f71808532dd140239b194e25f78e2ee4 : SUCCESS in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ee17c2dcb14c359afad802a7ce2c04 : SUCCESS in 13m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc8b7b3a161b42dc8f8d84a221a780d5 : SUCCESS in 25m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63eb662ce7d140eb9bd7a2c51ae7ac5b : SUCCESS in 27m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40475d13344b49fe93b56d4dbc0abe2a : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47751b184fba4aeba8f59963555249f3 : SUCCESS in 13m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/efb7af43a8354901af9305257e410745 : SUCCESS in 55m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b56f855b3e541a7b30ea3eb41442881 : TIMED_OUT in 2h 12m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b39d74cc96de44f187ceb6ed286c040d : SUCCESS in 1h 00m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/470422e8555247bda6c6d21bf2014007 : SUCCESS in 53m 51s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/272954c0b9ef41589ca75e748c99dda0 : SUCCESS in 1h 13m 24s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d88d65a84a56406893ca6a2f8508d469 : SUCCESS in 53m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8ed85e9dc1134ad0a0c29d8b5afa96ec : SUCCESS in 1h 09m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aedc6d9087b2429287586256936a553b : SUCCESS in 1h 04m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/72829e7476e040879fbb33e8f29d08cc : SUCCESS in 58m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f3351e06d1aa4ca9b44eeac6dd4108bb : SUCCESS in 1h 27m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3e448d35162e4fa38541a7c63f3e8371 : SUCCESS in 1h 30m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/be811def4efc41c281a80d361e971d77 : SUCCESS in 1h 12m 35s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b8a43c9923af4d22b9642c34df68100f : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":22},{"id":"c1d3916ee261cbe472f7d48bbac40f25383b29fa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-12 16:55:01.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"2bb132fa54a8fe34f6d9b6463eecaf4407f34060","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 18:47:48.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34a2e20aab65476688ed6c975afd7945\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2705f6eec964c1b94fdbbdc425e9bfe : SUCCESS in 52m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ddc6af7d1c7a4ace8d634611d3f9df0d : SUCCESS in 1h 07m 00s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"59945a72e1cb5894c2cdb8207ca47fb71c7fac66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 18:50:14.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd6305fb016d42a1b63c020377a4486a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7e13d9f6d5de44be810aef75837e1e15 : SUCCESS in 1h 25m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e346f4ba923b422db9a164eb73f20892 : SUCCESS in 1h 26m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/043347ef2e6d466c81a6eb54484f2eb8 : SUCCESS in 1h 11m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ffd293dd11244ffc9dd7636cd8d0bb5c : SUCCESS in 1h 03m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/faa1327699df4393813388b18c5c1115 : SUCCESS in 25m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fbd8ce6f9a1461fa6cf109a262b17e7 : SUCCESS in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33dd02daaa04690bddb668915816bba : SUCCESS in 12m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d87eddde81454fb6a37879e8dfd146a5 : SUCCESS in 24m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9af322b05a7a4874af89bd0ccb9b450b : SUCCESS in 23m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2374171d3250464093daa4b430bf24b0 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90aba95195b94ee1abe39e1e64557c8b : SUCCESS in 13m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/25dd92de880f4257a0f034c3c3a06b58 : SUCCESS in 48m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/260927412f1f443c9fd398fc26206a22 : SUCCESS in 1h 36m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/acd01109694942be85cb8f795983d218 : SUCCESS in 1h 03m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f4364dbebe5d4c9f8be15e8f4e955374 : SUCCESS in 56m 48s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/51f4e360c9344242b113f83bf03879da : SUCCESS in 1h 07m 54s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/070918bbf462400cb681cb9c0b907e11 : SUCCESS in 57m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cd087bb8e33849f7b702f38f27b55826 : SUCCESS in 1h 21m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b70b576901974ba99498ddec056d0c5a : FAILURE in 1h 04m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7391b35460ac4020b8b7d11cc8ee0bb2 : SUCCESS in 1h 02m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/20bfa425341b460d9511bedbaf3131aa : SUCCESS in 1h 50m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0739a064c880441882dc6011467601fa : SUCCESS in 1h 47m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34ba96c9da264080bfefc1d4057da085 : SUCCESS in 1h 14m 59s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/6b7a4ed347bb47a0befdecf54746dee0 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":22},{"id":"2801a0e2719d5d8311409a4fd9888383f36b878b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-13 09:12:03.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"7f274520fc8b1c2917ee6b88647da3d5b8a2c0b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 10:07:16.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/272cbcea82194a39a309e34b8849864b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e49bb313f97240a1b11e7f3f830ad135 : SUCCESS in 50m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ffbab14f365446e1a5eeedafd834cf55 : SUCCESS in 52m 23s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"23e1b2607c02101e86c413fb7ae946315be1d919","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-13 10:46:55.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"cc4cdb6030c19f1e82f437ee2c562bf2fae8dc73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 11:46:49.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8447db04a35241a99dee79a5535123a7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c1b481745ac5446d8df3e8284a197286 : SUCCESS in 1h 50m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d57b3dbd82b241518e42ad1b23b9885d : SUCCESS in 2h 30m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2470314c9b24507b281c45ed3b1c0b8 : SUCCESS in 1h 52m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/057523721021429fb53b568904577cfe : SUCCESS in 2h 31m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/36ee89b7f4cb492094670742228c1db7 : TIMED_OUT in 1h 03m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04a1caf9d90045888d38f8a72d0f9b34 : SUCCESS in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f50abfb913f4d93a1a84797a14ee1ca : SUCCESS in 13m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1d30582906b4dbeb8417e4cc617b90c : SUCCESS in 26m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90c3e19981fd4295b7cd7c4815b8dacb : SUCCESS in 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3f864251b141f2b6c177331f94415a : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e701671337d946b3945c80fb5d29307a : SUCCESS in 8m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2122d4607f7a40fa8d6129f56b170b26 : SUCCESS in 56m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a561b7ce6f8c42ca9f814f4d8d95a4a2 : TIMED_OUT in 2h 12m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0967752e12f1426694bbaa9bb7811790 : FAILURE in 47m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/04d31fafab1444daaea3c72560b12744 : FAILURE in 45m 42s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4f15b075d29146beacee0291adc9fc44 : FAILURE in 53m 12s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1504c68c2b544f6f97f25ceda2942480 : FAILURE in 46m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/172a03c236654ec3a870ae38a4ec1800 : SUCCESS in 1h 33m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/58f9d8c107f54ec38b38737e40f09fe8 : FAILURE in 1h 37m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/de6a573db02e41299f0145269f64f4f6 : SUCCESS in 1h 25m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/272b54519ab74642bc06144e846161d4 : SUCCESS in 1h 41m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a17f0b8145a745cfb04a93c4bbe2baa9 : SUCCESS in 1h 49m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb407ada76e0495bb101a3e7dd189b15 : SUCCESS in 1h 33m 56s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/af6f78a414df42cba12b15af462b2b92 : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":22},{"id":"4791f090121e11ce55ac7d7112731ca1ab22a757","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-13 12:09:47.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"63c127ad910b0ad4466fe917cd36ae86276feb5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 13:00:10.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b5afcebccae415cba48d844ef873e5b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e02c75acb4444632a104711b9c509ee0 : SUCCESS in 47m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e53c9c2fa6d46f898edf95b30554325 : SUCCESS in 46m 40s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"b0deb9f9dda82cc810697506762cac659349fe6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 14:10:49.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/994e9c60b59541f8b811690194214aff\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b17f6315357a483f8f814a66d253b0e0 : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ae0291c53a1d49909c81b0439dd5b4a4 : SUCCESS in 1h 25m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d4e04dbc19c4637b112ec8f753bfd72 : SUCCESS in 1h 18m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0236cb80266e4af5902930742c4b25f5 : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d057f59d7ccf4276b40979960411c86b : SUCCESS in 42m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/361cd35c7bff46bda7ba08853859552a : SUCCESS in 32m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb65a37986d741a69e0de362878595bc : SUCCESS in 11m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdb7e3145022439b895485a48eccd53c : SUCCESS in 26m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e25f566b03474910b1fe130c4b38df8d : SUCCESS in 27m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8b3322d5e34fb5b1911d1b69b56cc0 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/647d1e35cc2e4147829f401c3d29939e : SUCCESS in 12m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c70bfdd6eef4d2394ee840cc89ee7a1 : SUCCESS in 40m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d57b4516749449c4b6ceb5f0e25bf702 : SUCCESS in 1h 38m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdaee81a4b734e5689f675911984f3d5 : FAILURE in 37m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/599ac797c1954a3cad9aad1d4838efff : SUCCESS in 1h 02m 04s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/67e5c7edd55a4223a7c2eeb42c89cf7c : FAILURE in 50m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8695a9c408724ecaa0f87838ad59d104 : SUCCESS in 1h 00m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1454ac2e89da40afad02a70daf998a9d : FAILURE in 1h 07m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41c8620039914980bda0fcce8054e162 : FAILURE in 1h 23m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cd1c254172544ac2b71c7de0f86c424b : SUCCESS in 59m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fe7c01ce8e8746ae890c63a54a9cfc9a : SUCCESS in 1h 52m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65b8d1d33fd64bf09d78bf01e16d0699 : SUCCESS in 1h 55m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d6b6035de772423d9f86872b9bee3a2e : SUCCESS in 1h 10m 16s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/cc96c5be7e91418697af57c4636dd52c : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":22},{"id":"bb212343a163ec5785b8979c94ea89c4979290aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-13 15:24:51.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"67b337619760a642d67b89f0d5e335313e195246","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 17:25:46.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7917442b8de3420a8252d84336d06fdd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/beb415b12f06455b80d256e4571c3289 : FAILURE in 1h 05m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8e515fc77864522ab988d8591146975 : SUCCESS in 1h 07m 34s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5d1d019b0adbe0d5c4d44f41f54dc54ec6c31110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 17:48:25.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ed236a563364f77ad1ff61de81e46a9\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/335aac318980434d90cd48fb035acce9 : SUCCESS in 1h 40m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cca85b43dfc34e95af465d8a01dd4de5 : SUCCESS in 1h 33m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bfa72791623e4502ab5efe47431dd97d : SUCCESS in 1h 13m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aaf54e45915b488191eae3adb19fe56e : SUCCESS in 1h 25m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/67c5dc7254bd40159727c6cc1f868cff : SUCCESS in 26m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d452820f4f444aab4403ac0a1fb86d8 : SUCCESS in 33m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a4751e265fe466b9682ec3a9cb60a9a : SUCCESS in 12m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f84a8229b8d3469ba1a5990f486c7286 : SUCCESS in 25m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdbc0b2d80d94d6390e9b597d976bba3 : SUCCESS in 28m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb3dcc7ca9a426bb9c22fa73571214a : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/033cd3d308994b3d9f5170b0d4215ad4 : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a57e16954b9444a81c1130a3b7f1bcb : FAILURE in 53m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f9a1cea188a4f88a5688e92f9dee819 : SUCCESS in 1h 29m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/40ade8b39dc14e8d86535907210b647c : SUCCESS in 56m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/125f7179b64245e582ca34d0f2640f2d : SUCCESS in 1h 12m 49s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/74cb2ed3dd474a1686150f44b4f2b6b7 : SUCCESS in 1h 07m 51s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/44837f9bde4d47e7b8e6b2172f49f04c : SUCCESS in 1h 07m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/12033915a37245e1a49a0077e6dbf099 : FAILURE in 1h 11m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9599455b14d24557aa41ad1c9e54c046 : SUCCESS in 1h 15m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/06facfee71984a648376b8cb23fc5fdb : SUCCESS in 1h 18m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/514a341bbfce4ee79cd6bd16b993d1fe : SUCCESS in 2h 04m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1bf9e1e9834742538255c70c2241a3b8 : SUCCESS in 2h 08m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b19847b315241bda88401c222804346 : SUCCESS in 1h 16m 53s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/0895d02a6ecf422b9aa669cc3776c3f8 : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":22},{"id":"d80f3578e0444da8438316bf7d4788d0f73b8ff7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-13 19:42:54.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"603c4d730e8828d420ba1ce4db1c868ca67e36a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 21:59:47.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4865e309f706453d8516b1715f1727b3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c05ab40e3df342e482a07ebf3aea632d : FAILURE in 1h 54m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7110f56d541d4e97b3c30e7f6fb965c5 : SUCCESS in 1h 41m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd9a99cec5f14ad8bf8a7e2d29c2f572 : FAILURE in 1h 31m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/560b69cbd0bb4d19a42e03eac7ad202e : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1c0c88d17114c94be54eeb69b8e1410 : SUCCESS in 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c723051fa3724342b255f9ecc1264612 : FAILURE in 55m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a3426f9dd245f1b08ba9f9ff0dc8bd : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ebf274587f4d8f8af58a7b649919bc : SUCCESS in 25m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7585346982724be3b3b4c6581ab35876 : SUCCESS in 43m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4326d811aa40d9875f45c493a2bc3b : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4c98f4dca5e41f5bde4b5d69f91bfa2 : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bf96bfe522548feae639a74915784a1 : SUCCESS in 1h 03m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10ecc28643af4b18ae0c6a9d959d1208 : SUCCESS in 2h 15m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b709bb518e54872b978555f571d287a : SUCCESS in 1h 06m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e5034d78145d4e1dbbc627fa3e12b25f : SUCCESS in 1h 26m 30s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c46eb8eb034445f4852d6b55905b3ab2 : SUCCESS in 1h 18m 26s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f4218aa6408446987b4aeffd580add0 : SUCCESS in 1h 25m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9cf9dd2ba82d434294852034fbdc0139 : SUCCESS in 1h 33m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c14e4ae69ec54d93a62705b1d5a477fc : SUCCESS in 1h 51m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2e3509cb4c474b548d53b89f3303c101 : SUCCESS in 1h 12m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6aa2ea095b59409ab6e664868fe0d0e0 : SUCCESS in 1h 55m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ca1e89244b34ac19face28469710f8d : SUCCESS in 1h 47m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2a62619d319f4f35abc8c47bc225c923 : SUCCESS in 1h 18m 45s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1479cba6325d4d5f85a640f962e1f5c4 : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":22},{"id":"ee812f3eb66ab8c3c1971332775178dbf63b389f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 22:21:05.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d342927a4f417fb363789c362c3a0a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1aa1839d6274635b96275a30b3017c5 : SUCCESS in 55m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be6c17b720f94ab083de7294f37104a5 : SUCCESS in 1h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"44dfc5e9c811e933a296469094fb192946ed3c75","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-16 09:17:27.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"0960a85fad44d706c15a10c62f2116164de54760","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:15:39.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79faf6035602462097e8058d4e5a3770\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cb6b607bfe8416e982742aac76dd6b4 : SUCCESS in 40m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b065d7621b9f4b808d8bd08b290bffc4 : FAILURE in 54m 54s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"f263e5843d51bfc995e0e4a5e30dcea588bc3d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:05:50.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ffcdff4341a4ba386d96343df1d6b3e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7af9790de0374ab89b60e6cb96afbae6 : SUCCESS in 1h 19m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/37c4267813bf4b4c825f8857c79d5d4a : SUCCESS in 1h 19m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/887d545a3a0e4c0093730f7e1b0f9a1a : SUCCESS in 1h 07m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5e7373fc101c4335af6e6461ee194843 : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b734db803ccf40c48e6cd49c8a1a2cd9 : SUCCESS in 24m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b6b1580847249b0a2d564d7b5fd2148 : SUCCESS in 35m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108615d45cdd46b5be8e054a2a592a09 : SUCCESS in 14m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea56ab1af0814904918cbd6d4bd4f2e8 : SUCCESS in 29m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c243b8dbc1d14b4f806a6f64de70e349 : SUCCESS in 27m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda2411f448f462d9d1b4d65db1d2e34 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/180fca10e19f442096ed9a4601c22289 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/16f6a2cfd5a54fa1a9d5e71d1dbac012 : SUCCESS in 55m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/078741bae4a34d0689302ed20c9d274b : SUCCESS in 1h 36m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1c83943d228476284e4286db815be18 : SUCCESS in 1h 03m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6c755cb9d3004a3491b3911274a06894 : SUCCESS in 58m 35s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ac241658a61a46cdaae10e82db21cb94 : SUCCESS in 1h 07m 33s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/76edcdf6982940da8ae8f5c323fefdb5 : SUCCESS in 1h 06m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de17fa6e41384cef9ab8d89af294576c : SUCCESS in 1h 37m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b4ddb6b9e014987bb05737c6cd32d2b : SUCCESS in 1h 00m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef39d81d40a24c3db7cc0efa3481add1 : SUCCESS in 1h 10m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d9891b2ce223401dbb823f1ebc705602 : SUCCESS in 1h 16m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8b76092afc0f4485a69321343ee64ec5 : SUCCESS in 1h 44m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/83106332249e4a1f9b6d2b177b83671a : SUCCESS in 1h 17m 28s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/46b43885872541aca1634436d6007fbb : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":22},{"id":"b6c48ecc9bb15ca8dd41e6aa1caf263e4fd182d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:06:41.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"718b959b7bb58dad2f4a25b491d8668e019dffed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:57:51.000000000","message":"Patch Set 22: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/225b2389176c47d88af4c928d46beaf3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4c0ca54497454d388d81a6a6d51a993f : FAILURE in 1h 08m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5b4994fee1d14fdc89a9ce6ba79ea6a9 : SUCCESS in 1h 06m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/02b4d0f016214a57a71c20c55c587ebb : SUCCESS in 1h 04m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9ac23a51d6b94c408fbd4ee0091fb6b3 : SUCCESS in 1h 08m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c9b0d77964548b6ae0bbf7fdf3c8f67 : SUCCESS in 11m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfc64db184324d1f86c3241334633144 : SUCCESS in 27m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d75f980a03a4abc99a2d9d31d8ca74c : SUCCESS in 26m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b827e126af4f9ea4e614c1b7e90046 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91dccea23762492387d4bc720a94ce88 : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3e9d3ddf8d641078c967436b93095aa : FAILURE in 44m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b8567f72b4442569a8c7d87469e18e3 : SUCCESS in 1h 44m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b03feacbbbac4936bb8a324189a5edc6 : SUCCESS in 1h 33m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf5116c9c6374b02a7b5f41006d14a5f : SUCCESS in 1h 05m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2937b7832da94cf1be96f5f9105c7588 : SUCCESS in 53m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02e89ff0ca894e8a8f0f72d06eaa82d6 : SUCCESS in 1h 01m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ebf7863fcef843e590efd0f669c57e31 : SUCCESS in 1h 12m 58s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/79e0abdf985a4dd3b13fffbe9b7f08ac : SUCCESS in 51m 58s","accounts_in_message":[],"_revision_number":22},{"id":"ae255384180fc1c1cc4165ffd577900c2956cfc2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-16 15:02:54.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"70c4c425e7aeb4dbb3e9376ecb549e7c928f96e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:29:30.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e045a530114b0db40bf2a667345e5c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2e5fa5f9dfc94ca18e45dd68b1253e4b : SUCCESS in 1h 17m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2544792109b54fcfa1922535fa847562 : SUCCESS in 1h 19m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bfc15e809e834bd09de27ca30c441973 : SUCCESS in 1h 07m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1e8cbfadc5594e97918971ea0c0d09ec : SUCCESS in 1h 07m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/962c3a5eeeae47fc9afe28d2ac3f67e9 : SUCCESS in 26m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/508370a8651141e687858cf7c1818cb5 : SUCCESS in 34m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe917e3a1a74f49959ed38ae7211a26 : SUCCESS in 12m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96a0213048c74f9e9d7e63f9b8d8c897 : SUCCESS in 24m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbc96d299d2a4b588d7933b88dea2426 : SUCCESS in 28m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb941756e04f493f97826b02fc520f4e : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fcb5da5fc29428b989f41d49a4218cb : SUCCESS in 7m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d07ebe6e4f64b199fd7db7f911092d5 : SUCCESS in 40m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/447bc74d3c7d4f03be09ea49831c0664 : SUCCESS in 1h 52m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a3789871a9448f489b861773d49a669 : SUCCESS in 57m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b74d59021eae4bb4ae9482edfb559efe : SUCCESS in 1h 00m 44s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0d0f4b0202fd4d50bc34601e78769cf8 : SUCCESS in 1h 17m 24s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4a443e9ed02e4f14b05e026a232df312 : SUCCESS in 57m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c27f12dfcaee484fb90162365e43593d : SUCCESS in 1h 31m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1338cce9b8c544fdadc23662c4efa056 : SUCCESS in 1h 13m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ec3de5b91abd48c8b4876ee6d2f39795 : SUCCESS in 1h 01m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ae8c54f4fe8440999c7f43fd42cf83db : SUCCESS in 1h 43m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07c493886f464bf4ae80b9e9230876b4 : SUCCESS in 1h 31m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f848c4e5b69d48b6b43fdab461fbbbbd : SUCCESS in 1h 22m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f32d0f28d3d24dbd84102260e6d05b3b : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":22},{"id":"02d13b0e1af5e56b7b29664355301f55fd8bd9b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:30:05.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"7065d8634f5ed5c996512b4dba5c233cb033fd81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 18:59:58.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98deb1b4bfe49a89d126ba7a3044145\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0eebc982bba641a68b60b7f1cd3394e5 : SUCCESS in 1h 06m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1f6c50ea83a143fabbcc06b6e9bde6f2 : SUCCESS in 1h 22m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c48bac66d8af4ff98c4c78d418e779d4 : SUCCESS in 59m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ca981642a264edbb8514a89f0ecd490 : SUCCESS in 1h 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e60ca1b421a4f819607bd6e8095434e : SUCCESS in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e10654829fd046eabc3f0271628bf386 : SUCCESS in 26m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb19e327f4624d21888ecc9298c18a39 : SUCCESS in 26m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066d367ebc8744a7960e0a619d4ebfb0 : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db31bf0f6b1b4cbca9cbecac06df2be5 : SUCCESS in 8m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c760d0234af24ea4a6e751c25fee5c4c : SUCCESS in 37m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f52285b861f44dac805ed46b4de9f2a1 : SUCCESS in 1h 27m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e2318b36c9c43e38d16f586f9a932b8 : SUCCESS in 1h 19m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/83eb2f3392e142bbb5c1d9b2a0e1aafb : SUCCESS in 49m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/479b14a0f1b64fbd81cb214e3c4e896f : SUCCESS in 51m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5d0a9a575a1b414b8a1b26a1c4bba9b4 : SUCCESS in 57m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a504f827c8484ac085257dc2cfb6ebaa : SUCCESS in 1h 16m 05s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fa4cee5f85fd4716a1985b8fb29ee73d : SUCCESS in 1h 10m 54s","accounts_in_message":[],"_revision_number":22},{"id":"931d15163c70bfedcfc3341ffc798c70cbbd462f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 18:59:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"8a5e91d6a9dff4ae9b6042b0bf35fe7405ca4774","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 19:01:21.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1555fb0f4dea4f87a02a6f2b0a59bba4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8090c50b145d4c6384930606b4dc8d6b : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d0ca65cdf2347adb8d5299aab1cabf8 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":22},{"id":"01c334121467a0cf63b6b9ce6a93f72731ed5ba1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 19:06:44.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ac506b09a5547cebbe95dc33d8594c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b32588d683524e70b26f7caee1138cf6 : SUCCESS in 1h 03m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/29162cfbe4f549678f9776aeed72a68e : SUCCESS in 45m 27s (non-voting)","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"3ebdfe612a931a044877b2041a4ab68489bde5b7","revisions":{"2ae940c10cce7304a1a58bf7aeb24f86b2599b7d":{"kind":"REWORK","_number":1,"created":"2022-10-07 08:14:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/1"}}},"commit":{"parents":[{"commit":"44c894cb3892093e3fc4b9862a1f07e77ee752d5","subject":"Merge \"Allow to pass EUI64 IP address as fixed ip for the port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44c894cb3892093e3fc4b9862a1f07e77ee752d5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"subject":"Placement tunnelled networks","message":"Placement tunnelled networks\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ae940c10cce7304a1a58bf7aeb24f86b2599b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ae940c10cce7304a1a58bf7aeb24f86b2599b7d"}]},"branch":"refs/heads/master"},"d95d002257b64e13e7ef6966397c2bca5220cc38":{"kind":"REWORK","_number":2,"created":"2022-10-13 13:56:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/2"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-04 15:52:41.000000000","tz":120},"subject":"[WIP] Placement tunnelled networks","message":"[WIP] Placement tunnelled networks\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d95d002257b64e13e7ef6966397c2bca5220cc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d95d002257b64e13e7ef6966397c2bca5220cc38"}]},"branch":"refs/heads/master"},"2d738bca813637142e2704ced3048fd986694bc2":{"kind":"REWORK","_number":3,"created":"2022-10-13 16:29:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/3"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-04 18:32:47.000000000","tz":120},"subject":"[WIP] Placement tunnelled networks","message":"[WIP] Placement tunnelled networks\n\nTODO: OVS agent code\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d738bca813637142e2704ced3048fd986694bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d738bca813637142e2704ced3048fd986694bc2"}]},"branch":"refs/heads/master"},"386c9ea37f210fb2be5dae41621dc8c3fbef92fe":{"kind":"REWORK","_number":4,"created":"2022-10-14 16:17:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/4"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-14 16:17:19.000000000","tz":0},"subject":"[WIP] Placement tunnelled networks","message":"[WIP] Placement tunnelled networks\n\nTODO: OVS agent code\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/386c9ea37f210fb2be5dae41621dc8c3fbef92fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/386c9ea37f210fb2be5dae41621dc8c3fbef92fe"}]},"branch":"refs/heads/master"},"94cecf1232f2c9fd0379f931934a469e971fd066":{"kind":"REWORK","_number":5,"created":"2022-10-28 10:49:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/5"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-28 10:48:42.000000000","tz":0},"subject":"[WIP] Placement tunnelled networks","message":"[WIP] Placement tunnelled networks\n\nTODO: OVS agent code\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94cecf1232f2c9fd0379f931934a469e971fd066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94cecf1232f2c9fd0379f931934a469e971fd066"}]},"branch":"refs/heads/master"},"59d4066213fdc6dee0f18953785b389fab225c78":{"kind":"REWORK","_number":6,"created":"2022-10-28 16:38:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/6"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-28 16:37:07.000000000","tz":0},"subject":"[WIP] Placement tunnelled networks","message":"[WIP] Placement tunnelled networks\n\nDONE:\n* OVN mech driver is now working (need to check again)\n* OVS mech driver is now working\n\nTODO:\n* OVS agent code: qos plugin needs to discard those ports\n  with min-bw rules belonging to tunnelled networks\n* Fix UTs/FTs\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59d4066213fdc6dee0f18953785b389fab225c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59d4066213fdc6dee0f18953785b389fab225c78"}]},"branch":"refs/heads/master"},"b10520e5218c1cfb63a40fb7a55621340feb30f7":{"kind":"REWORK","_number":7,"created":"2022-11-02 10:38:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/7"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-02 10:38:30.000000000","tz":0},"subject":"[WIP] Placement tunnelled networks","message":"[WIP] Placement tunnelled networks\n\nDONE:\n* OVN mech driver is now working (need to check again)\n* OVS mech driver is now working\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b10520e5218c1cfb63a40fb7a55621340feb30f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b10520e5218c1cfb63a40fb7a55621340feb30f7"}]},"branch":"refs/heads/master"},"8eed319b55402bf410ee94435b192a1d6ea1314f":{"kind":"REWORK","_number":8,"created":"2022-11-03 10:49:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/8"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-03 10:48:08.000000000","tz":60},"subject":"[WIP] Support for minimum bandwidth rules in tunnelled networks","message":"[WIP] Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing:\n* Fullstack tests that will be added to this patch.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8eed319b55402bf410ee94435b192a1d6ea1314f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8eed319b55402bf410ee94435b192a1d6ea1314f"}]},"branch":"refs/heads/master"},"31407baa6070df609bf5c133d2cc6dc82cb29807":{"kind":"REWORK","_number":9,"created":"2022-11-04 11:19:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/9"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-04 11:19:31.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31407baa6070df609bf5c133d2cc6dc82cb29807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31407baa6070df609bf5c133d2cc6dc82cb29807"}]},"branch":"refs/heads/master"},"cd2fa0837efa45cd7e5ee066c0ac4a0ba94ec373":{"kind":"REWORK","_number":10,"created":"2022-11-07 14:16:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/10"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-07 14:15:28.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd2fa0837efa45cd7e5ee066c0ac4a0ba94ec373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd2fa0837efa45cd7e5ee066c0ac4a0ba94ec373"}]},"branch":"refs/heads/master"},"85b4f170e26db8e1713471d1808d637fdeecf8c5":{"kind":"REWORK","_number":11,"created":"2022-11-07 16:06:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/11"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-07 16:05:54.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b4f170e26db8e1713471d1808d637fdeecf8c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b4f170e26db8e1713471d1808d637fdeecf8c5"}]},"branch":"refs/heads/master"},"13571abfa396bfbc8188dd92367d35fec402a831":{"kind":"REWORK","_number":12,"created":"2022-11-08 11:35:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/12"}}},"commit":{"parents":[{"commit":"98d76ec6b480f895a619ff191e8a86b40af662f5","subject":"Merge \"Add \"neutron-ovn-tempest-ovs-release\" job to the check and gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d76ec6b480f895a619ff191e8a86b40af662f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-08 11:35:31.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13571abfa396bfbc8188dd92367d35fec402a831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13571abfa396bfbc8188dd92367d35fec402a831"}]},"branch":"refs/heads/master"},"04047ef121eb8dcb9c6f397057f276330db474e4":{"kind":"REWORK","_number":13,"created":"2022-11-15 16:58:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/13"}}},"commit":{"parents":[{"commit":"320f54eba1a82917e4f02244ea8ddf9757d8f39f","subject":"Merge \"Code move for metadata signature function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/320f54eba1a82917e4f02244ea8ddf9757d8f39f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 10:55:25.000000000","tz":60},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04047ef121eb8dcb9c6f397057f276330db474e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04047ef121eb8dcb9c6f397057f276330db474e4"}]},"branch":"refs/heads/master"},"5976c3b079c2b03b05d35696ad809382a3f95d87":{"kind":"REWORK","_number":14,"created":"2022-11-25 11:27:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/14"}}},"commit":{"parents":[{"commit":"ad742401b70394eba8408752c78a18710e57b58d","subject":"Merge \"Fix missing [designate] options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad742401b70394eba8408752c78a18710e57b58d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-17 09:16:04.000000000","tz":60},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5976c3b079c2b03b05d35696ad809382a3f95d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5976c3b079c2b03b05d35696ad809382a3f95d87"}]},"branch":"refs/heads/master"},"97e37fa64cd4152b3a3edbc5e0d30cdac4aaf1e1":{"kind":"REWORK","_number":15,"created":"2022-11-25 14:56:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/15"}}},"commit":{"parents":[{"commit":"ad742401b70394eba8408752c78a18710e57b58d","subject":"Merge \"Fix missing [designate] options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad742401b70394eba8408752c78a18710e57b58d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-17 12:42:48.000000000","tz":60},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97e37fa64cd4152b3a3edbc5e0d30cdac4aaf1e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97e37fa64cd4152b3a3edbc5e0d30cdac4aaf1e1"}]},"branch":"refs/heads/master"},"ce442db9220d13f799495fe7f4b3618685377557":{"kind":"REWORK","_number":16,"created":"2022-11-29 10:12:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/16"}}},"commit":{"parents":[{"commit":"f75cfbbbb55392ce96e6d6ccdf62eb490896af21","subject":"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f75cfbbbb55392ce96e6d6ccdf62eb490896af21"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-29 10:11:58.000000000","tz":60},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce442db9220d13f799495fe7f4b3618685377557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce442db9220d13f799495fe7f4b3618685377557"}]},"branch":"refs/heads/master"},"d92d86d17c099e5ff544ee2646f241c9e03ee3bd":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-11-29 16:16:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/17"}}},"commit":{"parents":[{"commit":"f7ca9d561d81de7682840ee8846841f0c6aea08d","subject":"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ca9d561d81de7682840ee8846841f0c6aea08d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-29 16:12:39.000000000","tz":60},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d92d86d17c099e5ff544ee2646f241c9e03ee3bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d92d86d17c099e5ff544ee2646f241c9e03ee3bd"}]},"branch":"refs/heads/master"},"4e883e7b871b8a4fb952f4c7706096e67248acd9":{"kind":"REWORK","_number":18,"created":"2022-12-02 12:37:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/18"}}},"commit":{"parents":[{"commit":"f7ca9d561d81de7682840ee8846841f0c6aea08d","subject":"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ca9d561d81de7682840ee8846841f0c6aea08d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-12-02 12:37:15.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e883e7b871b8a4fb952f4c7706096e67248acd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e883e7b871b8a4fb952f4c7706096e67248acd9"}]},"branch":"refs/heads/master"},"9f958da6340b3a1125696d5c573706db1d453a65":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-12-02 14:43:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/19"}}},"commit":{"parents":[{"commit":"49eab7d308b598e4c4966110d6aa61cd091cc3f6","subject":"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49eab7d308b598e4c4966110d6aa61cd091cc3f6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-12-02 14:43:43.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVS mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nNOTE: spec still under review:\n  https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f958da6340b3a1125696d5c573706db1d453a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f958da6340b3a1125696d5c573706db1d453a65"}]},"branch":"refs/heads/master","description":"Rebase"},"5a0f46618cb931994bc936aa2ec4f94516c9cbf1":{"kind":"REWORK","_number":20,"created":"2022-12-12 08:49:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/20"}}},"commit":{"parents":[{"commit":"e2d14b4209bdeeb80093a28a33f2939cb056532e","subject":"Merge \"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2d14b4209bdeeb80093a28a33f2939cb056532e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-06 22:31:52.000000000","tz":60},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVN mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nRFE spec: https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a0f46618cb931994bc936aa2ec4f94516c9cbf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a0f46618cb931994bc936aa2ec4f94516c9cbf1"}]},"branch":"refs/heads/master"},"9232617f763148f389206740928b3ec7750c4a01":{"kind":"REWORK","_number":21,"created":"2022-12-12 11:05:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/21"}}},"commit":{"parents":[{"commit":"e2d14b4209bdeeb80093a28a33f2939cb056532e","subject":"Merge \"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2d14b4209bdeeb80093a28a33f2939cb056532e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-12 11:05:20.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVN mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nRFE spec: https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Fullstack tests.\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9232617f763148f389206740928b3ec7750c4a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9232617f763148f389206740928b3ec7750c4a01"}]},"branch":"refs/heads/master"},"3ebdfe612a931a044877b2041a4ab68489bde5b7":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2023-01-02 11:14:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/860639/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/860639/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/860639/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/860639/22"}}},"commit":{"parents":[{"commit":"e2d14b4209bdeeb80093a28a33f2939cb056532e","subject":"Merge \"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2d14b4209bdeeb80093a28a33f2939cb056532e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-10-02 15:29:32.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-01-02 11:14:50.000000000","tz":0},"subject":"Support for minimum bandwidth rules in tunnelled networks","message":"Support for minimum bandwidth rules in tunnelled networks\n\nThis patch adds support for QoS minimum bandwidth rules in tunnelled\nnetworks. Now the ML2/OVS and ML2/OVN mechanism drivers can represent\nin the Placement API the available bandwidth of the tunnelled networks\nin each compute host.\n\nBoth mechanism drivers represent the compute VTEP (VXLAN) or TEP\n(Geneve) interface as an IP address. This new resource provider\n(by default called \"rp_tunnelled\") represents the available bandwidth\nof this interface. Any new port created in a compute node that belongs\nto a tunnelled network, will request to the Placement API the\ncorresponding bandwidth from the resource provider inventory.\n\nThis patch does not provide backend enforcement support for minimum\nbandwidth rules.\n\nRFE spec: https://review.opendev.org/c/openstack/neutron-specs/+/860859\n\nWhat is missing and will be added in next patches:\n* Tempest tests, that will be pushed to the corresponding repository.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/863880\n\nPartial-Bug: #1991965\nRelated-Bug: #1578989\nChange-Id: I3bfc2c0f9566bcc6861ca91339e32257ea92c7e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ebdfe612a931a044877b2041a4ab68489bde5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ebdfe612a931a044877b2041a4ab68489bde5b7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
