)]}'
{"id":"openstack%2Fneutron~860659","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I2f40dab6af63439c2103a3a9ef23f97aa17aa8d7","project":"openstack/neutron","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2022-11-18 10:28:59.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-11-18 10:28:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2f40dab6af63439c2103a3a9ef23f97aa17aa8d7","subject":"Remove unnecessary bullet points, unify formatting.","status":"ABANDONED","created":"2022-10-09 03:16:31.000000000","updated":"2022-11-18 10:28:59.000000000","total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"cherry_pick_of_change":860769,"cherry_pick_of_patch_set":1,"meta_rev_id":"18e0a319ae56b4dcae07a2af3216ca0e658ad392","_number":860659,"virtual_id_number":860659,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-03 01:54:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":-1,"date":"2022-11-03 06:31:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2022-11-04 10:59:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 06:57:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 08:57:18.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-11-01 18:29:36.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"CC"},{"updated":"2022-11-02 15:50:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-11-03 06:31:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-11-04 10:59:22.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"502b181fd08cb30b5d1352e69f692f55aeead99e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-10-09 03:16:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* doc/source/admin/config-sriov.rst","accounts_in_message":[],"_revision_number":1},{"id":"b806096623d04895639fd1c7ad885d34cd69354b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35025,"deleted":true},"date":"2022-10-09 03:19:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"0ba1b34b184b340ba2191f75eca4dd2a26d6958e","tag":"autogenerated:gerrit:restore","author":{"_account_id":35025,"deleted":true},"date":"2022-10-12 06:33:54.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"39d3bcb31205b9e93f0ef51820e59b3c82e2f685","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35025,"deleted":true},"date":"2022-10-12 06:34:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"ebd9a3f882ca796e8b4902537b598dbcd5f93d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-10-12 06:43:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b81c1ee1f0b59ac800bfe1e792d703e714d5b293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-10-12 06:45:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"65373e87ee69990adedfd4676a6e11f298f9002d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 06:57:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ce6e391c4864135994e365dd04642e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba081282746414c956d89eb41c3a777 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0420131e83f2462a9cfb87c5f6eb82f6 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":3},{"id":"4ffd6a329a74b3e7efcb014ae7d61dc12bf916af","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-12 08:57:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860659/3 : SUCCESS in 2h 10m 40s","accounts_in_message":[],"_revision_number":3},{"id":"9a2506694c46ff37548c8f56c60dbb2466262753","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-11-01 18:29:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7860ee91da4c90d28e7cb218205d4d75e4ba03b3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-02 15:50:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b1aa0e3c57b042a5f17d533ce19e1b70650d3a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-11-03 01:38:52.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0874b29886b64b9639db6c8641264e44d46015ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-11-03 01:39:34.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"115d05f9d592fd61f5a42fc8d25fc20344145fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 01:54:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2db90ec7eea5437bba0e0013a805d331\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff502e5cdea04a5eb11889e3b96a6e19 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac28df4180c4ca297cc3cf7b00b2aba : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":5},{"id":"5d13e120a767c3ebf68dcc5d556d124712b8f526","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-11-03 03:21:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860659/5 : SUCCESS in 1h 41m 30s","accounts_in_message":[],"_revision_number":5},{"id":"9843bc3303f50415fbbd25dae84f18325b54e3e7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-11-03 06:31:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"88d9d0477d655e63f4fd1d379759c1dc2eccc15e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-04 10:59:22.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"18e0a319ae56b4dcae07a2af3216ca0e658ad392","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35025,"deleted":true},"date":"2022-11-18 10:28:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"49f8b5b62c28793ef27873228a9dc7f8addbca9b","revisions":{"68109a3d8484c905719474afd9b784a9a8fe7ffd":{"kind":"REWORK","_number":1,"created":"2022-10-09 03:16:31.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/59/860659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/860659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/860659/1"}}},"commit":{"parents":[{"commit":"d890ad38a291fb735d7830e571cd71e193c7cceb","subject":"Merge \"Accept a port deletion with missing port binding information\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d890ad38a291fb735d7830e571cd71e193c7cceb"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-10-09 03:10:55.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-10-09 03:16:31.000000000","tz":0},"subject":"Remove unnecessary bullet points, unify formatting, and update broken links in other branches.","message":"Remove unnecessary bullet points, unify formatting,\nand update broken links in other branches.\n\nChange-Id: I2f40dab6af63439c2103a3a9ef23f97aa17aa8d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68109a3d8484c905719474afd9b784a9a8fe7ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68109a3d8484c905719474afd9b784a9a8fe7ffd"}]},"branch":"refs/heads/stable/yoga"},"64b17c544e6402856eb4fd957ddf2981c4ee1412":{"kind":"REWORK","_number":2,"created":"2022-10-12 06:43:35.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/59/860659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/860659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/860659/2"}}},"commit":{"parents":[{"commit":"d890ad38a291fb735d7830e571cd71e193c7cceb","subject":"Merge \"Accept a port deletion with missing port binding information\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d890ad38a291fb735d7830e571cd71e193c7cceb"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-10-09 03:10:55.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-10-12 06:43:15.000000000","tz":0},"subject":"Remove unnecessary bullet points, unify formatting, and update broken links in other branches.","message":"Remove unnecessary bullet points, unify formatting,\nand update broken links in other branches.\n\nChange-Id: I2f40dab6af63439c2103a3a9ef23f97aa17aa8d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64b17c544e6402856eb4fd957ddf2981c4ee1412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64b17c544e6402856eb4fd957ddf2981c4ee1412"}]},"branch":"refs/heads/stable/yoga"},"91763bf1808a493f4424927440c9a8909ca090ad":{"kind":"REWORK","_number":3,"created":"2022-10-12 06:45:59.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/59/860659/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/860659/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/860659/3"}}},"commit":{"parents":[{"commit":"d890ad38a291fb735d7830e571cd71e193c7cceb","subject":"Merge \"Accept a port deletion with missing port binding information\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d890ad38a291fb735d7830e571cd71e193c7cceb"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-10-09 03:10:55.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-10-12 06:45:59.000000000","tz":0},"subject":"Remove unnecessary bullet points, unify formatting, and update broken links in other branches.","message":"Remove unnecessary bullet points, unify formatting,\nand update broken links in other branches.\n\nChange-Id: I2f40dab6af63439c2103a3a9ef23f97aa17aa8d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91763bf1808a493f4424927440c9a8909ca090ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91763bf1808a493f4424927440c9a8909ca090ad"}]},"branch":"refs/heads/stable/yoga"},"58b7db4ae9d3bcca25d91707284f4ddd2b01490c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-11-03 01:38:52.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/59/860659/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/860659/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/860659/4"}}},"commit":{"parents":[{"commit":"d890ad38a291fb735d7830e571cd71e193c7cceb","subject":"Merge \"Accept a port deletion with missing port binding information\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d890ad38a291fb735d7830e571cd71e193c7cceb"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-10-09 03:10:55.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-11-03 01:38:52.000000000","tz":0},"subject":"Remove unnecessary bullet points, unify formatting.","message":"Remove unnecessary bullet points, unify formatting.\n\nChange-Id: I2f40dab6af63439c2103a3a9ef23f97aa17aa8d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58b7db4ae9d3bcca25d91707284f4ddd2b01490c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58b7db4ae9d3bcca25d91707284f4ddd2b01490c"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"},"49f8b5b62c28793ef27873228a9dc7f8addbca9b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-11-03 01:39:34.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/59/860659/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/860659/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/860659/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/860659/5"}}},"commit":{"parents":[{"commit":"d890ad38a291fb735d7830e571cd71e193c7cceb","subject":"Merge \"Accept a port deletion with missing port binding information\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d890ad38a291fb735d7830e571cd71e193c7cceb"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-10-09 03:10:55.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-11-03 01:39:34.000000000","tz":0},"subject":"Remove unnecessary bullet points, unify formatting.","message":"Remove unnecessary bullet points, unify formatting.\n\nUse bulleted documents to display problems, uniform formatting.\n\nChange-Id: I2f40dab6af63439c2103a3a9ef23f97aa17aa8d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49f8b5b62c28793ef27873228a9dc7f8addbca9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49f8b5b62c28793ef27873228a9dc7f8addbca9b"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
