)]}'
{"id":"openstack%2Fneutron~860767","triplet_id":"openstack%2Fneutron~master~Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec","project":"openstack/neutron","branch":"master","topic":"bug/1964342","attention_set":{},"removed_from_attention_set":{"35618":{"account":{"_account_id":35618,"name":"zhihua fu","display_name":"zhihua fu","email":"jxfu@163.com","username":"qingshanyy"},"last_update":"2022-12-16 02:07:55.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-03 09:58:24.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2025-02-07 23:40:29.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-02-07 21:10:35.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec","subject":"Add meter bandwidth limit support","status":"MERGED","created":"2022-10-09 01:50:35.000000000","updated":"2025-02-07 23:41:36.000000000","submitted":"2025-02-07 23:40:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860767-bug/1964342","meta_rev_id":"94b3beba5774fccbac0923d0d1b35c5785120212","_number":860767,"virtual_id_number":860767,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-07 23:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-02-03 09:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-02-07 21:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-02-07 21:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-09 02:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-09 03:20:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-09 03:39:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 05:02:26.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-11-22 08:47:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-12-07 14:21:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-05-28 20:42:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-02-07 21:10:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"00f63c624ddbe385e12fdd73b31bcbd01b77ddf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-10-09 01:50:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e69ab0f8a25af3f924e003f82600e5d0d1ef148c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 02:36:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f3c2542ede9411284dd4c049cef2b68\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/974495712aa243d5bd10f58223684574 : SUCCESS in 39m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/273b60849d984433a527580d7e969504 : SUCCESS in 37m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a7b593ef97d4c30283c2ce81fdc6d3bd0339e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 03:20:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c762a673047d45c48f1c409b74de818f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/deb1c1398f78421f8fefe1e8a3d1cda1 : SUCCESS in 1h 10m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cfc24501f098419bb0eead95752ec673 : SUCCESS in 1h 05m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12d6a1596a314f8b845d7544fdd95f85 : SUCCESS in 56m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/094dc2c543de40418a2815bcd5088a5b : SUCCESS in 1h 03m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc19d346a22a4b1d9ffc66c1e129845d : SUCCESS in 29m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe96e489dda43539935ead9465fab21 : SUCCESS in 32m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547de1139dc54b3296ebb231a1507589 : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2fe2592818142b1bcf86973875d1ca4 : SUCCESS in 28m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d4c32e63d6146b6afb4dab51c2955e5 : SUCCESS in 27m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d5774395d4744a78274df046ca4d374 : SUCCESS in 30m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a823fec5894645919958d187b9d25e : SUCCESS in 8m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c932b23d5eee4cbd89c93a4f76ec2d53 : SUCCESS in 54m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/821e10080d9d456c8503019d739b0ee3 : FAILURE in 1h 22m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/762fbfa7300c4e549595d54b338b47de : SUCCESS in 1h 14m 48s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a86ca3f8ed1145b4b24a3435b82d5c08 : SUCCESS in 57m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1f339732086435db0220added4f08c6 : SUCCESS in 52m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/18c6bef2e34042ebafd77d26056ca649 : SUCCESS in 1h 20m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b9a4a2fef6f24b01a51a9a516219f4bb : SUCCESS in 1h 23m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e9e5cebe734413880a001ea04015ade : SUCCESS in 1h 03m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51a0e8584cbd4cea90c138fd51c0135d : SUCCESS in 1h 17m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df109ea19ff606d2e936a46bfd9f80692a56fa5a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-09 03:39:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/67/860767/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e8b3fbb79824eec2e827ee1659a6625ebbe44507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-10-09 06:57:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"608d14d439758c99a18d50b122267cfa7805f545","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 07:40:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/324f7cb451e9451eaffbf9e380a2d2e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/136e28caa79b4a66a9c6d546f9a6010c : SUCCESS in 34m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15e4a038e4764fa6b880390ee918e5b2 : SUCCESS in 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf996f684e423cd29f54889dc03a6abb00198719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 08:32:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd27ce3fa38741b988f2993ab74e9851\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f9a5a94a158442b4bc597103ae184478 : SUCCESS in 1h 11m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f49fce60c6344f3bace24f29cf3c65b : SUCCESS in 1h 04m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d83f99b9f394b1f8afce39d067b5a57 : SUCCESS in 59m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d33cac78bc7948279d335bdc43e30ace : SUCCESS in 1h 11m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b22637866b6043e3b865a1ffe947c69d : SUCCESS in 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02601a4855aa417ea628cfe36f584ae6 : SUCCESS in 33m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82914ef81aa7485aaf2b6e88f333f8ae : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8901b57297e14d16879b7abe3732887b : SUCCESS in 20m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4e5850d0a2b463486a2af3fc8fb86e2 : SUCCESS in 23m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d1d5a5ffdc349a18ac1d78d8a8cfc4c : SUCCESS in 25m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3b764285424dbc8a6e7c1d60b710db : SUCCESS in 9m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a44057f6044046a7b174317a08499faf : SUCCESS in 39m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4b9da27f9a74bb7bb0204368cfc6ce5 : FAILURE in 1h 29m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb0cf53bfca24d89a1a586ef6a2608d7 : SUCCESS in 59m 20s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/77e2c71f1a564092b1a2f5bcdc07cfd7 : FAILURE in 43m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f7fd50038f0e40fcb389b30a06eecaa1 : SUCCESS in 1h 05m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ca48ae55ee6443c3be589d78bd246543 : FAILURE in 1h 10m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1046a360bf904560951449d8334f0034 : SUCCESS in 58m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8148979f2d6349afadfb843b08e4b6ed : SUCCESS in 53m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc57da51904c4d949702917560cc5d9a : SUCCESS in 54m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60b3551ae91835cbbaf024dd9d28374b850a0ad1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-09 08:35:52.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/67/860767/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7e7c26b4f26100816d4336573e8db35656385fa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-10-18 03:05:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"68c01ab6ad092e707c926e9d69e4dfc4bc8dcdaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:59:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e2935313bfd4bc0927ae5b7b0e20d7b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d46cfb509e2346dfa35ce437cadc2d55 : SUCCESS in 51m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b5a1688c17d84318bb950feeaabc45c3 : SUCCESS in 40m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ec0e98c72444563a8c2e91ef9221c08 : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":3},{"id":"5e32c02b39935ccedc934cdaa9a30e8d6d8b6ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 04:30:48.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e2f366a35c443e09add4b6f46503205\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4b167067a9aa44e7a7c7a425e6178895 : SUCCESS in 1h 06m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/192616acaea14b20bcf8fe8ced3e14de : SUCCESS in 1h 06m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9fad8d45f2aa4c499336c2f3547dcc6f : SUCCESS in 57m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/71be5db7576c4640bd1ea6dd0760f151 : SUCCESS in 1h 01m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/56feefe6c49d412a83abd3a8df31e308 : SUCCESS in 26m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f25d3103ab24f7799a8def48b4ffb5e : SUCCESS in 37m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/736124c601e644b8909e2c8699ef2f54 : FAILURE in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba789f7a08384a679c7351517a91a768 : SUCCESS in 22m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/973ab848f9f840e6882255d846f66eab : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436956a8384e4f7a8f473060bc0a14f2 : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8ef86da3003456595a4533a5e3575d7 : SUCCESS in 32m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a0706a2ff624e95be6e637056784553 : SUCCESS in 53m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0e39496d4fb4732a94dbc99e2e97dd6 : FAILURE in 1h 08m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/59c476f143f94c459f5c7b81f1e137c3 : SUCCESS in 53m 41s\n- neutron-ovs-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/bbfe82caec2940af9d5abd9e7e6137e4 : SUCCESS in 53m 42s (non-voting)\n- neutron-ovn-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/e56e77594b6e4375ab0bf349c743ca65 : SUCCESS in 1h 01m 32s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/832bc7110c41424f805c60a76da808f1 : SUCCESS in 47m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/44fc6d315d8a443c85cde662a82bfd1b : SUCCESS in 1h 18m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/464babbe4c1440339ef0b1a469cd7294 : SUCCESS in 1h 16m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/953e6a48a5e041c8a965f32660a6bc83 : SUCCESS in 59m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d2cd5664ec4b459cbb3534d63b2ff46d : SUCCESS in 54m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e3cfd11997b349e1bf5f779b9dad3ba8 : SUCCESS in 1h 04m 19s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/59110d9d5fb14bdface251f2ffbf4147 : SUCCESS in 1h 09m 43s\n\nWarning:\n  Comments left for invalid file neutron/tests/common/agents/l2_extensions.py","accounts_in_message":[],"_revision_number":3},{"id":"94a3a5959cd0a478e4563f98be31b1f956da12b0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-18 05:02:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860767/3 : SUCCESS in 1h 54m 51s","accounts_in_message":[],"_revision_number":3},{"id":"3f39981d2b0592fe2d58d26e7f3311170143cd8c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-18 10:47:36.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/67/860767/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"40280fbe225d3c0a8d170bd504e227ad4a0ef4b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-10-19 01:44:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c8e3a6a6e78cd4a0545659536e363312ebab940b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 02:42:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c1f6e2c2504cffaac5664e8e3f9bf2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48c40b08b81d4380b5a5e267d511aa2e : SUCCESS in 46m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95a89ef22c67484dad5c4d7ad5aca470 : SUCCESS in 47m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13c340e13ac84e8aa717c3dfd78bfe60 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":4},{"id":"243f079c25c70cf19e470fb5a6338351628350d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 03:27:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfa1b5314b244c1597fbb1259ff1f6d4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9ff5ba39df4d4363ad5c026478725e7f : SUCCESS in 1h 32m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7f4895d6a5eb4f52a90496e6d65ddd25 : SUCCESS in 1h 22m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3242e015940b46acbc7ec5338653c17e : SUCCESS in 1h 18m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6b94d5ca97b94f0cb9568bffd560e8ca : SUCCESS in 1h 26m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45efec0943e04143881a30bcb541832b : SUCCESS in 34m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dca8d51e5164a75a69ac77569893b6a : SUCCESS in 39m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/976616f089da419ebb635c866ed345e8 : SUCCESS in 13m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c67e6890821d41ff81206a3b520e74be : SUCCESS in 26m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f235391670cf42cdae53adf2cfb79b47 : SUCCESS in 27m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db51523c8e19428d9625fdf9496711dc : SUCCESS in 9m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7465cf6a387a45e6911846fb0e3ed805 : SUCCESS in 27m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/269e2f30666d4fc596e9b6e3056ab250 : SUCCESS in 46m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1c7b6b9f17d4069973f5ed7a2905dab : FAILURE in 1h 24m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e41e739db146494d8933d788dc24ab08 : SUCCESS in 55m 55s\n- neutron-ovs-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/2bcaafcbc90d4283b6e31567a6d26bf0 : SUCCESS in 55m 21s (non-voting)\n- neutron-ovn-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/f6c1cad55f9b4ea39583e9e3a1ee3c86 : SUCCESS in 1h 04m 56s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c1a24dcec6354d64b186e8529814d326 : SUCCESS in 55m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/34608ec261b74720863b648c429f9b6f : SUCCESS in 1h 21m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/838ad75852504e94b7d45b99d083d4cd : SUCCESS in 1h 28m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d5e1f4848f19478f8be5feddb818ebdf : SUCCESS in 56m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7cc024d33cad476994b1f6f45761196e : SUCCESS in 1h 03m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9757b3ec845e4cd98c5d33938de96a60 : SUCCESS in 59m 28s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a6f7bc4f547443789d4b6e04ee03aec : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":4},{"id":"2d9b53ddec8191af2c30b2d9aeed292a886d624b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-19 03:41:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/860767/4 : SUCCESS in 1h 54m 00s","accounts_in_message":[],"_revision_number":4},{"id":"1b74ecf45cab7af9bbe3b1e6100b8d2a868974f1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-10-19 05:45:49.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2022/Arista_Tempest_Testrefs/changes/67/860767/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"96375fdaec84e55b5ee7dd6dd087e44b77435aa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-14 01:40:03.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e514a4113adb3c3738f8d21b060a032aeb8729a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 03:04:20.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2666bf4753564904b21df3222068a7a3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ecb9ba0e6f34200ad673de1ef1940e4 : FAILURE in 1h 11m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/925dd8e521134511a0ad18853cae64c0 : TIMED_OUT in 1h 00m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d630bf7799044199ad01f7df666dd9ba : FAILURE in 1h 10m 58s","accounts_in_message":[],"_revision_number":5},{"id":"f21a39514c5d0c3b079f7e09152ae6215b2a5917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 03:19:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efe88bc441de4bd68ae9dbdcf43e690b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1fbcc3a420e64d0a8431b7558e0b23e7 : SUCCESS in 1h 01m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/89606229009f4f1ca95a2e1e0fe2e42a : SUCCESS in 1h 11m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2bb9b5b5accb4e2f8eb4a23b2bce0269 : SUCCESS in 1h 02m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b503abb47134b30946e2629b2098353 : SUCCESS in 1h 02m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/63eade021468412c94de0e384877c38e : SUCCESS in 25m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8587ad020694a2ab7866dbc68a97566 : SUCCESS in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf5ca3aed6644d891a303571eba92de : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b8896544fc84a15a6befef12024b7e9 : SUCCESS in 27m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5202f463a8c24c32a6986c89bf486347 : SUCCESS in 21m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27a6d11d5da4477a56fc7a1722c3e50 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b6144814f55479ba6fc86331d383730 : SUCCESS in 22m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b3e43fb26b64359bf6176c2144eb5f5 : SUCCESS in 49m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/80e0b07e2486460fa60686ee52f937fc : FAILURE in 1h 32m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5bb35b42a5724dfb90b2b0f85b84f716 : SUCCESS in 51m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/af274da6b3be4cf4915e9ab93843ff96 : SUCCESS in 49m 35s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9629c92477f744eb9024365dfbe5dcb4 : SUCCESS in 53m 51s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/644d7053a67b472e9128d239a94601d4 : SUCCESS in 53m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e6de9f08c25246cfac4f62d75b43a9db : SUCCESS in 1h 10m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c2e00e5feebb474aa9b5ac82347d7740 : SUCCESS in 1h 10m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/453ca4cbb12d47d5873b8a72fc91489e : SUCCESS in 1h 04m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f288aff9c594f868c37bfd4f639a191 : SUCCESS in 1h 33m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74237e1eff434852b1d53d023a890169 : SUCCESS in 1h 16m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b690eda49b0443449cae7167b35dd105 : SUCCESS in 1h 15m 50s","accounts_in_message":[],"_revision_number":5},{"id":"82950cef82dab784c4bdecb09b6ae1fb6a7a929d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-14 03:40:31.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"2289dd00aee82b5409784770a37dea62b132d6d0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-14 03:46:34.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"4971da2fee29a983f59ea7e29d5b4fed391eb3ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 04:29:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e40f03738d5d48eebfb0f56fc66d2087\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37233884505646a1af7901ec8360c250 : SUCCESS in 37m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc5f40abe7394de69a8c4962d19307fd : SUCCESS in 46m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c10a043d64e047bfa76c2f1f541c9ab3 : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":6},{"id":"35cfd023560c8f6ae9ef71f8c595b68b953d6232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 05:09:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/885e98ce73264c98ae538617deb68386\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9d72ce93840d4107807adf1514af697f : SUCCESS in 1h 03m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/73be09b9553545b7a1906811929cf04d : SUCCESS in 1h 04m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/246b083062d4443e9736a2c34940b71f : SUCCESS in 57m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/85e413eb3e864ef3968647027255cdbb : SUCCESS in 1h 07m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca42027a8421463385a255b7cc473def : SUCCESS in 25m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e931472058c148b2a602ee3fa4bcde68 : SUCCESS in 37m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce5e76e50b04e53815a4ef3a68fcb17 : SUCCESS in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5753cee89bda45a0a0d43ac944fee5ce : SUCCESS in 23m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcc590bdb18742b88a95b92646304456 : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aeb85dd638c4d50bbabeff552ce5c38 : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c77e24f3945437b819fdd1004d386dc : SUCCESS in 22m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad5e6030471042d0b52d1c2963caee12 : SUCCESS in 44m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74579feba8b549df80e4f8ab079f6ded : FAILURE in 1h 26m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/50bf594e43714848a36689a49c568f3f : SUCCESS in 51m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d9cc13fa3be54e068356452a6828c60d : SUCCESS in 52m 19s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/87260ad8f7064f0fb739b98dba467562 : SUCCESS in 1h 06m 21s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/633f8f87f5354b948612a03e975e17bc : SUCCESS in 57m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ce151df228f47e59814d62e3c5508eb : SUCCESS in 1h 09m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d027868ce91472997a18baa83a4326c : SUCCESS in 1h 12m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f0f6840f2b904ae1a25ca7c397efb568 : SUCCESS in 56m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d968370efa6c49a3bf530b1fd2aa9612 : SUCCESS in 1h 13m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf51c30a6cdc4a1bb3a79ee67980ec55 : SUCCESS in 1h 14m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/987c6ec9c7714b33be9a2b42b954a0f7 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":6},{"id":"a5e546bc8f569a99732e3fc08f0a4ff465afc447","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-14 05:39:42.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"43d846bf6a12f25cc115e5360ed260012cb9dff4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-14 06:47:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c12865c980f5e6477e2975e66204eaabcb84738c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-14 07:13:37.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"14f35171d32517f416535d3b288858088ac957c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 07:42:23.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4b163d6983c4984bbd8d274412f60e5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfe93f9d0c5e42bb86784f61220c0e22 : FAILURE in 46m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d54b8a3abde44a868185cf01175fac55 : FAILURE in 43m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de123235a2214d97b11a0c3e524803fc : FAILURE in 46m 34s","accounts_in_message":[],"_revision_number":7},{"id":"c496f7bbf3827528f7cb4123c8b443c217eaa207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 09:08:42.000000000","message":"Patch Set 7: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8640d953f87e46e191d6e247c4613c31\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/63a23f70560749489ae1538963734339 : SUCCESS in 1h 18m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/17c0775d1bd246e09e775b09beee8baa : FAILURE in 21m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f4395e0b83fa479eadd733cf81195958 : FAILURE in 20m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e91730dd0f44485fb32961fca25d88f9 : SUCCESS in 1h 24m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a922bd7f883d4fbd99d8b05c2cc1a6a0 : SUCCESS in 30m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ae4b966d6e546ac8fa985c425d42b95 : FAILURE in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd5b0c5b0dc643b39b24c230dcc9a19e : FAILURE in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f63d264e30764530b129b84f24566fb7 : FAILURE in 26m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb14b037257b45e0b5961934c058f99b : FAILURE in 26m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8e8946ee254f1c835ded47b771c66c : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/672da57943d645958e2c4dea90bbe8c6 : FAILURE in 27m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e67eaaa2bcfd428488b3bfe9e07ec813 : TIMED_OUT in 2h 12m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ebd9d75a4294ce2b34ae22280e8c1f4 : TIMED_OUT in 2h 12m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8199863c15724d439258b1f21790b57c : FAILURE in 56m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d0f484b94f00488fa594a63b468a4ec4 : FAILURE in 1h 02m 58s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/431412906f014960958a669f71506ae5 : SUCCESS in 1h 13m 29s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/76a7e87472e94d03a48d533666534c05 : FAILURE in 58m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/70b64a662d3c4e0781cd70d7dc18f431 : FAILURE in 32m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a2857e93c1442cda03c837fc4aa5f92 : FAILURE in 29m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a08b1f85815c431e95a7971918620358 : SUCCESS in 1h 02m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dddda03ee3ff40ad8cec9f2d9a91015c : FAILURE in 27m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df5cfce85a494f4db10316660235312e : SUCCESS in 1h 13m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20d7d1dfd4224869b2e50dcbfce02dd8 : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":7},{"id":"6bd1d137b74975ff02e548de842de6a6a820ffc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-15 02:21:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2957cf606c5a9c61b36ae8d09779e42636d63450","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-15 03:34:18.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"5783587bb958aad3c07014c1a9d44c89b74e4bc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 03:36:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc81cfb6774b4f4fa33ee01d4686ddec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d68e41889ae342d09b784959365ca0ba : FAILURE in 44m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/34bc7f99b4784e4f91f9b8ceb5d164e4 : FAILURE in 53m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/123827a32f2c469ca6fb30fd8ee37b8b : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":8},{"id":"12e6d3b74f2c6773ff138997b381375ad0f5c108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 04:11:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97495d0c35a14994b0d5eaab3608464d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/85130ddf300740a686e93afa3de7f418 : SUCCESS in 1h 14m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/84228096dedd44029f6d907390081099 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4079da8b3eb646f98bba617276d24390 : SUCCESS in 58m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9bb40a81c51e449fa112811bb6f2df36 : SUCCESS in 1h 06m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d8af877926744cc805e82fed6889056 : SUCCESS in 33m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36cd7d55911449909a643da50ca74d95 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3378b5e92482488084ad4b534701c4a0 : SUCCESS in 12m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91d648b4170e4aa286d78ea022c58f27 : SUCCESS in 24m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3b80bbd9c6345058f77406c956ac089 : SUCCESS in 29m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9fce2d393b44ba9ee5413e19d2b231 : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ae8b313e5f4d509d95034f7189dd96 : SUCCESS in 28m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7275d011888f431289a0339641e836f3 : SUCCESS in 41m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c70d4ddce0b48379dad3ebc2fd6aa0b : FAILURE in 1h 10m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee8090274d81427997b2d267d64a3b55 : SUCCESS in 46m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6453635d1f1b46f99a2116bf650bc276 : SUCCESS in 1h 03m 09s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/479e2afeb1c340d8a8a476cfc6d36b10 : SUCCESS in 1h 03m 33s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f2abfd1c322d4f299f84c6d29bf72a87 : FAILURE in 46m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/121307c73a454de281d0f22e692980b1 : SUCCESS in 1h 28m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f97d2941de345cdbcd9885587cbcb3d : SUCCESS in 1h 23m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d6eded7d206946938b33cf3c704e051d : SUCCESS in 57m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/456f379dd5a44a9aa30dec2a5593cfb9 : SUCCESS in 1h 19m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4698acd16df04387a57f79088487a26d : SUCCESS in 1h 15m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0294b28bc1da4e1892a4c3ff4a00fbed : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":8},{"id":"521570d7a0d04c39caa2efd120ea1c61be4fa777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-16 01:33:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e50f6db2f45a764b9cb70a5438fe93b6627c269c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 02:58:53.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d7df48a63254bfc91fbfae6007b5561\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47021a88f82c49f693cdd4e7bdb0338c : FAILURE in 1h 12m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/edc05d2c6e97427ab02425014632d7c5 : TIMED_OUT in 1h 01m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06a82e0858474db284c3640962fbe631 : FAILURE in 1h 11m 40s","accounts_in_message":[],"_revision_number":9},{"id":"05993e6fb4c4121146685372969d17a17dd47adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 03:14:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5c5cbe018e74ef5b3dee4ff38de59f2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dc08d021d1254d109c541ee70c95f52b : SUCCESS in 1h 07m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/70cdf4d19dc34a8fa7c6e0b927fd2006 : SUCCESS in 1h 07m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b54568384166459b950101463ec3b687 : SUCCESS in 1h 16m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d645b99354c049c68f1582414c1eee9d : SUCCESS in 1h 03m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1953050814194c95bded7aa00ff27cd9 : SUCCESS in 27m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a4216bcf2f14f259db2c49fc57baeb1 : SUCCESS in 48m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053f08599b2b41afae2d519f1dbe3f40 : SUCCESS in 18m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77f9c6e1b41f4ce080522e39bd7e8b4b : SUCCESS in 25m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc271fdf3a0849f698780d1d84b57311 : SUCCESS in 37m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00612a6afd304ca4a2944e3b524cae7a : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8881a13d669d4acfb558e9c9938a46d9 : SUCCESS in 27m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0f3fe3218a647418efbc07474e5ff59 : SUCCESS in 58m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e84db49d86643ff8fbd83cbacb84463 : FAILURE in 1h 26m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9e015e87f934fc2a8206d226908db16 : SUCCESS in 1h 12m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1f794e24c5a847499d02e47f9f9bf772 : SUCCESS in 1h 10m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/31204337b94f4b66b4161edffefb04f8 : SUCCESS in 1h 24m 22s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8b3e79bfbcf14074bd97e912b357c7c1 : SUCCESS in 1h 11m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ab22992ed1fa479abcd76e05815d1dd0 : SUCCESS in 1h 27m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1370802192c646ac92f01bcce4ef814e : SUCCESS in 1h 32m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e7db990764a44867b2806af6e2d5ece8 : SUCCESS in 1h 15m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fb9f3eaa70814576b0b6a7c38567d8f2 : SUCCESS in 1h 17m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65b0486a1ff5447f8b7e81768da9e44e : SUCCESS in 1h 13m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5058fa857547404a87edd415fca94ff6 : SUCCESS in 1h 28m 17s","accounts_in_message":[],"_revision_number":9},{"id":"d52111e452edeef0fc7242a8b273503b9047e070","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-16 03:28:14.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"da1825183a328e5b9de2de3c7a68cc66dcadf91b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-17 02:11:52.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c9714dbf81c3ec3fb79b2acdcb82a84e2f6a5a99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 03:24:46.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51f7c8e6acdc4ad68c475ac4800c334e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41beb29e7ff94dd480f435f95645441c : SUCCESS in 44m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83f7af17cd944e7a9f638cbe79bda5f4 : SUCCESS in 46m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d4f55b17a9844efbda1fdbbb2420d23 : SUCCESS in 36m 55s","accounts_in_message":[],"_revision_number":10},{"id":"22a5bd27d355012c609168c1b6d9117b74fc0209","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-17 03:34:20.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"28ff410e012781adad01ee9c52489c2788d691a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 03:53:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb801035d3a946ad95b036986b0a66aa\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1dea4b472fbb4f298a331df4713c3b4d : SUCCESS in 1h 04m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3f692d887984494695f8814d2c530f81 : SUCCESS in 1h 08m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1dabc2addcc4860ab815cf4edfd9fbc : SUCCESS in 53m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ed37fc38c8484d2ca56f7b62f096e9ab : SUCCESS in 1h 05m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/85baa95a55ab4a07ac19dceb4ac74ade : SUCCESS in 30m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80fae45b16a54e80b53139feb49387f6 : SUCCESS in 26m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce647fc9fb55440791415ed334c5b426 : SUCCESS in 12m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14a04bf1f52d4444b4ad204444b0aa36 : SUCCESS in 30m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ca937c86fed4b88a4f0f4a8d119a3f1 : SUCCESS in 25m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304532e937604a8380d7db3db5a79f1e : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ddae235267d4aefa297b3a537bc8a3c : SUCCESS in 30m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b9813d26ab04af39bc06e21a16b66ac : SUCCESS in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/918be490a3d84f6d82ffc752d5bd8c8b : FAILURE in 1h 17m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7572184f6bb7470789b08055d824b1af : SUCCESS in 1h 14m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0e3a867881a748c8817c9c3ba6d1323a : SUCCESS in 56m 50s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/95749e16ecfa4879be4a651adeb12ba4 : SUCCESS in 1h 05m 25s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8e4e6b2e92c74f8198bb6b1e2f6043ee : SUCCESS in 58m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/21e2c21e9504423f862a44b209e64718 : SUCCESS in 1h 20m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5057c55508f44592bb7805224f56b89d : SUCCESS in 1h 22m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0ce29ec860f4a778446cd54d6498a7d : SUCCESS in 1h 01m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5e2e06bedca9432ab018c1b8c5b1ff6c : SUCCESS in 1h 14m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e0c2973b0bac42dfb75d33ffa04daf19 : SUCCESS in 1h 21m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a52e43d0483414abe2fc64e04bc958c : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":10},{"id":"e4177a76c12628e8a4ef4ebb36bb70699b93a0fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-17 06:36:54.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"842a981560159662fd1c75cfdf581c551d2573dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 07:27:10.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31bc7058c7474cdcb5efa590c0cd351e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65211389e6194389b657224938e3217e : SUCCESS in 48m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bbdc3fabed4243a19c7e017d4722c5a6 : SUCCESS in 47m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c007f03e19ff4b1db680b48d35e6ea32 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":11},{"id":"3ac8aa340e66fe9032127e684421675c13656934","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-17 07:58:10.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"093f153b3325b530dfe0c435707b809819d51c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 08:04:12.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dadf285f8c44a16a446a1ec99f337cc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5a3122bff7364754ab8c660b5df99f1a : SUCCESS in 1h 21m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/367ded45bcf442bda694c2dfd6276021 : SUCCESS in 1h 05m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1ec96646b4c4e4b874987a26768bf1b : SUCCESS in 1h 03m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/753ec2970c8f432cb5c1e147fbf3f0f4 : SUCCESS in 1h 02m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce0945f43566428d92e5e9a8ad539f1b : SUCCESS in 25m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e407b4bab674fadb10bbe4d2565ea7e : SUCCESS in 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f934590e0264c54b51dfbb87ac3dde2 : SUCCESS in 15m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/981aa75431944e2599f5090e9b503979 : SUCCESS in 24m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afad37feb83648f7a82ef02e31810dd6 : SUCCESS in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3a90879c0e483a9da3f8244408a223 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d86d4441f0044dc58788f22ac4ae7555 : SUCCESS in 29m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27b026645ffb4a51ab57a0e4de4aeb65 : SUCCESS in 45m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/19f39e9674c34cae8a75bf93371f22af : FAILURE in 1h 13m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff78953126a54e6aaaf80f93698d817f : SUCCESS in 50m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/71d88b1f398240ca8deb5bf5a48abc66 : SUCCESS in 53m 16s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aba5148fd3f546419b8b208e2acab816 : SUCCESS in 1h 01m 57s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b69df6cb0c134e21aa4958f42c5e7bd8 : SUCCESS in 52m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fceb66a3346d4bc4a87b3b55ec2258fa : SUCCESS in 1h 09m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a3db485c47e45679c9e263e22bf72fa : SUCCESS in 1h 08m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1fbacd6b2b704b3bafbbd0fc3171ca4b : SUCCESS in 1h 07m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a2c8945f17a946f3b7e85b48ca1b7dbf : SUCCESS in 52m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eee843341b064ee68c9117f3209416ec : SUCCESS in 1h 16m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10e7fc02aaec4ab9a40b90d5ca40e2eb : FAILURE in 55m 01s","accounts_in_message":[],"_revision_number":11},{"id":"7524ee491458dd64bc25e8dd0e9cf70a9779e89b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-22 08:47:50.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d0c767c1a9b480aeb8d161e71fdabe97d210c4d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-11-24 01:31:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bff2f648b460c3579c39d4496cf1bee880271405","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-24 02:07:16.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/67/860767/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"e0d50e44aa418669319973740fa4eccb8fa0a3ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 02:13:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e01296c159ca4870be30f448068ff3ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8a5fd82653f4cf2be9756cb2ec744c2 : SUCCESS in 35m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94fa128fcfb24a6eb3d56397f77025f0 : SUCCESS in 35m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c8c566f264b4f4bb23f2a454b9e057c : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":12},{"id":"572a30fa0438dbd852838e9123dbb63acbea56c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 03:11:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e4da04fbce84b95b5580f18fdca4fa0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c3f491da86014bdbbefa69301a1459e2 : SUCCESS in 1h 11m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/762ca2f03000408fb9fd7fdcf6bed499 : SUCCESS in 1h 13m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2736f23b3e6e4fb6b07763de669ababf : SUCCESS in 1h 08m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a491c172da5c4e638ab51b9d4e79eaeb : SUCCESS in 58m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/14d59eb61e2d48a1a92441aa8654b34a : SUCCESS in 31m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd6f4c778bc14355a379190e7a2da02d : SUCCESS in 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34322b5c83574a82a13f74e06fe4aa06 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6374e1aca46040c0aaee0d031bc30d24 : SUCCESS in 29m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/726cdc82f5954407affd39c7fec5ad58 : SUCCESS in 22m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a48d6dd002d46e0b0e87031290978fc : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0881e25a497443fa8cc6d3d5f8ca30a3 : SUCCESS in 32m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/74598212f781418e9049197f57875948 : SUCCESS in 50m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9766ec109b4643aaaa3256e7a9b654df : FAILURE in 1h 18m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ae1cd30e18c4b94b265cfebfe4a62ef : SUCCESS in 59m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2dd8aedbab304f928dc66bf7a3cff315 : SUCCESS in 1h 00m 15s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0d90ce2847a6454189f54a5a53d2d1d7 : SUCCESS in 1h 15m 54s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1576393f4f1d424c8cdaae1a0b7b7990 : SUCCESS in 1h 03m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d96fa0f885f41308ac5647f529b4306 : SUCCESS in 1h 32m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41d625da48004d01bd8dfb26d0be4766 : SUCCESS in 1h 23m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/986765f426bf41eabd650b268e3a5777 : SUCCESS in 1h 10m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2094f06c0bd54506ae16e3dad8cf389d : SUCCESS in 1h 34m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d10edaf57bf74478b9328862109dd5ba : SUCCESS in 1h 38m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e8b8c886f094433d86dd9883e2085e86 : SUCCESS in 1h 13m 19s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/83fca9a88f6743f49443e72463814426 : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":12},{"id":"467f27e77dc1e9ddcf4f832445a94333f3cc94a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-05 04:18:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0a0212253e3550a1b50ff13c0f5fbe43e03e7625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-05 04:20:47.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"6f15846ef55c320f97117f3ec7408de7b7fc0885","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-05 05:32:21.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"b46096c61a2f30846ed8a3ac75b05c0c5de59bd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 05:40:02.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c12165cbb0e7468d9dc7b1e59b61b633\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b15ff18ed9d4f099bab5ca5bb190846 : SUCCESS in 50m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0e245bf7d4d42bd9abac0e61520d234 : SUCCESS in 51m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/403dd9bfdce6481ca4beb97e85240d06 : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":14},{"id":"596bbfe108a99b39eec50ef6dbbcabcff32c6103","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-05 06:05:22.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"97cd7498f8b88f95d787b5300ac7507c6f0a2167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 06:17:51.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f6f0dc2aeb49929b54386fa994f85a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b1802b66ece24438816c93be2afc12cb : SUCCESS in 1h 00m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bcd8c00d0dec4bba84d99e16c40c0c05 : SUCCESS in 1h 23m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9933b8fa5074cdf9b8ddcf032644903 : SUCCESS in 54m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c0168e320ba044768c3047a48f891732 : SUCCESS in 1h 18m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3ea0847b16e47c5b916327b89b1249e : SUCCESS in 28m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3c7acde00e746d7a589855412a5be68 : SUCCESS in 33m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da680ef1d28a400c86a153909bba6b4f : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61b8a9e8ed024d518691b54cca94a192 : SUCCESS in 28m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/845420ddd971420ba3a4e5a1faa7eb29 : SUCCESS in 29m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0bcac0e59c24a86a1f2e76b8771a570 : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88563119a79e44e395c34d4f912208fc : SUCCESS in 26m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89aa0b60dc0d41eb88571826e6cd53c0 : SUCCESS in 52m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e14d455be524f788c83bccea84fa6ad : FAILURE in 1h 36m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d87c3a59a97c4e08ae6a2e4e7c122533 : SUCCESS in 47m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/96cdaac69f434527ac45555e0e80470d : SUCCESS in 50m 45s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7bc20c1536624e94a319e69769a1b52e : SUCCESS in 1h 08m 44s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82599aa6c2e2448fb36380c7406f4363 : SUCCESS in 1h 01m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f4d6f9bcdcb34eac91e99f0d83d971bd : SUCCESS in 1h 08m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9aa4bd4508c74f9f850cdfa13e0724e2 : SUCCESS in 55m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ad349e4b799e4713a99045e249226abf : SUCCESS in 1h 14m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a68ef617ac724c908c152d1a454ea47f : SUCCESS in 1h 32m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/304eb9bd5582466f9231b09decb8f6b4 : SUCCESS in 1h 52m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8aefd6ba441c4e62bea050929733dc32 : SUCCESS in 1h 31m 47s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/48e22de4559f41b4aa2cb616fe6339ae : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":14},{"id":"32fafe607f26d92b7b328e9f84421d05d1f060a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-05 11:49:01.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"14f14ba533396310fe21ecfff42e08b4f588d2a7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-05 12:27:05.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"22756955b1a744e7b752f1c28d3c835f4f37f550","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 12:46:13.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6250d331f7114f378339940a8f096756\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b35ad8b7ef041228a523234a0ffc4d4 : SUCCESS in 54m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03b6ad3735f541e490a6b065ad202718 : SUCCESS in 54m 15s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6f8b02d3a11917cba4e6960837519d3b17b43170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 13:58:23.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/049a86133293469a92746f719943c18b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f10249d2a9f946588f58e4b17648177d : SUCCESS in 1h 28m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9746517491fd421fbc0d1667ff77627a : SUCCESS in 1h 16m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4cf532f5f0fa43da9cfcdcf851f1b1d0 : SUCCESS in 1h 10m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/66338cd785684ece8f99810c2ac5d660 : SUCCESS in 1h 10m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/67d8ce19fe2540ed8035314cfeea3064 : SUCCESS in 29m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dfc41b96eb946858f92a9bc980f60f1 : SUCCESS in 40m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9a5d763c47422faa64609528892497 : SUCCESS in 13m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/062f040784f049ad9eb23b15cb7f8268 : SUCCESS in 24m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a319e56f020e4ac1a3bd8e1a59b318bb : SUCCESS in 22m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c85abad0b747dcbb16d2e7053606aa : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccf757d5a7794e0da31a5dd6a537d8d2 : SUCCESS in 54m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8498a26ee5d64917bb432412abdc76b2 : FAILURE in 1h 30m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8520aebc0b12439c83641175e44fc25a : SUCCESS in 1h 04m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e0ecdb7271744d8bb1f2e955628cf1c7 : SUCCESS in 44m 47s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f1be94fcf2984b4f90476c4dd8e9b59f : SUCCESS in 1h 20m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6fd110857eea41868aadd210ad2c7389 : SUCCESS in 1h 04m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/37a273205acf49a7a4cccaa0d117db64 : SUCCESS in 1h 02m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d18469d6f37245f1a75eb0e5da95244b : SUCCESS in 1h 28m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ec58cdac9aa647ea8c885787d1bad12e : SUCCESS in 1h 15m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/684c9bf527064d76922581a0f2890ee5 : SUCCESS in 1h 57m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed316dccf41c4a49a5d0b23dcba81abc : SUCCESS in 2h 01m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d038e25738f04e158a69e2276ed2e33e : SUCCESS in 1h 19m 50s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/cafa68f78bf14e59a646a0a3036cbfd9 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":15},{"id":"d45eebde19f5642b2c2727977239a1c40e94996f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-06 01:14:22.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c63b7c63b3c88b2758db0a99ead161297e6a7440","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-06 01:52:21.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"6632773f0a652da9551ec91aeb17e08f3fe9f060","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 02:02:55.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/000de4833661445fae23e36854c9eb42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c151436e71b04cbf9d51a594d70210e0 : SUCCESS in 46m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f2c62a578764f048ea72501368891af : SUCCESS in 46m 07s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"332958ba3176221166580d59fe45e074ddaa6c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 02:57:03.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8753cc2a7a65427899f6d186c867419d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f4a214eb84714597a59dfa3af825702b : SUCCESS in 1h 29m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c5fca42ed5cf4fe6844eae3e3127aada : SUCCESS in 1h 16m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a4eaa1012af647719dcd48cca6f71d9e : SUCCESS in 1h 25m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6daa38839444a1ca8168d0fa8380360 : SUCCESS in 1h 08m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d39622cc36b64f10b662d358dee89f60 : SUCCESS in 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6448e2b46bf4f5196f65dc90d7ea21f : SUCCESS in 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0f136c5c354e88b5a4535beb688581 : SUCCESS in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf0b31f0c478475f8e6896a7b51ad2b1 : SUCCESS in 24m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d94b5ef6952470b821ef76cfcec0b81 : SUCCESS in 22m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebb1d8d8bca4e5e829bd1099c29c0af : SUCCESS in 10m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/339d106691b149f9a66b25ee1a25c492 : SUCCESS in 42m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/17e8c58497dc4403843b5b70dbd8c825 : SUCCESS in 1h 25m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b931c817e48b4c2d82ee8bf6ae4bee95 : SUCCESS in 48m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1d56336fa8674f879d020d8cbcdba765 : SUCCESS in 53m 11s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d2ed3c21cb7a4de6954ae59eaa3fb3ec : SUCCESS in 59m 54s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2acb4390e1d1456aacadda86893b0624 : SUCCESS in 50m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1f4d80a57dc542c5824aaecadbcfb540 : SUCCESS in 1h 10m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/49742533ed344c078e278733485b0e34 : SUCCESS in 1h 19m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f086e99b49614f71a27843b66230924a : SUCCESS in 39m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/69ec58a98f3d488eb0296c979361afec : SUCCESS in 1h 20m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f69065f1ba84c428a8dbd52984258e2 : SUCCESS in 1h 40m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9567510096fa482ca0a2568d677f1b47 : SUCCESS in 1h 02m 59s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/807efcf7dadf4fa09bd84a3beb5c7745 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":16},{"id":"89a79ee546123507fb992d9d10185f35f63b44c5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-12-07 14:21:00.000000000","message":"Patch Set 16: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"d3ae3747357bb0b3317837969d41f03c8deb8ad2","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-14 03:07:19.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f51cd1f4abfa7b6f863249c58e45169a7f53e10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-14 03:07:51.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a8c2ce80389bdcdc980629ebaf6251db3f0e9f75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35618,"name":"zhihua fu","display_name":"zhihua fu","email":"jxfu@163.com","username":"qingshanyy"},"date":"2022-12-14 03:09:25.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"ac5d2021bc7653f0bbaaa15f73c1599626bd6a34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35618,"name":"zhihua fu","display_name":"zhihua fu","email":"jxfu@163.com","username":"qingshanyy"},"date":"2022-12-14 03:09:51.000000000","message":"Patch Set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"4498e4f9370561e6cac2f11e1e41f96887ee57c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 04:38:29.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c267ddefcd4fabaa6eb0c875602eca\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5204e1b4fea42f2bedd915ca01552cf : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6053500371e647a78b45a5338b39d5b8 : SUCCESS in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d4dc84cf67a0d256c1c5bbda44cf91656ff9875b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 04:50:19.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f37ce620f3804ceda6531c18249ddf5d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c4986d301653435da1911547a9784c3a : SUCCESS in 1h 20m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/deec424318c1462e9b69943369952f38 : SUCCESS in 1h 14m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8c0969cac10c40d4b2ce2f87079b0a2c : SUCCESS in 1h 01m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/90ba20e36cd0403d881f8d034b40db39 : SUCCESS in 1h 16m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1df75bdd9e88407796e0bcda41b38577 : SUCCESS in 29m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9bf3fa9e11c408495f5154e0792813f : SUCCESS in 35m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0eea55b01744128916274e24f0b8e8 : SUCCESS in 12m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bda5f8a55b98456881476422b1165d25 : SUCCESS in 26m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26d2de1e1b374898bcb0c7e5fde86454 : SUCCESS in 25m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b2f6ff185d4d51b408ca1fe60a4ddb : SUCCESS in 10m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c50447adbeb43edb21609ad01411eda : SUCCESS in 43m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a651e66783ac44d1b4534a493defb11c : SUCCESS in 1h 38m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/749a8e78b7814de6b688670c228b1948 : SUCCESS in 59m 33s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/16f2b0c093fe4a04b748d00b0442c8e4 : SUCCESS in 1h 07m 47s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0cd10b038c4d4c4ab97a7f38515951c2 : SUCCESS in 57m 21s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/93874b5afbaf46488061fc4a4ec17848 : FAILURE in 28m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/83ee6d9fd4484df587ed4ea504a88594 : SUCCESS in 1h 22m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a135b55898ec4940a10d03bba5545690 : FAILURE in 1h 05m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/34dd5a124f234df396bc4ccbce03cd42 : SUCCESS in 53m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/72faa3d8c55e4960a7142413ee1259e4 : SUCCESS in 1h 35m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/16b4a44dd5f042bbbed843b72d7d7616 : SUCCESS in 1h 24m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6abc5e37e0b34f8daefb70c47c180652 : SUCCESS in 1h 20m 58s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bd3eb83ec9ac45c8a996bd302f7e10a9 : SUCCESS in 1h 09m 37s","accounts_in_message":[],"_revision_number":19},{"id":"c8e87e26de494c7ea0320f68d6d2571e2613e808","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-14 04:56:28.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"a072f59cbc938632de6c10971f730532f4c692c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-14 06:11:27.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"2f5aed61901213aa121b5a39aced471c96774b3b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-14 06:33:10.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"35ff5328120527377e6f296d848f4f26771830d8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-14 06:49:10.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"7fedeeff99952ff6cfb78b7e00b7a49465f52ddf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 07:17:47.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8a5c123a7d4ac28b96e5e4eed60675\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b514bb7b33c4f6d981ee348d0b0e493 : SUCCESS in 42m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3c0e95f30724139935a59285cd35796 : SUCCESS in 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"37d63fbcf91c27cdc2cb713a79711e8c998c8558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 08:31:47.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54aeab97eac74a1185a2f7decea2f72a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d5ed9d3869de4a1f960ab6ef37da6608 : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/43cf5979111a4d07a0a859b4242dd3db : SUCCESS in 1h 30m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33f9d41b57ca4da8be0bac2b6dd5f7c1 : SUCCESS in 1h 29m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a91829111b84b5e8d9c9953dbe12219 : SUCCESS in 1h 23m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c598a88a4d940aea7ec7cd06be16a8a : SUCCESS in 36m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9434685c12684136bb7f846344786ce1 : SUCCESS in 30m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a503c6a6acb441ee8818e0db0f236c36 : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c758b47831d94961b3023e19e74c8f37 : SUCCESS in 21m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49317f51c7964752a7ea176198267c50 : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af204bbd6314d12a96117746d97e608 : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1b7690b512d4ad092c0c1a43d7f0852 : SUCCESS in 42m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6df270fb7a946cdb97fee41d5d25dd6 : SUCCESS in 1h 34m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3170a06e15ca4b999b85e93f2e354555 : SUCCESS in 55m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/74f65b89c8484272a411ad57598c0cfb : SUCCESS in 57m 49s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f3771c70695f4858979c91a3ce65291f : SUCCESS in 1h 19m 30s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4b603f515a814192a2a604c2541d10a0 : FAILURE in 40m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4dfb7e43c676478cb76950f4e3924379 : SUCCESS in 1h 15m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a563cd4fa45f4ceca1b58016e04e4a0f : SUCCESS in 1h 31m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c4c953982bd3487182f06cf022650624 : SUCCESS in 1h 08m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc24619dda41493181a7034ddfbb0b18 : SUCCESS in 1h 50m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/13d96bdfc5be4432af26e8575972e502 : SUCCESS in 1h 38m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77ce2f9d36a64309b89a39ff775703b9 : SUCCESS in 1h 11m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/eed18f185152404b9d31b0db9f44862c : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":19},{"id":"4f8cbd990bb583d0c5c460babf2707c7f9512413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-16 00:55:24.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a979ea8a9b63990770981584a555415e2e084a80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 02:07:55.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a016626be04e40b7a556e6080c21684c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31bfcae23f024edf99e578ab6aa254e2 : SUCCESS in 1h 10m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70efa969652b48d9a200aaa1dde3a8db : FAILURE in 1h 09m 58s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e744d6ee3d155c2bcd69b9e44a65201f3480c1c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-16 02:11:32.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"ab83b1e0b880ec31e94c0ce2275a253dc9e742d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 03:59:38.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e187caa850b457f87a64952ad94452a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e248a4f6f10644d3b9f7599dcd31e5d8 : TIMED_OUT in 2h 49m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e6c31560e18b4e149c022c4cb0b77fb0 : TIMED_OUT in 2h 49m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9bd8af7948bc448a88dccd1aeda823a5 : SUCCESS in 1h 09m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b724155cc66d438b8be1324f0ee7469f : TIMED_OUT in 3h 02m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/50deef2c76564a7e973e98343e5f9d16 : SUCCESS in 36m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e3c0c747723473da87dddc1b5575bad : SUCCESS in 33m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b875bec90c44cbbf2e6b5706ae7713 : SUCCESS in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10345c5a391e499b93edf15c718d549b : SUCCESS in 23m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e5dd8f5884b4d8d9907fefb40e60ddd : SUCCESS in 24m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f2be6817f94a4794fb4d6beabcdcb3 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c95dc19f2d741418f390a6eab744d67 : SUCCESS in 46m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d896a67507548389429f9ecc310e8f5 : FAILURE in 1h 56m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f274fe30ead4c99b7aa96cca765c80b : SUCCESS in 55m 22s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6da2be11cc8742e987fdbaff4903f6ee : SUCCESS in 40m 41s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f1ed547fd36f4c4280b8e8ca92493df7 : SUCCESS in 1h 11m 42s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6cc6ac0576f475db67df2be8961f76e : SUCCESS in 52m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c3f5ec122007452b98ff9e7fb59d903d : SUCCESS in 1h 23m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c088650e31ba469a8dbf1a30e15c4468 : SUCCESS in 1h 18m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5f79f97d33ef46cf8a0bbc62ff462179 : SUCCESS in 50m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f98dd4702f0d41fe870c1f5543e7396a : SUCCESS in 1h 41m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed2e9b22bb684fd9920068b4e0732ae0 : SUCCESS in 1h 47m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1810adec14d24e4d918af1c4ecfe8688 : SUCCESS in 1h 30m 47s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/beee5ca9761a465194e91566653c90cb : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":20},{"id":"b66004fd730bf99c30bb9e0af39088b9c081cefa","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-16 07:42:55.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"cadbc0bb13093c60090387914ce6fdfeaf8c9e6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 08:36:58.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d370338d94e643a68e7ff699902ff4c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b1825179bc946c2acae5bce64e4086b : SUCCESS in 51m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fdbcdb55142242cc965711f143a7f7aa : SUCCESS in 52m 18s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"9ccb1d69c8b28cd6bf015ea4ac21ff807b57ef82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 09:50:19.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/811a46cd84af4df391169e66094329ea\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5479b8641d7444e8ae7d21ff138f73ab : SUCCESS in 1h 43m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7706f588827a496784aecc665963ea4d : SUCCESS in 1h 45m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8bc1a6538fea48368f6d58650195b525 : SUCCESS in 1h 34m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3fcf9d6c62084678ac2de762c02c39d9 : SUCCESS in 1h 42m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be59af3d11c84702bcb762db014b98f2 : SUCCESS in 37m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d212fd040c249d1a7ef5bb3d29e8db8 : SUCCESS in 35m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0749c95b5cce41ef914fddcbf1ff176f : SUCCESS in 14m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/080360e6b70142a89945157f956d3747 : SUCCESS in 26m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efa71b441e374b2fb0f5e7b830fae410 : SUCCESS in 26m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83fc6f9316394226a850cab771f2671b : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a74d52762c374ed49dabc95df12aefbc : SUCCESS in 1h 06m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f428f9bf6594071aef417c5034970e8 : SUCCESS in 1h 51m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd0d435708d74fac8ffe6be0cc336be4 : SUCCESS in 57m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a61c2aa7146e46d1adc96941c239d1fa : SUCCESS in 59m 27s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1ac6e928c0314dc6a4bb43648158aeb8 : SUCCESS in 53m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6ffc073c770146c489aecd730f9f16df : SUCCESS in 1h 03m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c633cd9af2e642ae8502d5685a86fcd7 : SUCCESS in 1h 35m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1b8109c3dec24b2298f618bb8b859ee7 : SUCCESS in 1h 06m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4afa53d312024fab8a16b1381a28c23d : SUCCESS in 1h 00m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c3f5640d5a12449082cb65714cb586bc : SUCCESS in 1h 38m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2309ee7cb9e8423fb95f20539526f7cf : SUCCESS in 1h 26m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aeedcac7e6334edea541da0c8286cf87 : SUCCESS in 1h 20m 15s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ed0160f707894d2bb4d51677d87cf5fb : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":20},{"id":"ff0fc162c20f5c50b7dfa51152edfa4753b6dc26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-21 02:05:05.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"e6dded7db9858c765719316f95067d3cae7643b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-21 02:59:05.000000000","message":"Patch Set 21:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/67/860767/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"e69c607e9b11225a264fdae612f1f991f4c2bbd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 04:04:26.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/113ebc9ac55e438aa678df7a9e107cdf\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8736929bf0464360aa78391552cd687a : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d8f9660a836841fa997744000fb2c4fd : SUCCESS in 1h 06m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dbd93c09596748a988288e0623fd6faf : SUCCESS in 1h 00m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a11e4fde40984b379b87bd6a7af1e87d : SUCCESS in 1h 06m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/884678d2e070448e8ea58c5a25ca0488 : SUCCESS in 26m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60cd297a493746c988d5bcc378e916ca : SUCCESS in 36m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812357e9ddb84f979b0f68d2d78ad611 : SUCCESS in 12m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bcd96043af2478e80ceb6e32f59a370 : SUCCESS in 28m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6ada6df19a14611a1e2f19a82c7d729 : SUCCESS in 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf6c1347d65498ca1374b0f6ff392c0 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4525ac0667d24345a30b6469417328e2 : SUCCESS in 56m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8a0a2b419164320bdec43247c06cdda : SUCCESS in 1h 41m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/88f28e2210c74f8e8f842e32139fe57c : SUCCESS in 1h 11m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b0b158c035cb47d3a2455264ddf6b265 : SUCCESS in 59m 58s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/755a0e9bd1b646ebb5de42d5c98d5a78 : SUCCESS in 1h 17m 33s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/20f6b33b762a4987ab9c6f2ef6326df4 : SUCCESS in 1h 02m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ed39e8c4e754b96912208ab27f40d4b : SUCCESS in 1h 27m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4895dadc69e946ffac07d937f2756134 : SUCCESS in 1h 25m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9e6ccdd5583d4b5e9bd33adadc2621ef : SUCCESS in 1h 08m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6a479aa60dd74cda8dbc9e2e7cd78b85 : SUCCESS in 1h 41m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/836138844a5e4490bca2d181579e6191 : SUCCESS in 1h 32m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a32c488d36a74a1aaf981dd9f92c986d : SUCCESS in 1h 25m 07s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2ebcc064eedb48eb85a6b8163f574141 : SUCCESS in 1h 11m 54s","accounts_in_message":[],"_revision_number":21},{"id":"6ebf192e0a44e31b239da8caa388e519e509c78a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 04:13:09.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4bf341ff55643dea8df76244f57f86a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/308664b2cbad40b7bdd03cde04a1ea70 : SUCCESS in 1h 00m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca01a629133541e8bb41a6a062b7baa9 : SUCCESS in 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ebaf46f53c1d580afe9e1dfba31ddf6befea90f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-03-09 01:34:13.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"bc20b6a6cf34a0754ba299cf58fa086f07777550","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 02:38:18.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c65d3fe6f5c44f8ba9630338fa7b981e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e4604b985d54adbb03e35d67749ed46 : SUCCESS in 1h 00m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a348209c883c47a5aff0ff550aaed1bc : SUCCESS in 1h 00m 04s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"628f77e25cff4d66bf9e9fa732670606a8ffdd4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 03:50:56.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20c605fcaed475193e0135bc6fc5b9f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/89682a128f8441cdb2b43f718feb548a : SUCCESS in 1h 46m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/af35c624c682443f84f7a812004454f1 : SUCCESS in 1h 29m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/71818c45fb044ae381b4cb1ad7420598 : SUCCESS in 2h 13m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aab9bb4d5ecd47aaa1e64d6e0dbce378 : SUCCESS in 1h 46m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aa5992da8b644759a44c9b3e556fbb41 : SUCCESS in 1h 45m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44b1784f640747fe9e1f369ac7528545 : SUCCESS in 35m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da4d01ef872e497fbb11000747651d21 : SUCCESS in 39m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8afa47b30914813a34fd8eeb9b604fa : SUCCESS in 12m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77149714119842b182d57ea962f9a983 : SUCCESS in 23m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01c71d1cbc0f4f00a8c857750e956b89 : SUCCESS in 24m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd16a453aaa4beaa58519d9097b909f : SUCCESS in 10m 54s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b6bbe75322fb47a0aa2faeae4eaabdab : SUCCESS in 24m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e73fc0399ca84404b163e211a82b3967 : SUCCESS in 1h 28m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdcdf4b83ad4494fbe39e738fc907eed : SUCCESS in 1h 46m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e792c0d2d0e40ec98dc1d5e170bfb64 : SUCCESS in 1h 42m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/958c7d988df046c1807cf154f3d04a2b : SUCCESS in 1h 05m 15s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/498bcd7408444145a0b834094f3fb64f : SUCCESS in 1h 16m 58s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/27f467580bc94e00a4c62d2a2987b120 : SUCCESS in 1h 24m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/77fb2beb0244484eba56b0cf85f83239 : SUCCESS in 1h 20m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/545fbc995f0f4bf1892ec0c91647178f : SUCCESS in 1h 29m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/db0485358af244da93887105bf0cdbce : SUCCESS in 1h 23m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2ae02ac2451d4b4cb55a7b5f16e7ec1e : SUCCESS in 1h 41m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cfd958c29e6a4589b692ccd6e968a679 : SUCCESS in 2h 01m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ace529f89ec144e2a05729d5a8656e82 : SUCCESS in 1h 20m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/9a7ca29104894844a4f0e877c51419aa : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":22},{"id":"9cb3037f1f63128ec4a0ec512a556c5ae5c8c4d3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-09 05:52:44.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/67/860767/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"b8abf08a82a8e8f581f9a1160e9c0d33d66e9bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-04-11 01:52:51.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"558786504437536a93c27a002681dd2cfb3f04b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 02:51:57.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43d9eb56e7b241b79911be3f18c19667\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48a308c755be48afa6b128a1f8aa5e69 : SUCCESS in 55m 03s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ee79da6993643cbbc1a89d4e5bc81d5 : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ee726aec0f6461a84e47a8c0783705f : SUCCESS in 56m 37s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"9e8e54c12b79795f393a981bd26117cfa54d98d0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-11 03:25:40.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/67/860767/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"c928240af7f85ca52dd270ec64c4501823261a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 04:37:34.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01c2cfb286684a5eab9e1ebf7e63adef\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4b273f3817ca47b092cbea70baefd38c : SUCCESS in 1h 50m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/50a636d8bca34b668a170d3e78f5c263 : SUCCESS in 1h 06m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b5511e5fa02544b6adf6d6b2744e7186 : SUCCESS in 1h 38m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0569a1b29982431aa6f90a1cdd22c491 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cbee74d0210f4d6599db7113857adb92 : SUCCESS in 1h 35m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d49de6189fc48088435ef145d7caa70 : SUCCESS in 35m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73182b2597a24d84a4c972f8fbdd07a8 : SUCCESS in 48m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7dd0d96d6e44bcb8b102112db0bb5b : SUCCESS in 15m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f354b85dcae246c59778f1d36636edb5 : SUCCESS in 29m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e24fc754f3e14ac3820b3f47d21f6734 : SUCCESS in 29m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50e6721312ed462a97f281e36137d5e7 : SUCCESS in 35m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b86fb123d2c647789ca24f101ce4c511 : SUCCESS in 11m 22s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8da4dab60b9b41a98fb9391a1bc0fced : SUCCESS in 31m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1dd9c5aa3e864e71874f18cafb079731 : SUCCESS in 50m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a95494094378404191f6a5792f101988 : SUCCESS in 1h 40m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e301e9898096479f8cb988e56ea346cb : SUCCESS in 56m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4c3e2998f4e2453f9ac4bd612468521c : SUCCESS in 53m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/36a354e340c147aaafbf95ad5dfea8b3 : SUCCESS in 1h 24m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7450faca64ce4b2caa3c48f058b235a0 : SUCCESS in 1h 37m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e99397f49d074cb18c131b7cde341b80 : SUCCESS in 1h 31m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7414ebb3751449578703233a6b5e0173 : SUCCESS in 1h 31m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b5503746957d4717bff4aaa1dd99ea69 : SUCCESS in 1h 51m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c1f78f4ea74484287aec267924e61ec : SUCCESS in 1h 26m 34s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e1571e52120e438b834e55d9f335892e : SUCCESS in 1h 12m 47s","accounts_in_message":[],"_revision_number":23},{"id":"5f2810b1fd2b21b6cbf7f2bfb6871d0d8232640b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-06-01 07:14:38.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"4e6a23b2f58d0eefa153a4d1fcec07bde7adc9ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 08:13:25.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c760ffd2ca0245e8bff1f7cfab85ac55\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c308eb91c5c4e46b7d0a1d272520f91 : SUCCESS in 55m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bcb4d033cbc43b098a211d5d69efb65 : SUCCESS in 56m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d45872a18dc48c486edd7cea6d0aeff : SUCCESS in 56m 02s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"8b71350d1b63876dd8b0d2b2af80df41fd776813","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-01 08:35:25.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/67/860767/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"f7a0d51ae807b0b52f7b3f6bc755f2973ed352a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 09:46:54.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af7141ff2d28465e8ec000c7d071c12c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b94932e544314a54ae8dfafe59cdeab0 : SUCCESS in 2h 01m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2cf2e036c4e649beb3709f899c7bbe4c : SUCCESS in 2h 08m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3efb9eb1973541b6be041e74dde509ed : SUCCESS in 2h 24m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/34ddba0c4782466c96a176809fa2941e : SUCCESS in 1h 38m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c0a3bd85ee5c4f7288ccd1b0a78657cc : SUCCESS in 1h 50m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e72a9114da57444ea6e535b82d6e33e5 : SUCCESS in 35m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bfa885bbff747559ef9032bca19be5c : SUCCESS in 42m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b00b1703334afb99214c3623805ac6 : SUCCESS in 9m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d13bd7b6589f4bdc87cf7f2ee1db3590 : SUCCESS in 28m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cc1fe211e574c78809801c5ede8259b : SUCCESS in 31m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd54b4ff7d5b44789903aa25734850a8 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3215124cc9c44e1f85dbc6444836b092 : SUCCESS in 10m 07s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7963f7c2b46741ca9486076ce7fe2a24 : SUCCESS in 26m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ebbe19339b947e495342634fe5141f3 : SUCCESS in 59m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ce75b1085b44cccbbc2cdc330a83f2e : SUCCESS in 2h 30m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e605f3b954f41f499bca4223f74cb1c : SUCCESS in 51m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f38d5dbc368b43a7acdd4dd1867bc7fb : SUCCESS in 55m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2d01ab05825d48d79dcf1562b607c797 : SUCCESS in 1h 44m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ccef402344e640fd9b336c9f378491ee : SUCCESS in 1h 29m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/290eb7c003e54e16bfe4e53f31b75164 : SUCCESS in 1h 00m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bc94cf947f674acfa5c1ed9b1c4b64f2 : SUCCESS in 1h 38m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c5d86843c3534fb89de3078b68485726 : SUCCESS in 1h 48m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ee0f3c92005b4ae9b5e1bde475244a4a : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":24},{"id":"613735991c449711009c2e45699e41a6dd53aac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-07-05 09:51:06.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"d2e5b7f9bf2b59efadec30499add2f8fa1d27908","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 10:17:43.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/543da3774c214ad483779c81945f49e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/271b4b4274ea4437b15373b639b22b22 : SUCCESS in 23m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c83fa58dd2014252bd0d3277df9e16b5 : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93b8f921e1ee42b3b9011126a21b1cb4 : SUCCESS in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"398398ed1947ec7c60b112b67f2b1c6749ffc9c0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-05 11:10:42.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/67/860767/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"ba1a35cc8a9fd35931480e9d5a7f04f3ec46769e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 11:52:31.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e52c36f972e40b1888414c1921e820a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/29985d30854c4aaf8bc34321ff42d5ac : SUCCESS in 1h 24m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1e811b84dbd4273a138f419fc253c85 : SUCCESS in 1h 27m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1568f6eb2e624c2897fece85defc3b21 : SUCCESS in 1h 09m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a95197be96634b51a402ed4ceb12d78c : SUCCESS in 1h 26m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31477e9e80b44f7c9da52af68dcab167 : SUCCESS in 32m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f26918ae7e294d05b145f97cdf0ae75e : SUCCESS in 47m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24264f2308024573acfd068f617a0b7b : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0957ef1ca2ce49a29c36c21b67e26018 : SUCCESS in 31m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e70d9dd59aa4694ad6cde269a07eb13 : SUCCESS in 34m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/720cbc3488f14e598045c063d8869227 : SUCCESS in 27m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f2257f2144407fa134381d0177c394 : SUCCESS in 11m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1f9f2240dd4747d489cc0400a150e36e : SUCCESS in 25m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00c3f61c9138434db2133f034b1fa49d : SUCCESS in 57m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4b275f70c0c49b48c8eddad15fc8485 : SUCCESS in 1h 59m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b3aef7bb8144205aff19d19198cfbc3 : SUCCESS in 1h 10m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/de301a1e3dd949beb0ac2cee57ffc726 : SUCCESS in 52m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f5ae74f32a7d488b9b1b339cb32eca24 : SUCCESS in 1h 24m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/221670a5c05b4479824faf9941704234 : SUCCESS in 1h 31m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/864634545ada4ece8372133461ce8c4f : SUCCESS in 1h 14m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1d493aa773a445808402a628a2589f28 : SUCCESS in 1h 34m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/17016b93102349e0b839e9cc782b10f0 : SUCCESS in 1h 37m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c2f366af617f41899fa5501703078595 : SUCCESS in 1h 46m 54s","accounts_in_message":[],"_revision_number":25},{"id":"c5535fe5c118c2558038060db27ae61fb56b62e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-07-11 00:45:21.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"b1ee2274e6b1b6b2d31c046bd6a6281369037a81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 01:17:50.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a84f91b7c04d7a80455f62e6b5952d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08d1f69e46c94fd7b20dc24a158d4075 : SUCCESS in 25m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad7e71051e6444998ae270ef3ad1ed9f : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ece65d503293425eaae2dd02ba3a2bcd : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"1e56ab05f7b4ef27404456742ff40856d770e609","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-11 01:26:10.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/67/860767/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"6bfd958e2e0cc2cec0930ec6c28651082baaddc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 02:56:50.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8360128e0837417f87c71be27544e225\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9320923f1ed142f39b117e83e676b7d2 : SUCCESS in 1h 43m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/590d5236a96941eeac38ee7c819fafe5 : SUCCESS in 1h 53m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd53e96fee4f48b1a81d91b8af7096e9 : SUCCESS in 1h 24m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/05fba73a3e5640cd9cec09e3f10a8182 : SUCCESS in 2h 09m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/299dd9ccacf64c65b90b95a7d5c609c0 : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72b4621c7dc1469d80d97ecb4987c8c1 : SUCCESS in 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c216e6c957bc4dc49621d0fecda69871 : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96bd2053964745bd9a7d165a4a9baa1f : SUCCESS in 27m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a59c22ab4034def9bb939610aa31aca : SUCCESS in 28m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13e0e08ccdfa4f8590f125631430b81d : SUCCESS in 28m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f794adc3e7e4fa782bdd8f69463c8a2 : SUCCESS in 11m 11s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9f8dc10dfaba462f960a6b50f930d838 : SUCCESS in 23m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbea25b1081c4933a37b71fff34de20f : SUCCESS in 1h 00m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b076d8ded3084186a2c08be4bde010e0 : SUCCESS in 1h 49m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c81625b28554479b7ef40aebaa9875a : SUCCESS in 56m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4af55cb350ba48beb5c1d4676f7cf2e3 : SUCCESS in 58m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ace136574344b9eb8a5fd9e3eb75143 : SUCCESS in 1h 25m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8fde4fd0278c4820979ad18bd92ffdfe : SUCCESS in 1h 24m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a872c0d568d34e6eacd41c2aa67d0c76 : SUCCESS in 53m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9084988ef8e447eda44f0de5baacbee6 : SUCCESS in 1h 51m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c59e843cdd444fcdafbefb777221be9a : SUCCESS in 1h 26m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ea157a636dd1480eb790c1a57eca51c7 : SUCCESS in 1h 13m 45s","accounts_in_message":[],"_revision_number":26},{"id":"1bdc5667fba21993670d5bacd03009672175010e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-12-05 08:30:20.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"2eaed59f8cdf4e892909f447ec0adf0e22ea5e13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 09:10:39.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb0ab18ef80146068478a9e9c81579af\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd58abed1357462abb4a673997605d53 : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30a5cb99439b4217b03eca66c7a23b62 : SUCCESS in 38m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/838bbecc2f6144918bbb688909a2d6f9 : SUCCESS in 29m 14s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d66c995b1f66bf4ab81cf6ce5db93c6d7befef7f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-05 10:16:42.000000000","message":"Patch Set 27: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/67/860767/27 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":27},{"id":"8f1eaee96d5ee13ab4cbb306e50df07a0ed829a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 10:53:35.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/576c0fb0cf5c4060b1365d2f3116528e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/023ee18f371f418eb53fb1383fc4fb3e : SUCCESS in 1h 47m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/393e8728fa7d4d09b27a2be7f340dbef : FAILURE in 14m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/98d6a8fadb5d4ea9af0aedeb5c9d411e : SUCCESS in 1h 28m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ba601ae17d24a11b315627813b2a8c6 : SUCCESS in 1h 28m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89eb9ad009ed4ddf80da740826ee1841 : SUCCESS in 44m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fa7051717944323a848718cae2c1500 : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bfb8333450647978997a83db9b93914 : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00a39aff70404c008218079299337e0f : SUCCESS in 27m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18e3ed952fb74576ba8d4cb1da704a71 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17345beb523d43dea9b4b74d7728a1f2 : SUCCESS in 10m 26s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9cda77202d28483383a5487b09e7f7c0 : SUCCESS in 28m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e92ff0d9d85345fdb219fb2891a7ca61 : SUCCESS in 1h 09m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/98655eabccca4fdf9a9333e6d92723f1 : SUCCESS in 1h 07m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e341cac83c0241afb782cc81c059f7a8 : SUCCESS in 50m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d5b3b7f853d4aac810f5fe1adb4aba1 : SUCCESS in 1h 35m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0cd57b59db743309dbd13a408a3499b : SUCCESS in 1h 04m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9560f85e3a4a4b97a85a13e3f3e447de : SUCCESS in 51m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1d0468fbc46d4905ae81185b2fa4b35c : SUCCESS in 1h 30m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d9d43fa937f748bf93a2d3bf8cc5e58c : SUCCESS in 1h 21m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5b0f8b8dcd446ab82ce57e686b0e988 : SUCCESS in 1h 07m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a5523bf2c44b40aaa9df166ac95898e0 : SUCCESS in 1h 48m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9984c80efae84e4092b00fd7ad2bb88b : SUCCESS in 2h 16m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e8a70b0c0663427a8f5a7302af16089d : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":27},{"id":"3281bdbf9dbc50015b60893481aae66ebc044ca2","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-12-06 00:51:40.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"6d3a8a49cc852397569906591eb0c2a87508bc0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 01:33:57.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8542723580934443afdadafdf8261608\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c246c19cfd274fcebb64a01d40e7db72 : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/345c9ef1e9124abfbb8434d72fef30a6 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa549caacf094c0c9d707bde29e1b3a2 : SUCCESS in 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"fdf5a976e5de2c1c18f8fa8ad9ddb08111387eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 03:07:10.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc131ffc42b14bc39225ae4427d43c6d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a32636a61cb4893813567b1788964a3 : SUCCESS in 1h 22m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9571e994094048138327ab8195c95912 : SUCCESS in 1h 25m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f5d1342b64594bf7b09f910b8e53dfe3 : SUCCESS in 1h 20m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/542cf4e3639e4df39f12d1de8fb9d20b : SUCCESS in 1h 45m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/41780707f9a6488ca2ca0fc7a268989d : SUCCESS in 38m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c87ce601e8c47ce82d6d73e6f898f39 : SUCCESS in 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25fbe39db93a40c692e28c95fe310040 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64e8a20533ef4f2ca9a12cee83a1016f : SUCCESS in 22m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c7c6588678b4311b270ccd06b853e6c : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbca2765071040ac89f0fbea69c005e2 : SUCCESS in 8m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c36a72191fe04c919eb8cbe7dfe21ca2 : SUCCESS in 19m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/79f79443e63247cf84f0235698e29246 : SUCCESS in 1h 01m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/651d38fbe8c44d009074f916d5417e1f : SUCCESS in 49m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/116c7158c05b407ab9f3544adb364749 : SUCCESS in 1h 02m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e13b976ec814c808fcca66b03d04bd6 : SUCCESS in 1h 42m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f4752e8ef234bf4bb963c5489a9de3e : SUCCESS in 41m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ed9ee586aa748d1961780e8ca2d628e : SUCCESS in 1h 05m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ec8e4a00403d46bc9cefdfedbc09016a : SUCCESS in 2h 05m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7d24c052a23f4d18afb69e36beedd487 : SUCCESS in 1h 05m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/541db52922db4412b518a3d9bc666acb : SUCCESS in 43m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fac38743189449e88ebd92b126d2d2ac : SUCCESS in 2h 00m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a67acc6d6f6c44b9a55ab3f27b1f3ed1 : SUCCESS in 2h 09m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/68194272c11444a3b75ecd89bf3334d9 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":27},{"id":"2adf703f8e1f1abb5f8ed6bd66da1bee4f8d80ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-12-13 01:03:16.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"ef2ebb608d4594b74140bb15790fd3a3af023d88","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-13 01:38:53.000000000","message":"Patch Set 28: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/67/860767/28 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":28},{"id":"65034262ffb4fedfc5b446eb40b8ba5b969fa53a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 01:48:22.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5a7fae78f14a24b8fb492c71f1005f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad0ffdd26eff4bf084b433879098b602 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95c9943746b74e6eb616f125714f6a69 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2109dc9365d043eba710ce4cf20551d1 : SUCCESS in 41m 58s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"45b5eaa926da5156ab220f68d90fb6cb3be0a562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 03:06:26.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e71c82626f4b77abf4e7531f1521d6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/424603809b8c42f4af6edd8d7442d796 : SUCCESS in 1h 53m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b6371005e504319999a1d2f2fd6827e : SUCCESS in 2h 00m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e8e7f796412441cab086e79fe878b1f9 : SUCCESS in 1h 42m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/71c84a1625294648a05f916c522b5e2f : SUCCESS in 1h 53m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b75349a2af2e403ca49bf1a2f6e6f371 : SUCCESS in 48m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d37392a2e4e40499a970f4a1a5212fe : SUCCESS in 36m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0089eaa2b2314649a149a4baec9b12eb : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f3a4229df3341b18fa86fea146c1efd : SUCCESS in 27m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34df7f4d91be4b3eabea3b73bc64792e : SUCCESS in 26m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18148fd5a5d489c8d3765ba6ac90658 : SUCCESS in 11m 43s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/44ab7e67623740bf8f88ef98c61a0859 : SUCCESS in 28m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c2a6637ca1e64fcd9ccc8703fe6d01f0 : SUCCESS in 45m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/44f8b8b91b8941b8ae533a544216866d : SUCCESS in 48m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/489417ba967e465e947640b3187add47 : SUCCESS in 1h 01m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/edd2f41fb5fa4aa5bc8fcbeb6914b2de : SUCCESS in 1h 46m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c07f56160a041728c31d29d255a970f : SUCCESS in 1h 21m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/13ac6deb38a940bca8d07ed4361c93ea : SUCCESS in 49m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c6d4bc879ec642a9b4bbe752c70f45dd : SUCCESS in 1h 23m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/29aaeccf57944cb79bb6aadaf56013af : SUCCESS in 1h 32m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/995c37b74ae742d38f1560984be34b5a : SUCCESS in 1h 06m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d5ea5dbbbf4a4bd984f17a7e834c9f23 : SUCCESS in 1h 27m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/87e49bce2b5d49d18cea155e4e305362 : SUCCESS in 1h 29m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d78088e8fdcf47dbbf6e00fbd7d07295 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":28},{"id":"3e3181b03628786a94e90c47bcbf1c1061c82641","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-28 20:42:15.000000000","message":"Patch Set 28:\n\n(6 comments)","accounts_in_message":[],"_revision_number":28},{"id":"98d8e362b0de3ee031a9d19d73305452cce9648a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-12-20 01:56:56.000000000","message":"Patch Set 28:\n\n(5 comments)","accounts_in_message":[],"_revision_number":28},{"id":"a45cb163aca648b5d9a3393c0ba43da3d2c82f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-12-20 01:57:09.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"1ead4198ba40353cbe73f4ec48f1b6a4a0b015ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 02:30:41.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91bdab693c44475f8617dd76cef5fb4a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7cb3bb317604992a8b930dbc88cd3d5 : SUCCESS in 21m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/34ad811425544031a9de308a24c33e0d : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"f48a2a582da900b2d2a6133d0950b5c0be21fe61","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-20 04:01:44.000000000","message":"Patch Set 29: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/67/860767/29 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":29},{"id":"5a34196c10f54f690f260523ec02d155d6a464d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 04:51:02.000000000","message":"Patch Set 29: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0190850818e24ea682e128893d06bf47\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1fc4b21cfe214527b56b59ef37dd3ae7 : SUCCESS in 1h 44m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8717bfa3aca4b6ab7ac10da37b4adc1 : SUCCESS in 2h 38m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f76613f0c74c4afabd1064fa39a6e664 : SUCCESS in 2h 16m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6fb5aa63cf94ad0adebe7b6ce867fb7 : SUCCESS in 29m 03s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7cf6643d2bd49b19144850c98e7d43b : SUCCESS in 1h 35m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9987509a36a4271a420c45c2594a172 : FAILURE in 2h 51m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f648962517e845e3a721fecd417d4724 : SUCCESS in 1h 04m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812e32a3d4fc4ac5862831fb765bfad6 : FAILURE in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aef3feca641a4d78bc5abe37d1de6b62 : SUCCESS in 24m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3e993a6653a415e87579e8d87e1b3cc : SUCCESS in 19m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382cf0f1c7ee46d1bd9a5bd378c5f22f : SUCCESS in 10m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/39b1aa49c547451bba8953d42c7ad27a : SUCCESS in 52m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aff0aa5e7f7046e081d49c9ed273ea1a : SUCCESS in 1h 18m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/825c262004e14fc9800dbbcdc5ecfb6d : FAILURE in 58m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/73a6cdf316aa4b0ba4cdaa0e9d23c854 : FAILURE in 1h 53m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c6556b1f1dd4f91ac3c2f61bfd144c3 : SUCCESS in 1h 20m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/462e826af39f44a38b560aceb02ab671 : SUCCESS in 1h 12m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/71abbc3884954ddc932b74d94c78a30e : SUCCESS in 2h 01m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6c87edba61754170b15a97b72634d460 : SUCCESS in 1h 24m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ea881263f227442e9fe8c686cc71a430 : SUCCESS in 55m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4e0b7252791242049f13f086552c987f : SUCCESS in 36m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c8e2bc9b3a8f41618e1fc08cce353735 : SUCCESS in 1h 28m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a970466b14394fd9b993fa98e271b905 : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":29},{"id":"ae98e530e5f4375014fcb653003bcbe6ad2042c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-12-20 09:00:53.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"ac0995def9fd1f69baa25fa7cfe635786260cccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 09:34:55.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34003f957f4b49a6a4927badac8e0d81\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ccfb4f0ce7c4b2c9cd76a7c4f119eb0 : SUCCESS in 32m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/377c56b7a6a249f1a2324d31f0be667f : SUCCESS in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"e8b7eacb7eb52a62d7da343890314e800ef1ba8a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-20 09:56:18.000000000","message":"Patch Set 30: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/67/860767/30 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":30},{"id":"d6549c39a0ed4914530c14007fec8b494b33897a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 11:40:03.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c31b0530491b48b0a212f1152e601f80\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e622a7b0c4e4360945725ca04c539c8 : SUCCESS in 1h 55m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/719ed473a7a84fe8bbf3b8946d98fff4 : SUCCESS in 1h 43m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b44ceb41eabc4bba8433184a01f38905 : SUCCESS in 2h 37m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45d9ba6766c3478aa0df4f9676e38f8b : SUCCESS in 32m 40s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/310876e0212042459e3a7442e0ce6467 : SUCCESS in 1h 20m 27s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee4cf812e22242a080b7f6862a118e40 : SUCCESS in 1h 57m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ad598740fe049b593723a50ccc00be8 : SUCCESS in 39m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5938d5863ee54085a105877e360c6fd6 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6bb5db72e624aae841c4388908d3c4d : SUCCESS in 18m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9dc85852945481e9c8981896c49850d : SUCCESS in 17m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fad2eb8f8e5148e8ba01a5eb41f5ee76 : SUCCESS in 10m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ebd2beaa760f4865be6e21b0bd799696 : SUCCESS in 1h 00m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e8d0f6753dc74a8fb077e3c1757e06cf : SUCCESS in 42m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b0b78d4db6df4cd7a7a53eed18e17719 : SUCCESS in 43m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8798f7695e594eb19d8f08b09a21224b : FAILURE in 1h 49m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4da85fb55a154c818dc762d6321301ed : SUCCESS in 52m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/300b8fa93e9040f4ba43ca224fda536e : SUCCESS in 1h 02m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/42b573cfcbb94da9848ae4c9c4d6e734 : SUCCESS in 1h 42m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c25df53cf1b94c22855793a711097f0c : SUCCESS in 59m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1100949d7e9242c29be958eef3aaef3d : SUCCESS in 56m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b0ea5518e44845168ef7fb22236bd661 : SUCCESS in 1h 06m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/58786a048d0a4c6e885e106ef7aabf8e : SUCCESS in 1h 23m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e88585f5fa4443ec80a4879b0e2715a5 : SUCCESS in 1h 26m 32s","accounts_in_message":[],"_revision_number":30},{"id":"2cfbfb6061ea833efa2ab6210e6c141980362f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-12-23 03:37:22.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"ac78c821dc96f6ed2322c682ca8a993bb38b22a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 04:02:10.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abeaa456ca83462598f8b04628d4faa4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08d84c74bcee48c39773435aa5374d78 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/597d9f442d2b435da9e7b3e59840e538 : SUCCESS in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"5ae55e9b944e7ff02912ebd54e4e4d8f27dfaf60","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-23 04:09:46.000000000","message":"Patch Set 31: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/67/860767/31 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":31},{"id":"7c37e892d5608f14a08c4c66e5a69e6554e47839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 05:44:30.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1485c3c71a4b4aaae0aadc53dd0457\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b8f1104881bc43e099b57c6bdeca65ff : SUCCESS in 1h 44m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a72c7545f7743e898ae0386d5c0646c : SUCCESS in 1h 56m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fb8427a44bad444ea4a040e1f6447766 : SUCCESS in 1h 51m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2910bd01c3f84d0a8aa1a0addc5590de : SUCCESS in 43m 21s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d53b600798d84ac1bb387c826d305a10 : SUCCESS in 1h 26m 39s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02f32a19874849639f4817cec2c68987 : SUCCESS in 2h 05m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7ae0da01ec348e580b5a36fc9b27969 : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf2ebe73cfb4e7eb1a9ad4bf0d31150 : SUCCESS in 9m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5eec01419424853a4d79a4daf97b1c0 : SUCCESS in 18m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a552599a771d47c39dbedaa49775bb32 : SUCCESS in 19m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54296a9fdc44592a69ad9f66c900aea : SUCCESS in 8m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3bc1360c31bd4554bee20b8d920a68c7 : SUCCESS in 41m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/76b473294f274001bf34ce0d1da979ce : SUCCESS in 40m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/432802493484485d809d1b2350a65a0b : SUCCESS in 1h 01m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/96c2123dbd6c4c9e9f9385efccab255b : SUCCESS in 1h 30m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7f24e5e336b4b729943168b86ddc25e : SUCCESS in 39m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1eeb16e6b86d4bcc83aa6dded28bc93b : SUCCESS in 49m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a27eab914a7e4cc7a03a4f942c9a3314 : SUCCESS in 1h 18m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aff5a46a65a34ecd896762b92490e6d7 : SUCCESS in 1h 01m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/117ed3d27a4740c0b80cbf69b78912cc : SUCCESS in 1h 03m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c87282ea95e743efacfe300676a70c6f : SUCCESS in 41m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5ab84b371d8b465caed671ada8dd696a : SUCCESS in 45m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e268f88b59bb406fb66f8dde2d0f41e8 : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":31},{"id":"a1fb01088c5f988383432437f81494ae07774975","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-06 09:34:25.000000000","message":"Patch Set 31: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"a35931f09807f70bf831a7db6820480a6abd7db0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2025-01-07 09:54:48.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"9811d5684fa77ed3e01d45e9f8c3f3a09cbc7164","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2025-01-07 09:55:02.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"3b793307b67eeea3fe9222afa67c54a729ec6569","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 10:19:23.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe451c9114324e4a9d8b981c37787fc9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c2a6b33444c44718d1882b87a39e159 : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f6ce1c7569e489fa1fbaaa52ac63805 : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"4df25873f2b7940802a9e56bb4f03b0de47fd80a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 12:25:31.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6d2ac13ddc948898648716ac8b5e925\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dae6a992142347b2a7d9750ffef955f8 : SUCCESS in 2h 03m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/96e84b2ee5274f28b06950e5a841947d : SUCCESS in 2h 05m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9403a8ede0a34898888819039c722b81 : FAILURE in 2h 28m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5effcf18d90c4dfc95a9988a8b0651e3 : SUCCESS in 45m 00s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4a44d169ed448499353242960753a38 : SUCCESS in 2h 06m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/791f32d18fd64a1aa246562b3e8ff4ac : SUCCESS in 1h 47m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5a1f81d874c410abb6a74e61768c19a : SUCCESS in 51m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/881b0b6aa4cb46599de98e378f577983 : SUCCESS in 12m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/437a0a6b1bff44d7b2eb75f545765b5d : SUCCESS in 21m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9aa46e34847344f99ea584b9519a007d : SUCCESS in 19m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf99964989642ef858df2cc397aefd2 : SUCCESS in 11m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a49c306d443841278d8fe602632cc2ee : SUCCESS in 1h 01m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9354b16d86c349cd86a4fea5c8348955 : SUCCESS in 1h 08m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2d2c44e20d524431b9906f996f8f2993 : SUCCESS in 1h 03m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d69b06ebb795444a8f350be07abecf4b : SUCCESS in 1h 51m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c242568020da457fa77966e09fad419c : SUCCESS in 54m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ee1e04696d54f998fc77d6130319723 : SUCCESS in 1h 15m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/060bf045dbdf4b75a734d880ff871a12 : SUCCESS in 1h 42m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e710572d36ef43998880c5f8b841c540 : SUCCESS in 1h 25m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/80e09fc580ef4e47ab6e9a0cc98b15f6 : SUCCESS in 56m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/800a84b972bb43b1b42551747a2f4d56 : SUCCESS in 51m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a966e20080d2428abc4e0f48813d1eaa : SUCCESS in 1h 01m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d904addc8f7b4797bc3a85160f22fd1f : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":32},{"id":"295cd76767ae80bc1994fd3fc110ed6dede9ced0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-07 13:21:40.000000000","message":"Patch Set 32: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/67/860767/32 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":32},{"id":"d6620e5ae4c7963c6fac2c10aa336f037df436f1","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2025-01-09 01:22:19.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"dd5dd089be85260346b833bc12b0af38b4f0fc5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 01:48:56.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf0caa924f8d4779b1fd7d17d4d7805f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ddd36ebe0fa54dd08aba5101e99571b7 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7341c1b5c991422fa7defd392768a653 : SUCCESS in 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"59075642f024b52efc7a6e5910b68546fc09a9d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-09 01:53:59.000000000","message":"Patch Set 32:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/67/860767/32 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":32},{"id":"d1245db23497317f152b004ba441736eedc9d648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 03:53:45.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae1d0772ab84b52806c208db77bf61d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3ec33b52018d4fe1a57c198941d5eb55 : SUCCESS in 1h 51m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cbe2596f92984de79c11fe947a5285cb : SUCCESS in 2h 00m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3029d07293cf47a7b4fff33c8ebdecbd : SUCCESS in 2h 20m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d574a40f7fc44767add3120d296dc29d : SUCCESS in 30m 10s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75072e6329594fa297d457cc69591752 : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/672eee57aa304d699373dc90f5ce9bea : SUCCESS in 1h 49m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819409588682460c84a45c0b3ddc8061 : SUCCESS in 33m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be73a8d4452d489cb9eb4bb5f90d166c : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09f9ecb30c6c4b2f9e47683a5435d797 : SUCCESS in 17m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/144fbea1c6c348feaedb415bd2b7f3aa : SUCCESS in 15m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dcd814de13340b899cc432838ca4701 : SUCCESS in 10m 02s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa97ff68fd1e4d7ea4469d8836d0f24f : SUCCESS in 33m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/adf586e1da01422e93f497a0e15e6d46 : SUCCESS in 1h 17m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dee8a73e7fe642d5b32e01ebc6a1e2c7 : SUCCESS in 1h 17m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/06610b02a4874a429a448a537954f8b4 : SUCCESS in 2h 30m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/44ec43b575f045868c965ca6b2727f9b : SUCCESS in 53m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e04b6baa4c784d14bbc6c182be145837 : SUCCESS in 49m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0d7e38764c6840009b2346e7ab6cfb6d : SUCCESS in 1h 45m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bdd28d23102549eebc8d398d3ed73626 : SUCCESS in 1h 35m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ead9c57d2d249adae59e40a8486bf38 : SUCCESS in 55m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/314e1bb1bce64dfcb6c350f78d52e18e : SUCCESS in 52m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c85bf3b7e78649d982ec0c1ee4239b31 : SUCCESS in 46m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fe0884717d024378ba8bb4e430ca5ad6 : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":32},{"id":"c85406a23bdd45b003a7e078042f9d92cc51072b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-09 10:51:41.000000000","message":"Patch Set 32: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"cb601e5aadd111d3f915fac29aa75bb60f7a827f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2025-02-02 11:52:31.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"94d2d933e0e786e798b8dd5d9bfe37caecfc976d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 12:33:02.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d22d6d6a314205acafc2287914fc5a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a938e63d7e14f559bb0dd8b4df353a6 : SUCCESS in 38m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/009862e0d5fe41c0a4b1396119e1c1ab : SUCCESS in 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"2c0fd027c7833deeda01305436e4ced92c432932","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-02 13:12:48.000000000","message":"Patch Set 33:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/67/860767/33 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":33},{"id":"007505b11c8193681ad4e739aee8595ede21b2a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 14:09:34.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e39718f0daa41bc852b102431c2f95a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4faa74ba0c6d4bb9a5f11664b4f9d997 : SUCCESS in 1h 48m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c87b6c6c5c1a46e3a0d50f550c7e7e56 : SUCCESS in 2h 07m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/990add9b59304d21b5fd7e1c8ac3de26 : SUCCESS in 2h 14m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ccafe9b6bdbf4d74806748c4f2798754 : SUCCESS in 27m 33s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/484fe66c9d49435c900fb5e7eec59ee6 : SUCCESS in 1h 50m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e434fecdeff745278115993a6b8e52ee : SUCCESS in 2h 03m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9b10f02b83d4444a8cf6e2f42d9ab44 : SUCCESS in 54m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3556fefb7e14d998147d10557ec64a7 : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8010c7ab96c14b318de12eca41f472a3 : SUCCESS in 22m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/866885fcf8f04ef98216e7c3310a2cf9 : SUCCESS in 20m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04a4a2306774fd1ae57f0ad78e2f31b : SUCCESS in 8m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bcda7e0eca25463bb65de088cfc54dbc : SUCCESS in 59m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3cd8723d0d144e99a6bff64ef2bc2f60 : SUCCESS in 40m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c8d79388c28445c6af63bd38787119d0 : SUCCESS in 1h 10m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/51221656840245f7b20cf7ce37392294 : SUCCESS in 1h 53m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b19a6e1fb2c47b692f1da07e94cd87b : SUCCESS in 47m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/81bdeab662894e1a8f4413a0058a2d9b : SUCCESS in 1h 10m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5362507155f846cd889b27a91aa5ba66 : SUCCESS in 1h 49m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/42b03fff6a70439592792055d77e6477 : SUCCESS in 1h 36m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/545bc5bea4294cd1baa6fa36d14184a5 : SUCCESS in 54m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8ed4f77f7a134ccdb916df1dfeaed61d : SUCCESS in 53m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b8b04b3808e413eace2639e69c666d2 : SUCCESS in 56m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b29e36cd0b904b16809db4e979ce6d19 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":33},{"id":"abc03120b4bedd1b5c7a27bc4dfa5072434ccd94","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-03 09:58:24.000000000","message":"Patch Set 33: Code-Review+2","accounts_in_message":[],"_revision_number":33},{"id":"17c5d2da891c57c7e8a2924181dca60d5db1ec0b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-07 21:10:35.000000000","message":"Patch Set 33: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":33},{"id":"64dd4f0c06bc1b9159ab262fb59b9479b40b4fa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 21:11:05.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"9357d31fa4fd4c4ce9071e9fce89697c2e3893e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 23:40:25.000000000","message":"Patch Set 33: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c210d49c0f964aed90badeca61cb16f2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bd485e16194948e5a5e49b844c932655 : SUCCESS in 1h 33m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a21c90a4cecc458d9701e35888dde11e : SUCCESS in 2h 20m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26619c75e0404ea39ea8e7732a44d4ea : SUCCESS in 1h 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc1f2f944f040648c892f584421af5d : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8101274e8e4243fb9b66c8477da3c977 : SUCCESS in 19m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70434926b97247b5a932f6cb0b20a40b : SUCCESS in 21m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25debd59e02b4506a9ac898cee172459 : SUCCESS in 10m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba43ea9c79c24e18942a32518ea8fbf0 : SUCCESS in 1h 11m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1e067318ed5645a7b49f04a3175c2342 : SUCCESS in 1h 54m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e4b8700ada2843a0beb721ad18bb7da1 : SUCCESS in 1h 37m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f580c202c8b34d6db1a1a1862423fa5b : SUCCESS in 1h 19m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02a4a5365cf14aad9ae787d6da212617 : SUCCESS in 1h 07m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a7207f0f6e9d43dc8e2b5ee3418b3626 : SUCCESS in 1h 10m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1b00546c00ba49879644db148182dbbf : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":33},{"id":"b5f6072a00cb75c1480ffed051aa14783aa74364","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 23:40:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":33},{"id":"94b3beba5774fccbac0923d0d1b35c5785120212","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 23:41:36.000000000","message":"Patch Set 33:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcbb1d3d2f0b4e6d9305386982d661c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b6d7e2b3c940579a3640d320ccdde9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"be259520ce70b060c7a61c104d9e982eb0766c9e","revisions":{"a4f64a346e2d6994760e6afbd5885ae820c2f930":{"kind":"REWORK","_number":1,"created":"2022-10-09 01:50:35.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/1"}}},"commit":{"parents":[{"commit":"2a36d111e057daff514a944f8b881dc0265a7003","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a36d111e057daff514a944f8b881dc0265a7003"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-10-09 01:49:22.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4f64a346e2d6994760e6afbd5885ae820c2f930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4f64a346e2d6994760e6afbd5885ae820c2f930"}]},"branch":"refs/heads/master"},"679b0070365a41ab700ab6e49c290bf2cf844345":{"kind":"REWORK","_number":2,"created":"2022-10-09 06:57:00.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/2"}}},"commit":{"parents":[{"commit":"0dc07ac0570ed57a6d5ae3636b6d4e1ff4cae658","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dc07ac0570ed57a6d5ae3636b6d4e1ff4cae658"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-10-09 06:35:20.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/679b0070365a41ab700ab6e49c290bf2cf844345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/679b0070365a41ab700ab6e49c290bf2cf844345"}]},"branch":"refs/heads/master"},"f50588ac5f611c9b125ab73c23c787b36efbf554":{"kind":"REWORK","_number":3,"created":"2022-10-18 03:05:46.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/3"}}},"commit":{"parents":[{"commit":"3b0a5e493aef282de8af924a9746fca2769fc406","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b0a5e493aef282de8af924a9746fca2769fc406"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-10-18 03:02:16.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f50588ac5f611c9b125ab73c23c787b36efbf554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f50588ac5f611c9b125ab73c23c787b36efbf554"}]},"branch":"refs/heads/master"},"83c72e94ed1e5934e8172678ec565d228c5c3561":{"kind":"REWORK","_number":4,"created":"2022-10-19 01:44:52.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/4"}}},"commit":{"parents":[{"commit":"d66cd08f372ea03f568fa1a2a7edf467cb723844","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d66cd08f372ea03f568fa1a2a7edf467cb723844"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-10-19 01:38:31.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83c72e94ed1e5934e8172678ec565d228c5c3561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83c72e94ed1e5934e8172678ec565d228c5c3561"}]},"branch":"refs/heads/master"},"6ef3762662d8fb07a89e92d7ee5dfa8bc83689ec":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-11-14 01:40:03.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/5"}}},"commit":{"parents":[{"commit":"129d08ecbad5697731897cd723d326c3cd1f21b4","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/129d08ecbad5697731897cd723d326c3cd1f21b4"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-14 01:38:17.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ef3762662d8fb07a89e92d7ee5dfa8bc83689ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ef3762662d8fb07a89e92d7ee5dfa8bc83689ec"}]},"branch":"refs/heads/master"},"e322fb114e68275e3e0bdce3e15f9eaf45bb4810":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-11-14 03:40:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/6"}}},"commit":{"parents":[{"commit":"1c3c447cfa8dfc85b9a16359478c4e3d4d31ec5d","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c3c447cfa8dfc85b9a16359478c4e3d4d31ec5d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-14 03:39:15.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e322fb114e68275e3e0bdce3e15f9eaf45bb4810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e322fb114e68275e3e0bdce3e15f9eaf45bb4810"}]},"branch":"refs/heads/master"},"aaaba2e02882ce6c7f5bb906954a2ba52f14fa75":{"kind":"REWORK","_number":7,"created":"2022-11-14 06:47:08.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/7"}}},"commit":{"parents":[{"commit":"5ee6434540da04132bc2568036a84bb406281472","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee6434540da04132bc2568036a84bb406281472"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-14 06:46:22.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaaba2e02882ce6c7f5bb906954a2ba52f14fa75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaaba2e02882ce6c7f5bb906954a2ba52f14fa75"}]},"branch":"refs/heads/master"},"ee47b19d8138764a17f30d77f3042c4b13e56dda":{"kind":"REWORK","_number":8,"created":"2022-11-15 02:21:21.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/8"}}},"commit":{"parents":[{"commit":"beff37583ba5129f334c9989b4dbc5e1289a3d56","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/beff37583ba5129f334c9989b4dbc5e1289a3d56"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-15 02:10:17.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee47b19d8138764a17f30d77f3042c4b13e56dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee47b19d8138764a17f30d77f3042c4b13e56dda"}]},"branch":"refs/heads/master"},"ceb18b263757ae8da461ad81d81e1a6faceffe8c":{"kind":"REWORK","_number":9,"created":"2022-11-16 01:33:37.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/9"}}},"commit":{"parents":[{"commit":"15f67175aa97f32bb9925c74183b739f264b98f0","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15f67175aa97f32bb9925c74183b739f264b98f0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-16 01:32:07.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ceb18b263757ae8da461ad81d81e1a6faceffe8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ceb18b263757ae8da461ad81d81e1a6faceffe8c"}]},"branch":"refs/heads/master"},"af2b952e821f03ba6d542ab106001e2b9be1442c":{"kind":"REWORK","_number":10,"created":"2022-11-17 02:11:52.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/10"}}},"commit":{"parents":[{"commit":"15f67175aa97f32bb9925c74183b739f264b98f0","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15f67175aa97f32bb9925c74183b739f264b98f0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-17 01:48:46.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af2b952e821f03ba6d542ab106001e2b9be1442c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af2b952e821f03ba6d542ab106001e2b9be1442c"}]},"branch":"refs/heads/master"},"ba5ff9b7855750e762d22284c7a5adcf486bfb6b":{"kind":"REWORK","_number":11,"created":"2022-11-17 06:36:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/11"}}},"commit":{"parents":[{"commit":"15f67175aa97f32bb9925c74183b739f264b98f0","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15f67175aa97f32bb9925c74183b739f264b98f0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-17 06:34:55.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba5ff9b7855750e762d22284c7a5adcf486bfb6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba5ff9b7855750e762d22284c7a5adcf486bfb6b"}]},"branch":"refs/heads/master"},"33c67dc4d2087fc6bd0e2685e71b244e38e1db1c":{"kind":"REWORK","_number":12,"created":"2022-11-24 01:31:12.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/12"}}},"commit":{"parents":[{"commit":"15f67175aa97f32bb9925c74183b739f264b98f0","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15f67175aa97f32bb9925c74183b739f264b98f0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-11-24 01:30:35.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33c67dc4d2087fc6bd0e2685e71b244e38e1db1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33c67dc4d2087fc6bd0e2685e71b244e38e1db1c"}]},"branch":"refs/heads/master"},"41d50350fbe5c1e9cfc5bbe4a197992523165e42":{"kind":"REWORK","_number":13,"created":"2022-12-05 04:18:41.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/13"}}},"commit":{"parents":[{"commit":"874b3e03f61fcd49070ab05cb2fa205a9503231b","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/874b3e03f61fcd49070ab05cb2fa205a9503231b"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-05 04:13:42.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41d50350fbe5c1e9cfc5bbe4a197992523165e42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41d50350fbe5c1e9cfc5bbe4a197992523165e42"}]},"branch":"refs/heads/master"},"c3087d9a07e7900d587f4694e420e807de252a6b":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2022-12-05 04:20:47.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/14"}}},"commit":{"parents":[{"commit":"dc3d1dfa29cac770de925549b6094add9ef95e9e","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc3d1dfa29cac770de925549b6094add9ef95e9e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-05 04:19:59.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3087d9a07e7900d587f4694e420e807de252a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3087d9a07e7900d587f4694e420e807de252a6b"}]},"branch":"refs/heads/master"},"f8c0571a37400bdb71eb438538ca59c83101dd66":{"kind":"REWORK","_number":15,"created":"2022-12-05 11:49:01.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/15"}}},"commit":{"parents":[{"commit":"dc3d1dfa29cac770de925549b6094add9ef95e9e","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc3d1dfa29cac770de925549b6094add9ef95e9e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-05 11:47:37.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8c0571a37400bdb71eb438538ca59c83101dd66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8c0571a37400bdb71eb438538ca59c83101dd66"}]},"branch":"refs/heads/master"},"935e0419657b73f701280638e7c103f44cf04cc3":{"kind":"REWORK","_number":16,"created":"2022-12-06 01:14:22.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/16"}}},"commit":{"parents":[{"commit":"dc3d1dfa29cac770de925549b6094add9ef95e9e","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc3d1dfa29cac770de925549b6094add9ef95e9e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-06 01:11:15.000000000","tz":480},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/935e0419657b73f701280638e7c103f44cf04cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/935e0419657b73f701280638e7c103f44cf04cc3"}]},"branch":"refs/heads/master"},"ba3d77ee301690025e7a0fe5cb04010ad15972ec":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":17,"created":"2022-12-14 03:07:51.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/17"}}},"commit":{"parents":[{"commit":"11295e130f135ebc6d65818dc0b7754b02a1b900","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11295e130f135ebc6d65818dc0b7754b02a1b900"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-14 03:05:25.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba3d77ee301690025e7a0fe5cb04010ad15972ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba3d77ee301690025e7a0fe5cb04010ad15972ec"}]},"branch":"refs/heads/master"},"5c0647843e81bf4babceb875538148641bf65f68":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2022-12-14 03:09:25.000000000","uploader":{"_account_id":35618,"name":"zhihua fu","display_name":"zhihua fu","email":"jxfu@163.com","username":"qingshanyy"},"ref":"refs/changes/67/860767/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/18"}}},"commit":{"parents":[{"commit":"11295e130f135ebc6d65818dc0b7754b02a1b900","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11295e130f135ebc6d65818dc0b7754b02a1b900"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"zhihua fu","email":"jxfu@163.com","date":"2022-12-14 03:09:25.000000000","tz":0},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n111\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c0647843e81bf4babceb875538148641bf65f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c0647843e81bf4babceb875538148641bf65f68"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6600b32db2257016292b5710165063b4263048a5":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2022-12-14 03:09:51.000000000","uploader":{"_account_id":35618,"name":"zhihua fu","display_name":"zhihua fu","email":"jxfu@163.com","username":"qingshanyy"},"ref":"refs/changes/67/860767/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/19"}}},"commit":{"parents":[{"commit":"11295e130f135ebc6d65818dc0b7754b02a1b900","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11295e130f135ebc6d65818dc0b7754b02a1b900"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"zhihua fu","email":"jxfu@163.com","date":"2022-12-14 03:09:51.000000000","tz":0},"subject":"Add meter bandwidth bandwidth support","message":"Add meter bandwidth bandwidth support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6600b32db2257016292b5710165063b4263048a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6600b32db2257016292b5710165063b4263048a5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"33ba3e40acb3ecdb23b1bb39e5843800d617e1f9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":20,"created":"2022-12-16 00:55:24.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/20"}}},"commit":{"parents":[{"commit":"334016e0141862c8c409251649578c9a0f9aee54","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/334016e0141862c8c409251649578c9a0f9aee54"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-16 00:54:17.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33ba3e40acb3ecdb23b1bb39e5843800d617e1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33ba3e40acb3ecdb23b1bb39e5843800d617e1f9"}]},"branch":"refs/heads/master"},"5dc5dc1d121218a6dda5953f516cc3ebb4fcfc96":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-12-21 02:05:05.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/21"}}},"commit":{"parents":[{"commit":"20b99eb63d0402139bb965bb863e3dc60b68610a","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20b99eb63d0402139bb965bb863e3dc60b68610a"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-21 02:01:42.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dc5dc1d121218a6dda5953f516cc3ebb4fcfc96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dc5dc1d121218a6dda5953f516cc3ebb4fcfc96"}]},"branch":"refs/heads/master"},"41c5945f1d2bd09c6f5f8e4befe5b6d0d869568f":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2023-03-09 01:34:13.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/22"}}},"commit":{"parents":[{"commit":"dd886cf8b049a7f62db78cf75095561989f21744","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd886cf8b049a7f62db78cf75095561989f21744"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-03-09 01:32:57.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41c5945f1d2bd09c6f5f8e4befe5b6d0d869568f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41c5945f1d2bd09c6f5f8e4befe5b6d0d869568f"}]},"branch":"refs/heads/master"},"c3c54a76f4f13091cd01c2f94a2889a060a34239":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-04-11 01:52:51.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/23"}}},"commit":{"parents":[{"commit":"e39145073e3d782f8a02b9771e9c512ed7389577","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e39145073e3d782f8a02b9771e9c512ed7389577"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-04-11 01:50:18.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3c54a76f4f13091cd01c2f94a2889a060a34239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3c54a76f4f13091cd01c2f94a2889a060a34239"}]},"branch":"refs/heads/master"},"d7a1874e6fc3ece8f6f154d0f44d66168ac021a2":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-06-01 07:14:38.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/24"}}},"commit":{"parents":[{"commit":"e839702fea6c29865ac35c7188e61e64161fe7a4","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e839702fea6c29865ac35c7188e61e64161fe7a4"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-06-01 07:12:18.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7a1874e6fc3ece8f6f154d0f44d66168ac021a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7a1874e6fc3ece8f6f154d0f44d66168ac021a2"}]},"branch":"refs/heads/master"},"13850abc1063650a4374be9ab72d046f75c68fbb":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-07-05 09:51:06.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/25"}}},"commit":{"parents":[{"commit":"02b12b09175c8608ac7d2032baa5d6caf01c660b","subject":"Refactor for ovs qos driver meter limit features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02b12b09175c8608ac7d2032baa5d6caf01c660b"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-07-05 09:40:10.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13850abc1063650a4374be9ab72d046f75c68fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13850abc1063650a4374be9ab72d046f75c68fbb"}]},"branch":"refs/heads/master"},"22313f4655b49b7c37db9e9fd5df490d4f0cbff4":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2023-07-11 00:45:21.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/26"}}},"commit":{"parents":[{"commit":"6fa9142ced0eed8de9a89042747438b93de7202d","subject":"Merge \"Return back the test_dvr_router_interface_mtu_update test case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fa9142ced0eed8de9a89042747438b93de7202d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-07-11 00:44:05.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22313f4655b49b7c37db9e9fd5df490d4f0cbff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22313f4655b49b7c37db9e9fd5df490d4f0cbff4"}]},"branch":"refs/heads/master"},"44877c27d416dd6015b7ed9c150f828c38bbab35":{"kind":"REWORK","_number":27,"created":"2023-12-05 08:30:20.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/27"}}},"commit":{"parents":[{"commit":"729baeb4e7eba4fa5775e4620cc5ef72b91a9da4","subject":"Merge \"Remove deprected and unused argument from update_port_status()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/729baeb4e7eba4fa5775e4620cc5ef72b91a9da4"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-12-05 08:28:39.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44877c27d416dd6015b7ed9c150f828c38bbab35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44877c27d416dd6015b7ed9c150f828c38bbab35"}]},"branch":"refs/heads/master"},"dd95b246709ecf29fd059605361f113cd51afd4f":{"kind":"REWORK","_number":28,"created":"2023-12-13 01:03:16.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/28"}}},"commit":{"parents":[{"commit":"a09e0f202577eaf7518df8c837887ed73642ae41","subject":"Merge \"[Fullstack] Remove SecurityGroupRulesTest.test_normalized_cidr_in_rule test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a09e0f202577eaf7518df8c837887ed73642ae41"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-12-13 01:02:40.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd95b246709ecf29fd059605361f113cd51afd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd95b246709ecf29fd059605361f113cd51afd4f"}]},"branch":"refs/heads/master"},"e90066b4923c594d1666af66504a229679e09fc9":{"kind":"REWORK","_number":29,"created":"2024-12-20 01:57:09.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/29"}}},"commit":{"parents":[{"commit":"559672f777312d00c4358626650cc577e49ec0b6","subject":"Merge \"Fast exit when initially creating tunnel allocations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/559672f777312d00c4358626650cc577e49ec0b6"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-12-20 01:45:27.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e90066b4923c594d1666af66504a229679e09fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e90066b4923c594d1666af66504a229679e09fc9"}]},"branch":"refs/heads/master"},"4a871cca22479c103fe9876e792002be7685123a":{"kind":"REWORK","_number":30,"created":"2024-12-20 09:00:53.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/30"}}},"commit":{"parents":[{"commit":"559672f777312d00c4358626650cc577e49ec0b6","subject":"Merge \"Fast exit when initially creating tunnel allocations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/559672f777312d00c4358626650cc577e49ec0b6"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-12-20 08:59:59.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a871cca22479c103fe9876e792002be7685123a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a871cca22479c103fe9876e792002be7685123a"}]},"branch":"refs/heads/master"},"bb74ff44221ae6e02520bb866a8bb3cb52738977":{"kind":"REWORK","_number":31,"created":"2024-12-23 03:37:22.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/31"}}},"commit":{"parents":[{"commit":"559672f777312d00c4358626650cc577e49ec0b6","subject":"Merge \"Fast exit when initially creating tunnel allocations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/559672f777312d00c4358626650cc577e49ec0b6"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-12-23 03:36:50.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb74ff44221ae6e02520bb866a8bb3cb52738977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb74ff44221ae6e02520bb866a8bb3cb52738977"}]},"branch":"refs/heads/master"},"602c49882e6dc0bc38b2cea2ab1518187aa6b0d1":{"kind":"REWORK","_number":32,"created":"2025-01-07 09:54:48.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/32"}}},"commit":{"parents":[{"commit":"f2c3e3e68c0fa77cff5cdd22d847e3f6ea7cdd84","subject":"Merge \"ml2/ovs: improve log message for ports without VLAN tags/net-id during OVS restore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2c3e3e68c0fa77cff5cdd22d847e3f6ea7cdd84"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2025-01-07 09:54:23.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/602c49882e6dc0bc38b2cea2ab1518187aa6b0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/602c49882e6dc0bc38b2cea2ab1518187aa6b0d1"}]},"branch":"refs/heads/master"},"be259520ce70b060c7a61c104d9e982eb0766c9e":{"kind":"REWORK","_number":33,"created":"2025-02-02 11:52:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/860767/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/860767/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/860767/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/860767/33"}}},"commit":{"parents":[{"commit":"62f39610edfa43c79015097e7c1c14d6d7cc5a5a","subject":"Merge \"[eventlet-removal] Use non-eventlet metadata proxy in OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62f39610edfa43c79015097e7c1c14d6d7cc5a5a"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-09-02 02:13:20.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2025-02-02 11:51:44.000000000","tz":480},"subject":"Add meter bandwidth limit support","message":"Add meter bandwidth limit support\n\nBecasue many tables are still goto table PACKET_RATE_LIMIT\nby default, in order to reduce the refator work, packet will\nstill go to packet rate limit table\u003d58 first, then\nbandwidth rate limit table\u003d59.\n\nDepends-On: I87da4960a92bb807203534965bac2b0a42c5d4a4\n\nCloses-Bug: #1964342\nChange-Id: Ia7f842f7e2fb6b218f8b1dc930bf1450e6f810ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be259520ce70b060c7a61c104d9e982eb0766c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be259520ce70b060c7a61c104d9e982eb0766c9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
