)]}'
{"id":"openstack%2Fneutron~861844","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Ic3b5a0ac7d832b162848b363396264ed0bfc4a25","project":"openstack/neutron","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2022-10-25 13:53:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3b5a0ac7d832b162848b363396264ed0bfc4a25","subject":"update the nova host aggregates links","status":"MERGED","created":"2022-10-20 02:13:19.000000000","updated":"2022-10-25 13:54:17.000000000","submitted":"2022-10-25 13:53:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861844","cherry_pick_of_change":861844,"cherry_pick_of_patch_set":1,"meta_rev_id":"1a95059b4cc5459215bbc6e4e5292170c9b18c43","_number":861844,"virtual_id_number":861844,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-25 13:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-10-25 13:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-10-20 09:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-10-25 13:32:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 03:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 09:11:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-10-25 13:32:34.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"52352041275f02f3c32dd4103e0b11c5fd378328","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-10-20 02:13:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"b9db4dc65095fb89199b0f210eabd32bc26da82a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-10-20 02:13:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1283d667367a06d26dd71e92b3d4f653fda05ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 03:04:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29ef4c66dabd420a9e3035aae019f357\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c7659870464ab8aa0ddbdb0bffc8fc : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8a5f24ee084979864cc30fd5efcd3f : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":2},{"id":"00782056e5e5ab23ed234858b78523459e5e4199","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-20 09:11:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"748f11709c68f0163e5b0133d326fd6b00bbd3e7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-10-25 13:32:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c10582c9011ea79e997a195d895e3e081bc41e23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 13:33:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e85ef9a7610ae4eea3534a7a635220cb04e31e74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 13:53:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed93b614547c4a17a6a409d599a30473\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25037b21ac224b1bb06e693230517fa0 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58482f852c0d4a4c8770dc145e1ff143 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":2},{"id":"aa91e91a1777eab13f8c19b12c275b12a1a8d2ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 13:53:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1a95059b4cc5459215bbc6e4e5292170c9b18c43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 13:54:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918f23e5946e4d7792ed9bbb1ea73dbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22259b461359428d920cf78b0b892087 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98e2fd47fe6bcb0d4870c58b1533746fe8c57608","revisions":{"d9a3d0b63c8f4bd74caa9c79bcf7e00201cd21b1":{"kind":"REWORK","_number":1,"created":"2022-10-20 02:13:19.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/44/861844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/861844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/861844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/861844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/861844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/861844/1"}}},"commit":{"parents":[{"commit":"32a8b2d3388b0e1ed4b19a5f3b68a99762f9f70b","subject":"Disable in-band management for bridges before setting up controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a8b2d3388b0e1ed4b19a5f3b68a99762f9f70b"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-07-04 06:00:55.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-10-20 02:13:19.000000000","tz":0},"subject":"update the nova host aggregates links","message":"update the nova host aggregates links\n\nhttps://docs.openstack.org/nova/latest/admin/aggregates.html link failure,\nFrom openstack U version, this link is not in the user directory. Currently, only the latest version has been changed.\nYou are advised to change all the links\n\nChange-Id: Ic3b5a0ac7d832b162848b363396264ed0bfc4a25\n(cherry picked from commit 210f5297f5ac5c180677d3f6419f436f77e56f1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9a3d0b63c8f4bd74caa9c79bcf7e00201cd21b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9a3d0b63c8f4bd74caa9c79bcf7e00201cd21b1"}]},"branch":"refs/heads/stable/victoria"},"98e2fd47fe6bcb0d4870c58b1533746fe8c57608":{"kind":"NO_CHANGE","_number":2,"created":"2022-10-20 02:13:49.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/44/861844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/861844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/861844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/861844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/861844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/861844/2"}}},"commit":{"parents":[{"commit":"32a8b2d3388b0e1ed4b19a5f3b68a99762f9f70b","subject":"Disable in-band management for bridges before setting up controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a8b2d3388b0e1ed4b19a5f3b68a99762f9f70b"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-07-04 06:00:55.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-10-20 02:13:49.000000000","tz":0},"subject":"update the nova host aggregates links","message":"update the nova host aggregates links\n\nhttps://docs.openstack.org/nova/latest/admin/aggregates.html link failure,\nFrom openstack U version, this link is not in the user directory. Currently, only the latest version has been changed.\nYou are advised to change all the links\n\nChange-Id: Ic3b5a0ac7d832b162848b363396264ed0bfc4a25\n(cherry picked from commit 210f5297f5ac5c180677d3f6419f436f77e56f1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98e2fd47fe6bcb0d4870c58b1533746fe8c57608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98e2fd47fe6bcb0d4870c58b1533746fe8c57608"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
