)]}'
{"id":"openstack%2Fneutron~863013","triplet_id":"openstack%2Fneutron~master~I41ba7d6929b28317622dbf868f265eab4c6fd84e","project":"openstack/neutron","branch":"master","topic":"fix-url-and-package-name","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-04-13 13:38:39.000000000","reason":"Change was submitted"},"33787":{"account":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"last_update":"2023-04-13 13:38:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41ba7d6929b28317622dbf868f265eab4c6fd84e","subject":"Update url and package name","status":"MERGED","created":"2022-10-31 06:54:23.000000000","updated":"2023-04-13 13:40:06.000000000","submitted":"2023-04-13 13:38:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863013-fix-url-and-package-name","meta_rev_id":"af8886780e64005c8cceed3687f3537e18ac1341","_number":863013,"virtual_id_number":863013,"owner":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 13:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-04-04 13:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-04-13 09:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-04-13 09:57:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-31 07:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 08:53:41.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-11-01 18:01:04.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-11-02 15:53:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-11-03 06:45:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"3b65df84ab467b16c29da5321fb04ea79ff90011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2022-10-31 06:54:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b43155a2217a8a7cd12f5b9f9f3232b075da8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 07:04:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369f216f794746458c6263a5aa293f9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700b446ddfa24068ac9a0d107e322469 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3a97d5c6584c5586c6df2c1948d87d : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1},{"id":"79e0e8e97b636c8032aa691faffa4cec0143e6aa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-10-31 08:53:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/863013/1 : SUCCESS in 1h 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fd10c5ea942e0d8e40a0761b93767d3d217d50bb","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-11-01 18:01:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"757c0009ce4cf3171b4a7c00ef5ef46f4e3b7429","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-02 15:53:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a01870c18c0d5642d85e3f8e44995b1014f0e52f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-11-03 06:45:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"35adc3f76e20d278da594fa194770686bb8b277f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2022-11-18 09:36:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a786f375396b8da4f8c7f15794104b85e401974","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2022-11-18 09:48:12.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f9c8f05ea7e0240e43ad0f33bc8ccaa98083e27f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 10:45:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff63fd49f214efca6b688e511f12c11\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a132a8bf291b465bb1b0148ee25a6cb4 : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e0799f0ada04504b088312d74e841c5 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8558e7ab3b6943f889e6d594f12aed6f : FAILURE in 1h 05m 36s","accounts_in_message":[],"_revision_number":2},{"id":"2de3e57c8c5da21274d0a712a076316fd24472ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 11:12:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954ad0d745174f21818055e12b1c9503\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/399f6a53bc4c4b9581439c65ab3fc009 : SUCCESS in 1h 28m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/998401c4d9ba4776a3fa25af25246169 : SUCCESS in 1h 24m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c30129ed55024680a3672b8b8bd94edc : SUCCESS in 1h 20m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/41e3d5cc0585477581ee755c45318276 : SUCCESS in 1h 18m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/321f78431e90407c8974399274d068d0 : SUCCESS in 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4935e1c78d884b2da66c48f1282d0107 : SUCCESS in 29m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f5d7cbb3e7240c1b1e466631a0f3a14 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37c526ec1e2f4c3483a7743e06e7eb23 : SUCCESS in 23m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d7f7f5668a245249b6ac43c81049b6a : SUCCESS in 22m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d293b15884e14d0ba2eac899945859a4 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/361296aa919846f9b932591561977397 : SUCCESS in 23m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/053df73763084e73a16264dc99e3fb12 : SUCCESS in 45m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d331d083e4fb4caea0d2ab24fa273708 : SUCCESS in 1h 34m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e65d0e2ab0d4ff8b50d1c6f57aa671b : SUCCESS in 51m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/172d40ead2db45ec994f072b663a8dda : SUCCESS in 52m 39s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/24515db3c9274cbb978833d3e2663694 : SUCCESS in 1h 05m 06s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/645dc67acc8b4a318647cdb739d12e25 : SUCCESS in 52m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/22b2736a77484120a24e985c5dd4b543 : SUCCESS in 1h 09m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/43b61e10acc24ecc9a9a682df04ab78e : SUCCESS in 1h 03m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bc0426d1e38b4f7fa6072b0c1ee4a5b6 : SUCCESS in 54m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/91f8a4a099a0473dbc96c3f51178dc31 : SUCCESS in 56m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f4a71e6889d49068530c8a98d000853 : SUCCESS in 1h 07m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3147f828a9e34863b5dc4154bb7b18b2 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d0539595cfa1830f80d5f71baeee9b874bbfdc73","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-25 14:17:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5c8ab74f9504b4a5229d4b23d8da3c2cbf88079f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-11-25 14:31:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ab9a451a9abbeb61f4b5ba54427291a79fe1709f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-23 06:20:34.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"af2637f77404c2d15f6987d949f6811f6624c9b2","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-23 06:23:22.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"28b000724430c414156eb3f6d9bb5332edae83c5","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-23 06:24:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eebbe9fcd07adfcc1f37e78857b523d09c6ab0a1","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-23 06:26:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40ad1961bde80af1bc1ecec415a59ca964bee871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 08:40:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd7706c8642c46979a42db0aacc596b7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a76feef3c3074b46a68e2d2e4d131d66 : SUCCESS in 1h 46m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99617fbd4b524d39a63ab3d218122c7a : SUCCESS in 1h 36m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d10946e9178e4b6fbcfdde5fdea69def : SUCCESS in 1h 41m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5d1053a9da474e2c8a4f222749773d01 : SUCCESS in 2h 01m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f76ad36ea5b34dfd81580672820a8edc : SUCCESS in 1h 37m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f998312a42c48b49f1222c11e6c9bb2 : SUCCESS in 29m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3df58eb68a004f47bbd29f6aba334d19 : SUCCESS in 41m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b02d9028fe4f21a045de2e90c863c3 : SUCCESS in 13m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5c0daf80b6d4e92b11044c0310df6c0 : SUCCESS in 33m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d543c3c44fe4695a7345ba500160d65 : SUCCESS in 33m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e92dc694488d420f9caa1dec4865a271 : SUCCESS in 28m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ef5647cac644d5a06f22669abdeef7 : SUCCESS in 10m 42s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/13241e87ce4b43048c9adc3187a98e7c : SUCCESS in 23m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6c991f836414bad88fe3271e517e5c4 : SUCCESS in 1h 00m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/21235f7786d241bcb89551732ac5a17d : FAILURE in 1h 36m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/86e0d29bea5d4020875999af1897a0d3 : SUCCESS in 1h 03m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/60971c0edfa14bc1b2a0068e20f5f96a : FAILURE in 22m 34s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/72ab1b66fe814b7a9ed5b7476ffd3252 : FAILURE in 22m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/478d0ebc765541efada5ca9dc9547929 : SUCCESS in 1h 06m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ad5ab45a816404b94cafde3f9e1bb7a : SUCCESS in 1h 31m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e79307469e554d2a8f76cc9e648ecd53 : SUCCESS in 1h 26m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c76a5695356b4dfaaf5a917528fb315c : SUCCESS in 1h 10m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8336534079c1428f8054a4b9afae1434 : SUCCESS in 1h 51m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/acf92d0265c5473c9379bb55a08ca3c9 : SUCCESS in 2h 11m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b13ad891db048fabe5947417e507aa4 : SUCCESS in 1h 25m 35s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3dea3b0e741746b1a9323684cc351096 : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":3},{"id":"6f7647fd365ecfda0528fc8c6f617b83ec36bccb","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-24 06:47:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83a77d569e1b320ac344aa706f4019593b857788","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 07:07:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f46dec6abaa44dc1b733647bb90b12d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/417363b83dcd4ac5a175f21ff1afe73e : SUCCESS in 18m 19s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b93f8e45f694535b96bbb96536cd8f7 : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dcfe3fc844b5406ba30de007ad97d99a : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e674f7c88336b4898d1d8ab9b981b08eb67e1315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 08:41:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76a9354eb4514ff8923ff69b162f1d27\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/17438b9cd1fb450bb8f75ab149e7d940 : SUCCESS in 1h 31m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/10049d8666484032a82bb2329dbcd155 : SUCCESS in 1h 19m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6b0bd1290ec4ea9b3843f39afbda7d5 : SUCCESS in 1h 13m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2635dbc45d2f4e59b3c071524511a7d3 : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e8f97159849341a7ab3883ed094412d7 : SUCCESS in 1h 16m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6ac5470b07b49d4a3e09a50ea6514d1 : SUCCESS in 25m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c62db9f0bb7e437bbeccd5fefff4b087 : SUCCESS in 40m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c1787a4ffd493e9f14f441811074ba : SUCCESS in 14m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45f6936d91544891bf214648d4812a52 : SUCCESS in 24m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dcd314e3d6240c8bc74390218cf5d72 : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14d1920796c04f4eab02e841d6e7ec34 : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35898aea5aa14552bbe4f229c13c8a0b : SUCCESS in 10m 13s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/323569d8f0d14bf3a86ee2ca31ce850e : SUCCESS in 24m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec8898ce1a2c4f3eb6e3f7bd5a67e804 : FAILURE in 56m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebc5237e0b344b20aa06a7f062174b6d : SUCCESS in 1h 45m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca62a31e72c1449abf5db7acbb9bb47e : FAILURE in 44m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f69ed3797e614f7e8d71d2041a5f775b : FAILURE in 21m 02s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6b894dcf44f140ce95d5fa105a4dc43a : FAILURE in 21m 04s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c85e1a3ba8734ecbaa5584e78c21f4d6 : SUCCESS in 52m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2d048acb321248ea862e0b73885a19da : SUCCESS in 1h 26m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/109ed6e22c674e899a21e15c743d97f4 : SUCCESS in 1h 15m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b7616326f8b9441cbe07fb518b763c38 : SUCCESS in 1h 06m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/485764bafdcf4987b6a0ed7404b22ba4 : SUCCESS in 1h 48m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7e1a76b96aa40a58ef1c5e762b8dbb5 : SUCCESS in 1h 32m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d7683437daed4953a18d8bc1529154f3 : SUCCESS in 1h 28m 09s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/185726aee84640d2b3aa90093752b92d : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":3},{"id":"d3b1d7048dd77b540923aa5dc025acd55b577d9d","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 02:06:05.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9074eebb5ad342f367cc5a94c23ea88faadd8cc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 03:20:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5610bc025eea4709b73f2e3d7e8de815\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f226682f344c4bd1b2a61ac142bfdaf7 : SUCCESS in 1h 06m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f993e925f4344e99b34067663482c6fb : TIMED_OUT in 1h 01m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/675630ae47fd472f802a2546abafdce7 : SUCCESS in 1h 06m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e36d01ae4317b307d5eb3d138ab07f7fe5fd56d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 04:54:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18dfa9397bbc451b80b253cdb79dbeba\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c0abc981eefe43658cbfb04f15950bf5 : SUCCESS in 1h 23m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b14e44a1f944477d8f7ca458a19f38dc : SUCCESS in 1h 16m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0bf59a4b865a471085bdd2965a7dc988 : SUCCESS in 1h 07m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a3a4165f2f424263bcd51d6416d451c8 : SUCCESS in 1h 07m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0aeb7e04a98b47e4a423ab79780d5d76 : SUCCESS in 1h 28m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ed0ad68855d743aa9c500d835898ae32 : SUCCESS in 26m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fca558117a64a6b93aed43f74c9b977 : SUCCESS in 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7449953ff2469fb74f5bac601316a7 : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b449f587e51490798d1e0d07becd57f : SUCCESS in 28m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/029c287e375f47e18d7fb9b4afa3d859 : SUCCESS in 36m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a854807436294f01b753c49169db06ef : SUCCESS in 39m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1bf169b86e74de6af8ae515a49b57d3 : SUCCESS in 12m 23s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cc8b6288cdcf4dc5800374d459134ce0 : SUCCESS in 29m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85af32ecd85a4285bc5f792697c53cab : FAILURE in 58m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d45737eeba9e48ff8ba67e84d8d0e8ea : SUCCESS in 1h 42m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ccd28c5252744c0bee53c67ef16df8b : SUCCESS in 57m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/60f70ac21737418397aa20bd8569e8ff : FAILURE in 35m 33s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e98b68411cbd407ca72efd33efa26632 : FAILURE in 26m 47s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3d765bd9fbc6496cad54586238e127de : SUCCESS in 1h 19m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54af6eef6b8c48fd800e89a1c0734981 : SUCCESS in 1h 10m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/118a04795e9648ceb82fb771bf852581 : SUCCESS in 1h 20m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e4c1da220c314badafbc2606753054ab : SUCCESS in 1h 12m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/584f2f69505648558f092fdc079d14ff : SUCCESS in 1h 44m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/624a4b6f84ef433bb57157ccf18ef847 : SUCCESS in 2h 07m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/13f4a8e47fa442b8a586786ff587f095 : SUCCESS in 1h 27m 29s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3d461503164e4b3eaa5ebf5c54cb95be : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b480e334b0ac10a4e3b404aab4106071bfae7294","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 06:20:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6fa0e0ea2df51c0b52e9819e34611391c1a22bf2","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 06:21:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"243c3804b25d188960e021fc2fd6b63b4f79c4a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-28 06:49:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e439120ffdf9fd256d0fd0b0174e49597006d887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-28 06:59:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"766f859b00827c54759cb447c652a17025c2e88d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 07:41:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56b84d08f7af4afbbf320da23107db04\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5375cbc8bc8e456f9350634ee282b6ff : SUCCESS in 39m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"89d4ee61039aa911d4d6dbbc7e8604446085abfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 08:05:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e2486a89994bba86482c641eeb00da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6cf00b2f9b04c1ab6369248672b126c : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/639d9b91c437448cb5c2daf0f604ff46 : SUCCESS in 25m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d750e94210ee4a568f0f67d81827ca3f : SUCCESS in 10m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/715396bf6a664343adc683fb68682365 : SUCCESS in 1h 03m 48s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/060a102112de4bd99565e365dbb7c7a6 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":5},{"id":"8ae08eedf2f8aae448340265972206fd6c67048d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-04 13:00:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"405d54d583bbdc96a42770ed83f4c49f0d350c13","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-13 09:57:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5c44af0b3474641226ad149082316773d4608fc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 09:58:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"484d2a5516ee0ae382d14335bcaa99fac15568d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 13:38:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ee0da93639741fb82ac404eb8c607d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95479235b40842ff97e306a106e0a152 : SUCCESS in 12m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5f42d0ef6fe458d89e724fb6017319f : SUCCESS in 24m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f60f0c744a94593a0ad3275c38a4238 : SUCCESS in 9m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/481e14f45b804ecc82d0d3b13dbd9d9c : SUCCESS in 1h 01m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1a1d085c98ee4c8b93458a0c8a08d7cd : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":5},{"id":"a8ded36ceaac4746837a6433fc57760ce1053f83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 13:38:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"af8886780e64005c8cceed3687f3537e18ac1341","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 13:40:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1596451ee29141529d7938f23d2cc5b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2755bd8a6b5476993da7e660cbdf386 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"33c4a2d97eeabe812598d651bde60556e2127072","revisions":{"8300ad7b9074951b9704b5a21f598dd4dbcdda24":{"kind":"REWORK","_number":1,"created":"2022-10-31 06:54:23.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/13/863013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/863013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/863013/1"}}},"commit":{"parents":[{"commit":"da8dfcd613453ae67d951b5f5a309c72147f53b6","subject":"Merge \"[Trunk] Update the trunk status with the parent status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da8dfcd613453ae67d951b5f5a309c72147f53b6"}]}],"author":{"name":"WeiLingfei","email":"weilingfei@uniontech.com","date":"2022-10-31 06:49:40.000000000","tz":480},"committer":{"name":"WeiLingfei","email":"weilingfei@uniontech.com","date":"2022-10-31 06:49:40.000000000","tz":480},"subject":"Update url and package name","message":"Update url and package name\n\nThe package name and url need to update.\n\n- Update package form \"python-neutron-ovn-migration-tool\" and\n\"python-neutron\" to \"openstack-neutron-ovn-migration-tool\" and\n\"openstack-neutron\".\n- Update url form \"github\" to \"opendev\"\n\nChange-Id: I41ba7d6929b28317622dbf868f265eab4c6fd84e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8300ad7b9074951b9704b5a21f598dd4dbcdda24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8300ad7b9074951b9704b5a21f598dd4dbcdda24"}]},"branch":"refs/heads/master"},"ec6eb5494966df8198ead4408cf6a523a1a23c1f":{"kind":"REWORK","_number":2,"created":"2022-11-18 09:36:33.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/13/863013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/863013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/863013/2"}}},"commit":{"parents":[{"commit":"da8dfcd613453ae67d951b5f5a309c72147f53b6","subject":"Merge \"[Trunk] Update the trunk status with the parent status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da8dfcd613453ae67d951b5f5a309c72147f53b6"}]}],"author":{"name":"WeiLingfei","email":"weilingfei@uniontech.com","date":"2022-10-31 06:49:40.000000000","tz":480},"committer":{"name":"weilingfei","email":"weilingfei@uniontech.com","date":"2022-11-18 09:36:08.000000000","tz":480},"subject":"Update url and package name","message":"Update url and package name\n\nThe package name and url need to update.\n\n- Update package from \"python-neutron-ovn-migration-tool\" and\n\"python-neutron\" to \"openstack-neutron-ovn-migration-tool\" and\n\"openstack-neutron\".\n- Update url from \"github\" to \"opendev\"\n\nChange-Id: I41ba7d6929b28317622dbf868f265eab4c6fd84e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec6eb5494966df8198ead4408cf6a523a1a23c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec6eb5494966df8198ead4408cf6a523a1a23c1f"}]},"branch":"refs/heads/master"},"f02e2eb3f37ed8b4708646e86e24413ce576966f":{"kind":"REWORK","_number":3,"created":"2023-03-23 06:20:34.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/13/863013/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/863013/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/863013/3"}}},"commit":{"parents":[{"commit":"da8dfcd613453ae67d951b5f5a309c72147f53b6","subject":"Merge \"[Trunk] Update the trunk status with the parent status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da8dfcd613453ae67d951b5f5a309c72147f53b6"}]}],"author":{"name":"WeiLingfei","email":"weilingfei@uniontech.com","date":"2022-10-31 06:49:40.000000000","tz":480},"committer":{"name":"weilingfei","email":"weilingfei@uniontech.com","date":"2023-03-23 06:20:34.000000000","tz":0},"subject":"Update url and package name","message":"Update url and package name\n\nThe package name and url need to update.\n\n- Update package from \"python-neutron-ovn-migration-tool\" and\n\"python-neutron\" to \"openstack-neutron-ovn-migration-tool\" and\n\"openstack-neutron\".\n- Update url from \"github\" to \"opendev\"\n\nChange-Id: I41ba7d6929b28317622dbf868f265eab4c6fd84e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f02e2eb3f37ed8b4708646e86e24413ce576966f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f02e2eb3f37ed8b4708646e86e24413ce576966f"}]},"branch":"refs/heads/master"},"29131b013363b679feb88316a6cc1573f2bc6d17":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-28 06:49:04.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/13/863013/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/863013/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/863013/4"}}},"commit":{"parents":[{"commit":"55a0d23fe312764e1f92b63fae2d1add942a1ced","subject":"Add plugin.spec to irrelevant-files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55a0d23fe312764e1f92b63fae2d1add942a1ced"}]}],"author":{"name":"WeiLingfei","email":"weilingfei@uniontech.com","date":"2022-10-31 06:49:40.000000000","tz":480},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-03-28 06:49:04.000000000","tz":0},"subject":"Update url and package name","message":"Update url and package name\n\nThe package name and url need to update.\n\n- Update package from \"python-neutron-ovn-migration-tool\" and\n\"python-neutron\" to \"openstack-neutron-ovn-migration-tool\" and\n\"openstack-neutron\".\n- Update url from \"github\" to \"opendev\"\n\nChange-Id: I41ba7d6929b28317622dbf868f265eab4c6fd84e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29131b013363b679feb88316a6cc1573f2bc6d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29131b013363b679feb88316a6cc1573f2bc6d17"}]},"branch":"refs/heads/master","description":"Rebase"},"33c4a2d97eeabe812598d651bde60556e2127072":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-28 06:59:20.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/13/863013/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/863013/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/863013/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/863013/5"}}},"commit":{"parents":[{"commit":"3cc28a004a33e9021be2907cf08e27857e6e085e","subject":"Add plugin.spec to irrelevant-files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cc28a004a33e9021be2907cf08e27857e6e085e"}]}],"author":{"name":"WeiLingfei","email":"weilingfei@uniontech.com","date":"2022-10-31 06:49:40.000000000","tz":480},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-03-28 06:59:20.000000000","tz":0},"subject":"Update url and package name","message":"Update url and package name\n\nThe package name and url need to update.\n\n- Update package from \"python-neutron-ovn-migration-tool\" and\n\"python-neutron\" to \"openstack-neutron-ovn-migration-tool\" and\n\"openstack-neutron\".\n- Update url from \"github\" to \"opendev\"\n\nChange-Id: I41ba7d6929b28317622dbf868f265eab4c6fd84e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33c4a2d97eeabe812598d651bde60556e2127072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33c4a2d97eeabe812598d651bde60556e2127072"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
