)]}'
{"id":"openstack%2Fneutron~863271","triplet_id":"openstack%2Fneutron~stable%2Fzed~I7cec8b53e72e7abf34012906e6adfecf079525af","project":"openstack/neutron","branch":"stable/zed","topic":"bug/1987666-stable/zed","attention_set":{},"removed_from_attention_set":{"8213":{"account":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"last_update":"2022-11-14 11:15:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7cec8b53e72e7abf34012906e6adfecf079525af","subject":"Check subnet overlapping after add router interface","status":"MERGED","created":"2022-11-02 11:06:58.000000000","updated":"2022-11-14 11:16:55.000000000","submitted":"2022-11-14 11:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863271-bug/1987666-stable/zed","cherry_pick_of_change":859143,"cherry_pick_of_patch_set":12,"meta_rev_id":"443aa09369e126c36dd7a5886d643325cd7b8884","_number":863271,"virtual_id_number":863271,"owner":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 11:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-11-11 15:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-11-04 10:55:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-11-03 08:17:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-11-11 15:48:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 11:06:58.000000000","updated_by":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"CC"},{"updated":"2022-11-02 12:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-02 12:44:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 14:11:53.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-11-03 08:17:20.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-11-04 10:55:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-11-11 15:48:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"9872043664cbe77b1fce2e6f2de2fb86af607e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2022-11-02 11:06:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1c67336bde103f8ad7fc38a7e37275bdc2f83938","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 12:03:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f80364e6b0d34a399a7a496df0c6f038\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f20072283ec941b6a671714babcfb819 : SUCCESS in 50m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e13e116de43f481ba0add8abb8698ed7 : SUCCESS in 47m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bf86c88c9c2b2c8e71e55f71bcbf1834a1c5466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 12:44:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b15f4ecc3704e8ba92e2648c3693888\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/327b9b48f3d643edab0ad0c715f2b4fa : SUCCESS in 1h 27m 04s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/92b8042ec10544f5992c50989b5e9298 : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/a053383a135c4b43baaed4da3d955ab5 : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/9d7abf2012cc49b38eec185ab1b49812 : SUCCESS in 1h 22m 35s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/835dd4316bf149258c330a187fa033c8 : SUCCESS in 44m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19f8f636c3e1444482e7409c0d774c09 : SUCCESS in 29m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da82052f2f2465a85cbbd74bfd5e71d : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/708e6f5ed9e140eea484d80a75fb0d83 : SUCCESS in 28m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36742d73214643aea08612a33d790d12 : SUCCESS in 30m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7caab320f0114e6f91cbc5001e32e968 : SUCCESS in 29m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25fa92e1ad01448e8629b2cce1a5dd77 : SUCCESS in 8m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdee979afa6047438bfe07d65d07055b : SUCCESS in 46m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc4ba6d1f8e347c2b0deb8fe7d39ddda : SUCCESS in 1h 31m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5454760eeb064dbb83afb8eff02f5186 : SUCCESS in 1h 03m 24s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/804d0420a885464d80235f2e7734bbbe : SUCCESS in 1h 02m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/166c957dd6e94ce08673c92d73eef3e8 : SUCCESS in 57m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/23e1f8745b594c79948661d9c9047cb2 : SUCCESS in 1h 15m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8d5d9830cc8c48528d5ecbbbf6401c99 : SUCCESS in 1h 16m 04s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaa7c28f71df430f90f43642f09fb756 : SUCCESS in 55m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/160bc0bbd52445b5b33507abe9d2178e : SUCCESS in 1h 06m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"401dd633be0635a804d57ad049d4c4f5c6367289","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-11-02 14:11:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/863271/1 : SUCCESS in 2h 40m 52s","accounts_in_message":[],"_revision_number":1},{"id":"833e9be30f7f88c7ad9d028828524eb9b9ff2d64","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-11-03 08:17:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ca885297f90798d82b16b620d73eb25b8716359","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-04 10:55:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9de75a00d314457a988631814f292bb4bf3f5c39","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-11 15:48:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13971dacb717036a6495446ad55b7bf39c4c41cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 15:48:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b99a9f5d99f2092ab359f60ba9cf73892f1bc60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 17:27:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18ab0c94b2ab491799ca5cc247fd47c4\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/d8b506dd13044d44914ff5077292a4bd : SUCCESS in 1h 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713cc431116749baae2b84e0b99b5bc1 : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1681ab92edc748ecb88d29b31e403ff2 : SUCCESS in 24m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ffd1560c4464a7a8989539242cb0d2d : SUCCESS in 22m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/082b3911221644c6b581d368486302b3 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/384b8f5f7dc545898601a3d3bb216b9b : POST_FAILURE in 43m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a48a05c3ac94d679c720bd841281498 : SUCCESS in 1h 34m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c732cbbc4c004f7cbfcbda7c1ef436c3 : SUCCESS in 1h 06m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/aacb9b089aef48c0bdf2594aa0eb7067 : SUCCESS in 53m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f9a07209ad3241b9a414f4eb8eec75c7 : SUCCESS in 51m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3caaaf91e2b3457695ed4786420e8a48 : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b4295226f9b7482008a2c090c3fadde0b2e8c464","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-11-11 17:41:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4090a3922364d89aba905aa75f825e2c6ade1e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 18:20:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/722410ad2ebe4ec983d621f561afbebf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e838ca86506437c8ae72e51b01a01bf : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2b1f66ec5864460bef6615f3a1a585e : SUCCESS in 34m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ab316c76d6748cab059d5f7787010c80be54b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 19:16:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b44cf3f93d4ca6ad595e32f6e5f347\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/7b1c89a2fd484e1ebda6bd7731d061b0 : SUCCESS in 1h 17m 26s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/ad556932d60045629df4a828e438a3f1 : SUCCESS in 1h 32m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/53c8dde364254baf93ae590812b5f795 : SUCCESS in 1h 12m 38s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/af406df6835642228ccbd13b7039544e : SUCCESS in 1h 23m 01s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/7f73c98f747346c29492d7fda17f5ef9 : SUCCESS in 29m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a70c98b4bc4e4a478a372a7fc0d548e8 : SUCCESS in 26m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a8a1f8930864e0898c2006d7a63a8d5 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/001e1792038a43c3bd4a72d212aad10d : SUCCESS in 21m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5787c200837749ad877925134622bd12 : SUCCESS in 23m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b85a5bc8df6e47f08dbd506d53560858 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb43ab76f0745868bcd2bf6ddb2fc74 : SUCCESS in 8m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a1aadb4c4f14ae1be675cb73db1ff20 : SUCCESS in 44m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/de0ad1ba667a4ffcb2ad6cfcbe9115e7 : SUCCESS in 1h 25m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/69b16cef5bbe407f8338f2a77f18f67f : SUCCESS in 59m 09s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7936256da934950a1c72dd79e672b15 : SUCCESS in 39m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/53aa38422d014f3ba0ea913175861a19 : SUCCESS in 54m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/75b6bc1ee9154f79b5d9419dc9feb100 : SUCCESS in 1h 09m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8fd0c19ec532462f8090a139d5ed1477 : SUCCESS in 1h 07m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a39e942ac2143f5bf6693e09cd7d935 : SUCCESS in 44m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/592f2a2a2d3d42929592b2a2be33b1c1 : SUCCESS in 1h 04m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adc3c3a2be26f788f2f360b5948279d0cbea423f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 19:17:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43a701aa03f6cc91ecb4a81f40c88e1ef805e8fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 20:57:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdd063061f854b9e8e289c09f79de71f\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/7a6226125d1e49b6b8ff0fad651a0b8c : SUCCESS in 56m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82195719ef0a4dbeb67b208f092c86da : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12c60ed840f34021a6fcd719977da661 : SUCCESS in 20m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fbac9cf7fc145b7978902ada83dd3c8 : SUCCESS in 20m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f84edd6864b44aca2615d8109033e7a : SUCCESS in 8m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad456d0418da4aeab497430021ef0214 : FAILURE in 37m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f92cacbb65624aeab206ab75c8526130 : SUCCESS in 1h 38m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a4eb7b201f244705b0f03470fb23633f : SUCCESS in 55m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/803f8f37a8c448288822d4078aa76e3f : SUCCESS in 52m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/917641ec7204445f951359a233218df7 : SUCCESS in 39m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aad6101baa194f3597795418b302b168 : SUCCESS in 55m 41s","accounts_in_message":[],"_revision_number":1},{"id":"751a298bc89d791fe7008db8924dd59fd31a8011","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-12 00:11:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fcfbe170b3683de7fcacc7d25109d322f45b935","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-12 00:52:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6eaafa4d72744549996f25deae94a87\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26fe66e472bb412886e808cc6f97b4f0 : SUCCESS in 33m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14b253012c3444d4b48a6eb7b67c3568 : SUCCESS in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"900e3ab5d5ebbc3d859745f6b31d10e7dace9860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-12 02:08:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a75ce54c1b5f47e282bc1ad23ad118e1\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/c3b0f535a69d4dd69653691a9bfcd356 : SUCCESS in 1h 19m 46s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/a800886869a845a983f8e9e49d3ea0be : SUCCESS in 1h 09m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/7e6315c27502498daae3f211b86efd55 : SUCCESS in 1h 00m 56s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/fac7a90395a14516ab137be2b11ad0dc : SUCCESS in 1h 04m 27s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/9aa63e21a5e546c3a56fccdb78843639 : SUCCESS in 33m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99ad096f48024ba2b015e0894b677901 : SUCCESS in 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e570c2ec94432a941aaf272dbd1561 : SUCCESS in 12m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90f77cd5c97e4d75a1fda7218ec8e468 : SUCCESS in 21m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccee5da9e2a44286826a71d86936c33b : SUCCESS in 21m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc54fd1049554ea09d3f334c9f6439f6 : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb39fa958f74a658362e19053c3e970 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a07d0bd463a472bac24bcec208dc259 : SUCCESS in 43m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a54d1c206f804b1bac8b613a624c2d9f : SUCCESS in 1h 35m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3a59e3268cf04491931aa09a26c35075 : RETRY_LIMIT in 13s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/69a5eb28b23140038fc41555fb52709d : SUCCESS in 1h 02m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/99cc4159907241f0aba25fcc507e1880 : SUCCESS in 51m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e3e7617bf7e845cf944a06c98a5d1c8b : SUCCESS in 1h 11m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/68d7c3a2a74c42749e570613d433b1ae : SUCCESS in 1h 12m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd307608e40a4849a9d8011edeafbcfc : SUCCESS in 52m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e12a9dca5a6d4613a0a5f47ca765bcb1 : SUCCESS in 1h 04m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbd1e32d413a2d6c341f8268c3a68e994935ca0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-12 02:13:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0416c0120cf35919df33522a8c09d4e303b81981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-12 03:34:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53660c3ae5b84c96a5a81e61ce9177cf\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/392dab7ed857487bbee274a70c470756 : SUCCESS in 1h 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86ae339e5cf4d489a2c862bb49874d8 : RETRY_LIMIT in 6s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eee2152984af44aeadfc7d5967387dc3 : SUCCESS in 23m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fdbf25cecf44793bbe878a3f07dbc87 : SUCCESS in 24m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23dd6bf780be47a980e12c23b7841959 : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f22cb94559040ddbb15207dea0d498b : FAILURE in 42m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0ddc775b1ef4983a1cdde8c981cb167 : RETRY_LIMIT in 3s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e50de46b43ed433ab84f14cd748d05e4 : SUCCESS in 1h 11m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccc277c142a341a8b43be1871159c7db : RETRY_LIMIT in 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9bbbae40d3184b99a7d24ce01256caa3 : SUCCESS in 58m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09f8fed1bbc04aeaa11e090329589782 : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":1},{"id":"eda985fc1ec8499d1e205d5908d19c3ebbcb5f4b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-13 20:35:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d871955dbd746696460f719bca4dbc7267a3b3a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-13 21:14:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e174c05d70a4f03aab366389ffffad9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/810e657c1e99456b98a68c65fba5dd85 : SUCCESS in 33m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67ff6f9f4005484e9d2545f1f6acc5e2 : SUCCESS in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea5d2a6a8afc20930d602e9af1b30be529f2ae5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-13 22:19:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4066b7c66a3c4e64afa8ff72171c387c\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/619df967bbb746dcbbdda8e7d3082e6e : FAILURE in 1h 17m 12s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/8e37d998289e47a9a1f28ab6ecf26fce : SUCCESS in 1h 04m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/1f0f34833c31412fab78de72d1c070b8 : SUCCESS in 55m 36s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/1e00f00fc1df41d59c1fde3f3a86864a : SUCCESS in 1h 16m 25s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/0000cf995a8a4858b9d301d7ae3e1d5a : SUCCESS in 26m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16ea92ea03b546bf8b16d8748d18c137 : SUCCESS in 39m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92beb77cdc4c49f79d5f74509060423c : SUCCESS in 13m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/360925757608458789a43d388091b11d : SUCCESS in 28m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d30c594d9d4649ed8f6222f8504bf2d2 : SUCCESS in 28m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5226db222c9e44e096da3f8043aeff60 : SUCCESS in 31m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338918ed31b6466cbcb6014a63b2a6ee : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e733747fc27c410892409decfe8760d4 : FAILURE in 50m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a23c148c8a54e14a4387ca36f3e4eaf : SUCCESS in 1h 33m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d5f0f9abe530445d96ea9570223fc55a : SUCCESS in 1h 12m 47s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f22f7e74df194d5f8f0b09c3704c6db3 : SUCCESS in 1h 01m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1a82c747db584223a464f6d8abd51853 : SUCCESS in 1h 05m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/98ddb8843b6f4e59aaf50fec5b08b56b : SUCCESS in 1h 36m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8df4555202c94b8dab940ecbdb58ee8e : SUCCESS in 1h 22m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55f989978c694be283f0202caa933336 : SUCCESS in 1h 12m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6878c09abd5b4432968b0b6220a2ef82 : SUCCESS in 1h 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8efb8a8f109b1060301c35c87a71a3d0faedd1b4","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-11-14 07:51:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2ee6008c8c73750ce183a3027c9cd87f10fd080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 08:28:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45114707d8442a892fe2a95989ae5a5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/851c0aac023b4733a51127126894a55b : SUCCESS in 32m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec46a954c95143e38142c455d40c94bc : SUCCESS in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f505bf3c842471cd49e0c75b8cf5a24f8b1b3e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 09:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cba0ffb82fff4519926d0c111d8f5ee8\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/96b95d43b21540a89142250ac11e0c63 : SUCCESS in 1h 09m 28s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/41abec9f1cf44bd889096e5999bea85c : SUCCESS in 1h 04m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/37bcc43eec204699b3d699cb0cff7c04 : SUCCESS in 1h 01m 47s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/b84a4d23bfcd449ab9a9ea6908bfca86 : SUCCESS in 1h 04m 16s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/c7359262daea42d48f34dfb9ee21b5e1 : SUCCESS in 29m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4b5c8bb266740508c14ecb3a8e879d4 : SUCCESS in 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df758ed5091a434c968cdc65eed64031 : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed7453ced03f4158ab1b88b2493b947e : SUCCESS in 21m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e2f6a98dfc5498abaa080809cb0a023 : SUCCESS in 27m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b62e2cf3251e476e88d6f2225500e6a9 : SUCCESS in 29m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493a53d57a864d85919598dc6737e674 : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/391088a6114e4ab3a53680fe5b9dcf3e : SUCCESS in 53m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/29912ff74c4e4efc91a60679a712fc9b : SUCCESS in 1h 36m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c68d90e508f46b3a036bc64c5908f16 : SUCCESS in 1h 31m 45s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3f56dea8aca446da22c8ba10827f132 : SUCCESS in 40m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a5a2a357d6984cc48237f761ae305c2f : SUCCESS in 51m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c3705b613b3042aab20f4678914d2082 : SUCCESS in 1h 25m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e58e846d98164384bcbcd134bf3fdedf : SUCCESS in 1h 00m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ab4a29ee07d43fba1f83028d329859d : SUCCESS in 52m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/233298166c2240fea9d5519ceceaec20 : SUCCESS in 1h 01m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc6fd0f6abf6276bbd237787e03d27c23d4828cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 09:33:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d3319a47dffd450308ff9fe39143718e253dab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 11:15:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7c5da9332e54719b2c032e9912ab78a\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/160dac115c9449369d4cab191c98b30d : SUCCESS in 1h 05m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725444819ffc4850bd481963550fd6a7 : SUCCESS in 12m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6c776f8e05244f695122912f8027c85 : SUCCESS in 29m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85b7dc1722ff4a4dbbd5270d69f54fe5 : SUCCESS in 26m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42bc732d3a4a49e2a9d70f5e214c3304 : SUCCESS in 9m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8bfad4cc0124077a10d1e00715194cf : SUCCESS in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/17e2daef5c1c4ef093b8ef6bcdf7e3e4 : SUCCESS in 1h 37m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3fc521ddc30340b28c80d54569ac93e0 : SUCCESS in 1h 22m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7783a41a8144f65909bf75a263e16f9 : SUCCESS in 1h 04m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d075619a37ec4f718db0fc27e5a41877 : SUCCESS in 1h 01m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96b173bbd88043638226cf71f7500a49 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7199ff106c4ad17b34b46b22a8b9bec0007e68f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 11:15:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"443aa09369e126c36dd7a5886d643325cd7b8884","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 11:16:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82d7066aee8c400bb7b0dd3e0cd1bbc8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0bb91b6b734b8b8b056383bb1eb0c2 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71801fba9143b4e341037695c9a95a6ddc722766","revisions":{"71801fba9143b4e341037695c9a95a6ddc722766":{"kind":"REWORK","_number":1,"created":"2022-11-02 11:06:58.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/71/863271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/863271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/863271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/863271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/863271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/863271/1"}}},"commit":{"parents":[{"commit":"8dd2dea65d3c90ba7be6907142b7ccbd05ea4238","subject":"[OVN] Set the default OVN metadata worker number to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8dd2dea65d3c90ba7be6907142b7ccbd05ea4238"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2022-09-23 18:53:09.000000000","tz":120},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2022-11-02 11:06:58.000000000","tz":0},"subject":"Check subnet overlapping after add router interface","message":"Check subnet overlapping after add router interface\n\nWhen simultaneous attempts are made to add an interface\nto the same router including overlapping networks in cidrs,\nboth attempts are successful. There is a check to avoid this\noverlap but is performed when creating the network interface\nand it is done over the ports already attached to the router,\nso at this moment the check is not able to detect the\noverlapping. Furthermore, the create_port operation over the\nML2 plugin  must be executed in isolated transactions, so\ntrying to control the execution context or adding additional\nsteps to the transaction is not feasible.\n\nThis patch checks once the RouterPort is created on the\nneutron database if there is more than one overlapping port,\ntriggering in that case the exception that will remove the\nthe culprit of overlapping.\n\nCloses-Bug: #1987666\nChange-Id: I7cec8b53e72e7abf34012906e6adfecf079525af\n(cherry picked from commit 1abb77d7a63cde2aa9640351f663870c14430919)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71801fba9143b4e341037695c9a95a6ddc722766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71801fba9143b4e341037695c9a95a6ddc722766"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
