)]}'
{"id":"openstack%2Fneutron~863280","triplet_id":"openstack%2Fneutron~stable%2Fzed~I54c8fd4d065ae537f396408df16832b158ee8998","project":"openstack/neutron","branch":"stable/zed","topic":"bug/1990229-stable/zed","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-14 14:20:42.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"29074":{"account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"last_update":"2022-11-14 13:32:02.000000000","reason":"\u003cGERRIT_ACCOUNT_29074\u003e replied on the change","reason_account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"}},"35462":{"account":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"last_update":"2022-11-18 10:31:49.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-11-18 08:59:23.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I54c8fd4d065ae537f396408df16832b158ee8998","subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","status":"MERGED","created":"2022-11-02 11:57:08.000000000","updated":"2022-11-18 10:33:28.000000000","submitted":"2022-11-18 10:31:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863280-bug/1990229-stable/zed","cherry_pick_of_change":860671,"cherry_pick_of_patch_set":2,"meta_rev_id":"b81312bf0a6af3fe4b27f43805f6517017d73323","_number":863280,"virtual_id_number":863280,"owner":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-18 10:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-11-17 17:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-11-14 13:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":2,"date":"2022-11-18 08:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":1,"date":"2022-11-18 08:59:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 11:57:08.000000000","updated_by":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-11-02 12:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-02 13:45:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 13:53:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-11-04 08:36:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-11-14 11:22:26.000000000","updated_by":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"reviewer":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"state":"CC"},{"updated":"2022-11-14 13:32:02.000000000","updated_by":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"reviewer":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"state":"REVIEWER"},{"updated":"2022-11-16 14:29:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-11-18 08:59:23.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"cb8b77c499cc5d43a5593b7cbb2eccdaa3c7a1a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"date":"2022-11-02 11:57:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"29ace7a12f01af8db145de06504bf9f6283b107f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 12:44:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c486ecb0e404476ca3df5e4a44bd045b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d260d863e3de47558a9245c3ee9c3933 : SUCCESS in 44m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dddc2ebf3d954dd99fc8f99979556f1c : SUCCESS in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15f2c9b8bf5fc79022e3448ab326b056f7282800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 13:45:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f182b62cffde4184b5a96f398a6b2e42\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/6839be119818481c9fb47a4a3f1df098 : SUCCESS in 1h 12m 12s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/551c9748d65140cc9f6f81690f695657 : SUCCESS in 36m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/881dabd79f284f85a84f9136f0d83562 : SUCCESS in 35m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c304c605170f4f18a0654e8cb56d2192 : SUCCESS in 12m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be2cd9fa50174a278c91ec2e7297a6bf : SUCCESS in 25m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f3156cf228c4f4bac7a6ba749c7016b : SUCCESS in 31m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41ab1ed63fbb4ac6a2806d58e3f70654 : SUCCESS in 27m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35e0ee317b8c45f3b9b38bf438e85ac2 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cbb287fb1f14a12a63802114c168bdb : SUCCESS in 12m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20ffd61085524d73bee7766a03c41549 : FAILURE in 53m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f84715cb7654e6e825caad438b07f4b : SUCCESS in 1h 41m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16a0101d38f148ca97a3ff98df3b3873 : SUCCESS in 1h 02m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1e92854d10d4732a7a42b6c10a89ec2 : SUCCESS in 1h 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f1aaad175760602515f690373279eba818789c2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-11-02 13:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/863280/1 : SUCCESS in 1h 56m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ce6b82eb4dd224c158c8032d53f5b49df1411514","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-04 08:36:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e551c70a551d75f1a30297de63ad75af9e85e4a","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-14 11:22:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93d5662103ca0ffe43300a61231184b72107fd18","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-14 11:46:00.000000000","message":"Patch Set 1: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0d597c8b8b96c7bba73e2ec45cf687ff2715d8db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 12:31:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d73fe54a161428abf3e723d91f4ee00\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec3d97f8debe48c6829e8d62ad6fca77 : SUCCESS in 38m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55ab344bfe56416da5e9c9d6e5e50e00 : SUCCESS in 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be2882ae8723769d3d16d90b58eca3862a0f70cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 13:29:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f5aa7dc0e044099b8caa2c8f5e0ef34\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/f5383a27735b4571a7393c5aecf38d87 : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/62ea7eecd4294cd09b6e99eb641f5b28 : SUCCESS in 25m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97618f8d408e4728a3f4287448d70b73 : SUCCESS in 42m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617fdb07dd7f48eabfa6146c536ea743 : SUCCESS in 14m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0bb2cf639ea40ef9fdd730767b223a6 : SUCCESS in 26m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e022e026b66045cb98352f6401d31968 : SUCCESS in 28m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c44ac3406f24fea8261375882ce908a : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6913533303d45efbc3c0618816cc48c : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b0722e12324c33b1faa4c03b60bad2 : SUCCESS in 12m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3f58c190cd241d594d79e4a64a72afa : SUCCESS in 59m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b445ad031994d84a101bcaf73e7e506 : SUCCESS in 1h 36m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a94100b5b3a64794adb57b28a8eb896f : SUCCESS in 1h 01m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/670d9bc42da144d685e30c7fd2aec605 : SUCCESS in 1h 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22ccd0b2b8143b6ed1b92d6c6ff4c064d67665cb","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-14 13:32:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"110f337d43b69863252f9ca2f89f1c01035ab4ac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-14 14:20:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe9c58f2e3a96e5d1c3004cbbe4be18df91e0932","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-14 14:21:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"586b32724244c90c66a9c78990086f05db0cdc39","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-14 14:21:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e31699a9a93b337b9bd81fafe62d65ea209fed7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 15:06:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f90216db8cc540ab8af05a9d92bedfe0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a164104a939a4277940d49234d733fb3 : SUCCESS in 41m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0884c39e29a44e7abf49b01df38f5bf : SUCCESS in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2082f3a1b1309ab94def71a281aa9c0b939e8e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 16:04:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f53a3e52c5f34ebcb1e890be33e64f93\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/118d397436c847d6ae4054ce3ba7535f : SUCCESS in 1h 11m 05s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/9b8d04d0654741efa927356cc75a7a12 : SUCCESS in 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5471d70f2ea45ee9d99ea36572873c6 : SUCCESS in 40m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b688aacb10b5462184ef3b42cde3d79e : SUCCESS in 14m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/492739c7682b4eab8204d9b9bdaf8e86 : SUCCESS in 30m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8def942519b487790af05731a6aae8f : SUCCESS in 28m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7374b0e471a04d2391d9077646795af5 : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6ace4284a6448188a6784d8316d0bc : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9e9a7c2d045416795bfb70b79e89d91 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1aa8a2dee96f4c1aa07a16562b2664d1 : SUCCESS in 38m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce5e3ee949614837b1c0478fde102b33 : SUCCESS in 1h 21m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46ca89f8ba0c422ebe24fa0ad3fa2f44 : SUCCESS in 44m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e80dd7a45914707a47c4f1608c6f9db : SUCCESS in 1h 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf6951a2f0643887d2a0cec95eb1527277c45639","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 17:34:22.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbde7fcbd4134be58debd05850b4a6ae\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bb21e92e65c1496f98b206ab6f9a92cb : SUCCESS in 48m 52s\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/c44e324548984db79c31741e13548f0b : SUCCESS in 1h 03m 34s\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ec396b523fd24411a2950846afa3df2a : SUCCESS in 50m 01s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/bdb936a26e0b4332b17234605f48020c : SUCCESS in 53m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/120329df13a542f394509a46cde6c8b6 : SUCCESS in 1h 43m 02s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/e9f86a914a6a4256849dc959c4428220 : SUCCESS in 1h 40m 45s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5ea86f29e77e42ebb7124968652a1896 : SUCCESS in 1h 29m 03s\n- neutron-fullstack-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/99372b7c59de4544a742d690711311b6 : SUCCESS in 1h 38m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/feaa4ab6fcef4985a1ba23852c76e959 : SUCCESS in 1h 02m 21s\n- neutron-ovn-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/eda67dad7a8e442d9253439570e9e3b8 : SUCCESS in 1h 04m 32s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b070005e43a44c428be09d1b2077c654 : SUCCESS in 1h 09m 36s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/f7ad764e5bb546b79e44c7e29cd0f360 : SUCCESS in 1h 11m 04s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/712e8f2320d3411e8b9ccb44c1768b97 : SUCCESS in 47m 11s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6e464d4554314646987d14b2034e25ca : SUCCESS in 1h 45m 23s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/a8bd2bd3052244c1895467b1320bfa2f : SUCCESS in 46m 54s\n- neutron-ovn-tempest-postgres-full https://zuul.opendev.org/t/openstack/build/2b34c2fd649b4dc991098d397fc18d9f : SUCCESS in 1h 03m 26s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/f244fb21ef5f4bcca52bb3ad8fe16a07 : SUCCESS in 44m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f15467d4d2a340358b78fbf0ec2d4278 : SUCCESS in 1h 04m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7a09c18945c14f5fa8538984cc9af1ee : SUCCESS in 23m 46s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/1929c29f869841f7bc535e83739dd48d : SUCCESS in 1h 18m 08s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/510c39df6ef6487882efd138fb702894 : SUCCESS in 1h 41m 39s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/16c3f60b500c41378d1e4e0b336d75b8 : SUCCESS in 42m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5116a66d8fd0402e82c44b9712ab4fef : SUCCESS in 1h 13m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0dbdb253323b48d6a900de55d8ef8fde : SUCCESS in 1h 21m 41s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/54371c66a8524ad2b048950001ffa8ed : TIMED_OUT in 3h 06m 35s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/832ec5478b1742429ddc374c2e81db89 : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b16fd4259ac4e058f350a3daf3ce2936cfc48638","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-16 10:56:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"760195e665bf18a4f6632a66fead84a43dbd044f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-16 14:29:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58cda4842c37e24ecdcdcc8dd5fd41eafca3a6f4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-16 14:31:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2dcdf88c10ba886c9996ed491e451f982a76dd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:16:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a9e94dc4c741a485a46a0d9522f933\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79e0aae2c73c44b494ebbc884df1e1ef : SUCCESS in 43m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa272e59e173482e8174a98c16921e1f : SUCCESS in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8427c83e002e180909ca311af287056858fa1837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:00:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d8d628eeb764ad58189d8628934046e\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/ef72c68b8f844065a4de93af7fb54483 : SUCCESS in 1h 04m 23s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/fa9c2a44104e445785f649d5e92c8247 : SUCCESS in 35m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49bd523e52594c69b6f188fc5e0f4730 : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48cc762d98c04147b4f0f78fa7dd4e27 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfde9c9d2eae4e4f976f8bb94c2a6205 : SUCCESS in 23m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10c64df28c554530b42d0f4335bdbef6 : SUCCESS in 23m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21712c047ba1415c906a29f3747dd47d : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374881ba584640f991c33799265646b0 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9055723a9ef94a8a969ca85981c0b4c4 : SUCCESS in 10m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/05eebe4fcc5f4beea142cb2d54ed120a : SUCCESS in 46m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3316536fbd284d6b87d754bf1e1b1d09 : SUCCESS in 1h 29m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f16c792ebd3b45dcb9db02b02dc6b8db : SUCCESS in 53m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d1f828315a5049b7a861cb111dfed32d : SUCCESS in 1h 05m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e66a4b35fafda2c3fcae5d24277ac2eb92757f44","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:31:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0f485d905540dc94fb72b4d9aa3005\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b69cea7b42fa4e85b6ed49f828e926e4 : SUCCESS in 50m 24s\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/0f6556c9b5d843dfb59e4e42c326a30d : SUCCESS in 47m 37s\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a41fa0b638754a46afcfdf0536047227 : SUCCESS in 53m 02s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/5759502c04f84f4cac388d5ab495f914 : SUCCESS in 48m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d64a24753b2e45a49aa3ed24a41a2aa7 : SUCCESS in 1h 33m 00s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/e74e16341abf41ca80e4c967397d2f3d : SUCCESS in 1h 44m 29s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/639cab807ea04185b3a8650ce583270e : SUCCESS in 1h 36m 31s\n- neutron-fullstack-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/2ec8b3affc4340448c45aea16718c47b : SUCCESS in 1h 30m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/279a574ebc1f4988aa547b6e3aac5413 : SUCCESS in 1h 06m 25s\n- neutron-ovn-grenade-multinode-tick-tick https://zuul.opendev.org/t/openstack/build/6c441f6675dc4d04a24690bb6a5fda76 : SUCCESS in 1h 09m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cf9c651ceeca477eb5c4b42f8b91059a : SUCCESS in 1h 25m 13s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/617f1116fa6c446bad45bda83f3e8923 : SUCCESS in 1h 08m 16s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/82ba6b4f30854cb3b70827ba57d52a5d : SUCCESS in 1h 05m 23s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/06945d6f7b8442b58183f5c402ab04ab : SUCCESS in 1h 47m 54s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/040de2272b8148f19a80364097158365 : SUCCESS in 1h 02m 38s\n- neutron-ovn-tempest-postgres-full https://zuul.opendev.org/t/openstack/build/f0570efd45ef4c808cdb5476ec59ef01 : SUCCESS in 55m 01s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/2b7d2db7a1434aab870e30f06b8626b2 : SUCCESS in 1h 10m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/67f438f1926f414ca91154b852b3e67e : SUCCESS in 1h 12m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/afd259378caa45b8841a1ce76f31d2c9 : SUCCESS in 24m 08s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/08d60c18373e46aaa80b8cf411eeffad : SUCCESS in 1h 11m 02s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/7f0e0949ff364385a135f0c0cec7d9d1 : SUCCESS in 1h 53m 25s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/751ae15453604074ad49f7b7b12d04bb : SUCCESS in 40m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8ba0a59dbe6b49e09eba25d3c42ecfb3 : SUCCESS in 53m 19s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/052f75f51ec34cc2a96d0eb82ad0bdd4 : SUCCESS in 1h 33m 17s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/13e1994d3dce4bfa89076aa09f4d0c28 : SUCCESS in 59m 01s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/0b59892658514f1d8eab8413471e34b9 : SUCCESS in 23m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8c51ca4edac647bbb07cc9c20b027c79 : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7f3eaaf6f6ef5995d0f60b3cf0f165620df91990","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-17 11:32:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e8262a0a8c453f601f98875f4a8247ac594cce2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 17:08:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccb14b2d85aff1b0476d6e31092ced23c3fb3914","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-18 08:59:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"377fcd25cba31cfb23f8109b241ff78bf69d9389","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 08:59:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4a2c28e929c46f9933a40a11b28e977377f61ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 10:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04eab408817d4f358e7da674cd8eb8e5\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e3339245f05e4fdfb2667ec8f46d42f1 : SUCCESS in 58m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd55118028e41779e2edbd1533d6a96 : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6041b0a578c4393be7e885b64795cb2 : SUCCESS in 21m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a27456e22fe7468ebfc490a732d146ce : SUCCESS in 21m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3897fe02d6f34beb9a88d4d3e5d32c66 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13c4fd0ec808475ca8ff1aef05ec6d00 : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc56dd09139a49b884644f16403ba812 : SUCCESS in 57m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffcbbede0adf4cfcb3c39c43432beb30 : SUCCESS in 1h 27m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3996ef78155746d2ac95842cd3a03cfe : SUCCESS in 49m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e9597cae76470145b9d6117af674e8cfee5875bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 10:31:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b81312bf0a6af3fe4b27f43805f6517017d73323","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 10:33:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9103a27d659f4d09a11f31e88cd3f481\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b4860b5a2140f3a464adafbdc6447a : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2134f5ceb350464dabc272d4d6d1c000 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af8a3a7f0810dc64ee51c8db38b49723fd5d3d17","revisions":{"af8a3a7f0810dc64ee51c8db38b49723fd5d3d17":{"kind":"REWORK","_number":1,"created":"2022-11-02 11:57:08.000000000","uploader":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"ref":"refs/changes/80/863280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/863280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/863280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/863280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/863280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/863280/1"}}},"commit":{"parents":[{"commit":"8dd2dea65d3c90ba7be6907142b7ccbd05ea4238","subject":"[OVN] Set the default OVN metadata worker number to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8dd2dea65d3c90ba7be6907142b7ccbd05ea4238"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-23 07:11:25.000000000","tz":120},"committer":{"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","date":"2022-11-02 11:57:08.000000000","tz":0},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\"\n\nSince OVN 20.06 [1], the OVN configuration is stored in\n\"Chassis.other_config\".\n\nSince OVN 22.09, the \"Chassis\" configuration stored in\n\"Chassis.other_config\" will not be replicated to\n\"Chassis.external_ids\".\n\nThe ML2/OVN plugin tries to retrieve the \"Chassis\"\nconfiguration from the \"other_config\" field first; if this\nfield does not exist (in OVN versions before 20.06), the plugin\nwill use \"external_ids\" field instead. Neutron will be compatible\nwith the different OVN versions (with and without \"other_config\"\nfield).\n\n[1]https://github.com/ovn-org/ovn/commit/74d90c2223d0a8c123823fb849b4c2de58c296e4\n[2]https://github.com/ovn-org/ovn/commit/51309429cc3032a0cb422603e7bbda4905ca01ae\n\nNOTE: this patch is similar to [1], but in this case neutron keeps\ncompatibility with the different OVN versions (with and without\n\"other_config\" field). Since [2], the Neutron CI has a new job that\nuses the OVN/OVS packages distributed by the operating system\ninstalled by the CI (in this case, Ubuntu 20.04 and OVN 20.03).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/859642\n[2]https://review.opendev.org/c/openstack/neutron/+/860636\n\nCloses-Bug: #1990229\nChange-Id: I54c8fd4d065ae537f396408df16832b158ee8998\n(cherry picked from commit 536498a29a4e7662a4d0b1bb923e2521509ad77a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8a3a7f0810dc64ee51c8db38b49723fd5d3d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8a3a7f0810dc64ee51c8db38b49723fd5d3d17"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
