)]}'
{"id":"openstack%2Fneutron~863281","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I54c8fd4d065ae537f396408df16832b158ee8998","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1990229-stable/zed-stable/yoga","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-30 08:38:11.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"29074":{"account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"last_update":"2022-12-11 21:08:55.000000000","reason":"Change was submitted"},"35462":{"account":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"last_update":"2022-12-11 21:08:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54c8fd4d065ae537f396408df16832b158ee8998","subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","status":"MERGED","created":"2022-11-02 11:59:24.000000000","updated":"2022-12-11 21:10:07.000000000","submitted":"2022-12-11 21:08:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"863281-bug/1990229-stable/zed-stable/yoga","cherry_pick_of_change":860671,"cherry_pick_of_patch_set":2,"meta_rev_id":"03183e2729030e6a609bb328de594a321324417f","_number":863281,"virtual_id_number":863281,"owner":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":0,"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-11 21:08:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-12-11 19:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-12-09 13:42:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":1,"date":"2022-11-25 12:51:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":2,"date":"2022-12-09 22:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-12-11 19:35:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":0,"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":0,"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 11:59:24.000000000","updated_by":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-11-02 12:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-02 13:20:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 14:17:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-11-14 14:00:52.000000000","updated_by":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"reviewer":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"state":"CC"},{"updated":"2022-11-18 11:07:49.000000000","updated_by":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"reviewer":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"state":"REVIEWER"},{"updated":"2022-11-21 14:02:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-11-25 09:24:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-11-25 12:02:09.000000000","updated_by":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-11-25 12:05:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-11-25 12:51:20.000000000","updated_by":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-12-09 13:42:02.000000000","updated_by":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"reviewer":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"state":"REVIEWER"},{"updated":"2022-12-09 22:38:52.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-12-11 19:35:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"53dc3fb8f531409045a6c9d2d02123971ed6ea53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"date":"2022-11-02 11:59:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.\n\nThe following files contain Git conflicts:\n* neutron/common/ovn/utils.py\n* neutron/tests/functional/base.py\n* neutron/tests/functional/services/ovn_l3/test_plugin.py","accounts_in_message":[],"_revision_number":1},{"id":"e9e97748cf7aeb7c5e85170be40cfbfe665de79e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 12:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/347d659e6b4348a28a1aee40f7bdebf5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f64ca4c23503464d8250370cea00c873 : FAILURE in 10m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdaa74d7305843cabd4b6a815b7d98c6 : FAILURE in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14d8c1244966548cdfbc70d04402c8d3ce72f5f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 13:20:18.000000000","message":"Patch Set 1: Verified-1\n\n(97 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dea489c646ac447cb7bf9729f205e1f9\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/03c44453fade4f8fbb32f6469d0066da : FAILURE in 18m 10s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/76bfcaf941794a48b2bbf6e5c74e4212 : FAILURE in 18m 00s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/48ca62b8d966427c9537cfc97a982fca : FAILURE in 18m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f981c61b6f4fadafe1e7c51e1844ab : FAILURE in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c67793b568e45c99d92e8899dc2d264 : FAILURE in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a19b93294eef45758712be49638104cb : FAILURE in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a3fea565fce4b059fa543b2b9ee0677 : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b9d8132fdc46c4b27bea4b38c03b2c : FAILURE in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c1db37c06ce40e293c65621c66a6d05 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a49adefc0677487c9d51eed6003b07c9 : FAILURE in 11m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/94b768d43e6b43fda19da3f0d290cde4 : SUCCESS in 1h 18m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fda02f08e03469ba9475643b6c4bbfd : FAILURE in 15m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99832172ed554ef3a48a3614853fa65d : FAILURE in 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0792a74493eab97b2f59f46dc6101561b2bfe9b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-11-02 14:17:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/863281/1 : SUCCESS in 2h 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"29b58872dcd891e65be4c064e2439a83437a5279","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-14 13:40:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"923028dc7dc2700b6c55b1866b5c2a9d6ba52909","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-14 14:00:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"884c70b510fd239a0c7a6726184433546c11132e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 14:15:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df3ebdb2b3e748e9a43184ab6c2f5bdc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84dce4dafcd9426aaa57c48d48c8cf0c : SUCCESS in 30m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6321b88984e049229bae0a584d41d665 : SUCCESS in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91b2ca1be7511e6a212ad36fba06e539795c4b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 15:22:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49405c9dc16417e9e8aae5836aafc4c\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/3dea61a73a1e4d49bb475ee03d9a33ef : SUCCESS in 53m 00s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/d5af01fa834f4eee8c97802d48a86820 : SUCCESS in 59m 32s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/f9b7736174ad470f8535633210e4a655 : SUCCESS in 34m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/233813f240c743eab4f6bdeaa68defc6 : SUCCESS in 37m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50de9b52e81c4b0482777d16f11a81cf : SUCCESS in 15m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33d8425ac2094fa08009b8e8564e9a6a : SUCCESS in 25m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eec86139c394c379930b444cefe2324 : SUCCESS in 24m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1386958d505040dd89f3009264cb2613 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b2135dc87243c6acb6ebe7376ab1dd : SUCCESS in 11m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/46126e59f07c40768e55056099e36ad4 : SUCCESS in 38m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd07f8cb4b124703a6082ddde8a9ad6c : SUCCESS in 1h 35m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f93637dfbb5a4a38915a2b9ae0a8eba7 : SUCCESS in 41m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6e88007a326e48e58280c49e99c2fb79 : SUCCESS in 1h 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5f6bdecb547266b9260996165948e0ec7963307","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-18 11:05:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a9ce830ebebd6a32fc17d1ec08ddc0e6334bdfa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-18 11:06:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b1e1a31dab3a41ec16f5fec012a1a321b9d0c348","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-18 11:07:49.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da14806bd5b44232e5c746c068228b442281a015","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 12:03:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22259530976d41ca942da3b3497db0f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba528bbe37d94a91ae4b2fd9116ace28 : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e55130d80ae64e9790fc893a17ee5b92 : SUCCESS in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"81eda70ab2d2da11a079848259fce6ba53b42abb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 12:50:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c718daf1812b4c248d974fd99da51f21\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/b46823fc181e4dd0ba0e567f31c86ccb : SUCCESS in 49m 22s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/5a560e2a12d3409c9b28895cd03c6a9a : SUCCESS in 1h 03m 21s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/f4f6cfc6e175420b92038fd4d7c59470 : SUCCESS in 29m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8bab1a7867b4b30946d3a5bd521a2bd : SUCCESS in 30m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c11b432c4e419d8005085099571bb4 : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51898a09c28a4555bd606a408316dc05 : SUCCESS in 33m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/481300a855464f4fb9ceec2012a12fe9 : SUCCESS in 31m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da39b60a471411d933aed03a311dca8 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1e01b84f9914d6e982ba747c185edd0 : SUCCESS in 12m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fe1c31cda0d489097c0496703a5b068 : SUCCESS in 40m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b828aa359e39402da8dd47a09b874f35 : SUCCESS in 1h 38m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61fb1ef956b84c6c95af2b25aec48e54 : SUCCESS in 1h 05m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/380b29fcd7f442f68a0a1dfd83f0a7ef : SUCCESS in 1h 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eae9e2e8ce3350c75158fe20499578c069cb25dd","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-21 13:56:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9049c5a002d75015f4527b9a5625b03e20578239","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-21 14:02:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"820aa1d6ce216edee4bdd621bc5fcd06c3d5273f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 09:21:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"99f9925e8a44af3ee6d33ad9d7300aea744ad4b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 09:24:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3cd37c8140fbf890700a8fca2a71a9f0dc8be023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-25 12:02:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1c36a225b6dfed281575cdc0886aadf373b94aaf","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-25 12:02:55.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b1d1d4423261254442e0f059e534ea6ddca0866d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 12:05:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43d855b69c0e47098802cf8b9793a0747f94392d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 12:06:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"add7d289a32cd447f484d3740ad917e96533e873","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 12:38:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e8012d38bd2402a8f6e6c56f4de29bc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c632d3026ab4761ae3c64298044ce2e : SUCCESS in 32m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2db4c8545a8464f83626f32c1c162aa : SUCCESS in 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"91b22983569beadfebfcc69cf57ba6dc2a5ca2ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-25 12:51:20.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"209be9abfd38ec57823edd3d20c42202824823ef","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-25 12:51:44.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4d6a411cf9c2417508db8b03f66c4df83d95bf0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 13:40:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f62baf6da4a54a24a5833442a2410dbe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad9b52ba2fe2491b95f59041be782182 : SUCCESS in 44m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f1a5b049c2a410f80349027c8afecc6 : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f10f25c53b1b1eba4d7d5721bb76ca8a1c479876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 14:36:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4de7271f624bfab03bc325b000146e\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/7c43f48acc7042d9aaf0ff16a2cd07d9 : SUCCESS in 1h 04m 33s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/f2fa69e69d55450bbd00155fd8249354 : SUCCESS in 1h 02m 38s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/7317d09ce6f34944a7f909da2b2cb2ab : SUCCESS in 34m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9f310e57c274b7f83d32b363e4a8fdf : SUCCESS in 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf7448100698456992b488734e79682c : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e77abbe7ebd74d88b721f3ff4f0952f5 : SUCCESS in 21m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/982f7477ceac458b806c209056ca9648 : SUCCESS in 22m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884135d5a84d4564aa22089baefb466f : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ac5fd153b2346ee97a1822892affb97 : SUCCESS in 12m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc212b18c78547848cb31b9edc6364b8 : SUCCESS in 59m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/27af6372cf774571a66986a7ab93ab9a : SUCCESS in 1h 38m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/239c37dd651a4745b5ef6fb5a66e78e5 : SUCCESS in 57m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61d71476de8d4e26ae54dc40e9a59666 : SUCCESS in 1h 09m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f14f3bc53daa629b1072de911cd3f51a3bcfb0b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-30 08:38:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"276e75e0d0ce02b89c1378326e72b0c318d62ef9","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 09:44:18.000000000","message":"Patch Set 6:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a36dd94b93f4d4589059cfe12481862\n\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d19eb447ebee472b97ff300ee894f970 : FAILURE in 46m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec6514fb0da841299910c3fa4ab5d200 : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":6},{"id":"a2b633630bb8a00418fbe846f847a6ad09349d1e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-30 10:15:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"29e083378d43fa9dfe47f63c1fd11627e790fdaf","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:19:27.000000000","message":"Patch Set 6:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/690add86cc6a40ca98ee5fd008c6a95f\n\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/91393cf1c90848dca393f7a40e54c097 : FAILURE in 42m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b5a177f08e846c591cbf3b04ee8b8bb : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":6},{"id":"76739ee71b3c716f2954d17672d74611edf776dc","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-11-30 15:21:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e9378d6e3aefc3e1b977636451efbe989144ce1f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 16:54:20.000000000","message":"Patch Set 6:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03c295f9c80f4556b48d23c4c81f79be\n\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1231bf7b849e422a8048065bbec8b848 : FAILURE in 53m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a59169c4d0845c29d16aaf3490f25c2 : SUCCESS in 1h 27m 46s","accounts_in_message":[],"_revision_number":6},{"id":"6cc92d364e02c0fa46f6f89e098e8f62c186cb0a","author":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"date":"2022-12-09 13:42:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"3bf7f294c707fc3837834e86a088c776afcebc0b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-09 22:38:52.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"31a48a8e0c657a710d9d060990a90b2b8b5895b7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-11 19:35:44.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"70d54662a842ef767c997ce2faeb8465a8f37216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 19:36:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a694a9e815c73e1fb1c06e19610713a65a80b468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 21:08:54.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f81ef51de54e1984c4b294788329d5\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/881c80bcbee94ec7a98d78c138e5d969 : SUCCESS in 47m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f19875eb85a4b1d8709a30ba0e41590 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02bb30823cc2430db7748bbabe8897b8 : SUCCESS in 21m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfce3c0238264353933f0c5117e4cf73 : SUCCESS in 22m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1419ca2f2884da8a2ad1a5750f19c13 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3a01ab102ea4915a315a7cb06832751 : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c4f7c49398d434ea4ad9dec67ed3484 : SUCCESS in 44m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c97b4a992c746ccb32b7f02e8cc7f2c : SUCCESS in 1h 31m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e14d946a5d23430ba92414b47cc7bfa7 : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":6},{"id":"440f1760a3986901a54546b2c2c3b11780c2ebd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 21:08:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"03183e2729030e6a609bb328de594a321324417f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 21:10:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29d42bebdbb146708d84f26681037f04\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72eae53f65042b097f3c44a12ae5e9c : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/671128e14f934a46a742f07201e6d9e6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8a4c62d094a5419d878c38aa9017e40931a08f21","revisions":{"147cedca53955e77a21e3e6a2f001deb16555064":{"kind":"REWORK","_number":1,"created":"2022-11-02 11:59:24.000000000","uploader":{"_account_id":35462,"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","username":"cosmin-andrei-patache"},"ref":"refs/changes/81/863281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/863281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/863281/1"}}},"commit":{"parents":[{"commit":"56841381ef327aa940bd0a630ec0bf8d7066056c","subject":"[OVN] Set the default OVN metadata worker number to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56841381ef327aa940bd0a630ec0bf8d7066056c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-23 07:11:25.000000000","tz":120},"committer":{"name":"Patache Cosmin - Andrei","email":"cosmin-andrei.patache@staff.tuiasi.ro","date":"2022-11-02 11:59:24.000000000","tz":0},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\"\n\nSince OVN 20.06 [1], the OVN configuration is stored in\n\"Chassis.other_config\".\n\nSince OVN 22.09, the \"Chassis\" configuration stored in\n\"Chassis.other_config\" will not be replicated to\n\"Chassis.external_ids\".\n\nThe ML2/OVN plugin tries to retrieve the \"Chassis\"\nconfiguration from the \"other_config\" field first; if this\nfield does not exist (in OVN versions before 20.06), the plugin\nwill use \"external_ids\" field instead. Neutron will be compatible\nwith the different OVN versions (with and without \"other_config\"\nfield).\n\n[1]https://github.com/ovn-org/ovn/commit/74d90c2223d0a8c123823fb849b4c2de58c296e4\n[2]https://github.com/ovn-org/ovn/commit/51309429cc3032a0cb422603e7bbda4905ca01ae\n\nNOTE: this patch is similar to [1], but in this case neutron keeps\ncompatibility with the different OVN versions (with and without\n\"other_config\" field). Since [2], the Neutron CI has a new job that\nuses the OVN/OVS packages distributed by the operating system\ninstalled by the CI (in this case, Ubuntu 20.04 and OVN 20.03).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/859642\n[2]https://review.opendev.org/c/openstack/neutron/+/860636\n\nCloses-Bug: #1990229\nChange-Id: I54c8fd4d065ae537f396408df16832b158ee8998\n(cherry picked from commit 536498a29a4e7662a4d0b1bb923e2521509ad77a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/147cedca53955e77a21e3e6a2f001deb16555064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/147cedca53955e77a21e3e6a2f001deb16555064"}]},"branch":"refs/heads/stable/yoga"},"a7578caf00923c321a0fb07379375cbec5902897":{"kind":"REWORK","_number":2,"created":"2022-11-14 13:40:13.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/81/863281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/863281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/863281/2"}}},"commit":{"parents":[{"commit":"4465987475c49b9fd557e4150c2e776be5067f26","subject":"Merge \"Port provisioning should retry only for VM ports\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4465987475c49b9fd557e4150c2e776be5067f26"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-23 07:11:25.000000000","tz":120},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-11-14 13:40:09.000000000","tz":60},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\"\n\nSince OVN 20.06 [1], the OVN configuration is stored in\n\"Chassis.other_config\".\n\nSince OVN 22.09, the \"Chassis\" configuration stored in\n\"Chassis.other_config\" will not be replicated to\n\"Chassis.external_ids\".\n\nThe ML2/OVN plugin tries to retrieve the \"Chassis\"\nconfiguration from the \"other_config\" field first; if this\nfield does not exist (in OVN versions before 20.06), the plugin\nwill use \"external_ids\" field instead. Neutron will be compatible\nwith the different OVN versions (with and without \"other_config\"\nfield).\n\n[1]https://github.com/ovn-org/ovn/commit/74d90c2223d0a8c123823fb849b4c2de58c296e4\n[2]https://github.com/ovn-org/ovn/commit/51309429cc3032a0cb422603e7bbda4905ca01ae\n\nNOTE: this patch is similar to [1], but in this case neutron keeps\ncompatibility with the different OVN versions (with and without\n\"other_config\" field). Since [2], the Neutron CI has a new job that\nuses the OVN/OVS packages distributed by the operating system\ninstalled by the CI (in this case, Ubuntu 20.04 and OVN 20.03).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/859642\n[2]https://review.opendev.org/c/openstack/neutron/+/860636\n\nCloses-Bug: #1990229\nChange-Id: I54c8fd4d065ae537f396408df16832b158ee8998\n(cherry picked from commit 536498a29a4e7662a4d0b1bb923e2521509ad77a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7578caf00923c321a0fb07379375cbec5902897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7578caf00923c321a0fb07379375cbec5902897"}]},"branch":"refs/heads/stable/yoga"},"b5b8b494f724d7d56bb4fc5ebd4924cc4858c6dc":{"kind":"REWORK","_number":3,"created":"2022-11-18 11:05:18.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/81/863281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/863281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/863281/3"}}},"commit":{"parents":[{"commit":"9a8301796bafca1b30bc27881a49b0d74680268d","subject":"[stable-only] Add \"tempest-integrated-networking\" job to experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a8301796bafca1b30bc27881a49b0d74680268d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-23 07:11:25.000000000","tz":120},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-11-18 11:05:18.000000000","tz":0},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\"\n\nSince OVN 20.06 [1], the OVN configuration is stored in\n\"Chassis.other_config\".\n\nSince OVN 22.09, the \"Chassis\" configuration stored in\n\"Chassis.other_config\" will not be replicated to\n\"Chassis.external_ids\".\n\nThe ML2/OVN plugin tries to retrieve the \"Chassis\"\nconfiguration from the \"other_config\" field first; if this\nfield does not exist (in OVN versions before 20.06), the plugin\nwill use \"external_ids\" field instead. Neutron will be compatible\nwith the different OVN versions (with and without \"other_config\"\nfield).\n\n[1]https://github.com/ovn-org/ovn/commit/74d90c2223d0a8c123823fb849b4c2de58c296e4\n[2]https://github.com/ovn-org/ovn/commit/51309429cc3032a0cb422603e7bbda4905ca01ae\n\nNOTE: this patch is similar to [1], but in this case neutron keeps\ncompatibility with the different OVN versions (with and without\n\"other_config\" field). Since [2], the Neutron CI has a new job that\nuses the OVN/OVS packages distributed by the operating system\ninstalled by the CI (in this case, Ubuntu 20.04 and OVN 20.03).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/859642\n[2]https://review.opendev.org/c/openstack/neutron/+/860636\n\nCloses-Bug: #1990229\nChange-Id: I54c8fd4d065ae537f396408df16832b158ee8998\n(cherry picked from commit 536498a29a4e7662a4d0b1bb923e2521509ad77a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5b8b494f724d7d56bb4fc5ebd4924cc4858c6dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5b8b494f724d7d56bb4fc5ebd4924cc4858c6dc"}]},"branch":"refs/heads/stable/yoga"},"6ef3b2cda28b7a4af8b9e2dd58f4556f6ace0734":{"kind":"REWORK","_number":4,"created":"2022-11-18 11:06:46.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/81/863281/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/863281/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/863281/4"}}},"commit":{"parents":[{"commit":"4465987475c49b9fd557e4150c2e776be5067f26","subject":"Merge \"Port provisioning should retry only for VM ports\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4465987475c49b9fd557e4150c2e776be5067f26"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-23 07:11:25.000000000","tz":120},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-11-18 11:06:36.000000000","tz":60},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\"\n\nSince OVN 20.06 [1], the OVN configuration is stored in\n\"Chassis.other_config\".\n\nSince OVN 22.09, the \"Chassis\" configuration stored in\n\"Chassis.other_config\" will not be replicated to\n\"Chassis.external_ids\".\n\nThe ML2/OVN plugin tries to retrieve the \"Chassis\"\nconfiguration from the \"other_config\" field first; if this\nfield does not exist (in OVN versions before 20.06), the plugin\nwill use \"external_ids\" field instead. Neutron will be compatible\nwith the different OVN versions (with and without \"other_config\"\nfield).\n\n[1]https://github.com/ovn-org/ovn/commit/74d90c2223d0a8c123823fb849b4c2de58c296e4\n[2]https://github.com/ovn-org/ovn/commit/51309429cc3032a0cb422603e7bbda4905ca01ae\n\nNOTE: this patch is similar to [1], but in this case neutron keeps\ncompatibility with the different OVN versions (with and without\n\"other_config\" field). Since [2], the Neutron CI has a new job that\nuses the OVN/OVS packages distributed by the operating system\ninstalled by the CI (in this case, Ubuntu 20.04 and OVN 20.03).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/859642\n[2]https://review.opendev.org/c/openstack/neutron/+/860636\n\nCloses-Bug: #1990229\nChange-Id: I54c8fd4d065ae537f396408df16832b158ee8998\n(cherry picked from commit 536498a29a4e7662a4d0b1bb923e2521509ad77a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ef3b2cda28b7a4af8b9e2dd58f4556f6ace0734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ef3b2cda28b7a4af8b9e2dd58f4556f6ace0734"}]},"branch":"refs/heads/stable/yoga"},"3a6a7115d1b575af5a80fd1a356794e81782b3a1":{"kind":"REWORK","_number":5,"created":"2022-11-25 12:02:09.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/81/863281/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/863281/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/863281/5"}}},"commit":{"parents":[{"commit":"11832401021a73dfe8b31596d0457d15207dbbda","subject":"Merge \"ovn: Use ovsdb-client to create neutron_pg_drop\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11832401021a73dfe8b31596d0457d15207dbbda"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-23 07:11:25.000000000","tz":120},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-11-25 12:01:07.000000000","tz":60},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\"\n\nSince OVN 20.06 [1], the OVN configuration is stored in\n\"Chassis.other_config\".\n\nSince OVN 22.09, the \"Chassis\" configuration stored in\n\"Chassis.other_config\" will not be replicated to\n\"Chassis.external_ids\".\n\nThe ML2/OVN plugin tries to retrieve the \"Chassis\"\nconfiguration from the \"other_config\" field first; if this\nfield does not exist (in OVN versions before 20.06), the plugin\nwill use \"external_ids\" field instead. Neutron will be compatible\nwith the different OVN versions (with and without \"other_config\"\nfield).\n\n[1]https://github.com/ovn-org/ovn/commit/74d90c2223d0a8c123823fb849b4c2de58c296e4\n[2]https://github.com/ovn-org/ovn/commit/51309429cc3032a0cb422603e7bbda4905ca01ae\n\nNOTE: this patch is similar to [1], but in this case neutron keeps\ncompatibility with the different OVN versions (with and without\n\"other_config\" field). Since [2], the Neutron CI has a new job that\nuses the OVN/OVS packages distributed by the operating system\ninstalled by the CI (in this case, Ubuntu 20.04 and OVN 20.03).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/859642\n[2]https://review.opendev.org/c/openstack/neutron/+/860636\n\nconflicting files in cherry pick:\n- neutron/common/ovn/utils.py\n\nCloses-Bug: #1990229\nChange-Id: I54c8fd4d065ae537f396408df16832b158ee8998\n(cherry picked from commit 536498a29a4e7662a4d0b1bb923e2521509ad77a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a6a7115d1b575af5a80fd1a356794e81782b3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a6a7115d1b575af5a80fd1a356794e81782b3a1"}]},"branch":"refs/heads/stable/yoga"},"8a4c62d094a5419d878c38aa9017e40931a08f21":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-11-25 12:51:20.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/81/863281/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/863281/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/863281/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/863281/6"}}},"commit":{"parents":[{"commit":"11832401021a73dfe8b31596d0457d15207dbbda","subject":"Merge \"ovn: Use ovsdb-client to create neutron_pg_drop\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11832401021a73dfe8b31596d0457d15207dbbda"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-09-23 07:11:25.000000000","tz":120},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-11-25 12:50:49.000000000","tz":60},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\"","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\"\n\nSince OVN 20.06 [1], the OVN configuration is stored in\n\"Chassis.other_config\".\n\nSince OVN 22.09, the \"Chassis\" configuration stored in\n\"Chassis.other_config\" will not be replicated to\n\"Chassis.external_ids\".\n\nThe ML2/OVN plugin tries to retrieve the \"Chassis\"\nconfiguration from the \"other_config\" field first; if this\nfield does not exist (in OVN versions before 20.06), the plugin\nwill use \"external_ids\" field instead. Neutron will be compatible\nwith the different OVN versions (with and without \"other_config\"\nfield).\n\n[1]https://github.com/ovn-org/ovn/commit/74d90c2223d0a8c123823fb849b4c2de58c296e4\n[2]https://github.com/ovn-org/ovn/commit/51309429cc3032a0cb422603e7bbda4905ca01ae\n\nNOTE: this patch is similar to [1], but in this case neutron keeps\ncompatibility with the different OVN versions (with and without\n\"other_config\" field). Since [2], the Neutron CI has a new job that\nuses the OVN/OVS packages distributed by the operating system\ninstalled by the CI (in this case, Ubuntu 20.04 and OVN 20.03).\n\n[1]https://review.opendev.org/c/openstack/neutron/+/859642\n[2]https://review.opendev.org/c/openstack/neutron/+/860636\n\nconflicting files in cherry pick:\n- neutron/tests/functional/base.py\n- neutron/tests/functional/services/ovn_l3/test_plugin.py\n\nCloses-Bug: #1990229\nChange-Id: I54c8fd4d065ae537f396408df16832b158ee8998\n(cherry picked from commit 536498a29a4e7662a4d0b1bb923e2521509ad77a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a4c62d094a5419d878c38aa9017e40931a08f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a4c62d094a5419d878c38aa9017e40931a08f21"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
