)]}'
{"id":"openstack%2Fneutron~863468","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Ie8dd684a7b79b0a322b1f2d17fffb4d58cfe94fc","project":"openstack/neutron","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2022-11-15 15:57:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie8dd684a7b79b0a322b1f2d17fffb4d58cfe94fc","subject":"Update documentation link for openSUSE index.","status":"MERGED","created":"2022-11-04 08:47:43.000000000","updated":"2022-11-15 15:58:25.000000000","submitted":"2022-11-15 15:57:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863468","cherry_pick_of_change":863467,"cherry_pick_of_patch_set":1,"meta_rev_id":"91b984a4415bc58ee5747a006567fa3406ac8362","_number":863468,"virtual_id_number":863468,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-15 15:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-11-04 10:54:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-15 15:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-11-15 15:38:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 09:06:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 10:54:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-11-15 15:38:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a22e00527cb14aa180f8a3c91c73a73da1f8a048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2022-11-04 08:47:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"a4784ac557b93ea82ff50def1545995f3003f60a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:06:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad9b1fb37c184c1ea788c0a29078082a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154c6d1507674c4a9c3c105ebafa2972 : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aee1151358a49d98dbdddcd14d5e480 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9c8dbdddb2e096c666db21c5a3693c59f827179d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-04 10:54:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6d3c58850cd23fa0c5eb9fd3f2241f9b54da2d7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-15 15:38:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7f213f2860bb5d4664b050e2a4fc5760df33e9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 15:39:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cadf61a91ce593618b3bac00d4b6332ab2c5560","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 15:57:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2666bad593432fa331c51386e96bf7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ef2462c393493f865860f2dc6bb6ce : SUCCESS in 13m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21dd6d8875c4b46b43ff5fab1320128 : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8886ca6dcb89e1a2fcf5ea6c519eb0b83432aa36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 15:57:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91b984a4415bc58ee5747a006567fa3406ac8362","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 15:58:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb45cbe239d0411c9784cb33c57f577e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b69824f194e494292077b6821d98782 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebcbe4aec7f21d4fc9ee82ecc9a31b8fe09c5529","revisions":{"ebcbe4aec7f21d4fc9ee82ecc9a31b8fe09c5529":{"kind":"REWORK","_number":1,"created":"2022-11-04 08:47:43.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/68/863468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/863468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/863468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/863468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/863468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/863468/1"}}},"commit":{"parents":[{"commit":"c968f40967221f1b8e001a20b4868cbf53422693","subject":"Revert \"[OVN] Set NB/SB \"connection\" inactivity probe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c968f40967221f1b8e001a20b4868cbf53422693"}]}],"author":{"name":"renliang17","email":"renliang@uniontech.com","date":"2022-11-04 05:58:22.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2022-11-04 08:47:43.000000000","tz":0},"subject":"Update documentation link for openSUSE index.","message":"Update documentation link for openSUSE index.\n\nChange-Id: Ie8dd684a7b79b0a322b1f2d17fffb4d58cfe94fc\n(cherry picked from commit 562e9704f8ed292889f5786d373f9f09cf87ae8a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebcbe4aec7f21d4fc9ee82ecc9a31b8fe09c5529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebcbe4aec7f21d4fc9ee82ecc9a31b8fe09c5529"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
