)]}'
{"id":"openstack%2Fneutron~863671","triplet_id":"openstack%2Fneutron~master~I1258b04f64a18036407e1d9de9ddca7472af0d11","project":"openstack/neutron","branch":"master","topic":"pylint-c0330-api","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-11-15 19:09:37.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2022-11-08 02:42:28.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"I1258b04f64a18036407e1d9de9ddca7472af0d11","subject":"Fix some pylint indentation warnings","status":"MERGED","created":"2022-11-04 17:28:37.000000000","updated":"2022-11-15 22:11:07.000000000","submitted":"2022-11-15 22:09:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863671-pylint-c0330-api","meta_rev_id":"46ce0ef91711cd3ec80c038314fd3b88bb8ba6c7","_number":863671,"virtual_id_number":863671,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-15 22:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-11-15 15:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-11-07 10:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-08 02:42:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2022-11-15 15:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 18:00:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-04 18:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-04 19:09:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 21:08:02.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2022-11-07 10:53:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-15 15:46:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"713ae7acd1aa9c1f16f6b76215ad7244966f940e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-04 17:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b61e3fc642c3e00c17e5cda590ee7e90e9081004","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-04 18:00:18.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/71/863671/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"62ef33fc199310d2ba19c28cda17a3e071f31e2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 18:22:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2ca11316acc43d0a2d3a4f7fe259327\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a91acd7bd3e94d15a6fa6c7506a01c9e : SUCCESS in 50m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e0014478be94004b28fea807be802ce : SUCCESS in 51m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b992fcca4d76499f943b6dbec50d0052 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":1},{"id":"26ec8bafe3ed40c3b4ed7e73af522619d827fa82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:09:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c19117026b2f4a52a621daa4139a84dd\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d35c7c738b1b471e8fa5c0751a4eb250 : SUCCESS in 1h 32m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5801c1068e8e47e6a555dd8cf0e0a907 : SUCCESS in 1h 29m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b55910ee5ceb4643bf2001cb15a4b1df : SUCCESS in 1h 18m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/91e96995201d472eabfd876442eba0fb : SUCCESS in 1h 14m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9030aea37bd48608244b1616ac2fb3f : SUCCESS in 35m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1680e98fa68d4368b1018cb73eedfadb : SUCCESS in 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a451b28ae094081b1b716e32bcd7990 : SUCCESS in 9m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84ca3ae9735a4c8ea66b349f6708485d : SUCCESS in 21m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4f080c0d62541148dfb6cde00f01106 : SUCCESS in 19m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f68c41dc0fca49ef96139d54576205e0 : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c207e1bc25ae43d8bae2fa57bfa164c7 : SUCCESS in 22m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/890737fa82a843e29653f39704fd8ea4 : SUCCESS in 43m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3f78b5a427a4bd4be04747595b70cb2 : SUCCESS in 1h 24m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f11882947df64e33b3116d2790e38d2b : SUCCESS in 51m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ef9293a193d143b2aef33d141d62df19 : SUCCESS in 48m 41s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5bae048daf494c5c890c27bd55107698 : SUCCESS in 52m 35s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a5fb47ed6e464794a37a3f14fbc42f72 : SUCCESS in 49m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8185191edd114ca7841445bc8964132a : SUCCESS in 1h 08m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/316730a4ad254a3a98876bb1ff150b06 : SUCCESS in 1h 03m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/17fd3b06448047499a098c7a41dca98f : SUCCESS in 1h 05m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/723791ed8bd348da99336c2de9b8a292 : SUCCESS in 57m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a66ffb0822064c47a460b4244281f67e : SUCCESS in 1h 18m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0900e18683c4754a049568a8347cef6 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2f1f61c644ad7f612669f93ef6db4b379e4e7f63","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-11-04 21:08:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12d42cbf3475afd04cf68eb1eddca464f15aa510","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-07 10:53:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f07c088374f6a09276e1d981f2b93f72e79889a1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-07 17:48:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"164e959c2ad2f8b631206f8895c0d373cb9cc807","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-11-08 02:42:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc0dd2b02f8860c62276cc1e8624608da360d2f9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-15 15:46:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9b12e7ff12ed646843b2fa4bc48a6eafb659e45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 15:47:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3d89c4606ae884ec884eb63856c6c9e1ea73dd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 17:48:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd5abe421ace4608b60509a4ebd54704\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f3dcabce0c9f4f60993e036762d38f9d : SUCCESS in 1h 19m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/67214588310644f6a99fc3147f792866 : SUCCESS in 1h 45m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5e150ee40b1467ba7828e74bf663532 : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/94459868560a4215b17517523ba56861 : SUCCESS in 1h 31m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e590f4a47a14d83b8977c79fb54eb54 : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff84ede6ea9145478cb24c41bd3071fe : SUCCESS in 26m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7192da79e5d4a1ab03888a2cdf715fa : SUCCESS in 24m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e51f1e7735249419ec841d813be8040 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/900b28935bb84913ba69e8c3d2a9fdf5 : SUCCESS in 26m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/30768dac79ef4e72af6cfb4a620d0a4f : SUCCESS in 45m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b58e06c9bfd4c5a98ab2eba0bf89b7b : SUCCESS in 1h 22m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/735d0c9441ba4944a23046abd02c0654 : SUCCESS in 1h 18m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fdbbb133340471c80aad20172f3dbee : SUCCESS in 50m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b4240c7693d34b8f8f306cd434c1ecf6 : SUCCESS in 53m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/505b88c0e0cb40ae8dfc9e833ace8dfd : SUCCESS in 56m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/332169cd9ff14cdfbd3b55bf43b3786f : FAILURE in 1h 12m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e247e6806e6a359eda330f9f19325be728431e65","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-15 19:09:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7707440d664c08cfdca3349bde0c0ee4e12fd3a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 20:18:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2181f189974b46e4b593ae76799927e9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09ef6b8d64b34eef90d0ee2e03176e0e : SUCCESS in 36m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e7a6f4a93b98493a92118c2b7f3faf70 : SUCCESS in 38m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/370fdd3ec9d544de9f272bb13b6f05be : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f5f31761a1f9410819a97732291ca8ebbc97ac71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 20:50:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7761bd9be6d244d3bd648b950019ca62\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2e95f230c0e64199b894082a0c552827 : SUCCESS in 1h 32m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d04207fa54541f68b004d2cbc17d62b : SUCCESS in 1h 23m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d7f0f6b7adb4bdab8b32c932fc8d2f3 : SUCCESS in 1h 06m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21195f8149534ff48406db6fb99d206f : SUCCESS in 1h 19m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/63c8bb55bf684ae5a6275c9c5d9324c4 : SUCCESS in 29m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3631bd879c96466083a5a0f5945dcba0 : SUCCESS in 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1fb62ecfd94003bbb44927edaa0481 : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/788d711632134c8f857a043f54c690c6 : SUCCESS in 30m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e103b49c0984ff88912762433155cea : SUCCESS in 26m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d73ed272997404da675a6cdf7055ab8 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/881c0cd810ec4bf2b6796cacde1cebd0 : SUCCESS in 29m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c10c328c4b5d4e24bded0d6df1dba719 : SUCCESS in 52m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/60a8adb1ba56443784e3b39b627ed0be : SUCCESS in 1h 34m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b9b66a5791142118637b9494b225ac4 : SUCCESS in 56m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/309e8e3247524d14b9072588703f19fb : SUCCESS in 1h 01m 14s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9bdbaa919244a74b9fcced8afd9a4af : SUCCESS in 1h 09m 32s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/057129e7ca514622ae5937434f3ddd9a : SUCCESS in 1h 07m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c885c6bf4a724ffb8d698a2a96298b49 : SUCCESS in 1h 20m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c37a5b42488545b4aeb39adbf09cc167 : SUCCESS in 1h 27m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a215c33a5da04a8fbb75d8dfa784a0a5 : SUCCESS in 1h 10m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f36a746aa7d488bb4b5154b54587b0d : SUCCESS in 49m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/495edd9b3f3f484aa3431f9174eaea94 : SUCCESS in 1h 23m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/54dae097361a4942b397747d6319cb6e : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6d8bf0dadebde2d6b2957080e1f4b8ce1103602a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 20:50:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c814e7a35dbb3e8e18fa57078a320045d45c9177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 22:09:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99024a9887954f9898d02c9305c020b7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3cd2ffd7f01343e5857ebac053fd7b5b : SUCCESS in 1h 04m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cfa35fd47bf74865beaa4bcc3cd0d496 : SUCCESS in 1h 16m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b372267f7b9840e7bd7621aed95343bf : SUCCESS in 1h 02m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdd6b778c55840e0aeb243e96802658e : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4246929d747142babed734cab5cda2f0 : SUCCESS in 11m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/628514f9229c41c5b675955b31f6b41c : SUCCESS in 26m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff05b598683241a89b879c5f197c6e54 : SUCCESS in 29m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59dd299b77ec4ee388c9c4450e53e810 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ee6fbc137c94146a7579572f5e44bb8 : SUCCESS in 29m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eda5b8fe787047eca1d43d15afb25f5e : SUCCESS in 36m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63dcb1dc384042c7b64ae0cff160bff0 : SUCCESS in 1h 16m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bea267fe21784c4ab1b013b9a8d58d16 : SUCCESS in 57m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36b8c7e416f44f2db3a23d2b5e7527d4 : SUCCESS in 50m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd5fe0714c864806b5d9679d9cc1c89c : SUCCESS in 44m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a98e044ea7e44c72b4273c4da3094828 : SUCCESS in 52m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0940a4949a5c4819b41babb0361eeb02 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a6ebe6eedd676c8cd0ff5158a9b6738c84672171","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 22:09:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"46ce0ef91711cd3ec80c038314fd3b88bb8ba6c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 22:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbea53fbed79425bb6a3e9a642195e7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab54614ed834e72896ec558edd8547e : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"291eabb0b686992f5124bb80e749c02d13c0dc05","revisions":{"291eabb0b686992f5124bb80e749c02d13c0dc05":{"kind":"REWORK","_number":1,"created":"2022-11-04 17:28:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/71/863671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/863671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/863671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/863671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/863671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/863671/1"}}},"commit":{"parents":[{"commit":"ba795c669228a32cee0618950a5355b141bbe497","subject":"Merge \"Check subnet overlapping after add router interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba795c669228a32cee0618950a5355b141bbe497"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-02 14:11:46.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-02 14:11:46.000000000","tz":-240},"subject":"Fix some pylint indentation warnings","message":"Fix some pylint indentation warnings\n\nRunning with a stricter .pylintrc generates a lot of\nC0330 warnings (hanging/continued indentation). Fix\nthe ones in neutron/api.\n\nTrivialfix\n\nChange-Id: I1258b04f64a18036407e1d9de9ddca7472af0d11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/291eabb0b686992f5124bb80e749c02d13c0dc05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/291eabb0b686992f5124bb80e749c02d13c0dc05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
