)]}'
{"id":"openstack%2Fneutron~863950","triplet_id":"openstack%2Fneutron~master~Iac1d4c4004e1348ced388ee4dd44711f0e27bcc2","project":"openstack/neutron","branch":"master","topic":"bug/1995901","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-11-08 12:18:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac1d4c4004e1348ced388ee4dd44711f0e27bcc2","subject":"Use --exclude-regex in tox.ini","status":"MERGED","created":"2022-11-07 23:43:24.000000000","updated":"2022-11-08 12:20:05.000000000","submitted":"2022-11-08 12:18:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863950-bug/1995901","meta_rev_id":"a0d11a753a03629f65cb1e9f5f1727f5c708ddb4","_number":863950,"virtual_id_number":863950,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 12:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2022-11-08 10:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-11-08 05:26:26.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-08 06:00:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2022-11-08 07:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2022-11-08 10:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 00:36:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-08 01:29:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-08 05:26:26.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-11-08 06:00:52.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-11-08 07:18:07.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-11-08 10:28:07.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"1d09d4e56b1f2287b29c3f0298638e3ce8b1c016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-07 23:43:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"212691e28cc10ffa685e31bb140f2d1f5fa16e66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-07 23:46:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8f3668a7ecafe44b59835d7124dc9dcaffbebe7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 00:36:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d72a0adaf44fdd9589b9963c684572\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c3517fe5ce748b7874f769ae429f3cc : FAILURE in 43m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fdb841fc64fc41879c3adf0edb99fdd3 : SUCCESS in 43m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4eb74b3e6934f7a9218779dbb0f2f3f : SUCCESS in 43m 22s","accounts_in_message":[],"_revision_number":2},{"id":"62ca0969072d3aaabf73c38b797d66ce68184a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 01:29:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df5d620ab0ad4e5295f58320b6843ffe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a11b7d1253c34d63b7d267f65d3f8565 : SUCCESS in 40m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/880a696640c842d1a172f7f166b631f0 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34912bbc70fe49559d2104c7c9456aac : SUCCESS in 32m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58110c77026f431da8d722e906f61e5d : SUCCESS in 25m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1315291a877043aab9a5c7182e3bcca9 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d2130b5495a4a9289b710ca2fcefa69 : SUCCESS in 10m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fec20f9a3f324a6990975e8c705558bc : SUCCESS in 39m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/900bd8b84f8743a1a21de0612932d233 : SUCCESS in 58m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f42de562dc943538d0aa4d8ecdd53bc : SUCCESS in 1h 37m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b0cdd2b59108415522ca12578fcb76437dd7f321","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-11-08 05:26:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ee509c66f37f80234059044f13744f7e28be1aac","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-11-08 06:00:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1929ce4a9aebf5faa4fd46934674d5202ae682fa","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-11-08 07:18:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"87498fa13adbb0b6f1444c089475626700982848","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-08 10:28:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa330b833c70507a0b2083acec541d55ed071296","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:28:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e86ec2ec39e330609cd5ebbd0b44bf08db5944f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 12:18:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/184d1b230dc34023a2008952b7a298e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3461cad3114fc48d45c85a5020e2a8 : SUCCESS in 10m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55868c4d12c14a698eaf3fa6dd027501 : SUCCESS in 21m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/760bc82eb4024b1d8dca94432b5b0859 : SUCCESS in 22m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9e6aaef7d04c7392bf4a81f7afb9e0 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb14231475f449bab5add7a8c49ede79 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd33ae25f9784bb38189977df51582c1 : SUCCESS in 25m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e76c904a40cb45858354d27038f39cd1 : SUCCESS in 39m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09d4c1c963204dc09200ae0565075ff0 : SUCCESS in 1h 34m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a2695ad854e8fbd6639449d1279e0b8ebec2446c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 12:18:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a0d11a753a03629f65cb1e9f5f1727f5c708ddb4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 12:20:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d52bf7f574634b7e8ec8c380a96cf561\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e109bf37faf94b5f93cc770cd4bfb030 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8602e65d5f2744caa83fca5ac392315d : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9e2a0ac058305dd16d6dedb883d305cd162f5fe4","revisions":{"b41c1d1cc845d845d9fc781c243dda5502c2a179":{"kind":"REWORK","_number":1,"created":"2022-11-07 23:43:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/50/863950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/863950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/863950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/863950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/863950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/863950/1"}}},"commit":{"parents":[{"commit":"c8aa65796f335e0efdd4c9f14e0accad655154e4","subject":"Merge \"test: fix docstring validation function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8aa65796f335e0efdd4c9f14e0accad655154e4"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-07 23:41:56.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-07 23:41:56.000000000","tz":-300},"subject":"Use --exclude-regex in tox.ini","message":"Use --exclude-regex in tox.ini\n\n--black-regex has been deprecated and using it can sometimes\ncause failures in the fulltest job.\n\nChange-Id: Iac1d4c4004e1348ced388ee4dd44711f0e27bcc2\nCloses-bug: #1995901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b41c1d1cc845d845d9fc781c243dda5502c2a179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b41c1d1cc845d845d9fc781c243dda5502c2a179"}]},"branch":"refs/heads/master"},"9e2a0ac058305dd16d6dedb883d305cd162f5fe4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-07 23:46:45.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/50/863950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/863950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/863950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/863950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/863950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/863950/2"}}},"commit":{"parents":[{"commit":"c8aa65796f335e0efdd4c9f14e0accad655154e4","subject":"Merge \"test: fix docstring validation function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8aa65796f335e0efdd4c9f14e0accad655154e4"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-07 23:41:56.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-07 23:46:30.000000000","tz":-300},"subject":"Use --exclude-regex in tox.ini","message":"Use --exclude-regex in tox.ini\n\n--black-regex has been deprecated and using it can sometimes\ncause failures in the fullstack job.\n\nChange-Id: Iac1d4c4004e1348ced388ee4dd44711f0e27bcc2\nCloses-bug: #1995901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e2a0ac058305dd16d6dedb883d305cd162f5fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e2a0ac058305dd16d6dedb883d305cd162f5fe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
