)]}'
{"id":"openstack%2Fneutron~864073","triplet_id":"openstack%2Fneutron~master~I9311cfe5efc51552008072d84aa238e5d0c9de60","project":"openstack/neutron","branch":"master","topic":"pylint-c0330-db","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-11-14 13:10:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9311cfe5efc51552008072d84aa238e5d0c9de60","subject":"Fix some pylint indentation warnings","status":"MERGED","created":"2022-11-09 00:50:23.000000000","updated":"2022-11-14 13:12:14.000000000","submitted":"2022-11-14 13:10:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864073-pylint-c0330-db","meta_rev_id":"c1800a18667229b597a431c2f9163b1926acc0d9","_number":864073,"virtual_id_number":864073,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-11-09 20:00:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 13:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-14 08:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-11-14 10:12:33.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-11-14 10:12:33.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 01:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 02:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-09 20:00:37.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-14 08:53:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-14 10:12:33.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"b07dce3d178646646374bba0386bd6ba175d9e39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-09 00:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc24ee7fb3e2a70bd0d7c8fb51e8b0a52f29238a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 01:41:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa10ad234a0247d59ba75e77293f3ad4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83179bfae6ba4215a0d68289ea7014dc : SUCCESS in 42m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1044e6696e654c06b4c13349310725e7 : SUCCESS in 43m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3352ea476cb8431c85a694391827e0a9 : FAILURE in 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c959303e5841f648af47b82f8642b1cdd69badb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 02:31:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fb480af723041f4b1f54d5ab240f13f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/95e4916b7dfb4db5bc2061342e2bd535 : SUCCESS in 1h 08m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8fcb9b023aa943aa95d2f9dc85e09d0e : SUCCESS in 1h 02m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d38c7fbc9c084ce59c0edacdb6934829 : SUCCESS in 54m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/355612a850fc4c93ba8732b46b47a0d3 : SUCCESS in 57m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8cfaad7e07a84fbe97b0233bed81a47b : SUCCESS in 30m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a69d78fbebe4642a5a1b5ca4df5e6b2 : SUCCESS in 28m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7470a22a4e460b831bf72a40cfb093 : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc20f3ab3bb44a4a8ab2f69bcbbef4cc : SUCCESS in 21m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40d68684d39b4126a162e132aeb0980f : SUCCESS in 20m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfd369cc62d44139dc0ab78afbb0b78 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/848e093e778a450d8e921965f4b0d39b : SUCCESS in 21m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38f8fe51be8841a59b1ab2a20829ab71 : SUCCESS in 41m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/974c1e47e83344689f4d68db52fc2398 : SUCCESS in 1h 39m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b22afa39b0e46deb8e231f746d50535 : SUCCESS in 38m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d5026f57aaa248218cdcbd809ba520a7 : SUCCESS in 51m 21s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7ddd72f091f242459581bce2fc24765b : SUCCESS in 49m 26s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/732f5431c5484a7eadc8b02e57d22a2c : SUCCESS in 52m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f3e2803240bf4ab199bc8fdd0f0ed78f : SUCCESS in 1h 19m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bfdbaeab6b2744ff96ccde74c2a86070 : FAILURE in 1h 15m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6579cab5e5474b7c9601e64ad1cbd8fe : SUCCESS in 50m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/faaf8485410d4dada81175b223b4ae21 : SUCCESS in 47m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aaee13402cd840efa1602a11068b5b5e : SUCCESS in 1h 02m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a8af3bfb85a4183890b24ca247afe37 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8f3499ff79bde8a5a15197091a4aa5d0a52ce631","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-09 13:59:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9891dc1ec9cfab803cba411053b690fdf1856c68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 14:40:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce2dfd2fef2b46129d52795d2675800a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a74511ed6d1e49ffaf04ecb39c5186e9 : SUCCESS in 35m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/00cd8b3dd2714f569f9ddd09ee20a7d5 : SUCCESS in 35m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8892b2864c449769ab56a4a1d5bad85 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"72210e853b995997ed9762e088d8c3ecec5ec474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 15:47:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba872d7df52473e8f8b54949eed9c63\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/180be79914d742ccb06b729ff9d820d3 : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3547574e06784c0da109eff922c47699 : SUCCESS in 1h 15m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ae8af117bda846fa9a24383623d172b5 : SUCCESS in 1h 07m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2f7ab9436c7f4962878e63c57066f2f9 : SUCCESS in 1h 26m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a09655b81d840c0a3c4fae45fba5662 : SUCCESS in 32m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c1326dd44ea49768b38aeb632e8bd2a : SUCCESS in 27m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b91fdcf9094083af5a8ceb35c95330 : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5438e1c6f7ad4c7d8cac9a7244e12a8d : SUCCESS in 21m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a26c858cd8d14f4883495b425028ce30 : SUCCESS in 28m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e937b9627ca44aaebb4b2459a5968945 : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e6f5488e7ca40c694c346d65b1c96bc : SUCCESS in 26m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/98f081f8625146baa06aa51245dbafc3 : SUCCESS in 46m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/90e208bf301a4d30b7f87b0937f8f1b7 : SUCCESS in 1h 26m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e447b61cca354d2d942e845aa60f1fae : SUCCESS in 50m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7c4e03951cf84207bd45b0bfc8f43275 : SUCCESS in 1h 03m 09s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ee72b8df7388446fac91ddc38cc9b0b1 : SUCCESS in 1h 11m 14s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7c2b7391926e41dab4def86f7297652a : SUCCESS in 52m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b812dd8472e147a68a28f038e053a1de : SUCCESS in 1h 05m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2983384e34084f8fb5ae15bb7fa12327 : SUCCESS in 1h 36m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c7e80737f7b946fcbaee4d3e9dc09393 : SUCCESS in 55m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1ab69dfb4bbf4f9e817be101d0b8e887 : SUCCESS in 58m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9fbe4f2e53b445249193c06b12d47779 : SUCCESS in 1h 01m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c0a1596dcd1047e392a1ab847ae2da05 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"629d1340fde71745760ad33cb4b820c62f997821","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-09 20:00:37.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/73/864073/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"8a8f5e46d715182a2a2084272ed99fd7c2ecf326","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-14 08:53:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ebc2940ffade9197337b6b0272927025087a481","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-11-14 10:12:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe91897e4dea1e916f23d0d1f5e668945cf9f209","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 10:13:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"762af915a800bcf0e5191eec17307fede2ea8995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 13:10:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d8a5d620924c2b8bf37c46e6330ef8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e0dc9e1a04fd412d8a9487ba7449a0c5 : SUCCESS in 1h 23m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/49c01585748343fd9a9eb24f99b68b42 : SUCCESS in 1h 32m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7ee56d0a9dca468484e8c975db79dbc1 : SUCCESS in 1h 14m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f8d13b4999c436793eb1adcb9e32c42 : SUCCESS in 1h 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93bbdb5de54040e791ae7eb0b241299b : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5f7f916c8174774b5880887d38faeda : SUCCESS in 28m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdf15f72b5de4630b94eccfb80670fea : SUCCESS in 25m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec595bda03904cba8fa25d376a1d3dff : SUCCESS in 9m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1f8a98d50674a85bf9a7396fd7a623e : SUCCESS in 25m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4fef6554aee4b93afae5bca0353cf7e : SUCCESS in 52m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/84eb68b2b7d0469793185618f4357bae : SUCCESS in 1h 39m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32b17f5d007a457ba0e54fe7613079c4 : SUCCESS in 1h 30m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b42b816c078b491bbfa993f1e10f9804 : SUCCESS in 49m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/af3045eae36647d58a50c10e69143614 : SUCCESS in 1h 01m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/101b5ae0ee92452c824e9a4840242e40 : SUCCESS in 51m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/499876e770b74ec6b1bfba8081e842b0 : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7304f54f3471dcbaf439d67662ee55740e3487a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 13:10:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1800a18667229b597a431c2f9163b1926acc0d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 13:12:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2515d819bb4144629ec3c7f76f37cbfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58e0b9c76654337a722a5c2d49bacea : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55b16d7b7c5e85422c4db47ba666f2de95ef0e15","revisions":{"55b16d7b7c5e85422c4db47ba666f2de95ef0e15":{"kind":"REWORK","_number":1,"created":"2022-11-09 00:50:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/73/864073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/864073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/864073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/864073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/864073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/864073/1"}}},"commit":{"parents":[{"commit":"ba795c669228a32cee0618950a5355b141bbe497","subject":"Merge \"Check subnet overlapping after add router interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba795c669228a32cee0618950a5355b141bbe497"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-03 23:50:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-11-03 23:50:54.000000000","tz":-240},"subject":"Fix some pylint indentation warnings","message":"Fix some pylint indentation warnings\n\nRunning with a stricter .pylintrc generates a lot of\nC0330 warnings (hanging/continued indentation). Fix\nthe ones in neutron/db.\n\nTrivialfix\n\nChange-Id: I9311cfe5efc51552008072d84aa238e5d0c9de60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55b16d7b7c5e85422c4db47ba666f2de95ef0e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55b16d7b7c5e85422c4db47ba666f2de95ef0e15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
