)]}'
{"id":"openstack%2Fneutron~865039","triplet_id":"openstack%2Fneutron~master~I902cc542f2d0e81ff7d1c4379199a5cf92af907c","project":"openstack/neutron","branch":"master","topic":"bug/1869244","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-28 17:25:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I902cc542f2d0e81ff7d1c4379199a5cf92af907c","subject":"[DNM] Testing patch","status":"ABANDONED","created":"2022-11-18 15:32:10.000000000","updated":"2022-11-28 17:25:00.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31ae6f35e72c180fbc2efa45620730c74d9ea770","_number":865039,"virtual_id_number":865039,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-11-25 13:59:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-25 14:40:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-18 17:33:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-18 17:39:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-18 18:11:02.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"CC"},{"updated":"2022-11-18 19:16:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d46650b55d25b0d08abbf4c1d0cdb76c838e99c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-18 15:32:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37a3c662c202aa9530d28bb8009ef6aa35c720f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-18 16:29:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da56f88e63d2f8996db4423c80548b91615f2127","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 17:33:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59892c86f12e45b19c9e9d819c028f52\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c1f3de166c347cca14755e990edf004 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7659d9edff8c47d7b44f68a5fb03a430 : TIMED_OUT in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3352e3b538c7d044b303c1f001d857ebc66ee2f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 17:39:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/721940e1695c470eb593ef038120014d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f653449037243eaa3858fe16a530b9f : SUCCESS in 12m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55829b446ad5469098485ff591756968 : SUCCESS in 21m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8eddae746cf643cf9c7012fe5737d4dc : SUCCESS in 22m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a5cdd81c5b474108a61bbd5ef95bd121 : FAILURE in 1h 01m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e2e539867f82e97301192978ad581c3429f8be23","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-18 18:11:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bce26602e98e4978117925a78b911946f6de8e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 18:58:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77096a68c5e94e558756c97f6c5f8b18\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c9bde7234c847ffa959e042af72a916 : SUCCESS in 45m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a00b3b8fb9e34ddc95cfd096b2b506d2 : SUCCESS in 43m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c93e76b259d314ea176591e5171387c5cd6a2b7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-18 19:16:00.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a1cfa1c4af42ff7cbfe306ade6c0db70fad16973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 19:16:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8be10ff950949bc9627bb17f7ad4d92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d9003b537944b594c0e1798e3bd96a : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dcd316aa8ec4a31a1321877e8e0ed6e : SUCCESS in 23m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bea236b764614b3bb126bae66876f541 : SUCCESS in 21m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ae9baa059a214add8d4b0f99d2dcf9de : FAILURE in 1h 03m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6cca7f10d0ec851fff50c191b84b6099f3c5ccad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-19 00:39:22.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6eab53de1cb27799ee498afe7db97cc111b8049b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-19 05:08:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dce67b27e1896fe99c99ea730c817ac3fcf17a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 05:47:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f713b8f19283409b839dca4057a61ac1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2168086eee842df960c7e6d4837f3a2 : SUCCESS in 34m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/44a65b769745478786ecbc7b7a7f400c : SUCCESS in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"883b4ccb3ec1a01620cb5b6a545c871cfce3cb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 06:04:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b78dd65a27534866b50ac9521963c4fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74589dd7dc0d430bb66daf28fb534fb3 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc8ef7c3bfd848b39c7e3746c4a80fd3 : SUCCESS in 19m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72f59bf3808e4368b92138c7eb7904a7 : SUCCESS in 19m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d6ad4894b2f448a2a33bf5540b23fc38 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c5096e0215703ee70e9c1ea86411dec811e557c6","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-19 21:54:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7767a048c6ee6f56a419a28e6c5a7dd913ce196f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 22:41:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff4c46fac7a46cd9df3f38ac31ff99f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5f81cb943f7453c8c366f70572297d2 : SUCCESS in 39m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b5d42ed6fda44cfa38fb42429d8ef07 : SUCCESS in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d01e3da36b8ccf20ed376f0fa59d0d60eba24194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 23:13:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9e7328c3c24744b620b759f2dc220c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16b0a73a7c14f0588b68bfb7cdf49f0 : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be964e3aff2a444a879d1107c37cab64 : SUCCESS in 33m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c24f23c3d064f8ea6582729e758b74a : SUCCESS in 32m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd7ca709580b44a79d699a36ea71a863 : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8ee743355bf42e08e994b1d197172d030573ba67","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-19 23:53:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d018b9e727aef7b9dd4983babfa6a03387ad60a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-20 00:39:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde67ac105bd4049bcf885554cd57eac\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bebac3ec08d4abe8330b062684b3e23 : FAILURE in 43m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/416e2f6513e14771bd015eac64a1c228 : SUCCESS in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c9c5e91a41e5742617f58e233ce2e61107972a7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-20 01:00:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45c4066ed4d740eea580e245ee58de9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd3a96aae7f14f27b2da620a18687ce8 : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65d12cbdfe4e469089c7fcad7ad69fac : SUCCESS in 40m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47a1cfdcfb434854bddf01f12cf05459 : SUCCESS in 28m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d0cb0a13f093459993499f74f699df03 : FAILURE in 1h 01m 36s","accounts_in_message":[],"_revision_number":2},{"id":"87725b519dfcc9a98b1f808bec86e559a3cb711f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-20 15:54:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3aa127e1eec470ecb53b51f870add7253622f86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-20 16:33:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c0cedc80384189a6b485c4e792233f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e19cd27dcd54a3dadda6ea2e4879245 : SUCCESS in 32m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41e69eb4055149d6865d84f70e327df5 : FAILURE in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ccf457d4a96e62b3682cad988543c15cc56c924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-20 16:42:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10e4322f6b6c48478ca9f226f12cffd7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e7be8b54874d9ab8d454fd3c99e860 : SUCCESS in 12m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58ae0fdc9e7341c4acc40d10cb9f4b84 : SUCCESS in 21m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1d9f9f94e6c45f29b04b9a9ba937694 : SUCCESS in 24m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cac984273b024fc0b9fa1fdce57641de : FAILURE in 46m 01s","accounts_in_message":[],"_revision_number":2},{"id":"76906b23c76b985a33c3bf92eb9f90901f7aed06","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-20 19:56:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be9465f0f7488ff2c19d1a48895cc22c98fcfeb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-20 20:42:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b97548af0554d69a1cf16782f75f82b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6420c800a7534ed7b3ec2e840fcf87ed : SUCCESS in 31m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2212fcb244e43a5aa970854ac0e4268 : SUCCESS in 42m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b5e23c3702f9ccc73332a6523851b2d660384b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-20 20:43:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab1009603a354b03a038f6fee13c7017\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c85b986af274d74a52965ce1601c816 : SUCCESS in 15m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ef3a78d94cb488297e7ca6ef73cbc4d : SUCCESS in 20m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39affcd266da49bda512a2a1c05f0bc6 : SUCCESS in 19m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ea1415efe3b4f4ba340298bc7ba058b : FAILURE in 44m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8dbfbaf2fa1c4ff3f815bd4be07e37252f5b04b7","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-21 00:18:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b295a3f9a241e715f47edfef752aea8564df045","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-21 01:59:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc149038c9f4d1e1878ca0d76ddda6352fedd9a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-21 12:57:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"11ac293205db1d7b0d912662aec9fad7e4a45e94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 13:48:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1f6bdb446f4066a35e0181fae95068\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cad298862f34429abb39a4ea87314b47 : SUCCESS in 45m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/307f505f7ff04216a3a2ab6c267df6c1 : SUCCESS in 47m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c8cdaf7d74b8f45a47e05bcb32072778ae6e676e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 13:58:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c7ae5d9fb3e476b82e5edcdb1cd502a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11edd03390e84a18895c9acffb442368 : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2015565df7c4c9d90f81954af87272d : SUCCESS in 25m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e5cb0e124b64befa5f1b460ad96abcb : SUCCESS in 27m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d24b974a40974aa4bbe0e85ec5bd31b3 : FAILURE in 58m 17s","accounts_in_message":[],"_revision_number":3},{"id":"813c0251aba559a7fe315fd00ea5345f2c2d43df","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-21 14:15:38.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b560e5fff82cfd8736d9f4d489ab1deb89214002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-21 17:16:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5be1160ba407eb002dbd59bfb4212e273bc17a15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 18:26:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84328aed798f4e399e50d83a19bed131\n\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32632674ce2a45afad0f684083e73ec4 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":4},{"id":"f3ab493a60743db7cbfdcf080da2c1e80739f2db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-21 18:31:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"20a1566f2ff0c16535be278fcaa287bc45f15345","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-21 18:41:15.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"5184b886d73a7bfa155e2fe75dfbdc1b48ee221a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-21 19:38:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4efa752528d44d3638bc2877e88b2deba6157da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 20:54:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2fd9b2473484ab5b7a9094789f5eda0\n\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0da7050f6f374422af8dbbaf2d0a3d90 : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":5},{"id":"db292cb808b7b42c7c7beb7c4de6da22069df1a4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-21 23:03:31.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"9a865df266d05da9526bc4ac967a6042e759a3fe","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 07:42:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d36a2228b3a2e7358b5434a42a314e674bd0fcb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 09:08:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628301cfbc6040f396bea35e9bf27838\n\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f567266bff0943f3ae9cacdeb544a0ec : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":5},{"id":"33fdcb328ace4e190c5e623f989d921f26fb480b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 09:24:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"60e245a9b6968a3b2416e212601435c454363fa0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-22 09:57:28.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"10b85a7581d9eb860a2dfafdebbcc7d0edb7c52c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 11:00:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3bfa35d33d74bc7b1b7f5d818193f3f\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/7ef9ca6fb98d480eb048da593699a880 : FAILURE in 57m 34s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/f5c4fbf48607442ead49ee26daae33a2 : FAILURE in 1h 10m 29s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/d6e9dc3d52634aef97fd4630c868413a : FAILURE in 49m 27s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/6ba1ad742cdc4e648ee5ad66f47f6f3d : FAILURE in 51m 53s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/c7ebb40f4901437b9e49a6516444ef64 : FAILURE in 1h 06m 55s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/b126653092b847ca9284761b8c7e44d1 : FAILURE in 1h 08m 24s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/340171989e1644bcbebd5d312b17b54e : SUCCESS in 1h 28m 04s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/c9091ec15c88445186d602b8ff319c83 : FAILURE in 1h 10m 48s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/89d96413ce2943c0b0559ca0938cc08b : FAILURE in 1h 00m 30s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/99128e009bd943d79c55a74fbc110f0e : FAILURE in 58m 10s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/a7c4b2bb5eaa4e00bff45119a9e38de8 : FAILURE in 56m 44s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/d3b07d1baf514a79b7e9427a8fa8ac3d : FAILURE in 49m 56s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/0411d2ba4b504debaa955d6ce1719359 : FAILURE in 55m 16s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/a4dab65016454723bc7b5a99dc4f5160 : FAILURE in 47m 14s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/178e2acff5f141769dbc60dfe04179ee : FAILURE in 53m 02s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/37591e0179a54a598aceb38a55b70732 : FAILURE in 49m 52s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/20e500a6fc0d4e4f9ef8c0d98d3922a5 : SUCCESS in 1h 22m 07s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/1ea8a128112143509b604ebef22d7aac : FAILURE in 55m 47s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/9d5d03ac4ac1417dae72c9de5cdb525c : FAILURE in 1h 04m 55s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/34a78c38ff3f471cbde07b825d767ce0 : FAILURE in 55m 07s","accounts_in_message":[],"_revision_number":6},{"id":"b475d12fe4c7f48d92e246c61b1aaa44e98fdd50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 12:03:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"40f9cc9c902a231f5795ea59028f5052291619c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-22 12:39:16.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"78dbb4b3e4010c17e0fe3f30bccae298dfdd46d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 13:28:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/302f4d5e5f2f45e487d47268970e83b6\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/5ed1b422d80d40cfb7a835efb66a15e3 : FAILURE in 1h 00m 05s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/d39d2d1f509a4d2eb0d2c22525d6672e : SUCCESS in 1h 03m 15s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/a46c47e2598a47ddb5ce7014e242630f : FAILURE in 53m 50s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/f079dd8a63d6462da44ad47c6279b6c3 : FAILURE in 48m 58s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/a908386062a84b148418d9ddc8732607 : FAILURE in 1h 05m 27s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/e87ab1d00af642139228b9daa9dc80bc : FAILURE in 58m 00s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/f34f48dd1acb40beb36d3c9822613b66 : FAILURE in 55m 13s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/8cb5506a7ba94ad69f5b4fa301e800d4 : FAILURE in 49m 29s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/87d8841159fa4cba8f6e4715dfe2fcb1 : FAILURE in 1h 14m 20s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/8e07632ca22d4499b08cc1e56990c222 : FAILURE in 54m 48s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/19d6d27a8ac9451a911a8b15207e9349 : FAILURE in 57m 02s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/f2cce7d14734491d8e6e4d03c461ebc7 : SUCCESS in 1h 13m 28s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/f754d4050d114f4cbe3e3e846addee31 : FAILURE in 49m 24s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/f59a01e7812e47fc8dc3507e7161f879 : SUCCESS in 1h 16m 01s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/aaa12d275fbd4ed8a4e82b96ca74cf4d : SUCCESS in 1h 10m 13s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/1ca85e3ada624111baa540e09e050d81 : FAILURE in 53m 15s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/c5ab840f2f0541f0ac28b300d7b75b6a : SUCCESS in 1h 10m 10s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/c8acef6abaa94a1ea479c9dba5614c76 : FAILURE in 52m 33s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/3472c106b89f457897d45ce0c7a98f84 : SUCCESS in 1h 14m 28s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/70356e5989f84ffeaab545e1498086ec : FAILURE in 52m 29s","accounts_in_message":[],"_revision_number":7},{"id":"fc930e669366f8f08be79e083269a17824de5540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 15:34:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a452b30a17db0b8d6e92bd72113a2a8fcbad0f8c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-22 16:09:10.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"21114eca2b02c05831df7de38c28e5ec2b5bfba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 17:01:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0277d5218b1be6edc1470ed1a60b191fa0db1656","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-22 17:46:46.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"d475d773987ac44ac08b43f5ebfa2f901069c60a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 18:50:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/080c8d8390b64f61b27e215f7701c7a6\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/a61f4c4d0db74c2998cf657f2d77fcf8 : SUCCESS in 1h 17m 33s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/3118d16382804043a373eb9e9cf49e27 : SUCCESS in 1h 13m 06s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/4b8c4a3e42cb4b1f97fc6d386d3f29fc : FAILURE in 1h 04m 45s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/2356abca857e484ba291374c71e4d502 : SUCCESS in 1h 20m 35s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/692610f897fe4e90a66e0ade690f4ca3 : SUCCESS in 1h 25m 39s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/b417d36458ab4ac1b60c419604fd326c : SUCCESS in 1h 08m 44s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/64caf5e73ff540e9af114a817256cbd2 : SUCCESS in 1h 22m 14s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/cf15693cfc9346deab271494b9e67a45 : SUCCESS in 1h 38m 30s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/6ec61d26587c40828834dd42ea5ebee7 : SUCCESS in 1h 24m 31s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/34ae4424815442b497b7afeeefadcf6a : SUCCESS in 1h 10m 52s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/aafa03f98c7d44dfae72682e125564b9 : SUCCESS in 1h 35m 27s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/8f925d628575492bb7bb486be897a22c : SUCCESS in 1h 25m 09s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/b98149a587c4465b845623fc19212fef : SUCCESS in 1h 15m 07s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/ab93eaeda07d4a45801a4cb2f672e230 : SUCCESS in 1h 13m 31s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/23eec8ff34fe47b4a3a9e19828f7be31 : SUCCESS in 1h 08m 11s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/cf9c232030874973871aba33b0667211 : SUCCESS in 1h 35m 53s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/30aa09b2d0c74c99afe54648f321141e : SUCCESS in 1h 41m 39s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/71a48cb01bfa466295e67b37e610b5f5 : SUCCESS in 1h 13m 09s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/61d4f2c718574eca91ef1b372fd09a4a : SUCCESS in 1h 15m 49s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/3a308ac38de34f81b6923cc43e0d735b : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":9},{"id":"95679b530b9e52f25c09ad364b2b376d278c1de2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 19:46:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"112547ec5aaec8b40d349976903d2482ea072b58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-23 09:05:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4fa4ea601b832ffc1efd45b02e80e56b1e5ffd4f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-23 10:44:40.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"ead0ad109b43c815fd5cdf6721992d5f71590fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 10:48:37.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3a715dde8cc43c39cb8cb2b8f85a4fd\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/dd150310799e4172b669e1d6e1b865ed : SUCCESS in 1h 28m 33s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/87cc64abc235498499da9912939dfb80 : SUCCESS in 1h 37m 19s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/4fc1dda155994fb2811159c11847db6e : SUCCESS in 1h 26m 03s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/300b177bceb94deaa170f6cc96acde4c : SUCCESS in 1h 25m 07s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/f0925b0df10d402daca961b498193266 : SUCCESS in 1h 18m 44s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/06f648657f894f1abaeb67d1a2d20d6a : SUCCESS in 1h 30m 37s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/e55cbe6e6d8c402ebccf5fb13711c6d4 : SUCCESS in 1h 18m 28s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/867fb1e84fa3475fa8acaf5393036bce : SUCCESS in 1h 34m 34s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/069ce3dcd05646a2854001e861039b26 : SUCCESS in 1h 18m 57s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/31b80a98302a4cc696239b6dcaa5d396 : FAILURE in 1h 09m 34s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/55a8ad908cc24a498c18141cfd779d0d : SUCCESS in 1h 35m 14s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/1f3ea0f2215c46d8b462f6f1fce150a0 : SUCCESS in 1h 15m 30s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/838dd60385844413b6f0dfc6ac9c8932 : SUCCESS in 1h 17m 53s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/da502225747d4c65974cc5a9b61009ba : SUCCESS in 1h 20m 25s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/28a7e4d345324a429ddb02a4f528bdbe : SUCCESS in 1h 17m 00s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/3bb2062088164a98bd3fc9ead5390a1d : SUCCESS in 1h 18m 59s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/78cc4695cf924c05925295a4e736ef45 : SUCCESS in 1h 32m 48s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/9f7d418a1e1b438c9a6ac282bcc55b72 : SUCCESS in 1h 20m 36s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/4aeeb5734ee44960a993c7dfe978a989 : SUCCESS in 1h 16m 26s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/089a07eaec6648e6a8b177c1f13b9be6 : SUCCESS in 1h 21m 24s","accounts_in_message":[],"_revision_number":10},{"id":"ee64596ad23be2e6471e058fa675988846974f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-23 10:55:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4d78da3a6a68a4bdd71fbc5f58209e64b2317cf2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-23 12:32:40.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"16de8e74df03ebd730951b2560e70dbd07fd6584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 12:54:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e2eac014b974c2a960f8e5ab79f9a6d\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/a4b88d3dbf9146819fdf64b4c1a11937 : SUCCESS in 1h 24m 04s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/dd5fd6c275c044c09ddbe583836c444c : SUCCESS in 1h 26m 46s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/f43390f4c36e458a994ade793ad1c22b : SUCCESS in 1h 31m 39s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/bad6efce9bce4da59cf850f61399d8a0 : SUCCESS in 1h 19m 36s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/036e98b95f5c41a7a06ced538957ac8b : SUCCESS in 1h 32m 20s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/141cfc59ca4e49528a75633f7ee761c1 : SUCCESS in 1h 22m 24s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/ce808e6ffad6402db371bcccebbdfd23 : SUCCESS in 1h 15m 51s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/45cdcdf575364dc0a1292dcf3ec37c45 : SUCCESS in 1h 25m 18s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/3d8dfbb9b68948b29b15982b5db855d0 : SUCCESS in 1h 18m 36s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/21705760749c4624aac4d8cad8c427fd : SUCCESS in 1h 26m 43s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/24e8c248abf44f7d8f0b35c169766108 : SUCCESS in 1h 46m 00s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/f57d518ba2cd49d4b8906c02c84afb72 : SUCCESS in 1h 30m 10s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/17910456be0b4f70a15695e748423f20 : SUCCESS in 1h 44m 17s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/602756f9f4ed4b308870e695a97fc331 : SUCCESS in 1h 24m 20s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/1c7ec635485f4d60b9c8d356436728c4 : SUCCESS in 1h 22m 20s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/9b8a96eb39824d1696863db8c5833378 : SUCCESS in 1h 15m 49s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/af12142fa64f4e6f804cde402a40a4b9 : SUCCESS in 1h 21m 03s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/1e4d53e3591b452b9319374c6ae3b87b : SUCCESS in 1h 26m 44s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/c653bb06d70f4e1396f922e6c471571b : SUCCESS in 1h 16m 26s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/0ff9769c709c445ea5b5f0068625ce18 : SUCCESS in 1h 29m 09s\n- neutron-tempest-plugin-ovn-01 https://zuul.opendev.org/t/openstack/build/48f0d33f12544c8395019bdac00c9168 : FAILURE in 1h 02m 10s\n- neutron-tempest-plugin-ovn-02 https://zuul.opendev.org/t/openstack/build/7609732ee7554319965f0355251420c5 : FAILURE in 54m 24s\n- neutron-tempest-plugin-ovn-03 https://zuul.opendev.org/t/openstack/build/a1376ee0c8464e7b90b0c5cda23e5783 : FAILURE in 1h 05m 57s\n- neutron-tempest-plugin-ovn-04 https://zuul.opendev.org/t/openstack/build/799d8ef7d36547fcb0373732e4498104 : FAILURE in 59m 16s\n- neutron-tempest-plugin-ovn-05 https://zuul.opendev.org/t/openstack/build/0a8890ebbc5c4fd18dd544d35f9f2192 : FAILURE in 1h 00m 39s\n- neutron-tempest-plugin-ovn-06 https://zuul.opendev.org/t/openstack/build/565c7b293e8746ae923c3bc6f7cefeeb : FAILURE in 57m 22s\n- neutron-tempest-plugin-ovn-07 https://zuul.opendev.org/t/openstack/build/68fa5f0a72ce4fecb254ea6777ec1c15 : FAILURE in 1h 01m 09s\n- neutron-tempest-plugin-ovn-08 https://zuul.opendev.org/t/openstack/build/16e2f15e091c4658ad17a2c3b6f9d200 : FAILURE in 1h 07m 09s\n- neutron-tempest-plugin-ovn-09 https://zuul.opendev.org/t/openstack/build/708fbb2150474277814ed199b5879e09 : FAILURE in 1h 00m 56s\n- neutron-tempest-plugin-ovn-10 https://zuul.opendev.org/t/openstack/build/d2a86123c16548e9904481053e8f1071 : FAILURE in 58m 39s","accounts_in_message":[],"_revision_number":11},{"id":"4265bafce0fc836988eaea5672c7cd1dddd653df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-24 15:33:02.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"e8bb86fd1a8ddf744e2724236ad696dcb9ab2356","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-24 16:10:30.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"f3fb9da63ab2229c6a24045f2201eca77ba381ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 17:26:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47bbc1f7c11b4e918d5799804c4cbbe6\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/a9b6bab643f049b29f99381e603f1af1 : FAILURE in 29m 51s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/13317aeab7594719835c1289d2af8da6 : FAILURE in 31m 22s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/2a86308a23f04952a52bdb33876563de : SUCCESS in 1h 15m 09s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/827b77604de54bb3b5a9974395a0f01a : SUCCESS in 1h 12m 55s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/71b38ae3736546ed835f4ba79e4e6f51 : SUCCESS in 1h 19m 13s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/5379948c5a784ab2bddf89fbf7bd822c : FAILURE in 29m 43s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/083fc11ca32c442db87d8fd2d77017e5 : FAILURE in 32m 56s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/8ec295c90dad490fbd346909293d65d2 : SUCCESS in 1h 08m 18s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/c7c09e0b49984f76bb1659038a3d71fb : SUCCESS in 1h 22m 29s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/c7bcf1f5ee7b41bfb91813949ae1dcc5 : SUCCESS in 1h 18m 51s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/2549f1e33cfd48919710212bc646566d : SUCCESS in 1h 17m 04s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/9b361b096b124664a83a77353a217266 : SUCCESS in 1h 21m 11s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/90c9b84c65974df4910ad8309f55e9bd : SUCCESS in 1h 20m 37s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/54f51857fffd4a88a3c71c1e016917a3 : FAILURE in 32m 14s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/d2352d95aa124406ac8e474edcb63732 : SUCCESS in 1h 22m 32s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/957d21ca47ea48a7beef2f6571a2ab4d : SUCCESS in 1h 18m 26s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/fcf4f94a69b840f0bd44d163f625fa9c : SUCCESS in 1h 31m 51s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/bc03b976e3dd416b90eebf9bff461630 : SUCCESS in 1h 24m 42s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/9aa222c27a8343febc5e32b9a2a95403 : SUCCESS in 1h 18m 39s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/c3cfcb967de84be881393e22018dfa61 : SUCCESS in 1h 17m 01s\n- neutron-tempest-plugin-ovn-01 https://zuul.opendev.org/t/openstack/build/9acb8f5520ba4074bada0b36b3a62e20 : SUCCESS in 1h 03m 52s\n- neutron-tempest-plugin-ovn-02 https://zuul.opendev.org/t/openstack/build/49f0dc6933a6430e9fa4b6ce228f198f : SUCCESS in 55m 55s\n- neutron-tempest-plugin-ovn-03 https://zuul.opendev.org/t/openstack/build/c5d5330e9e6b4669a06fc268350841c1 : SUCCESS in 56m 38s\n- neutron-tempest-plugin-ovn-04 https://zuul.opendev.org/t/openstack/build/af4a06873f0d4aa7a0a7435a63558576 : SUCCESS in 56m 54s\n- neutron-tempest-plugin-ovn-05 https://zuul.opendev.org/t/openstack/build/50f32dec22f64ac18f1682493833c5b5 : SUCCESS in 53m 14s\n- neutron-tempest-plugin-ovn-06 https://zuul.opendev.org/t/openstack/build/4782a79ab85243fe9a582c1ddc596ff7 : SUCCESS in 59m 38s\n- neutron-tempest-plugin-ovn-07 https://zuul.opendev.org/t/openstack/build/b44c17811af540bb9273bf569f5158a9 : SUCCESS in 55m 47s\n- neutron-tempest-plugin-ovn-08 https://zuul.opendev.org/t/openstack/build/670659d60bfa4804b045b81de9c5464d : SUCCESS in 1h 02m 20s\n- neutron-tempest-plugin-ovn-09 https://zuul.opendev.org/t/openstack/build/0d69fe986aec468d86e0ce98657ed592 : SUCCESS in 59m 34s\n- neutron-tempest-plugin-ovn-10 https://zuul.opendev.org/t/openstack/build/4e640fb13b7348408efc2f6412611dd8 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":12},{"id":"cdcbed91ea4f91a49bb4cd8787896b6340e802ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-24 17:32:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8652661b7540845892806b715b5f3cb5c8871fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 19:04:14.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6518d58ad13045af8fa990780e58042a\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/9bc3978aab4c4a168d50e92517609272 : SUCCESS in 1h 17m 37s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/c9eb32bb8db44a19bb976f353e1b9a8c : SUCCESS in 1h 19m 18s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/5be83f52fbf545daa07f8064f44d523c : SUCCESS in 1h 16m 24s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/7fd1f020cc454c1a9e2c176cb7c159b1 : SUCCESS in 1h 19m 29s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/728bc1ecdf064f8b9e4e74942e1cfe72 : SUCCESS in 1h 17m 43s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/28953ca2f8654d2681ea6864415af18f : SUCCESS in 1h 15m 45s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/78385eab2beb4fc7b47774b8dad1b0c7 : SUCCESS in 1h 17m 55s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/54d03bae88804a2ab3c8e1c1acd57044 : SUCCESS in 1h 18m 11s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/6376834f270a45b5b9294d172b6e7cc5 : SUCCESS in 1h 16m 07s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/ff109be178d447eaaa2809feff706a6f : FAILURE in 57m 26s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/a6fa2f657b1c4e459f3a42bc94922943 : SUCCESS in 1h 23m 26s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/98ce07633de441b5b11ccdc26e713d27 : SUCCESS in 1h 02m 48s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/4c6c8189d1934079ae5d2bd8b60f30ee : SUCCESS in 1h 29m 29s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/50257436f1574c91a222b0da63a08140 : SUCCESS in 1h 21m 18s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/d1f89800cac649d8b5a76739416fa521 : SUCCESS in 1h 24m 03s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/e041202531364a459f1ffe4ac5f9505a : SUCCESS in 1h 19m 31s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/9ac3ce00c4854390963d6645fd6e914a : SUCCESS in 1h 25m 03s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/46712924e676452898644bb6e78848ac : SUCCESS in 1h 17m 24s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/8fbbcade69c4448d852f6dbac38198ba : SUCCESS in 1h 23m 29s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/e0737e469482495b998db0b6e982e2ac : SUCCESS in 1h 15m 44s\n- neutron-tempest-plugin-ovn-01 https://zuul.opendev.org/t/openstack/build/4bffbc66b1604894bbd1f324914c4c92 : SUCCESS in 1h 05m 42s\n- neutron-tempest-plugin-ovn-02 https://zuul.opendev.org/t/openstack/build/e5eb3625da7b4f719d7cad3d4d315a9b : SUCCESS in 1h 03m 14s\n- neutron-tempest-plugin-ovn-03 https://zuul.opendev.org/t/openstack/build/0ae3155812464bbd873d4dbf37f3f912 : SUCCESS in 1h 10m 49s\n- neutron-tempest-plugin-ovn-04 https://zuul.opendev.org/t/openstack/build/4e51451bc09648e4b903d7cd7ca5b259 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-ovn-05 https://zuul.opendev.org/t/openstack/build/b556834b94894391a0ba78a80232b1d5 : SUCCESS in 1h 02m 18s\n- neutron-tempest-plugin-ovn-06 https://zuul.opendev.org/t/openstack/build/72bd9a7ecbf84ba386962149a6a8a73d : SUCCESS in 1h 03m 09s\n- neutron-tempest-plugin-ovn-07 https://zuul.opendev.org/t/openstack/build/535d885a96734d658174f232f9729b7b : SUCCESS in 58m 05s\n- neutron-tempest-plugin-ovn-08 https://zuul.opendev.org/t/openstack/build/3f8d877b652c403b9109b3ba861ec43a : SUCCESS in 58m 45s\n- neutron-tempest-plugin-ovn-09 https://zuul.opendev.org/t/openstack/build/2683082047a8439ca0d656404e128f40 : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-ovn-10 https://zuul.opendev.org/t/openstack/build/a051d39c3387427cb2c54cc4aa578fe5 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":12},{"id":"71c76d1511db7a0f6a6846c7c41b45d164f129cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 10:23:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a02fa37f9b43c6d131f91c0e86451b2197678958","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-25 12:03:12.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"12bd82c3021d04661eb92408ae11f1d3d992f7b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 12:19:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f821b3db1722486688200ac0c8d07459\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/ab949cf557e14ef6ad1ddab0bec25745 : SUCCESS in 1h 16m 22s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/f2b74ea7700f4e28adda0944cc6ccce7 : SUCCESS in 1h 18m 22s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/b83909fc10ad411a90d55d4c8cffa5c8 : SUCCESS in 1h 25m 59s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/8cef4829b0fb401e974513d2f70b5c7a : SUCCESS in 1h 18m 41s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/8b10c0f690674d0ea47ba6488c79ba9a : SUCCESS in 1h 18m 09s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/4939b024888041e896ba1f8e1e19fa3e : SUCCESS in 1h 39m 23s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/deec15e318a740c1868b062acb22137f : SUCCESS in 1h 30m 32s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/8ea44646535c46059535a471ba72ee0d : SUCCESS in 1h 26m 16s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/99b9108275404415965eb250eded25fc : SUCCESS in 1h 20m 37s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/4555e87a673f4af9a2c7c57e7b1834df : SUCCESS in 1h 32m 02s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/741399479a8a4be0b1b0e942b825d0d7 : SUCCESS in 1h 26m 01s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/d1c8cd001ede47feafa1b22dea24c0c3 : SUCCESS in 1h 26m 27s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/52f1d0f7ebce470e90a1f6b57054d99f : SUCCESS in 1h 17m 31s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/b6c63c70c1434b1ca7d9c54cc742cf8b : SUCCESS in 1h 27m 38s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/aafba7c74b084a1d9dcbde34f641cf49 : SUCCESS in 1h 15m 17s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/41af938a0a4d4b93971daf1e8e7dcb75 : SUCCESS in 1h 19m 17s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/8b8f34e4b1134acf80c9c463ebdc4c7e : SUCCESS in 1h 19m 26s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/f3e862fc80624d708c3b3fc8e55d43ae : SUCCESS in 1h 08m 48s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/696195e22207430f8b5d198577ed2f25 : SUCCESS in 1h 29m 58s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/233df11fd0364b429824c74d383e1abf : SUCCESS in 1h 10m 33s\n- neutron-tempest-plugin-ovn-01 https://zuul.opendev.org/t/openstack/build/f24a5bed136149e78f0f613afab3f6f8 : SUCCESS in 1h 34m 52s\n- neutron-tempest-plugin-ovn-02 https://zuul.opendev.org/t/openstack/build/4ba472596a1c4a6b9061bd6458fcf8a0 : SUCCESS in 1h 11m 18s\n- neutron-tempest-plugin-ovn-03 https://zuul.opendev.org/t/openstack/build/73f8e397ddde4ccabbe40e6f5839d544 : SUCCESS in 1h 11m 15s\n- neutron-tempest-plugin-ovn-04 https://zuul.opendev.org/t/openstack/build/d7c8f40aa24c46839c5a68ef5a091cc3 : SUCCESS in 1h 41m 11s\n- neutron-tempest-plugin-ovn-05 https://zuul.opendev.org/t/openstack/build/d4f795001295420ba2be325bcaec40b3 : SUCCESS in 1h 18m 09s\n- neutron-tempest-plugin-ovn-06 https://zuul.opendev.org/t/openstack/build/9584d06637a243f38c09fdc30e00ae49 : SUCCESS in 53m 28s\n- neutron-tempest-plugin-ovn-07 https://zuul.opendev.org/t/openstack/build/e181df0a4e9646f6a068aba80c68c8d8 : SUCCESS in 1h 08m 23s\n- neutron-tempest-plugin-ovn-08 https://zuul.opendev.org/t/openstack/build/715edcbfdb5943d5a52df6600a76e03e : SUCCESS in 1h 19m 58s\n- neutron-tempest-plugin-ovn-09 https://zuul.opendev.org/t/openstack/build/7bf2ca2b2cbe496c8ad38cff6607ca16 : SUCCESS in 1h 31m 52s\n- neutron-tempest-plugin-ovn-10 https://zuul.opendev.org/t/openstack/build/649e3577f7ce4e1e9830124d0706abbf : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":13},{"id":"d176ba15fdc7d30cdb868d4c84d045b8e1e2402c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 12:48:59.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"f4651c68fe18c3279f5748685becb9b15fb329bc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-11-25 13:59:32.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2022/Arista_Tempest_Testrefs/changes/39/865039/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"6ef933f3fb658bead1d0ae0d2096d32456aed860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 14:40:27.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4932a6a5b27f489da6edc5784cc7ebe4\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/82fd25afa46c47bb93b8418908808384 : SUCCESS in 1h 25m 25s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/8952a7ccbcdb43cb953feb0fde7b3092 : SUCCESS in 1h 29m 03s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/8012fe6ec3fd480783d0abfbf177f8cd : SUCCESS in 1h 27m 47s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/dfbd9aa1221049fcaeaf730376e8f37b : SUCCESS in 1h 15m 48s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/13fa70f5a78d4e6c97005893fdcb5187 : SUCCESS in 1h 30m 41s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/1a7780a0487b460e984d42e40864e824 : SUCCESS in 1h 42m 54s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/e3b37fa3fa9f44ec96cbd45aac6d5fe9 : SUCCESS in 1h 15m 58s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/3ec9f919a59c4798bc78088817e7b9d9 : SUCCESS in 1h 37m 48s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/46b13d72635a42a9919edf40c03abfb8 : SUCCESS in 1h 14m 27s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/b5275be56d0d4cb7bceb5e0aece176c1 : SUCCESS in 1h 29m 48s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/0d2e5c268ba242ebadacd8a744ea1944 : FAILURE in 57m 28s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/59190a76992d4a36a3d7819c1c49999a : SUCCESS in 1h 26m 54s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/acb856c6189e429ca0f28a56e2a340fc : SUCCESS in 1h 09m 16s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/fbae45625e2840169435218cc9bc75e7 : SUCCESS in 1h 29m 38s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/4ffe961d869f40c9a295190586fc7f8d : SUCCESS in 1h 14m 18s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/22818af5ce544cef98ff8b6fccbbd17d : SUCCESS in 1h 25m 19s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/b1d879c2139b4a71a69c56f80078c73e : SUCCESS in 1h 09m 17s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/5932f4d2e2094543896e462b412eeb77 : SUCCESS in 1h 22m 15s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/fcce05be00494dc6934d9b45498f9c8a : SUCCESS in 1h 11m 39s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/2c6de7d33fc34e0589cee5e7e5f80264 : SUCCESS in 1h 23m 58s\n- neutron-tempest-plugin-ovn-01 https://zuul.opendev.org/t/openstack/build/1fffb5c89ebb44f39d884df675a8f8eb : SUCCESS in 1h 00m 08s\n- neutron-tempest-plugin-ovn-02 https://zuul.opendev.org/t/openstack/build/d769330d05744933af716c1ca62c0c6b : SUCCESS in 58m 37s\n- neutron-tempest-plugin-ovn-03 https://zuul.opendev.org/t/openstack/build/f1a66e67305f4bceb1388d354d5b4cac : SUCCESS in 1h 10m 31s\n- neutron-tempest-plugin-ovn-04 https://zuul.opendev.org/t/openstack/build/385fceb066c3448fa6482b9cab8fe0a4 : SUCCESS in 1h 07m 20s\n- neutron-tempest-plugin-ovn-05 https://zuul.opendev.org/t/openstack/build/4f40f2b2a5df44f9acc3b00948efb64c : SUCCESS in 1h 05m 34s\n- neutron-tempest-plugin-ovn-06 https://zuul.opendev.org/t/openstack/build/6b45b269b8a9469f9e910dbce669d9b6 : SUCCESS in 1h 06m 02s\n- neutron-tempest-plugin-ovn-07 https://zuul.opendev.org/t/openstack/build/ad4d40030a6c41b28fbe887b87b31393 : SUCCESS in 1h 10m 49s\n- neutron-tempest-plugin-ovn-08 https://zuul.opendev.org/t/openstack/build/20a8bde8fbee48ae93a880039a57edee : SUCCESS in 1h 00m 27s\n- neutron-tempest-plugin-ovn-09 https://zuul.opendev.org/t/openstack/build/a5438567a0b042ed97fb7da48d5997ed : SUCCESS in 57m 51s\n- neutron-tempest-plugin-ovn-10 https://zuul.opendev.org/t/openstack/build/56289d45e15744edbd5fa3d9495fb4ba : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":14},{"id":"14bec20c3cbb0641d0a5cb94e937e526f6d3212c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-25 14:55:11.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"10a79be0593fa884b1c28f4cafb6f058c104ee7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 16:41:03.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46be16a339e04b29a48bac828aff848d\n\n- neutron-ovs-tempest-multinode-full-01 https://zuul.opendev.org/t/openstack/build/1594adf15eda469e9cd4d530dce28caa : SUCCESS in 1h 21m 48s\n- neutron-ovs-tempest-multinode-full-02 https://zuul.opendev.org/t/openstack/build/24460433060143a9bf68edd84d06a17a : SUCCESS in 1h 37m 47s\n- neutron-ovs-tempest-multinode-full-03 https://zuul.opendev.org/t/openstack/build/d757244f8ea84c16a249faf840c69ce1 : SUCCESS in 1h 31m 15s\n- neutron-ovs-tempest-multinode-full-04 https://zuul.opendev.org/t/openstack/build/ade89254893946fca34a806452fde06e : SUCCESS in 1h 34m 44s\n- neutron-ovs-tempest-multinode-full-05 https://zuul.opendev.org/t/openstack/build/2b5cec6dcb3d4ac1af0df1c81898b08d : SUCCESS in 1h 29m 04s\n- neutron-ovs-tempest-multinode-full-06 https://zuul.opendev.org/t/openstack/build/b23b089f14774d8f962484ec2d44dbf6 : SUCCESS in 1h 32m 05s\n- neutron-ovs-tempest-multinode-full-07 https://zuul.opendev.org/t/openstack/build/790e155ae92f4a4aa79d39ef107ca642 : SUCCESS in 1h 19m 15s\n- neutron-ovs-tempest-multinode-full-08 https://zuul.opendev.org/t/openstack/build/bc153dc4a2a84d91a2878661cae8ca2c : SUCCESS in 1h 33m 42s\n- neutron-ovs-tempest-multinode-full-09 https://zuul.opendev.org/t/openstack/build/d958429669f144b3930a3a47bbf706fd : SUCCESS in 1h 30m 30s\n- neutron-ovs-tempest-multinode-full-10 https://zuul.opendev.org/t/openstack/build/677b8091e3f44b8090a8068a6b81705c : SUCCESS in 1h 14m 02s\n- neutron-ovs-tempest-multinode-full-11 https://zuul.opendev.org/t/openstack/build/8ad6a7bdc9c64c0b9a17e2f17783ea72 : SUCCESS in 1h 15m 51s\n- neutron-ovs-tempest-multinode-full-12 https://zuul.opendev.org/t/openstack/build/de7863eb8319403fa5d8a8cc5e7fa1be : SUCCESS in 1h 19m 41s\n- neutron-ovs-tempest-multinode-full-13 https://zuul.opendev.org/t/openstack/build/63e2275279cc4ac09836f1cb1ee142e9 : FAILURE in 1h 00m 51s\n- neutron-ovs-tempest-multinode-full-14 https://zuul.opendev.org/t/openstack/build/7a89c6f624104f238af946153b82b8f1 : SUCCESS in 1h 30m 54s\n- neutron-ovs-tempest-multinode-full-15 https://zuul.opendev.org/t/openstack/build/936d4d2466e54e9a971b3d50b4fa196f : SUCCESS in 1h 32m 16s\n- neutron-ovs-tempest-multinode-full-16 https://zuul.opendev.org/t/openstack/build/bc77627901a84e19a1384024f449dfcf : SUCCESS in 1h 14m 52s\n- neutron-ovs-tempest-multinode-full-17 https://zuul.opendev.org/t/openstack/build/828fa80ff27649c7b15b6ac5939af97f : SUCCESS in 1h 36m 19s\n- neutron-ovs-tempest-multinode-full-18 https://zuul.opendev.org/t/openstack/build/0e3353076b504b0094cb7f9d1a6b91d7 : SUCCESS in 1h 12m 59s\n- neutron-ovs-tempest-multinode-full-19 https://zuul.opendev.org/t/openstack/build/55cb854fe73c4c0b92b7f18d88cd5e7a : SUCCESS in 1h 11m 50s\n- neutron-ovs-tempest-multinode-full-20 https://zuul.opendev.org/t/openstack/build/94d1fc86c24845a783787ae43102f9b4 : SUCCESS in 1h 15m 03s\n- neutron-tempest-plugin-ovn-01 https://zuul.opendev.org/t/openstack/build/668983d4ca37468d8b6825b32da0f308 : SUCCESS in 1h 07m 03s\n- neutron-tempest-plugin-ovn-02 https://zuul.opendev.org/t/openstack/build/41d815fa8ce2444e894a294393a36707 : SUCCESS in 1h 02m 34s\n- neutron-tempest-plugin-ovn-03 https://zuul.opendev.org/t/openstack/build/9748a26a15af4547910ec7133a14bc92 : SUCCESS in 1h 10m 43s\n- neutron-tempest-plugin-ovn-04 https://zuul.opendev.org/t/openstack/build/0eef67c2d69a42e2abebd77c0366b830 : SUCCESS in 59m 38s\n- neutron-tempest-plugin-ovn-05 https://zuul.opendev.org/t/openstack/build/d3227d5c67f041abae787a4f11f539cc : SUCCESS in 1h 05m 35s\n- neutron-tempest-plugin-ovn-06 https://zuul.opendev.org/t/openstack/build/690e102b5e214ae49efb6b30d693f4fd : SUCCESS in 59m 29s\n- neutron-tempest-plugin-ovn-07 https://zuul.opendev.org/t/openstack/build/4c3a5645ac934fcb922c2e90fceb1b4c : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-ovn-08 https://zuul.opendev.org/t/openstack/build/f465e3cfb4234a65a50a388f909ef765 : SUCCESS in 1h 18m 10s\n- neutron-tempest-plugin-ovn-09 https://zuul.opendev.org/t/openstack/build/2dea8939e0264aba930430df96d5a96d : SUCCESS in 1h 03m 43s\n- neutron-tempest-plugin-ovn-10 https://zuul.opendev.org/t/openstack/build/a25e6b2a239b40729501a447b70c3140 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":14},{"id":"31ae6f35e72c180fbc2efa45620730c74d9ea770","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-28 17:25:00.000000000","message":"Abandoned\n\nPatch https://review.opendev.org/c/openstack/neutron/+/837780 merged","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"d8c53cef1d14b8242c23eab4e46c174044a05021","revisions":{"13c35ad5ddda254678fa918387109bf89553143f":{"kind":"REWORK","_number":1,"created":"2022-11-18 15:32:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/1"}}},"commit":{"parents":[{"commit":"89d1f52b7a08b697d2f6fcfd0eb6aab23d083278","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89d1f52b7a08b697d2f6fcfd0eb6aab23d083278"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13c35ad5ddda254678fa918387109bf89553143f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13c35ad5ddda254678fa918387109bf89553143f"}]},"branch":"refs/heads/master"},"36a42c574df6bb7557e087ab37edeca11723ca70":{"kind":"REWORK","_number":2,"created":"2022-11-18 16:29:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/2"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 08:29:48.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36a42c574df6bb7557e087ab37edeca11723ca70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36a42c574df6bb7557e087ab37edeca11723ca70"}]},"branch":"refs/heads/master"},"1440c8feb503f1548c8ac657c53ad866fb0b06fc":{"kind":"REWORK","_number":3,"created":"2022-11-21 12:57:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/3"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 13:27:14.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1440c8feb503f1548c8ac657c53ad866fb0b06fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1440c8feb503f1548c8ac657c53ad866fb0b06fc"}]},"branch":"refs/heads/master"},"140a8c6fd4b754f26564744ab557bed7b1df9d93":{"kind":"REWORK","_number":4,"created":"2022-11-21 17:16:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/4"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 17:03:13.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/140a8c6fd4b754f26564744ab557bed7b1df9d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/140a8c6fd4b754f26564744ab557bed7b1df9d93"}]},"branch":"refs/heads/master"},"e7f61163e21d6f4055b3a3bd757a7de559dded64":{"kind":"REWORK","_number":5,"created":"2022-11-21 19:38:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/5"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 20:08:33.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7f61163e21d6f4055b3a3bd757a7de559dded64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7f61163e21d6f4055b3a3bd757a7de559dded64"}]},"branch":"refs/heads/master"},"5e00993109f7b1ee044644401832d3c23d6b4359":{"kind":"REWORK","_number":6,"created":"2022-11-22 09:24:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/6"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 09:33:23.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e00993109f7b1ee044644401832d3c23d6b4359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e00993109f7b1ee044644401832d3c23d6b4359"}]},"branch":"refs/heads/master"},"d6733fdd7261d5dc1ca9c06ed5edc80e085817a1":{"kind":"REWORK","_number":7,"created":"2022-11-22 12:03:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/7"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 12:33:37.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6733fdd7261d5dc1ca9c06ed5edc80e085817a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6733fdd7261d5dc1ca9c06ed5edc80e085817a1"}]},"branch":"refs/heads/master"},"348b8cf559655543418352bc19062141c101c4eb":{"kind":"REWORK","_number":8,"created":"2022-11-22 15:34:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/8"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 16:04:30.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/348b8cf559655543418352bc19062141c101c4eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/348b8cf559655543418352bc19062141c101c4eb"}]},"branch":"refs/heads/master"},"b7555bc39d5f55801e6e77442654989405b3433e":{"kind":"REWORK","_number":9,"created":"2022-11-22 17:01:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/9"}}},"commit":{"parents":[{"commit":"ca52af597b04243e37795c6b6c231d768091a0c1","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca52af597b04243e37795c6b6c231d768091a0c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-14 07:31:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 17:31:17.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7555bc39d5f55801e6e77442654989405b3433e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7555bc39d5f55801e6e77442654989405b3433e"}]},"branch":"refs/heads/master"},"9ee4169aa1d5a89b529290f3b95367d758b3be0a":{"kind":"REWORK","_number":10,"created":"2022-11-23 09:05:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/10"}}},"commit":{"parents":[{"commit":"59045b54df67831363a925e0e3c7388494bd21e0","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59045b54df67831363a925e0e3c7388494bd21e0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 20:59:06.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 20:59:06.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ee4169aa1d5a89b529290f3b95367d758b3be0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ee4169aa1d5a89b529290f3b95367d758b3be0a"}]},"branch":"refs/heads/master"},"7da7c53442dbfd729a882d23f075695da5cd2883":{"kind":"REWORK","_number":11,"created":"2022-11-23 10:55:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/11"}}},"commit":{"parents":[{"commit":"adc79d90b3cfba01c2d9fb3fb4c14a6d62ae8114","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adc79d90b3cfba01c2d9fb3fb4c14a6d62ae8114"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 20:59:06.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 22:48:30.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7da7c53442dbfd729a882d23f075695da5cd2883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7da7c53442dbfd729a882d23f075695da5cd2883"}]},"branch":"refs/heads/master"},"c5b7828ebdd6377fae88f34c91eebf50b74115a4":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-11-24 15:33:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/12"}}},"commit":{"parents":[{"commit":"3a6678e3c746de11772cbe0fe70c227f44f0b3be","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a6678e3c746de11772cbe0fe70c227f44f0b3be"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 20:59:06.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-17 03:26:55.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5b7828ebdd6377fae88f34c91eebf50b74115a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5b7828ebdd6377fae88f34c91eebf50b74115a4"}]},"branch":"refs/heads/master"},"a04d5557ad92c6fc99f8fcda0b6a0ba7ca64bd7b":{"kind":"REWORK","_number":13,"created":"2022-11-25 10:23:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/13"}}},"commit":{"parents":[{"commit":"d0834eeb73ba906c5876d7d294d13048a4805b9e","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0834eeb73ba906c5876d7d294d13048a4805b9e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 20:59:06.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-17 08:12:33.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a04d5557ad92c6fc99f8fcda0b6a0ba7ca64bd7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a04d5557ad92c6fc99f8fcda0b6a0ba7ca64bd7b"}]},"branch":"refs/heads/master"},"d8c53cef1d14b8242c23eab4e46c174044a05021":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2022-11-25 12:48:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/865039/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/865039/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/865039/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/865039/14"}}},"commit":{"parents":[{"commit":"33de608f04dcc8117eeba63876598dc2ae93013a","subject":"Avoid race condition when deleting trunk bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33de608f04dcc8117eeba63876598dc2ae93013a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 20:59:06.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-17 09:40:05.000000000","tz":60},"subject":"[DNM] Testing patch","message":"[DNM] Testing patch\n\nThis patch tests https://review.opendev.org/c/openstack/neutron/+/837780\n\nI\u0027ll recheck this patch several times in order to run\n\"neutron-ovs-tempest-multinode-full\" several times, testing that\n\"test_live_migration_with_trunk\" test is now working (as expected).\n\nChange-Id: I902cc542f2d0e81ff7d1c4379199a5cf92af907c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8c53cef1d14b8242c23eab4e46c174044a05021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8c53cef1d14b8242c23eab4e46c174044a05021"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
