)]}'
{"id":"openstack%2Fneutron~865895","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ic35b5d4b9110b832c10345c45ec62c0923237cfd","project":"openstack/neutron","branch":"stable/xena","topic":"fix-duplicated-routes-stable/xena","attention_set":{},"removed_from_attention_set":{"34380":{"account":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"last_update":"2022-12-06 19:13:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic35b5d4b9110b832c10345c45ec62c0923237cfd","subject":"Fix duplicated routes exceptions","status":"MERGED","created":"2022-11-28 20:56:01.000000000","updated":"2022-12-06 19:14:35.000000000","submitted":"2022-12-06 19:13:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865895-fix-duplicated-routes-stable/xena","cherry_pick_of_change":865525,"cherry_pick_of_patch_set":8,"meta_rev_id":"7987da92ce5e719a1752c1f5113b815ae796603e","_number":865895,"virtual_id_number":865895,"owner":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-06 19:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-12-02 17:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-12-06 15:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-12-06 15:09:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 21:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-28 22:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-02 17:23:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-12-06 15:09:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"c4419c4dea8e9680f7ede2ed41c9cfee4399f454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2022-11-28 20:56:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6f467e781a207aca73f78664c990c0f431acdab0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 21:41:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/580e2bceae1841c4a976ca458d0ac1f9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4354c2cd4fc8434da2a2d7493e6fd477 : SUCCESS in 43m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70809bfe1e504e5eacb134e64ff87da6 : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b374993ca2fe6db824d7ae6dd76a6a6e6c592260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 22:30:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29e0eb4a07146bd836cd99b6b409925\n\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/c283cdb59b20410b89c436768ac47db4 : SUCCESS in 58m 20s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/0551afd3fb0340a6a92ffd894c839da9 : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/b54b25eb21e741ab95ed81e85a4027c9 : SUCCESS in 46m 19s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/94e48a4dbc2f4a58be020d7a62905d73 : SUCCESS in 50m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/851741c0f16344868f5d7adfb67ce1d2 : SUCCESS in 30m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2362827cd4934daa80de1a28fb4f8c9d : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9dd25680c164732862846155dbc2ab1 : SUCCESS in 29m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a56a9f0eb3e42bdbeb1f05fc84aa898 : SUCCESS in 20m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/335e4f9142014743967fd6f1b81ac65b : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fefa5f79d34ae5ad7448d68d14fafd : SUCCESS in 10m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7265417d0e554bf29be8c7d3326e8780 : SUCCESS in 50m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6687810f76374aa587429440955b4a5b : SUCCESS in 1h 21m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7e7fbecadcd04f7cba33478b3c09755a : SUCCESS in 54m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/73dd43a2171e4bc3a3d724e6245edad4 : SUCCESS in 52m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d21b936ce5242b8a7e82b91cc522b4b : SUCCESS in 59m 22s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f0d3cfca83594d719c3f892a8bdedfda : SUCCESS in 48m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b22d57ccba445db878e13dd790dd824 : SUCCESS in 1h 27m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9887675f7ce342d2be59e8d9c1b642cd : SUCCESS in 47m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce81cbf3632d40cabbbbeb02d5c225ba : SUCCESS in 1h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fef453187848ab0236ee5adce7d84476a53f330","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-02 17:23:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2732a2639f72191fecfdb8df3f53aa2d847f30b6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-06 15:09:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7d73ea1b561d6e940cb203a609468d51a7f42ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 15:10:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4588027d784a5bcf36072f8346d3b54caf4526e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 19:13:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d68047192174367be73bd52f562d761\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0da5e93023ba4c0c835ef48ba5e7e6e5 : SUCCESS in 13m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512e15009c254831831534d77d6347ac : SUCCESS in 24m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2292772fb387430bb233c4531d7d110b : SUCCESS in 24m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8259313e4794ed6990ba9dafd7810ab : SUCCESS in 7m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b92e9e703fc4ef6a14e498162c748c1 : SUCCESS in 44m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6e4fa1f196146ca8d420aa2572747c0 : SUCCESS in 1h 25m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7251d51045ec46b8bf6e989ad24a88a4 : SUCCESS in 52m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e0a9027bc4b44a28c5356f14f341b6c : SUCCESS in 56m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20919dd3c6ae41738b676791affce09b : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5516e4d16db92180cbd6cd867c85f1e007916eb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 19:13:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7987da92ce5e719a1752c1f5113b815ae796603e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 19:14:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7867939ca5924875bdb867bb3eeab188\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/092cd2e1508b4357a66f7d0e165de325 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7559a2ea827b6f36ece8029cb3fceff0308fee03","revisions":{"7559a2ea827b6f36ece8029cb3fceff0308fee03":{"kind":"REWORK","_number":1,"created":"2022-11-28 20:56:01.000000000","uploader":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"ref":"refs/changes/95/865895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/865895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/865895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/865895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/865895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/865895/1"}}},"commit":{"parents":[{"commit":"61c43567629662f2c053a7176b0f21f969553725","subject":"Merge \"ovn: Use ovsdb-client to create neutron_pg_drop\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61c43567629662f2c053a7176b0f21f969553725"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky@gec.io","date":"2022-11-24 10:40:04.000000000","tz":60},"committer":{"name":"Maximilian Stinsky","email":"maximilian.stinsky@gec.io","date":"2022-11-28 20:56:01.000000000","tz":0},"subject":"Fix duplicated routes exceptions","message":"Fix duplicated routes exceptions\n\nSince the train release neutron adds routes with protocol static.\nKeepalived also adds the same routes with different protocols depending\non the keepalived version. This can result in duplicated routes inside\nnetwork namespaces. On l3 agent restarts those duplicate routes\nthen prevent the l3 agent from updating its router state\nbecause it runs into \u0027RTNETLINK answers: File exists expections\u0027\nwhen it tries to execute \u0027ip route\u0027 commands.\n\nThis patch adds the same protocol static to each virtual route of\nkeepalived\u0027s configuration so network namespaces do not run into\nduplicated routes anymore.\n\nCloses-Bug: #1956846\nChange-Id: Ic35b5d4b9110b832c10345c45ec62c0923237cfd\n(cherry picked from commit c813b658d0e5e0c00093f90f849fcf67ddca16cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7559a2ea827b6f36ece8029cb3fceff0308fee03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7559a2ea827b6f36ece8029cb3fceff0308fee03"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
