)]}'
{"id":"openstack%2Fneutron~866701","triplet_id":"openstack%2Fneutron~stable%2Fxena~I74b0d53bfc54b71d3e8b2d46739a944e5f5a6b6f","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1966394","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-12-07 17:32:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74b0d53bfc54b71d3e8b2d46739a944e5f5a6b6f","subject":"Enable dstat and memory_tracker in functional/fullstack jobs","status":"MERGED","created":"2022-12-06 12:15:00.000000000","updated":"2022-12-07 17:34:05.000000000","submitted":"2022-12-07 17:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866701-bug/1966394","meta_rev_id":"4f7e9710cc07dac6700b0698c19659e5f950d8df","_number":866701,"virtual_id_number":866701,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 17:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-12-06 15:06:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-12-07 14:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-12-07 14:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 13:18:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-06 13:32:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 15:06:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-12-07 14:37:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"5975e7e3d1dfed9ac9a2ed9adb5d27065656e085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-06 12:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"384da525d6a14fa403e8a854c5fba02271e8060b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 13:18:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6399580c8413425fa5368c97b7efba51\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/049fc45c934f47b59eb4c709af2b71cb : TIMED_OUT in 1h 00m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/322944954aa144c899ce7d4135bc1ffa : SUCCESS in 45m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6287931f5c87103bd859dff65545a28c73edfc63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 13:32:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d234ebdffe94f598dd36907f83c9f46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851a34a92ed54cd8b32bd6adac65db6e : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b5ddd7b05e4273bb5c7fcfc1c90389 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf9ff63b9364c5ebc49daf7f6959774 : SUCCESS in 9m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f195375ff044fd18fbdc95630f1b719 : SUCCESS in 38m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5257a34e0644432293a24ec573659055 : SUCCESS in 1h 11m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/04d69b972c674426a9543a2aa35cdfa7 : SUCCESS in 57m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7a86bb7ba8042d1954da4c5b56528a8 : SUCCESS in 1h 05m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1d1740c94d934d479acf8152fe309e46 : SUCCESS in 1h 08m 49s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b6e9696815ee458498b37a8ee81921c6 : SUCCESS in 1h 00m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/30c07e465ecb4d37ab61da42f3dc7e02 : SUCCESS in 1h 04m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2842d575b1d4529a2de37c6937886f7 : SUCCESS in 58m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/653ea3ec7fd849caa69f5240f6a5d5c2 : SUCCESS in 1h 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93e0c632e3940f764ee3d44b7194b1035fa1845e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-06 15:06:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0703da430347bcee9d420ae9054b3b04e5b00fec","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-12-07 14:37:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"651a8f285070433636cc7eee41bad3bd722c243c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 14:37:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b41ad7fc19dce0095aa0fd2942a406b61cff360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 17:32:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be8d6359a7b74e2993cc8944ff2e2672\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47236de77bcd4cb5b7c9e7075a861d6e : SUCCESS in 13m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c9d66dcfec49c892fa4fcc9c37a7fb : SUCCESS in 26m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458de93e58944a558f040fb07253191a : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/14a32d999dda4bb6a9baaab82a642704 : SUCCESS in 47m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63365181e0b94c139c97d7ea07823eb7 : SUCCESS in 1h 21m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e3014191a834d3bb824045c4042f3f4 : SUCCESS in 1h 00m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe7a6d29f4df475f919d9d6ce53dd194 : SUCCESS in 1h 02m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdfe98a7312e4773890c54cc4857f954 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5d8ace2f307b2ff0f9581ebcbfc29f24e9829438","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 17:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f7e9710cc07dac6700b0698c19659e5f950d8df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 17:34:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e527b500d5b34f89a200dac4acb3f541\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307563ef8f204d738781281ba0078743 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4716449325916477d8ddf20fe6b25a88aef9e3dc","revisions":{"4716449325916477d8ddf20fe6b25a88aef9e3dc":{"kind":"REWORK","_number":1,"created":"2022-12-06 12:15:00.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/01/866701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/866701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/866701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/866701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/866701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/866701/1"}}},"commit":{"parents":[{"commit":"0f499c104cafee4ce460d9566040f9d5207fe2b6","subject":"Merge \"Set bigger swap in the functional and fullstack jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f499c104cafee4ce460d9566040f9d5207fe2b6"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-01 07:47:55.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-06 12:14:09.000000000","tz":330},"subject":"Enable dstat and memory_tracker in functional/fullstack jobs","message":"Enable dstat and memory_tracker in functional/fullstack jobs\n\nThis Will help in troubleshooting failures related to high\nmemory or cpu usage.\n\nRelated-Bug: #1966394\nChange-Id: I74b0d53bfc54b71d3e8b2d46739a944e5f5a6b6f\n(cherry picked from commit 9bc447077bce642c25faed9306537b4f1bf57230)\n(cherry picked from commit 969a0db8b9acc468e994f437aede9952bb6acb78)\nConflicts: roles/configure_functional_tests/tasks/main.yaml\n \tzuul.d/base.yaml\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4716449325916477d8ddf20fe6b25a88aef9e3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4716449325916477d8ddf20fe6b25a88aef9e3dc"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
