)]}'
{"id":"openstack%2Fneutron~866702","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I74b0d53bfc54b71d3e8b2d46739a944e5f5a6b6f","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1966394","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-12-10 03:05:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74b0d53bfc54b71d3e8b2d46739a944e5f5a6b6f","subject":"Enable dstat and memory_tracker in functional/fullstack jobs","status":"MERGED","created":"2022-12-06 12:17:12.000000000","updated":"2022-12-10 03:07:21.000000000","submitted":"2022-12-10 03:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866702-bug/1966394","meta_rev_id":"e47f401b3b7fe5ff494dbd4291110a8a4df285f1","_number":866702,"virtual_id_number":866702,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-10 03:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-12-07 10:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-12-07 14:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-12-07 14:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 13:21:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-06 13:52:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 15:06:25.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-12-07 10:05:58.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-12-07 14:37:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"446faaf9d033947ff6d9a9c9b24a4e701621175c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-06 12:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84b6dff9287c3b9ea433a0a25464e9703e6156eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 13:21:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084489f842764f2b80d96f6003fd1150\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7816795e53304cb68d8c5f18957cf007 : TIMED_OUT in 1h 01m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/618c837746f74ae4b201df410621342c : TIMED_OUT in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8b894a3d0f9b90c363ce3faca4b5cb1e3c7ab43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 13:52:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0eb38e5134074073b49e2458acab9686\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530c1145222b4b83bde0fba1858cab49 : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8701ac33b1b4695a545c4a08b065717 : SUCCESS in 26m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0391a052ca34579b9e06331ef963350 : SUCCESS in 8m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc2ed7207b864d9f813489fc0990abde : FAILURE in 54m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ed5b74182764819a395dd5eea347c7e : SUCCESS in 1h 04m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a450857fa14145b88e08a11ac00e008a : SUCCESS in 1h 20m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73f9fd720ae4494da84b05f698dc3565 : SUCCESS in 56m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5bdc958ea849484d98e801c0355e209a : SUCCESS in 1h 00m 07s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/af1eb98ab6ad4aea8850a73df48d1a59 : SUCCESS in 1h 09m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f49f7fbe61a3402b9b85a188fe644eb1 : SUCCESS in 1h 29m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2e8c3250e3a4890beda4c276fc58795 : SUCCESS in 1h 13m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1bff6706944778906a25ca641c6e7f : SUCCESS in 52m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/08bc4a543faf46e19c6952a582f59e38 : SUCCESS in 56m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e69a53975a8a4bf68235bb3874a8ea43 : SUCCESS in 24m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/32805c44ccfe4d19a5ea4ebbc370b98b : SUCCESS in 1h 13m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/038f3680fbe1490487df333da0b78755 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"03baff8d67fd90a9d257115f57930269d46d431b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-06 15:06:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a78cbb951f51850b5ad42ea2f2b71f8481fe76a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 16:36:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e4213060a7b41a399258845a7f5f42b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e5cb7b52802467d9ad03aeb94992d54 : SUCCESS in 1h 01m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3253b52fc2b4cc0b56e0ca92bffe690 : SUCCESS in 58m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c4deb0d0bf5a5cab815525c1eddb2d562fb9dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 16:55:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbba069ca2bb47fb93002e3637f5fb5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81fcbece38c3418c8365ad37176dce0e : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b6c420acdde4385ac097dd2352f07df : SUCCESS in 23m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be558d481174b43933d5e526741c3f9 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88344d4d5c824d16b1dcb62a2cff31ee : SUCCESS in 50m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5638e49b0d5d4c5888f689452daa637d : SUCCESS in 1h 15m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/acd5a2b4cc0f4f03a86a6cc3d91fc953 : SUCCESS in 1h 12m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69b002dfa276485e8d7bc742ff2bcff2 : SUCCESS in 57m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e091bb0213554ffa8d559d9b84746c03 : FAILURE in 1h 04m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/41e0229ce30f4326b37e3624474a96ec : SUCCESS in 1h 24m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e22516bdce4f4c7abae0f199e220d5e4 : SUCCESS in 1h 26m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66fa610daeb2475b98759b40c4c82f41 : SUCCESS in 1h 41m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e139eae32e514a7e80707334935d3257 : SUCCESS in 1h 02m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/debe8ed46e7b4a6f98dad6ee6e97f954 : SUCCESS in 59m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/52a99f83d2c14a44a3e390d840f8739c : SUCCESS in 30m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50168c1fef7a49dab73983dc3d19fe09 : SUCCESS in 1h 24m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/605af15b0eeb4c6293c32a5667639f6b : SUCCESS in 1h 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b3d2b87807accf4f7f11743ee1f32c0499518aa2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-07 10:05:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84efb7b6e19ae80b555a6c6857b5fce63415e4e8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-12-07 14:37:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6af2ebd67af3421a03e76254bc46d08e2facb1f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 14:38:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d5cdfb7e05b72107df21c97e97f8c79523fa7b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 18:25:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7cb9cdc38b6405bbff6c5ffd3f43fad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce5fde5df0dd4c8db323487a4b350d04 : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fd9c6a1f4b4decb633e2f209aaffa2 : SUCCESS in 8m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c9a7bd5f88948d581449d78c618ca79 : SUCCESS in 46m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c0b1d813a9f4fb8b162cf98c601132b : TIMED_OUT in 2h 12m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a1c65ba045b6443b8ef9ad4d23ef8d61 : SUCCESS in 1h 04m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb739c6dd8a64d86ae9731db8b06730c : SUCCESS in 1h 00m 50s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71a39f9a143246198ea6f8ed6543d8ab : SUCCESS in 1h 25m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4926c05d33a549e59f0cdee14475e152 : SUCCESS in 54m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/927b49a6524e47d58c8dba3161fb66b2 : SUCCESS in 1h 03m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/07fa1014fd164a8293b257b6f5567ba6 : SUCCESS in 35m 26s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d4668a1958444b368dc6e6f8f556bec1 : SUCCESS in 1h 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4b47ceac61c7ab594be56060cc800bc12075aa0c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-08 05:52:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbae834dd4443b58be7bc9e19d9aeb8ad21d99f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 07:42:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c07c8b6cbc4741188981f07b7c36c370\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc0e64b2ee24251ba18f22fe50d8796 : SUCCESS in 13m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c0b4614ba084d9a82da953d326b37e8 : SUCCESS in 23m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2f9470af6cc4612b8ae57cd7cfd91b7 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1afde9dab82d4989a0a5096e048c27cf : SUCCESS in 43m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/52da6512eea74705b382ad1d6e8a3ede : FAILURE in 1h 06m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4ce51926ee5340409907fa7c95f746ef : SUCCESS in 1h 14m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81e427dd3e5247008fb8590d8ddc56ba : FAILURE in 30m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/537665cf63704a37a1d2cd589256be9c : FAILURE in 23m 32s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/18dc814068e14aebb52621baeaa77e77 : SUCCESS in 1h 02m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e025b4eef5fe4185b13aa77625dbcafc : SUCCESS in 1h 21m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a918193895c14a2184d289666d01ac1e : SUCCESS in 1h 43m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3976054d0b0459f943e6e3edb03123d : SUCCESS in 51m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/69444f2c5a2644d190a2bc2194572b99 : SUCCESS in 1h 06m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0f13c79b6f694d8ca9daf95b37d76fcb : SUCCESS in 31m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e530fa2bd43495a8988a25e3670dafb : SUCCESS in 1h 13m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/24467738664b415f8089e60b2d12d447 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8c894b673b0721891d586c697c1a876d233301c1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-09 22:40:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f10b6580a1ce8688c96162307cc7dbde8a66978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 00:32:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16a89687aef4809bca91ce91c98ce8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd93aa8fdd44e4086d24178926fe61b : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddf179e3c65340a984297966dad42009 : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beaa034f4ee64a958abdb1414459360a : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a38fde22fb34116984390b5c08bba11 : SUCCESS in 51m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f3e16ea1ccb45978dd87f7f82bcf5d4 : SUCCESS in 1h 24m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3cdbade45a0845fab70056c09d4f2702 : SUCCESS in 1h 03m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/83799975dd1949d08338554eb98d2574 : SUCCESS in 55m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4bc6b81d9f8f4a089b9076b050e8defe : SUCCESS in 1h 07m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/70e2e777266e4f79aa2549fe4e1dac92 : SUCCESS in 1h 07m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6f29dc8766fc462eacfbaeef446d34e9 : SUCCESS in 1h 06m 33s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c0552dd6a9d84f53bd7aa4e24b8f7218 : SUCCESS in 1h 46m 11s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1230ccfe34da425abcea05b4768e41d0 : SUCCESS in 57m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0657085b53234239857af233f36268b6 : SUCCESS in 1h 01m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/072c9e14002d4738b59c18005a04953d : SUCCESS in 25m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14eae124e0804a6d8d162e6ba1fcbff2 : SUCCESS in 1h 17m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dd7e9ebb0e8141dda0b0456298e6b3e4 : SUCCESS in 1h 09m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2d344f9f2c669a9ccf9109f39a54837031387805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 00:33:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01c3dbcb451dee80384eb17558caf7b1ba6ac667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 03:05:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293d5201e172461891620af919a55fa4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ec55c94df94596b7eae59aea829a5c : SUCCESS in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40a8c81300a74391b599591e8997822d : SUCCESS in 8m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/53a1f4865f794d1c858b768ae04e077a : SUCCESS in 51m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b46f607b4b274dfeb8b6edfa2632fac6 : SUCCESS in 1h 20m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0dc7b242517f43dca555a1c7d963d4b3 : SUCCESS in 1h 23m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc5386717b814aad8654b89b4325869d : SUCCESS in 1h 12m 00s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e305d382bfa4d8bbd1980aa9e8e05dd : SUCCESS in 1h 51m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c60d2ea568504121b645699ae1dbe653 : SUCCESS in 1h 06m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/96572d0322c94c62b82bcb61524d0fc0 : SUCCESS in 1h 09m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/55dad227f95b4402b4c3b583adaa3efa : SUCCESS in 37m 03s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/05181d350c724ec1bf4cb2bb7cd02fc5 : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":1},{"id":"057e7785babb51cd5218f28bab6c4e84a819fbbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 03:05:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e47f401b3b7fe5ff494dbd4291110a8a4df285f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 03:07:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c1ee7a0cf64390b3174b720fba6b53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af3d0883bde41d486840c18c608e82a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c23f17902350611fc60c2de5fb97d0b57190250b","revisions":{"c23f17902350611fc60c2de5fb97d0b57190250b":{"kind":"REWORK","_number":1,"created":"2022-12-06 12:17:12.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/02/866702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/866702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/866702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/866702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/866702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/866702/1"}}},"commit":{"parents":[{"commit":"566f7edd4aa6141b23edda52f747e3d146267965","subject":"Merge \"Set bigger swap in the functional and fullstack jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/566f7edd4aa6141b23edda52f747e3d146267965"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-01 07:47:55.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-06 12:17:00.000000000","tz":330},"subject":"Enable dstat and memory_tracker in functional/fullstack jobs","message":"Enable dstat and memory_tracker in functional/fullstack jobs\n\nThis Will help in troubleshooting failures related to high\nmemory or cpu usage.\n\nRelated-Bug: #1966394\nChange-Id: I74b0d53bfc54b71d3e8b2d46739a944e5f5a6b6f\n(cherry picked from commit 9bc447077bce642c25faed9306537b4f1bf57230)\n(cherry picked from commit 969a0db8b9acc468e994f437aede9952bb6acb78)\nConflicts: roles/configure_functional_tests/tasks/main.yaml\n \tzuul.d/base.yaml\n(cherry picked from commit 4716449325916477d8ddf20fe6b25a88aef9e3dc)\nConflicts: zuul.d/base.yaml\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c23f17902350611fc60c2de5fb97d0b57190250b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c23f17902350611fc60c2de5fb97d0b57190250b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
