)]}'
{"id":"openstack%2Fneutron~867964","triplet_id":"openstack%2Fneutron~master~Ib614fd5fcb4159642695ab0b380fbc8ae49c0776","project":"openstack/neutron","branch":"master","topic":"bug/1999209","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-17 08:21:05.000000000","reason":"Change was abandoned"},"28056":{"account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"last_update":"2023-04-17 08:21:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib614fd5fcb4159642695ab0b380fbc8ae49c0776","subject":"[OVN] Update fip while fixed_ip of internal port has changed","status":"ABANDONED","created":"2022-12-16 10:14:32.000000000","updated":"2023-04-17 08:21:05.000000000","total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"96d0425a694915bf1aa8d129a3d1aaf2713448d3","_number":867964,"virtual_id_number":867964,"owner":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-12-16 11:53:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-16 11:58:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2022-12-16 11:17:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-16 11:17:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-12-16 11:23:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-16 11:53:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-16 11:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 08:21:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"2d763e80c75bc487de3a60ec96accfdc86dce6bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2022-12-16 10:14:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e83eba40bb529b7b47ff7b5871cb2557872f0ae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-16 11:17:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0f389e5df3d63f77a33364f633d7547f4ab111f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 11:23:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4860a0a4d3e4406a2e1e9746dc0656a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39ae8070a2af477eabdd5561bad24b9b : FAILURE in 1h 06m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/965cfaf1ec88494bbfe2ceb0719e2b01 : FAILURE in 1h 05m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d1aa3249b2928d26e1d3bda3dfa8f6b697f89fb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-16 11:53:07.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/64/867964/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a948764e97283d4f319a255c79d7e73614f6da00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 11:58:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2beaaf016bda4522ad0aab8629b81444\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ab6dd38e722499ca08de51da4ca3c60 : SUCCESS in 1h 23m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8750c716f52c43458b6bddcd0df6c129 : FAILURE in 32m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f022611d4f4375a4d62711b1741ed2 : SUCCESS in 10m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7a2d5c7415f4cf39991f56183a5261b : FAILURE in 26m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebcd1c97b531492bb671dea5c93ccb6b : FAILURE in 22m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0747db93471b4fae96d71a3630a71b2d : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc4e7cd5288f494f95c86a5ee190ec56 : SUCCESS in 53m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/62aa5b5e88ff4241813c28d014aa5584 : SUCCESS in 54m 10s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b8f6c7fc103f4cdfb32b4d64c296d22e : SUCCESS in 1h 06m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9a6917cbed2433ab8209e60830d0d3d : SUCCESS in 1h 42m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3e35e99a502c47bbb2bcaa20c27599a2 : SUCCESS in 1h 10m 28s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ffd7e1252c0c4595933f313db84a669c : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1f34682ad19a0a8f4805776b5030960570011407","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2022-12-16 12:56:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"96d0425a694915bf1aa8d129a3d1aaf2713448d3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 08:21:05.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4c29187c90cdeefb3c046e19127456a6d785a43","revisions":{"e4c29187c90cdeefb3c046e19127456a6d785a43":{"kind":"REWORK","_number":1,"created":"2022-12-16 10:14:32.000000000","uploader":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"ref":"refs/changes/64/867964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/867964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/867964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/867964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/867964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/867964/1"}}},"commit":{"parents":[{"commit":"35ea844158d1e380c8e0dbf85a4d485aac460472","subject":"Merge \"Amend documentation for fullstack tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35ea844158d1e380c8e0dbf85a4d485aac460472"}]}],"author":{"name":"liushy","email":"liuxie_11@163.com","date":"2022-12-16 10:02:32.000000000","tz":480},"committer":{"name":"liushy","email":"liuxie_11@163.com","date":"2022-12-16 10:13:45.000000000","tz":480},"subject":"[OVN] Update fip while fixed_ip of internal port has changed","message":"[OVN] Update fip while fixed_ip of internal port has changed\n\nIf we updated fixed_ips of one internal port which associated\na floatingip, but the dnat_adn_snat entry in ovn will not change.\n\nWe fix it by that it always query nat entries related the internal\nlogical port from ovn-nb when update_port, then update fip if nat\nentries are not empty and logical_ip is not change.\n\nCloses-Bug: #1999209\nChange-Id: Ib614fd5fcb4159642695ab0b380fbc8ae49c0776\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4c29187c90cdeefb3c046e19127456a6d785a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4c29187c90cdeefb3c046e19127456a6d785a43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
