)]}'
{"id":"openstack%2Fneutron~868078","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1999517","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-12-20 10:58:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8","subject":"Check if port exists in ``update_port_virtual_type`` method","status":"MERGED","created":"2022-12-19 12:38:40.000000000","updated":"2022-12-20 10:59:44.000000000","submitted":"2022-12-20 10:58:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868078-bug/1999517","meta_rev_id":"f3443bcded0c15283b1048dc3cf1866c0bd4eaa0","_number":868078,"virtual_id_number":868078,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 10:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-12-19 23:04:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-12-20 09:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2022-12-20 09:19:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 13:26:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-19 13:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-19 23:04:19.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-12-20 09:19:56.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"be08a9ecd3972244f16e9b4adcc4ddc7c46766a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-19 12:38:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47fe2eed73faa3102cbb9e6637c92c430b0a9e43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 13:26:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a5bafc54ec443183b28ff39d59c29e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec3c228608d042c893e98a57b529e6df : FAILURE in 46m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/156577ea40ff4e129f6c4267ea3ac161 : FAILURE in 46m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2aa7e76cc04e122b03975d9877086bb4a812e16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 13:57:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef8f07cf6ab949e7974a2e24336701bd\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8215112c209d4230b990ae6e7d3e2d39 : SUCCESS in 50m 24s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/620b4e6fc6e9440d931cd719ece0d1ff : SUCCESS in 1h 16m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e7788ce8994c0dabeb5b4c9a1409ce : FAILURE in 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f0adf8815e423a8f7f85befc65df31 : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/677f4703082449dea45f000ed0c91964 : FAILURE in 30m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac9470ea31554e69a30216c4a78ddaca : FAILURE in 22m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbbf8ecdc825496db2e37dff1cdaad9a : SUCCESS in 8m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f80bca65627344ebad03e71426488339 : SUCCESS in 54m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/792af6babe8f4a3194a58d20d3178df4 : SUCCESS in 1h 01m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d8c102cd11a248399cc00004072fd0f1 : SUCCESS in 1h 09m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d667fb2df272561c219f232068e9e45aa9f1ab9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-19 14:39:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"191c5c8e775648b38163fb54b91568fc340c73dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:37:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ada5864cd9ac4afe9cfce9fc61d2b4be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c53df9f10f0849568407df58c927b6c3 : SUCCESS in 45m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1706c6ff84d246b68b112916030ca858 : SUCCESS in 55m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b5ca7d094202f706b248345cbee24ea754d39db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:58:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be414fe87404baeb644b78ec794aee7\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/04a0c8ecbfd141ffa12a9cdf8282829e : SUCCESS in 43m 14s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/a5fbd0a193a24f59b1d4dbdddd0a67b7 : SUCCESS in 59m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e263acef439471abf7c1a36aaef87b4 : SUCCESS in 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34f3617628274e8da7be2c95e160938e : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/062213377a894941a55341c69aeea6ed : SUCCESS in 27m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cca44b6848204e5d9733cbf5b8782e49 : SUCCESS in 26m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502a1a7d8184483e804d935c532d7c3c : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5bf69d8d15c245f083c8d2dcd9403089 : SUCCESS in 45m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4da993a74704ebc8e24bd55bbbe4355 : SUCCESS in 1h 13m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91755e493e0740728d5a83dc2bc40214 : SUCCESS in 59m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29b6073fdd0e687041796a077bde1ed795716b87","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-12-19 23:04:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"940e9a6c08933fb3e53458d4b8400011e31c461f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-20 09:19:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8eeec0e8054096a848a43886dcc3651f317bb5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 09:20:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9522026501844e74dd9654839977329bc27d3022","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:58:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d3ad0db3854c0e970f0ffc9f4c9017\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/44a8b5c215494eb19b6fb6b81925279b : SUCCESS in 48m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15f24044437d451dba32f10c0dd43ff7 : SUCCESS in 12m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc0a1f68f3e400ab65def49f5c7b727 : SUCCESS in 24m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adbce0ca96814d519ceea14182e8cc37 : SUCCESS in 27m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7bdbec83e3482da7021bc8288c5055 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/57f09edf917a49648d9cf91869627a6e : SUCCESS in 44m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/025f62e5ed704a2c99ed5169364856e7 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d21c0f1d012c0aef8f11a05fea2ac0378caf62df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:58:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f3443bcded0c15283b1048dc3cf1866c0bd4eaa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:59:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39133a36e1574adfaee75418027e4168\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d01f2f015c184b71987ec4b34d8a9839 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"951e2c74ae0baff84e98b5580071d09db95f00d5","revisions":{"1440615363a622895618e8c6fa3847e9d4cebc5c":{"kind":"REWORK","_number":1,"created":"2022-12-19 12:38:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/868078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/868078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/868078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/868078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/868078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/868078/1"}}},"commit":{"parents":[{"commit":"a4b8c97a2860d54609137f76e2d4e932804179a6","subject":"Merge \"Always create a \"router_extra_attributes\" register per router\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4b8c97a2860d54609137f76e2d4e932804179a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-08 05:30:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-10 22:00:54.000000000","tz":60},"subject":"Check if port exists in ``update_port_virtual_type`` method","message":"Check if port exists in ``update_port_virtual_type`` method\n\nDuring the OVN DB inconsistency check, a OVN LSP could not be present\nin the Neutron DB. In this case, continue processing other LSPs and\nlet other ``DBInconsistenciesPeriodics`` methods to resolve this\nissue.\n\nCloses-Bug: #1999517\nChange-Id: Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8\n(cherry picked from commit dfe69472a82d7eee96299122c8a9520aeed73ddd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1440615363a622895618e8c6fa3847e9d4cebc5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1440615363a622895618e8c6fa3847e9d4cebc5c"}]},"branch":"refs/heads/stable/yoga"},"951e2c74ae0baff84e98b5580071d09db95f00d5":{"kind":"REWORK","_number":2,"created":"2022-12-19 14:39:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/868078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/868078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/868078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/868078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/868078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/868078/2"}}},"commit":{"parents":[{"commit":"a4b8c97a2860d54609137f76e2d4e932804179a6","subject":"Merge \"Always create a \"router_extra_attributes\" register per router\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4b8c97a2860d54609137f76e2d4e932804179a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-08 05:30:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-11 00:01:25.000000000","tz":60},"subject":"Check if port exists in ``update_port_virtual_type`` method","message":"Check if port exists in ``update_port_virtual_type`` method\n\nDuring the OVN DB inconsistency check, a OVN LSP could not be present\nin the Neutron DB. In this case, continue processing other LSPs and\nlet other ``DBInconsistenciesPeriodics`` methods to resolve this\nissue.\n\nCloses-Bug: #1999517\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n\nChange-Id: Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8\n(cherry picked from commit dfe69472a82d7eee96299122c8a9520aeed73ddd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/951e2c74ae0baff84e98b5580071d09db95f00d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/951e2c74ae0baff84e98b5580071d09db95f00d5"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
