)]}'
{"id":"openstack%2Fneutron~868079","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1999517","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-12-20 10:58:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8","subject":"Check if port exists in ``update_port_virtual_type`` method","status":"MERGED","created":"2022-12-19 12:39:01.000000000","updated":"2022-12-20 10:59:47.000000000","submitted":"2022-12-20 10:58:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868079-bug/1999517","meta_rev_id":"5371469e2d4ea7352bea08e20d3203b9a65cf7bb","_number":868079,"virtual_id_number":868079,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 10:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-12-19 23:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-12-20 09:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2022-12-20 09:20:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 13:26:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-19 13:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-19 23:04:46.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-12-20 09:20:07.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9b3931fbf6d94d4b939821972a65bd72a2a05003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-19 12:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f0e8a0e2c61fcee1209ed07492edfa172d1dc52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 13:26:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eea9d92b7ae4fffa8135447a440dd6d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/316dce65ecc6461db904cbf8d0ff6cde : FAILURE in 46m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ca42f0b25244995b5a226009bbeddfe : FAILURE in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8852a16203c4ae6d45f0c3942c1f612d29b9af6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 13:55:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea1764e5f0094636ae992708eafd8e7e\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/25c5680da56d4592aeb570f1927ff443 : SUCCESS in 56m 00s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/5cd1ceb65f374143968a43de0a2906dc : SUCCESS in 1h 05m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c59774a9ff6e4be5adb008cf5ad6c256 : FAILURE in 36m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71988832edfa40e1a1502737bf5c7e03 : FAILURE in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb5cc3ccec24448c9feb32896efcfe45 : FAILURE in 29m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f5c6d2e7d4944fa96b4e10fe1e3332a : FAILURE in 26m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40d22a1ff974438f96a777cb6aa4bd56 : FAILURE in 21m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28d4ceac37494ab9b575b161a0ba2b9b : SUCCESS in 7m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/138d85f7140a4693805dea90ae8728d0 : SUCCESS in 47m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b72d5732fce47aeac08d81b6abc0622 : SUCCESS in 52m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf90ad7af2aa444cadba14a1e881f2cb : SUCCESS in 1h 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70356094fae3b15e6833d851e7ab6f6da1b55204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-19 14:40:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42ce010482eb30178fa2db969e98b5122ca573cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:27:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb793f560a074eec8481406a1cf7b530\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73fa7ff91ed14210b5b5aeef1fb9e5df : SUCCESS in 42m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6966d1cec994352a466f95f77b351e5 : SUCCESS in 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca3c08d508363341e920ef58f76221edd4ee53ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:57:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa8dd8232f06464a9c770556272d59e2\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/20024097ba0c43acb1ed35d1895c78ec : SUCCESS in 57m 03s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/29508cb672ad4c24bf16a6a23fe5cc60 : SUCCESS in 42m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffded3a3c01e49539b3c4e60b8d0526b : SUCCESS in 35m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1891c31bbd408b8aac72233e1cd80d : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9203ca4d012a40d59ce0ada467937764 : SUCCESS in 27m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5827318892e41d595e28f78e886f7c3 : SUCCESS in 25m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42434e52723d471fb115fe6b578411c0 : SUCCESS in 26m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd2b51aec3d416da586407bfb5cba84 : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6b4936db61b4d3681713736c89595ee : SUCCESS in 47m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46e18d4f066a43198fad64fadeb62686 : SUCCESS in 59m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/086ead7204c9445f9173f75b1d334664 : SUCCESS in 1h 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"476af432a531ceef37fb15cafda9d780fa88c559","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-12-19 23:04:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13d1fd70e62dc72016d5ab8dc9bc3310a4b43d7b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-20 09:20:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c2af628a114d852425d1041765cd2a3e25f740ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 09:20:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b238f9f58683693d0451a6a39fbc688da02e00ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:58:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ffc4efe9a544f09c6288834b8813f8\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/a210792d0bfa4d01989bd5d7653138ea : SUCCESS in 47m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31bbf0f2383e4f53a3f0810db35453bf : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81080d27cce943a0990a0f05b04a790a : SUCCESS in 25m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f63ca6328cb4d5c9470b72e754c2f37 : SUCCESS in 22m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc63be4eb59144de87c5ab47662594f5 : SUCCESS in 7m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d614f3ea3170414a8146273a9246f2f9 : SUCCESS in 46m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3da79adcb9848768bcb2974d81f53bf : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":2},{"id":"cee576c589a680f66d18ed9be15a55536990417d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:58:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5371469e2d4ea7352bea08e20d3203b9a65cf7bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:59:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aac46af57294a368292796a9a5b9a9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/565bd5cfba7f45ccabcdfb7b65e5f28c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"037ba074f01e317a43d0387684afd528453126b5","revisions":{"209df4c13c9c1d1bc9817ec4b1713f292599cfae":{"kind":"REWORK","_number":1,"created":"2022-12-19 12:39:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/868079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/868079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/868079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/868079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/868079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/868079/1"}}},"commit":{"parents":[{"commit":"023d2c6e1a237b723b700df77d297988d3066037","subject":"Merge \"Always create a \"router_extra_attributes\" register per router\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/023d2c6e1a237b723b700df77d297988d3066037"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-08 05:30:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-10 22:01:16.000000000","tz":60},"subject":"Check if port exists in ``update_port_virtual_type`` method","message":"Check if port exists in ``update_port_virtual_type`` method\n\nDuring the OVN DB inconsistency check, a OVN LSP could not be present\nin the Neutron DB. In this case, continue processing other LSPs and\nlet other ``DBInconsistenciesPeriodics`` methods to resolve this\nissue.\n\nCloses-Bug: #1999517\nChange-Id: Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8\n(cherry picked from commit dfe69472a82d7eee96299122c8a9520aeed73ddd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/209df4c13c9c1d1bc9817ec4b1713f292599cfae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/209df4c13c9c1d1bc9817ec4b1713f292599cfae"}]},"branch":"refs/heads/stable/xena"},"037ba074f01e317a43d0387684afd528453126b5":{"kind":"REWORK","_number":2,"created":"2022-12-19 14:40:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/868079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/868079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/868079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/868079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/868079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/868079/2"}}},"commit":{"parents":[{"commit":"023d2c6e1a237b723b700df77d297988d3066037","subject":"Merge \"Always create a \"router_extra_attributes\" register per router\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/023d2c6e1a237b723b700df77d297988d3066037"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-08 05:30:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-11 00:02:37.000000000","tz":60},"subject":"Check if port exists in ``update_port_virtual_type`` method","message":"Check if port exists in ``update_port_virtual_type`` method\n\nDuring the OVN DB inconsistency check, a OVN LSP could not be present\nin the Neutron DB. In this case, continue processing other LSPs and\nlet other ``DBInconsistenciesPeriodics`` methods to resolve this\nissue.\n\nCloses-Bug: #1999517\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n\nChange-Id: Ifb8bdccf6819f7f8af1abd3b82ccb1cd2e4c2fb8\n(cherry picked from commit dfe69472a82d7eee96299122c8a9520aeed73ddd)\n(cherry picked from commit 951e2c74ae0baff84e98b5580071d09db95f00d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/037ba074f01e317a43d0387684afd528453126b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/037ba074f01e317a43d0387684afd528453126b5"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
