)]}'
{"id":"openstack%2Fneutron~868151","triplet_id":"openstack%2Fneutron~master~I74f22015ee5fe167ffa8f76ee9dca859e99e3427","project":"openstack/neutron","branch":"master","topic":"bug/1998749","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-22 01:24:05.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2025-01-20 15:01:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}}},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-01-22 01:24:05.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I74f22015ee5fe167ffa8f76ee9dca859e99e3427","subject":"Adds a RPC for ovs-agent to get server side config options","status":"NEW","created":"2022-12-20 03:14:52.000000000","updated":"2025-01-20 15:01:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"bbb2da1811fd2d3bf3e8c3745589b49a39ce99d7","_number":868151,"virtual_id_number":868151,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-12-05 09:44:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-05 10:36:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-01-22 01:24:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 03:42:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-20 04:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-20 04:43:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-23 02:16:45.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-01 18:50:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-22 01:24:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-06 18:16:26.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2025-01-20 15:01:12.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REMOVED"}],"messages":[{"id":"36c2fca9b42f12efa414ad5e48db09d0d34e65cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-20 03:14:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2642ed68e63253a087de721672c76891380a79f6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-20 03:42:32.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/51/868151/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"25f6395b2cd79f52cfbe6040275bbfc8e7af2e6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 04:08:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/375ba49b60b44df3b54df230a9cc24bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b16d53bff8b4350b4afe884579ea0aa : SUCCESS in 49m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c7bfd498cf44397aa2bd40da5c64f88 : SUCCESS in 50m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4deb183875044504a3acf123f88f63e9abd89080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 04:43:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f69d6356dde419c8b1c523d48137a0a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1f0ed60c0c2649ceb1e5d1f23bae0d2a : SUCCESS in 1h 17m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a86bdb428114c57b66426dda68094ee : SUCCESS in 1h 04m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12a997dc73b440a8aa21a6d2d6717c03 : SUCCESS in 1h 18m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/57c8c4b69ed1462cbb503f4697c59890 : SUCCESS in 1h 19m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c21241af6d1c48e4bfc148b09d9f81b5 : SUCCESS in 28m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e73334cf62e432c8c5cce47300ae37f : SUCCESS in 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0215efcbde4a42f694ee518cec64e8bb : SUCCESS in 13m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/767a4a8575e04f029607723bb4b6562a : SUCCESS in 23m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d39628526354b83b282c4ae9fd49308 : SUCCESS in 26m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dfeeb60e314a9dab5d35f4a26e7089 : SUCCESS in 9m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ac6889b1906415f8dfad871bf67a4af : SUCCESS in 42m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/41d46693da7b442badc8c55a7dc54054 : SUCCESS in 1h 26m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/baa8e2de2d74414cbb78be4b5a5ee5b8 : SUCCESS in 52m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/34b3cd675e574c97a13939916d8affd1 : SUCCESS in 44m 34s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/da30383e4e1a4912b7e04f1824ca2a89 : SUCCESS in 1h 08m 07s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6309a7c3a8ee492c8f5c3c96700de575 : FAILURE in 58m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aeee2b641b5445d79a20e19f3f79ef81 : SUCCESS in 1h 10m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b645d68e1ae4392aeea0f3982ea5dd4 : FAILURE in 20m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e03615ff3137431fb1b57888de63a428 : FAILURE in 17m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/db9ff52f16ee4211b2cd7f5f0a74feb6 : SUCCESS in 1h 16m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/267edeec588a48f8816e93927bee62e0 : SUCCESS in 1h 19m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/68f134c5c57f44e8a0bfa40afb4e687a : SUCCESS in 1h 13m 17s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ed78740c36f3482da7bdb5a5e4730c92 : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a5fd9935bdc405ca6942bed4ea812b5abb499bdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-20 07:17:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2bb01f111c75be49d6e877853589ac9ad006dd6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-12-20 07:20:46.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"48af678a9a3b77812ea1bbef75423f2df049ff02","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-20 07:44:27.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/51/868151/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a7bed5ee9995e6f6d6e1dea2081ad7352aa317d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 08:14:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/858ff368ec394976bccb1045919dc32b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9adb4f68bb2249e1861430962b957e83 : SUCCESS in 50m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bdd48297397a446fb8897f49b64fc62c : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4618298ce5a7b1bb3cb2320bb2d5870d3f62a9af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-12-20 08:35:29.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2022/Arista_Tempest_Testrefs/changes/51/868151/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"6fd664b05833c74ccd75e9e8331f0eccae47b477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 09:11:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f078da3173ac49f59e11d1b38f81e7c4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7a32a4544fa54041b8c98eb51d2285ce : SUCCESS in 1h 03m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/45fcc7af7e20444e90ab89d54bfa560f : SUCCESS in 1h 15m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/03c64d1ad2e84ca79d4c573b5b1a78ac : SUCCESS in 1h 04m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c7f0bc9d1134d94b2bb18a2f677005c : SUCCESS in 57m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0af8e01744f04e0eb67056f78985e2cf : SUCCESS in 38m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c221a1628eea4e2487be73b9937a3fa1 : SUCCESS in 33m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b5efd79e4a40bdb19f25ac7574b2c2 : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/281e1e1d6dbb4c2088d8d829b5adbec0 : SUCCESS in 24m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef61d6126cd34f34b8376aea43ec29e8 : SUCCESS in 30m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e86ec5eddd364ac9876176878d4450d6 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3db0595eed041958a8ce48f468762f4 : SUCCESS in 59m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c650366945540d7bcdaf204829ff61c : SUCCESS in 1h 47m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7728090104e74247a385e2cf8b3aeba2 : SUCCESS in 39m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/84ea33202d0e41528de37c50d57c65a4 : SUCCESS in 56m 41s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/494b5b18b17c4cfd8e0dee08739fc743 : SUCCESS in 1h 03m 15s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/22d79d3c493e49b88ff7b9389a939226 : SUCCESS in 1h 05m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d7a9cb5aeb404aae893e8f9b1cdc5fad : SUCCESS in 1h 39m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/84fe9b21dd0a41959118d031b8b3b2fa : SUCCESS in 1h 19m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c45c517c16ec41c78043edda75870ee9 : SUCCESS in 43m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/28439b3276204e7c900b4b4b09b89908 : SUCCESS in 1h 36m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e0b16c7715e04e9ca13d5b21b7325099 : SUCCESS in 1h 31m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/82669f046a364b0f98c22469698cf4ea : SUCCESS in 1h 03m 16s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2cd29eee05914f5b91932fe3c4ed75c5 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":3},{"id":"5ee7fd2928dc41c931358f86ad9257bb219d93e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-05-23 00:59:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"78f38be44786c6b926aadbe6c54371b64971b0f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 01:58:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6711e642cb4354a0aca1f4e7fa3edb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57f53017959b4ba99001ed5ad6672e60 : SUCCESS in 54m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93abbd65ac8244b8a785d4b09b0f9745 : SUCCESS in 55m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d7c73063f71429f9aa1d28997a27397 : SUCCESS in 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8568028280b9d80068200c2afed1483387443f9d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-23 02:16:45.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/51/868151/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d5c778480e5c3776664fe719f00ff0327c6429be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 02:55:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e120deda3dc4fa48a1d2dfddfcf4dce\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/92ffb844dfa64b15b35aed26ece5ae49 : SUCCESS in 1h 45m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c281c217a2e94d6ea48a00fdb30f22c8 : SUCCESS in 1h 18m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/618fd66849284b34a6f8dd83d8d5655d : SUCCESS in 1h 28m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/50a8ef7af1fa4ce384f9f1f86035120a : SUCCESS in 1h 27m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f0d81a63ea944253bcde319cf9c9fcee : SUCCESS in 1h 49m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06020ffb9f26462fa8f2ec7aa3399389 : SUCCESS in 46m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d50fb1955eb84f86982c3015dd2672f1 : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f60d308ef94bcb9be9689bdef9a472 : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87dfd3e28ba44537b14fbc019b3aefe0 : SUCCESS in 27m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be4e1c64b25b4882bac4643c868fb36e : SUCCESS in 29m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26d9c626ce064330bfbddb1342a275ea : SUCCESS in 28m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef97a66ffba45f795105570616d8cd5 : SUCCESS in 11m 53s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bd2bf8bf69f946b98c4c73453b798ca6 : SUCCESS in 24m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4ddc8b5e2d64c7fa83f56ef0156b07a : SUCCESS in 46m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/48e9dc150ade428b891e97b354b634ee : SUCCESS in 1h 31m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/03c10930e98a452ba3e598d7b5788840 : SUCCESS in 48m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f0823e8c0f8450ba1a4070ddaada13a : SUCCESS in 43m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0047fcdb7d984ca59703b90a30fe5e96 : SUCCESS in 1h 45m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e731b5ab9d4a4a18884e455b7a56c3eb : SUCCESS in 1h 04m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3dfa952944a64de288c7adc302a86475 : SUCCESS in 1h 00m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c4612eb77ab74e94bad4cd53581fd0ff : SUCCESS in 1h 30m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dfedceee87f44eb6bb9c6ebd06e83f3a : SUCCESS in 1h 40m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/98dc291e4d1347f2934e4f80790e7077 : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":4},{"id":"a9cabdf68dee780e9e73644ef94c297424c55d43","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 18:50:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"072630f255b501f94ac13d6a4f44256ccb388104","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-12-05 08:19:10.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a2edb0066a88638846b3900c1fcedd2c13064386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-12-05 08:26:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ea5316a5b7b18d01174ab22a61af9fdefe52e69a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 09:15:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc1a68d20bff400b98340c8cf9e041e2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0acf8fd02d74de68de858ff142c164a : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/394113d860744ac896e4d149f38c6b6d : SUCCESS in 47m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a29fdc2e393b43b3b1683ee35a24656b : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"329837587d51c04a25a9b1a4831b90dfcd614c59","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-05 09:44:14.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/51/868151/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"79019e4de4323f7fe91d5e7e062723f7b94f241d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 10:36:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8f629f7bf04b7c89d598778be8e473\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f83609333b342b08f349161c346ac73 : SUCCESS in 1h 34m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bbd1f24bec19464cac3fd2521c09c9c6 : SUCCESS in 1h 34m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8be9452901274a2dbff8c358570d0762 : SUCCESS in 1h 43m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd1760d7a7d847e7ba0e499598394307 : SUCCESS in 1h 25m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0ccc8c426a94636ada30472b18aafa5 : SUCCESS in 38m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2bb819c77cd41a4b1269461aad7780b : SUCCESS in 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1514575f553b42e59e29910ec559ed9d : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69526d39e6f84f28917567dcc6dea9ec : SUCCESS in 26m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be0542062a26440bac15b663ba132be8 : SUCCESS in 22m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7cc416bb9b4a96834a573b7b432159 : SUCCESS in 14m 47s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e1a2db47e7364ab09657d7397ac68269 : SUCCESS in 25m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/39add340a2f548a1ab3ed1153d9c20ff : SUCCESS in 50m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fcc93ae6459049c9924c99710a5ca8f1 : SUCCESS in 1h 12m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/168828885d5c4e249609a324fb197156 : SUCCESS in 46m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0b95255cc51493baf6d54cc05c81742 : SUCCESS in 2h 01m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd3f63420cbf4c709c44dfb0b2c293cc : SUCCESS in 55m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/21b7ce0e1cb8493e832040e36e203a0b : SUCCESS in 1h 11m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f15baeea9b18400687f74a3c19590c45 : SUCCESS in 1h 21m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/45d080d46f5c45acac6f41ae5145bf90 : SUCCESS in 1h 20m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4babe23bf732405ab76e4be88cda7c8a : SUCCESS in 57m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5ef638a93ef943c29cd442c30fd2a350 : SUCCESS in 1h 32m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3df8753291ad41eb98f8de5dacc9ebd6 : SUCCESS in 1h 38m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4b8adaa96d8841e494186bc4e19299b0 : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":5},{"id":"deee15c79ec3e89bc45945f11078ed098be436d0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-22 01:24:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"228087b12b611639336de4fbd5f0771b39e36a97","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-06 18:16:26.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bbb2da1811fd2d3bf3e8c3745589b49a39ce99d7","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-20 15:01:12.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_9656\u003e with the following votes:\n\n* Code-Review-1 by \u003cGERRIT_ACCOUNT_9656\u003e\n","accounts_in_message":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"_revision_number":5}],"current_revision_number":5,"current_revision":"c0b26db8fa87e40880c61a058e9be204f68c6121","revisions":{"9d3c09a0a985e26880d6f4f18b9ee03166a3271b":{"kind":"REWORK","_number":1,"created":"2022-12-20 03:14:52.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/51/868151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/868151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/868151/1"}}},"commit":{"parents":[{"commit":"faa28faea5fff37be11fa0dfb4e49f4b2ee35cdb","subject":"Add bridge_port_mappings to direct output for physical","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/faa28faea5fff37be11fa0dfb4e49f4b2ee35cdb"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-15 01:52:32.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-20 03:04:28.000000000","tz":480},"subject":"Adds a RPC for ovs-agent to get server side config options","message":"Adds a RPC for ovs-agent to get server side config options\n\nThis is a starts for ovs-agent to get config options from\nneutron-server side. The L3_HA_NET_CIDR \u003d \u0027169.254.192.0/18\u0027\nis set on neutron-server side, but ovs-agent needs this\nto add flows.\n\nCloses-Bug: #1998751\nChange-Id: I74f22015ee5fe167ffa8f76ee9dca859e99e3427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d3c09a0a985e26880d6f4f18b9ee03166a3271b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d3c09a0a985e26880d6f4f18b9ee03166a3271b"}]},"branch":"refs/heads/master"},"c1ea359a2c23eea991b36c0d92eccda336405651":{"kind":"REWORK","_number":2,"created":"2022-12-20 07:17:09.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/51/868151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/868151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/868151/2"}}},"commit":{"parents":[{"commit":"faa28faea5fff37be11fa0dfb4e49f4b2ee35cdb","subject":"Add bridge_port_mappings to direct output for physical","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/faa28faea5fff37be11fa0dfb4e49f4b2ee35cdb"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-15 01:52:32.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-20 07:12:53.000000000","tz":480},"subject":"Adds a RPC for ovs-agent to get server side config options","message":"Adds a RPC for ovs-agent to get server side config options\n\nThis is a starts for ovs-agent to get config options from\nneutron-server side. The L3_HA_NET_CIDR \u003d \u0027169.254.192.0/18\u0027\nis set on neutron-server side, but ovs-agent needs this\nto add flows.\n\nCloses-Bug: #1998751\nChange-Id: I74f22015ee5fe167ffa8f76ee9dca859e99e3427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1ea359a2c23eea991b36c0d92eccda336405651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1ea359a2c23eea991b36c0d92eccda336405651"}]},"branch":"refs/heads/master"},"7d4f424633dd8a0c1b13b93895262ade6146443e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-12-20 07:20:46.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/51/868151/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/868151/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/868151/3"}}},"commit":{"parents":[{"commit":"b17a6409052bd54bd312c42dac657b5b3dbecb67","subject":"Add bridge_port_mappings to direct output for physical","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b17a6409052bd54bd312c42dac657b5b3dbecb67"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-15 01:52:32.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-20 07:19:31.000000000","tz":480},"subject":"Adds a RPC for ovs-agent to get server side config options","message":"Adds a RPC for ovs-agent to get server side config options\n\nThis is a starts for ovs-agent to get config options from\nneutron-server side. The L3_HA_NET_CIDR \u003d \u0027169.254.192.0/18\u0027\nis set on neutron-server side, but ovs-agent needs this\nto add flows.\n\nCloses-Bug: #1998751\nChange-Id: I74f22015ee5fe167ffa8f76ee9dca859e99e3427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d4f424633dd8a0c1b13b93895262ade6146443e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d4f424633dd8a0c1b13b93895262ade6146443e"}]},"branch":"refs/heads/master"},"aeb5a445661cd36d32ef87a3c1637816cd5d9492":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-05-23 00:59:47.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/51/868151/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/868151/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/868151/4"}}},"commit":{"parents":[{"commit":"6c911b13e7ed8731f1b964dd442df6ddae0e51db","subject":"Add bridge_port_mappings to direct output for physical","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c911b13e7ed8731f1b964dd442df6ddae0e51db"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-15 01:52:32.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-05-23 00:59:02.000000000","tz":480},"subject":"Adds a RPC for ovs-agent to get server side config options","message":"Adds a RPC for ovs-agent to get server side config options\n\nThis is a starts for ovs-agent to get config options from\nneutron-server side. The L3_HA_NET_CIDR \u003d \u0027169.254.192.0/18\u0027\nis set on neutron-server side, but ovs-agent needs this\nto add flows.\n\nCloses-Bug: #1998751\nChange-Id: I74f22015ee5fe167ffa8f76ee9dca859e99e3427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeb5a445661cd36d32ef87a3c1637816cd5d9492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeb5a445661cd36d32ef87a3c1637816cd5d9492"}]},"branch":"refs/heads/master"},"c0b26db8fa87e40880c61a058e9be204f68c6121":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-05 08:26:04.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/51/868151/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/868151/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/868151/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/868151/5"}}},"commit":{"parents":[{"commit":"c1b81d059025acf1c94c0228aed10cd14bee1dc5","subject":"Add bridge_port_mappings to direct output for physical","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1b81d059025acf1c94c0228aed10cd14bee1dc5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-12-15 01:52:32.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-12-05 08:25:33.000000000","tz":480},"subject":"Adds a RPC for ovs-agent to get server side config options","message":"Adds a RPC for ovs-agent to get server side config options\n\nThis is a starts for ovs-agent to get config options from\nneutron-server side. The L3_HA_NET_CIDR \u003d \u0027169.254.192.0/18\u0027\nis set on neutron-server side, but ovs-agent needs this\nto add flows.\n\nCloses-Bug: #1998751\nChange-Id: I74f22015ee5fe167ffa8f76ee9dca859e99e3427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0b26db8fa87e40880c61a058e9be204f68c6121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0b26db8fa87e40880c61a058e9be204f68c6121"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
