)]}'
{"id":"openstack%2Fneutron~868264","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Icec8fee93e64b871999f38674e305238e9705fd4","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1997982","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-01-17 19:49:07.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-01-18 21:01:19.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-01-17 12:01:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Icec8fee93e64b871999f38674e305238e9705fd4","subject":"Fix handling the restart of ovn-controllers","status":"MERGED","created":"2022-12-21 09:11:17.000000000","updated":"2023-01-18 21:03:39.000000000","submitted":"2023-01-18 21:01:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868264-bug/1997982","meta_rev_id":"2987a088e2c6d46c36403260a3356d74567eb807","_number":868264,"virtual_id_number":868264,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 21:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-01-16 17:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-01-17 12:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-01-17 12:01:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 09:11:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"state":"CC"},{"updated":"2022-12-21 09:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-21 10:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 20:41:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-01-05 05:53:46.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2023-01-13 16:11:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"478075100b22e4b4f1dcf0c0858cdc6dbf2d724d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-21 09:11:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba38312393d3aa888ac01024182c92a582f4a90b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 09:59:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876507f2a3624898aca234b286906c86\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bdfba29e96c44d8901fc09a7b54baba : SUCCESS in 45m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79e37c42688343b28069206a1e6719da : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aca4381a13ca0d8e480fa643ab0b5913326a6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 10:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7888bbd7ff8b48958a25802a19f68467\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/47faef2065854c9992043a14829eacd0 : SUCCESS in 46m 52s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/772aa7db27f1451cb6a5ce784944cc5e : SUCCESS in 1h 09m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85e6afd7c4fd4c1382b4fc258c5a913f : SUCCESS in 42m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55de174958c4c99bc254dfe49a7154e : SUCCESS in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4551c9c663574473a67db4cd058d71aa : SUCCESS in 24m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92289696a2cc4a0fb04375fd5dff5f5e : SUCCESS in 24m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7029ac1545504627bc28f0ca85765163 : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6da64d41db147b09543a38dd18eb985 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cf536de17ad4b14a604c891480ab187 : SUCCESS in 39m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0b95e6a39de7485eb1ffd5c099b858f2 : SUCCESS in 1h 11m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/815f8c67b9954abbbc4f505753606355 : SUCCESS in 28m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/57a193359e3c41458eed6e6d56b7b008 : SUCCESS in 1h 09m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bd7f0cf84b874f8ea615f47f5fc0a4dc : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"69730861c7c249f40f23cb7253e662ef387c6f87","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-04 20:41:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6bbaa45dabf56b305fc0335fb08cef436cd15e41","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-05 05:53:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e17e6aac6cb23258f299e6f86af97331062cdbae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-05 10:59:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43b763a1da8c16c00678b9c67ab9a4d4153a0b30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-05 11:08:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac83997f024db3712c9322f0d9ade98a4e77b93f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 12:23:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db4ff869f77a4122a53fdb7c666dd53e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b738935fe3e422eb8172483d5418b94 : SUCCESS in 37m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2ae57afb7c54f06a799b9c60bc64c89 : SUCCESS in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3539903fdf7b8669753a712d0eaa350f02561608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 12:29:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31d7cf1d7be74fa8ba198ab358d116be\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/6d960f103b4d431695a95233d348f62f : SUCCESS in 51m 01s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/ea58f98107c44778a947ca252f6c1830 : SUCCESS in 52m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52b04a3f647b4e89aff6bdd7062edcaf : SUCCESS in 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40c445833154cafb17744c647c13c6b : SUCCESS in 12m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9caa03b6789490fb7ad38dc7ff1bac2 : SUCCESS in 32m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/301ebc5a93464d5186f61d206f65e37f : SUCCESS in 30m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff458f522ad46bdad23de1cd40e8985 : SUCCESS in 30m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8ef68006ae4c06aac28ff25f704579 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b8cb9a018f542528980fb78a69ae206 : FAILURE in 41m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/473260002a5c4262902b3e510b5ac227 : SUCCESS in 1h 02m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/62c37f0101b445fe83b6c47769deb7dc : SUCCESS in 29m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d08b195b502a4c28a3d93615d44df72a : SUCCESS in 1h 18m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f817b9a93ae341f99e16220baf1198fb : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a37d81e70913771a39147bca9402fe510d93fa5b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-06 09:31:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba80c024007b95386c8a837be1065d90431cd0ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 10:17:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e29f0d27700d4f88b1ffc4ea1e0179f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c05d4995bca4a918810cd03444420a1 : SUCCESS in 43m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcda27d6df8948d2be7ba7268eb31397 : SUCCESS in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0725ee7ec78af29c02e91465f7eab04170cddc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 11:04:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57bac887d17844c4be72ce6d58ee0297\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/5eaaa620beff494fba342f67b5b6fe6e : SUCCESS in 1h 02m 47s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/99134622a6bc4dbea94ba52cc34a5ff0 : SUCCESS in 1h 21m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6669250bbcef487db2a931eecfe9a271 : SUCCESS in 44m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dce3fbbce8124b8f98026ad01b9c21a7 : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a177e05bf7b145c2b5dec7aa8a0fa1b1 : SUCCESS in 30m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9f296c4150d436f92e3ee9a0a432db3 : SUCCESS in 29m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37e9d2bbba5a4ad7a6f0bf51b6f24bc7 : SUCCESS in 24m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d60c06ccff0146858ef1cbd0f9511555 : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17071316ac534fceb1ab3b0c79e2a378 : SUCCESS in 53m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8543a023e0804155b10a097ba0f747e8 : SUCCESS in 1h 00m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c7c237e3e6f947d3bd0f1ab08cbe1441 : SUCCESS in 45m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d995391ec4634984a7c9ad98c91799e1 : SUCCESS in 1h 23m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/99154e2750914412bba66f868e059c35 : SUCCESS in 1h 27m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9b03c0bb594e41d6e2d548d15d7af44339974a07","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-16 17:07:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e2902a3a505e83691c0e94a42ff8a07aecaf4e67","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-17 12:01:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a5c69545eb95280469194c314e6644b3465c3ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 12:02:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"011dd44d7fe3a6fd5c4a7fe1b8331ea91813668d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:39:12.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/869352, https://review.opendev.org/869349 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"0e3043dfcec2ec2d856250737a1c54648916f505","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-17 19:49:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5130221d5f06371eef735524138382577f8003b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 21:35:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/483d56d06330469bb7ad5ad5cbbe766e\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/578a708f4e18446fb6961fb6c4c0d7dc : SUCCESS in 38m 43s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/67eadc797516414a955cc2b88f72ba4d : SUCCESS in 1h 20m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cb3e48dd3bb490daaabbb1d8fd7d478 : SUCCESS in 34m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1514168fc249bbb3db714c3a4ebf89 : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/529f18d888c64102af7b56b18b0a90ec : SUCCESS in 33m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17c3c56c053940248b1b88e77e56189e : SUCCESS in 27m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fa50703cc9a4fffac5ffc3fa3e016b4 : SUCCESS in 27m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2871e9edc9164efcbb858997a43172d5 : SUCCESS in 8m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca7d231abfaf44f790f20162402673c1 : SUCCESS in 49m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3c755c63c94f4523ad2897c4193a22cc : SUCCESS in 48m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6d4faeea704d49eb926024296ee71767 : SUCCESS in 44m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65aedb7d314f4e8f988f1faa86af1a99 : SUCCESS in 1h 44m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/acca8e64a6b94248af5c7319058aa517 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":2},{"id":"050b52a9a0ea6a09bf5fdff412cf964e42ea5849","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 23:03:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f762d95864f84ddb9ff0815dbdedd4be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0ac9d8f1c2f4fe79e8e9869155b3a34 : SUCCESS in 43m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/166a0f6f497841cc8fe263055acf6ab0 : SUCCESS in 1h 08m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c4c3630704cb01c4c0d489cf1b91871b2c6b9fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:50:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bade1e124be516aed304eaac403c632db182cc0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:01:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af5d6cec141545bba75bd15d3a26643f\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/0d9d469eb609462ca0dc7d02601b0f04 : SUCCESS in 1h 03m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c840e197ba44f7adf9ae48f4390562 : SUCCESS in 13m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7f3b3ceda204885acedfd01db9253cd : SUCCESS in 33m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ab6f0c513d74fea98692c6daffed317 : SUCCESS in 31m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4c849f77874664b441b3f57f1b4c02 : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/271e81a55f9e42e78646d0496d320332 : SUCCESS in 1h 00m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6ddb0c4a47e64b5390f3907cea81b284 : SUCCESS in 1h 17m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7ca15814c4da4048ae3e43a8d9b86d3e : SUCCESS in 23m 35s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fd7130cb1ae140c0ba4372802ce48576 : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b66f06e0f729000277624fe4adf9f1ecb226d6b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:01:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2987a088e2c6d46c36403260a3356d74567eb807","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:03:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd9b073924647559eb4b11037d903e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409bc5e87f3f4a4d8641e1d680deeb98 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f1811afa57f4959d6454865a72109f63424b250","revisions":{"848114c329cfa778ab0735ef8301ed98f85aa332":{"kind":"REWORK","_number":1,"created":"2022-12-21 09:11:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/868264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/868264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/868264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/868264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/868264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/868264/1"}}},"commit":{"parents":[{"commit":"de01f9622904a598e1832f54ebb5ffd1f6e389d4","subject":"Merge \"Check if port exists in ``update_port_virtual_type`` method\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de01f9622904a598e1832f54ebb5ffd1f6e389d4"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-11-25 15:39:31.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-12 04:35:34.000000000","tz":60},"subject":"Fix handling the restart of ovn-controllers","message":"Fix handling the restart of ovn-controllers\n\nThe previous `getattr(old, \u0027nb_cfg\u0027, False)` would evaluate to `False`\nif the `old` row either did not contain a `nb_cfg` value or if the value\nwas 0.\n\nAs 0 is the value set on startup of the ovn-controller this causes the\nneutron-api to ignore any event a ovn-controller directly sends after\nstartup. In turn this causes us to miss the information that the agent\nis synchronized, causing the agent to appear as down, until something\nbumps the `nb_cfg` value globally.\n\nCloses-Bug: #1997982\n\nConflicts:\n    neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: Icec8fee93e64b871999f38674e305238e9705fd4\n(cherry picked from commit 4cc611d319d0afe1ee04df6e4419014f1133df09)\n(cherry picked from commit bc96c6378df35860bbd78894abdfcffde6e5ba2d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/848114c329cfa778ab0735ef8301ed98f85aa332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/848114c329cfa778ab0735ef8301ed98f85aa332"}]},"branch":"refs/heads/stable/wallaby"},"0f1811afa57f4959d6454865a72109f63424b250":{"kind":"REWORK","_number":2,"created":"2023-01-05 11:08:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/868264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/868264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/868264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/868264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/868264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/868264/2"}}},"commit":{"parents":[{"commit":"938534ecefa941f5a1b83b5df2b0d2b9da5c04d7","subject":"[ovn]neutron agent show real heartbeat_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/938534ecefa941f5a1b83b5df2b0d2b9da5c04d7"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-11-25 15:39:31.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-04 11:07:12.000000000","tz":60},"subject":"Fix handling the restart of ovn-controllers","message":"Fix handling the restart of ovn-controllers\n\nThe previous `getattr(old, \u0027nb_cfg\u0027, False)` would evaluate to `False`\nif the `old` row either did not contain a `nb_cfg` value or if the value\nwas 0.\n\nAs 0 is the value set on startup of the ovn-controller this causes the\nneutron-api to ignore any event a ovn-controller directly sends after\nstartup. In turn this causes us to miss the information that the agent\nis synchronized, causing the agent to appear as down, until something\nbumps the `nb_cfg` value globally.\n\nCloses-Bug: #1997982\n\nConflicts:\n    neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: Icec8fee93e64b871999f38674e305238e9705fd4\n(cherry picked from commit 4cc611d319d0afe1ee04df6e4419014f1133df09)\n(cherry picked from commit 61c6adf5fed415ae82dbe15e38015f7ad5c9f3e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f1811afa57f4959d6454865a72109f63424b250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f1811afa57f4959d6454865a72109f63424b250"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
