)]}'
{"id":"openstack%2Fneutron~868396","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I8c6d226e626812c3ccf0a2681be68a5b080b3463","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/2000150-stable/zed-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-12-22 19:07:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c6d226e626812c3ccf0a2681be68a5b080b3463","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","status":"MERGED","created":"2022-12-22 15:45:45.000000000","updated":"2022-12-22 19:08:30.000000000","submitted":"2022-12-22 19:07:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868396-bug/2000150-stable/zed-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":868395,"cherry_pick_of_patch_set":1,"meta_rev_id":"56363429966d10b2087aa91a01e2886ec501930f","_number":868396,"virtual_id_number":868396,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-22 19:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-12-22 15:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-12-22 17:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-12-22 17:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-22 15:48:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-12-22 17:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-22 17:09:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-22 17:28:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"daf23cfb3d74f0df6764e51e1e605e23d407161c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-12-22 15:45:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"4118db128175047f31b2ba04dd8fb48ccffbff5a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-22 15:48:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1159353301af92d209ac06f6880770669ac40847","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 17:01:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b46e752279f459ab7058576b0871cbc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d9c52eae7974ef3b2372647ebd13c29 : SUCCESS in 56m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99e7e688493540c48981bc5797e41836 : SUCCESS in 54m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca41f2f88ba1804b40f01340c6238fe3b22c0281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 17:09:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3cb16e4bd9486fad41fe360a462bbb\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/dfd685638ebd451792e5d8ff42687b2c : SUCCESS in 58m 18s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/743e14781a8c4fc88198f7ce028c58d9 : SUCCESS in 1h 09m 57s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/59021d8693f645228bbc9d8a76d8298e : SUCCESS in 1h 09m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/89d2cb026d7e4e0686bfb7a6157c7c92 : SUCCESS in 1h 09m 13s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/6ca6acb5da02417db173f74f28c704a2 : SUCCESS in 58m 32s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/7f9fc1eb252b4c6e8cd9ce99010ff69e : SUCCESS in 32m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/accc25c289b14958a8fe2888425c6b3f : SUCCESS in 45m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdd4906add9423786e9af5d5b37c9f4 : SUCCESS in 13m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be6eccc5bf6a42fda0a17e3dc306a0e5 : SUCCESS in 29m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99c1cd976814425580d562d9f6d3a698 : SUCCESS in 27m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7fd679ae0f94870ab7d03aef282fbb7 : SUCCESS in 29m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597f881143ba454fa8d70b9a96045b48 : SUCCESS in 8m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/66109c480247420bb3ef672a64636e9b : SUCCESS in 55m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4866305c3194421dbf5b31353f2a8e09 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a6c7b04428a64322b171cd8dc6c726c3f92890dd","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-22 17:28:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d022567a1ee7447228600bad8c7c8eca27a743e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 17:29:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dabd154b968bd79d5a5e52ff2b161cd2b5bc19c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 19:07:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1441ece0367741c8815a2ec7963c7f46\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/a5e465d5aba94339ba2b336863f708ca : SUCCESS in 46m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4088f0d623d49efb0ea50d635538afc : SUCCESS in 12m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40814d659ce94b56a403fce22066c9e0 : SUCCESS in 33m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/537bfae1ca924635a206c653d56e5b80 : SUCCESS in 24m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d61a4b06721c476d93ef959730071ec5 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7fba81e70e454339a14d79a5921642b8 : SUCCESS in 42m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c7abc34e2854594b023faf916227583 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"61d66ecb84e720f429c75b75a7a0e9d8d7c83e9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 19:07:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56363429966d10b2087aa91a01e2886ec501930f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 19:08:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e691d47a541444609aa7c32cd06c54cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ddb03a28910490382ceccbaa636779a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bd24646ba3b59c62e130d40348b390f800d15ef","revisions":{"6bd24646ba3b59c62e130d40348b390f800d15ef":{"kind":"REWORK","_number":1,"created":"2022-12-22 15:45:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/96/868396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/868396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/868396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/868396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/868396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/868396/1"}}},"commit":{"parents":[{"commit":"de01f9622904a598e1832f54ebb5ffd1f6e389d4","subject":"Merge \"Check if port exists in ``update_port_virtual_type`` method\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de01f9622904a598e1832f54ebb5ffd1f6e389d4"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-12-21 15:59:53.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-12-22 15:45:45.000000000","tz":0},"subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","message":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured\n\nIn Linuxbridge and OVS PortFixture, when port is created, in the fake\nvm\u0027s namespace it needs to have correct mac address configured.\nIt seems that for some reason it\u0027s not properly configured sometimes and\nthat may cause failure of e.g. DHCP tests.\nSo this patch adds retries for 10 seconds to ensure that MAC address is\nconfigured to the one which should be.\n\nCloses-bug: #2000150\nChange-Id: I8c6d226e626812c3ccf0a2681be68a5b080b3463\n(cherry picked from commit 370d8bcea3ae728c1aacba7b36800ecd759f3f8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bd24646ba3b59c62e130d40348b390f800d15ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bd24646ba3b59c62e130d40348b390f800d15ef"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
