)]}'
{"id":"openstack%2Fneutron~868503","triplet_id":"openstack%2Fneutron~stable%2Fzed~I686a8bf41589a9a0f81f87183877b3008079c186","project":"openstack/neutron","branch":"stable/zed","topic":"zuul_errors","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-12-24 10:19:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I686a8bf41589a9a0f81f87183877b3008079c186","subject":"DNM testing patch zuul errors during \"Run tox without tests\" command","status":"ABANDONED","created":"2022-12-23 16:47:56.000000000","updated":"2022-12-24 10:19:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c38e1e2f308cbd2710ce22f6be2cdd6c1da6262f","_number":868503,"virtual_id_number":868503,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-23 20:06:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-23 16:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-23 20:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04fc04d153360271023ab6d7d6b0890580c75256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-23 16:47:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c695faf883d16d32af6c0facb51cb28a6735dd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 16:56:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da48e25f7224cad8ca41f1281d18a14\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ec3c5da53e64f46910149ccccf388a5 : FAILURE in 4m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e8f0a6fbf1f45168c95e1fad9b25580 : FAILURE in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33b486b83e7e7bb53ff9a8e1886c6b57a34a65f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-23 17:34:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a828559a688a3d88a54c2a350bf1d45990285df1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 17:44:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab12ca4dc1d4a7687a238ce1660c5e3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd3549a4a6e441dda8b698ee65868b9c : FAILURE in 7m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fd66ccd7a4140a1afdd5bd7d32365fa : FAILURE in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf62a80bf28c59a745b16891718fddb7cd0df07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-23 17:45:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5427898a202d6532487aeedf92a422c5510282b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 17:57:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09d6e0b5c97c425cb9d18029563066a1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bd2703b04af4e5cab8197c879cf7f74 : FAILURE in 10m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4ff64f94ee14795a882693bd48135d4 : FAILURE in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5352a086d4101fd3435f32f34bb88f654cf383b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-23 17:57:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0c04deb9cdf91f5eb81bd7c56ab9bd0bca56be5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 18:40:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/774ffe89b9194636bcbe1f985df8ee72\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29c6b128f3124f77a720ab7dec82a32f : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb19a886bef241faa4011ee4f36d87b9 : SUCCESS in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4f199260e890cdd5af666ed3982c637fce87eaca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 20:06:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6f4834cae5440efb701793b2260deae\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/9cc16ad1ff684509bd51968cb78880c5 : SUCCESS in 1h 12m 58s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/ab9e26133e4043e1b6e877db956e67df : SUCCESS in 1h 01m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/6f7b3c65c6e1435b9d70187d6001c1ae : SUCCESS in 56m 13s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/a11c35f05e2e49c8a65e5dc970bf2764 : SUCCESS in 1h 11m 45s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/e7677d780dae4216b518882e6a46a640 : SUCCESS in 32m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba594c6d46bc4559aafa182fe98ce524 : SUCCESS in 25m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a57b3542e54a52a834accd28b276a4 : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0be792368b37438f816d585a7e2e810d : SUCCESS in 23m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81a76f62504b44c0b65804c37223df4e : SUCCESS in 20m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ef3705a04074ed6bfff256a9994f172 : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a85b3e884844b093edc247053376d9 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbe84e74d09f4e05a312fe2d630cf7f8 : SUCCESS in 8m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5aafce933904375a6b81f505c4d1257 : SUCCESS in 26m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/42619caa36a54e26b70e07c6da72f7ae : FAILURE in 2h 03m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b7d8d3392c234dc492286cccff45ae95 : SUCCESS in 1h 08m 16s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9a2564c9ec64101a7926eefd442f505 : SUCCESS in 1h 05m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c9ecdbf619cd465c9d0af5c40cd1359e : SUCCESS in 40m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/beaa14ced5ba43e09181ab961275bbbe : SUCCESS in 1h 24m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2868374804274b12bd81ce169134af87 : SUCCESS in 1h 19m 18s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5989ce06713484fa6e2b79d071a8f55 : SUCCESS in 41m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/246ad12f21c64fb6bd62418404be64ea : SUCCESS in 1h 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c38e1e2f308cbd2710ce22f6be2cdd6c1da6262f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-24 10:19:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"df27dc789491a02cc63c8df9c25ac58b7439d0cb","revisions":{"ed4b6fe6f0ea6ceb983d161ffc8eaf7f15290f7f":{"kind":"REWORK","_number":1,"created":"2022-12-23 16:47:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/868503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/868503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/868503/1"}}},"commit":{"parents":[{"commit":"8c4005763b5dc41e2a8e363be9063b05e8f8b69e","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c4005763b5dc41e2a8e363be9063b05e8f8b69e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 17:46:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 17:46:17.000000000","tz":60},"subject":"DNM testing patch zuul errors during \"Run tox without tests\" command","message":"DNM testing patch zuul errors during \"Run tox without tests\" command\n\nChange-Id: I686a8bf41589a9a0f81f87183877b3008079c186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed4b6fe6f0ea6ceb983d161ffc8eaf7f15290f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed4b6fe6f0ea6ceb983d161ffc8eaf7f15290f7f"}]},"branch":"refs/heads/stable/zed"},"c99ff859e2353212f08bc1900652f8a86723774d":{"kind":"REWORK","_number":2,"created":"2022-12-23 17:34:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/868503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/868503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/868503/2"}}},"commit":{"parents":[{"commit":"8c4005763b5dc41e2a8e363be9063b05e8f8b69e","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c4005763b5dc41e2a8e363be9063b05e8f8b69e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 17:46:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 18:34:22.000000000","tz":60},"subject":"DNM testing patch zuul errors during \"Run tox without tests\" command","message":"DNM testing patch zuul errors during \"Run tox without tests\" command\n\nChange-Id: I686a8bf41589a9a0f81f87183877b3008079c186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c99ff859e2353212f08bc1900652f8a86723774d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c99ff859e2353212f08bc1900652f8a86723774d"}]},"branch":"refs/heads/stable/zed"},"a5dee4f19c6152d2f2bcf3ded73ad8aa04602963":{"kind":"REWORK","_number":3,"created":"2022-12-23 17:45:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/868503/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/868503/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/868503/3"}}},"commit":{"parents":[{"commit":"8c4005763b5dc41e2a8e363be9063b05e8f8b69e","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c4005763b5dc41e2a8e363be9063b05e8f8b69e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 17:46:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 18:44:43.000000000","tz":60},"subject":"DNM testing patch zuul errors during \"Run tox without tests\" command","message":"DNM testing patch zuul errors during \"Run tox without tests\" command\n\nChange-Id: I686a8bf41589a9a0f81f87183877b3008079c186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5dee4f19c6152d2f2bcf3ded73ad8aa04602963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5dee4f19c6152d2f2bcf3ded73ad8aa04602963"}]},"branch":"refs/heads/stable/zed"},"df27dc789491a02cc63c8df9c25ac58b7439d0cb":{"kind":"REWORK","_number":4,"created":"2022-12-23 17:57:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/868503/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/868503/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/868503/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/868503/4"}}},"commit":{"parents":[{"commit":"8c4005763b5dc41e2a8e363be9063b05e8f8b69e","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c4005763b5dc41e2a8e363be9063b05e8f8b69e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 17:46:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 18:56:45.000000000","tz":60},"subject":"DNM testing patch zuul errors during \"Run tox without tests\" command","message":"DNM testing patch zuul errors during \"Run tox without tests\" command\n\nChange-Id: I686a8bf41589a9a0f81f87183877b3008079c186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df27dc789491a02cc63c8df9c25ac58b7439d0cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df27dc789491a02cc63c8df9c25ac58b7439d0cb"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
