)]}'
{"id":"openstack%2Fneutron~868520","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I9c62d429bb819336da05055fecd08e3816986bf8","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1999558","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-01-02 08:16:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I9c62d429bb819336da05055fecd08e3816986bf8","subject":"Limit tox version to \u003c4","status":"MERGED","created":"2022-12-24 10:13:19.000000000","updated":"2023-01-02 13:26:13.000000000","submitted":"2023-01-02 13:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868520-bug/1999558","meta_rev_id":"608acef1e0316974cf48f7724e6ad53475953c71","_number":868520,"virtual_id_number":868520,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-02 13:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-01-02 08:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-12-27 06:45:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-01-02 08:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-01-02 08:16:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-01-02 08:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-24 12:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-27 06:45:39.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-01-02 08:04:34.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"d0b81bd65623558ee75ccbcca7a325f0c2ab4676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-24 10:13:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2584770a3c5e620e56eb16a6f2c9369a7c1ce1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-24 10:19:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cbbf837d34c964d3d1b56075c577e941fe115216","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-24 12:35:00.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"dc1237f91e25e14c6cc24f474f4bf2c1da3a0184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 12:38:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c652a4eda4747bbb757da13b009c148\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aa0e163f5fd4bf385dbcfd47289bfdb : SUCCESS in 1h 06m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db09ed675b94f6a9cd1931cdd67446b : SUCCESS in 15m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142df5263ac74192b0f48201446fe443 : SUCCESS in 29m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d385043b6724a378c68896b2eb17ac4 : SUCCESS in 33m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27e209dc10154605bd95bd8bd887e219 : SUCCESS in 35m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc6210aa495482bbdc40b4660b60a01 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eff689a17a3f42479b68065ca001a744 : SUCCESS in 12m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8c87063ec09d4ff997e2edb1373f9a13 : SUCCESS in 1h 01m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a85422c14bea44008d45d91dadc191b6 : SUCCESS in 1h 32m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/18929633c46e43b79448e800c0a1f48a : SUCCESS in 1h 32m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/abd96d1ea2b9400e802937ccb0c1c926 : SUCCESS in 1h 53m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/954d5b6f9ab042c5915946115842bc5f : SUCCESS in 1h 34m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/55f706f8c5c24160a6e53f64e3388ae7 : SUCCESS in 2h 00m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b570510f7a1646df8642a45a17dcbcb6 : SUCCESS in 1h 28m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2fdcdcd240e4a03b0191ec8f9c41d9a : SUCCESS in 1h 26m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/280cad15f2e64d7ebf59edeeb5ff2a26 : SUCCESS in 1h 31m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e58d856cceff471ab688d5447956a666 : FAILURE in 1h 38m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5448f62a6d0a4a96b90aa2f418d12a84 : SUCCESS in 1h 44m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/96416603d12145f0a41dd89524e29ff1 : SUCCESS in 1h 26m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1edf723f9fe6478ea0d064f79917cc6c : SUCCESS in 1h 03m 41s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9b5066cc98347d1bbd70379738ffe76 : SUCCESS in 2h 00m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d88cfe07ff2c411691f838d1a4842c28 : SUCCESS in 1h 21m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/57041efc143742719737ffad6ae10d82 : SUCCESS in 1h 06m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8e6d5a012f51492a87eb240b0319f877 : SUCCESS in 41m 03s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/749e72e0dcc943e6b1b8cfdd4ffdffae : FAILURE in 27m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94e7141214b347689bb2500c102f2f0c : SUCCESS in 55m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2c4673db671c4743b6cca34528c47801 : SUCCESS in 1h 28m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e3ec2b934d9a4c90bef0d544f321c1a4 : FAILURE in 2h 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bd1228c76194cd25965d084fd66931502d54aaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-24 14:11:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6eaf295312bfd0e1120e30c2c91f2fc15e81e69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 15:47:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23ed5296281b48cb806886dfbbdf62a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08f77e5510064730b67bc6b9a7c1c15b : SUCCESS in 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e48746d17f45f389ff110e53acb088 : SUCCESS in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2635e6fd904d42bc875b7200ccc47029 : SUCCESS in 25m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e19288adf2d4020bf3df4e74a6264a6 : SUCCESS in 24m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be34464012ca423a8a0d1803c4bf9497 : SUCCESS in 29m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c3f517f3fc64e0594688975278d2695 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7c04765d0e649fa824c9c6d6981d8e6 : SUCCESS in 8m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5c45629c1b824aba8f44567e7e3a028a : SUCCESS in 58m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ad46d236e56940dc81da9308f7bb0bd4 : SUCCESS in 1h 14m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7267d6aecbf244fd9321db2e1d4cec89 : FAILURE in 1h 20m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c27a7c5161a641a68d8b0a3c01af91bc : SUCCESS in 1h 12m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4681356e0bc1495684f232a8f0c59ace : SUCCESS in 52m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01ef1e2d60564a74bbc7aecb5739cc30 : SUCCESS in 54m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7d54359968ce4fa9bc080718915be203 : SUCCESS in 1h 24m 16s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1d30857d1a74a1d94a3697bcbf4e386 : SUCCESS in 55m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/41031b6b08d84ce9987489470133ede6 : SUCCESS in 1h 19m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e8513253ba294c73b9d5ac14c415f1db : SUCCESS in 1h 16m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5ba320e4df8f4b1f880a9b5778a46c69 : SUCCESS in 1h 04m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6185763da1f43f2adbb7798fa5227f1 : SUCCESS in 1h 02m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2c9bd72387584a008b09fa36a10f472d : SUCCESS in 59m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a872e09bc1244da784b749a6fde14161 : SUCCESS in 1h 20m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6ce748c2f2f4714b6728280da4f44d1 : SUCCESS in 46m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b06f6953723b4b08a7df7a72736438fd : SUCCESS in 1h 16m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/de14b252b96e4f2b96c7c761a7a0d309 : SUCCESS in 33m 02s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/68b1338c83ab48c8ad60a574a776daf4 : FAILURE in 24m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3777dda173fa496fa8235a3b34545426 : SUCCESS in 48m 43s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/58925ddeb36245b3a934768b4912ab5b : SUCCESS in 1h 13m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8e508fdf3cf9491987b090a19e873761 : FAILURE in 1h 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c7698682eac9cfcbe9990fdb1a77e5d45109817","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-27 06:45:39.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90185e8d3e475dbe7ae65828382e6568a3b346e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 08:23:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c50a9f394b84c7bb9214d75005e2086\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df44c6a8d8574039b225685751d4f0f8 : SUCCESS in 42m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4257db4bfd499a8d383b9faddd680b : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/506f7f73017b49c9b18faf9710fc38c6 : SUCCESS in 32m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1918e3008b9940e699ebb0bcfe581638 : SUCCESS in 29m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ae1f8daff17419d8a1377b15ae2568f : SUCCESS in 30m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164c565a94ec4289937d01a08eb83a50 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f89b4782cc547fc9be286b40328e6db : SUCCESS in 11m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/be218b2f2ddb4a27b190cc566e1e4349 : FAILURE in 42m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7e29978afab44c4da184a0240c76dd7d : SUCCESS in 1h 07m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/051d7cfa97bf4a23b4fee3cdb0eeb75e : SUCCESS in 1h 15m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c7d66cd865ec4ab78aee20e99650954b : SUCCESS in 1h 14m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c53b8014db6047dc90790e2fc34ddec6 : SUCCESS in 1h 13m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0e90c91a2734406491b7e54e7d65f83e : SUCCESS in 59m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5f95e1113f2a491790e6f867aa4e1a9c : SUCCESS in 1h 15m 45s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/24e3bf6bf162487c9aececa4fd15d6e1 : SUCCESS in 51m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cd4b2f2ff9e449ac89d4452bd3816524 : SUCCESS in 1h 14m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6d80dda966b247829095879b7f26daec : SUCCESS in 1h 16m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7419406991464b1eb9d591cdb8a1517f : SUCCESS in 1h 19m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f85cc900b43474da8837609f7042315 : SUCCESS in 58m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e4faf8f5492c4914affb00203d9d6437 : SUCCESS in 31m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd47539b4e4745e8ac2516d12f29aa62 : SUCCESS in 1h 33m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/336f069e47a4457595beaa6affe3ef38 : SUCCESS in 42m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ad48e061e84f4ef1ba39ccc9a7ad9b6e : SUCCESS in 51m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a14b1c10735b4e8eb0b8c66ddb52870d : SUCCESS in 26m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/aba86a335e1f4ec1b763c373534066fd : FAILURE in 22m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54d4cdfcee37411ea98097993bbfc43b : SUCCESS in 57m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4bdf32a68bd147c69c0ece234b6762d1 : SUCCESS in 1h 25m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/61e38993506a4a3b83a3efa5a0380a90 : FAILURE in 1h 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d267046310a76968b0091c9c0675e1b7b1b2a680","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-02 08:04:34.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c23db7b5eab7a529daeffce3e237e810dc3062af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-02 08:16:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f58fd383a8bead7fbaea2b44b1b882732be93564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 10:07:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f85dde7201041c884bb9d9a59256715\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd822288dbe7485895d02bcfe688fcfb : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/511276f8ce83482b8ed2cad43b0f28c1 : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5aeef25ab104405aa310467cb259938 : SUCCESS in 36m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1c5c69cf87149358c203bff192ca102 : SUCCESS in 38m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9f4c09473484b24b3eda0cda8316a88 : SUCCESS in 32m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2f88db9e854e07bf052543c3314bc1 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2c4ba66abed4cd9972e89fcd84fc06d : SUCCESS in 11m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aa777b21d3a244529734f45e94730bac : SUCCESS in 56m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/878eef39117648df9a1c682ebc47a8e1 : SUCCESS in 1h 36m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a5c3260ea6f4496ac2ec7bda2c6a17f : SUCCESS in 1h 37m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1c17141e21124b93bfd46f6311aae08f : SUCCESS in 1h 52m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c7111a74de44850acac0a2478fe1f21 : SUCCESS in 1h 18m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9367835ab7644a9c8c6ec0f32bca256c : SUCCESS in 1h 20m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a6119c94b12f492ea871f2145edc8341 : SUCCESS in 1h 10m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e43b15c9995d4beba8e6c071a47c6cab : SUCCESS in 1h 14m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/269b73863ceb448e8a9a7261871719d6 : SUCCESS in 1h 10m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0be4d2b46e754e3c94a2a126efe13fbb : SUCCESS in 1h 38m 00s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/32cd9b4ed14841b3b630aefdbad9e364 : FAILURE in 1h 11m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b48b4648a2a34cc694dce174047b1784 : SUCCESS in 1h 10m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/99afa2c8b2a3412fb7eaa215b1ebe170 : SUCCESS in 54m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/43af82892d974fb19af823cd76a17e8e : SUCCESS in 1h 45m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0cb1fa62d50411e85ac1f08f2e9ebe4 : SUCCESS in 1h 05m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/87a343f82c614edbb6b41829f6f5a627 : SUCCESS in 1h 21m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/78e679c26c294bdc954aed1d09a3f97a : SUCCESS in 46m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/caa4b67b868944e59f38c7c353222e4e : FAILURE in 38m 21s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2dd28753ab154f0491d4b1e3d943615a : SUCCESS in 58m 05s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d93da6c0b8f4f6d927e998bb6b69c00 : SUCCESS in 1h 45m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/02800eccfa354cfb8197deda88dd56c8 : FAILURE in 1h 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"beea53c90e9608f69551356ba4604217dde5dd7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 10:07:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec91536d8491a4ddd41a4de311fb62e8da83f875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:25:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b7ded23b0bc4e4387c8fe90f5ce2f4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36303d5634574052858a24ea52260dfd : SUCCESS in 10m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8270229df9314ddcb9b78ef2b576ce21 : SUCCESS in 24m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45d6096c7ecb4094937fbac7296e84e2 : SUCCESS in 28m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e30f53774a46479813738136c69756 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/775cb751c43341c29b21930b1905c205 : SUCCESS in 9m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/370e9911ef9f406baf9380b3587bfe22 : SUCCESS in 47m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b699b1a65642473d9cafb9019593c229 : SUCCESS in 1h 11m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7cf93828455c484e9986b358b7d8f797 : SUCCESS in 54m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e58f8585d51f41868594b76ee0b59cc4 : SUCCESS in 58m 00s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/613c5bdd50d044bcae57d399931d1e7f : SUCCESS in 1h 02m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3040336d056c448ca6643eb4fb8d323b : SUCCESS in 55m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4a3fde5a337444319f187ae6cbe38616 : SUCCESS in 58m 30s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d7066d25f1624329a3b9f5515c7744bd : SUCCESS in 1h 19m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa625f8af4ef447a981209afd0f1e157 : SUCCESS in 51m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/157f796f07f44265a43a4e9b307b518c : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":3},{"id":"dccd06158aff8956ce0f399a961c11aecd4a5c49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:25:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"608acef1e0316974cf48f7724e6ad53475953c71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:26:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73108ac1d1024ff6b288209352ff8667\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a25ca0ea07479f9d4f2cb5b7acd74b : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fc055a8d4de465eb09ea8bff2030b49 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e99d5b66ef00a6e16038cea5792d0b2ab78aa8f5","revisions":{"799fbdc0925c6fc2177ed83f957fae5bfbc1f66a":{"kind":"REWORK","_number":1,"created":"2022-12-24 10:13:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/868520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/868520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/868520/1"}}},"commit":{"parents":[{"commit":"bcaba9d791d17eccd8420bc5ee2dd330dfa299e3","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcaba9d791d17eccd8420bc5ee2dd330dfa299e3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-10 15:34:48.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 19:27:05.000000000","tz":60},"subject":"Limit tox version to \u003c4","message":"Limit tox version to \u003c4\n\nThis patch limitis the tox version to \u003c4 in stable/ussuri.\n\nRelated-Bug: #1999558\nChange-Id: I9c62d429bb819336da05055fecd08e3816986bf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/799fbdc0925c6fc2177ed83f957fae5bfbc1f66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/799fbdc0925c6fc2177ed83f957fae5bfbc1f66a"}]},"branch":"refs/heads/stable/ussuri"},"eb3ae89308cc64dbf4224939072d83526271f4db":{"kind":"REWORK","_number":2,"created":"2022-12-24 10:19:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/868520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/868520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/868520/2"}}},"commit":{"parents":[{"commit":"bcaba9d791d17eccd8420bc5ee2dd330dfa299e3","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcaba9d791d17eccd8420bc5ee2dd330dfa299e3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-10 15:34:48.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-23 19:33:19.000000000","tz":60},"subject":"Limit tox version to \u003c4","message":"Limit tox version to \u003c4\n\nThis patch limitis the tox version to \u003c4 in stable/ussuri.\n\nRelated-Bug: #1999558\nChange-Id: I9c62d429bb819336da05055fecd08e3816986bf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb3ae89308cc64dbf4224939072d83526271f4db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb3ae89308cc64dbf4224939072d83526271f4db"}]},"branch":"refs/heads/stable/ussuri"},"e99d5b66ef00a6e16038cea5792d0b2ab78aa8f5":{"kind":"REWORK","_number":3,"created":"2022-12-24 14:11:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/868520/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/868520/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/868520/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/868520/3"}}},"commit":{"parents":[{"commit":"bcaba9d791d17eccd8420bc5ee2dd330dfa299e3","subject":"[Fullstack] Wait 10 seconds to ensure that MAC address is configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcaba9d791d17eccd8420bc5ee2dd330dfa299e3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-10 15:34:48.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-12-24 14:11:22.000000000","tz":0},"subject":"Limit tox version to \u003c4","message":"Limit tox version to \u003c4\n\nThis patch limitis the tox version to \u003c4 in stable/ussuri.\n\nRelated-Bug: #1999558\nChange-Id: I9c62d429bb819336da05055fecd08e3816986bf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e99d5b66ef00a6e16038cea5792d0b2ab78aa8f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e99d5b66ef00a6e16038cea5792d0b2ab78aa8f5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[]}
