)]}'
{"id":"openstack%2Fneutron~868784","triplet_id":"openstack%2Fneutron~stable%2Fzed~I74883041c642b9498299ebf2b5bf885776e241e0","project":"openstack/neutron","branch":"stable/zed","topic":"bug/1998108-stable/zed","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-01-16 12:05:51.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-01-04 20:43:09.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-01-16 10:19:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"I74883041c642b9498299ebf2b5bf885776e241e0","subject":"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init","status":"MERGED","created":"2023-01-02 12:13:09.000000000","updated":"2023-01-16 12:07:16.000000000","submitted":"2023-01-16 12:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868784-bug/1998108-stable/zed","cherry_pick_of_change":865857,"cherry_pick_of_patch_set":5,"meta_rev_id":"00f9acbc5e0f4a8d66533e6f1f91d1f7a9cdb19c","_number":868784,"virtual_id_number":868784,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 12:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-01-04 20:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-01-16 10:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-01-16 10:19:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-02 13:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-02 13:35:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 16:21:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-01-02 16:21:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"62a0c8f87e7cad72c7be722c1e0d4bd4a285c3ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-02 12:13:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d4879865ad0de7c9a1efe67d39e0f7f2d1c63c07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:14:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1946c07c1f5b4001970979bacc6a4281\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3c39c0d6f1e4d518e6be95f21fa9186 : SUCCESS in 52m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72ee063af97b477a8a0df4da5a7b443e : SUCCESS in 57m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a568bce9d0f7ed45d806aea776b049859a433aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:35:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84952f0e5c734e61aa9ee7077c6e3fd5\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/bcdd4bb8ffaf4f20ae20df6c07c5c7a2 : SUCCESS in 1h 20m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a156d5e42e24f4daa118188905e81de : SUCCESS in 35m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba4f73e212645268321044fb1365bba : SUCCESS in 13m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8713a830ae9b47b68f80187899de6fa1 : SUCCESS in 26m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daba5ef0a2a84e85935b919a18d99c3e : SUCCESS in 28m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ad7282118044940baecec876422d0c0 : SUCCESS in 34m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb2859cd8bf4f8996cef247108ee9bc : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f77173cc8989407d8ee12ce6960c7877 : SUCCESS in 47m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46b70c4679ef43848a03bc2c29a292f1 : SUCCESS in 58m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/03256638574640beb09c5505b7b86213 : SUCCESS in 1h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cc0014a6d5aec6f9e53d8f60eec44d77cc29153","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-04 20:43:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d9499d46a8c8697bb06ad1cb9cdbc41b88b4f6d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-16 10:19:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f6656c873c85aa65255ffae3052e19dd601dcc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:20:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97ee5a39289f51a19540d5d51b82f84c410f9169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:05:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f3d04097534028877a4416834a81ea\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e439ea7319c14ed6a2ee197da757b201 : SUCCESS in 1h 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30aa5f6124f24c61a1bd01a5d3aa6a9c : SUCCESS in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe5df0eb95f4e51a21aca7bb3eabc97 : SUCCESS in 24m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf8b1236374429290dc4079207fb116 : SUCCESS in 28m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dceead27a335461aa6cecd598e7369fa : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4542700161a4da6a85607bff232213e : SUCCESS in 45m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c227b10f298c4c519428b0d8ba712c59 : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c5c7e49c5ed9fef4c768b5c2d476f4534844d43c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:05:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"00f9acbc5e0f4a8d66533e6f1f91d1f7a9cdb19c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:07:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94fe65945cea40b8ad8a8ffe201dc807\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4b8395d3b74b8e9d6adf00b32c50b8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74b6c1dd0bd9034578d7be81940b36b553ffe627","revisions":{"74b6c1dd0bd9034578d7be81940b36b553ffe627":{"kind":"REWORK","_number":1,"created":"2023-01-02 12:13:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/868784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/868784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/868784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/868784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/868784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/868784/1"}}},"commit":{"parents":[{"commit":"44ffbc97b72e60f46a8f471d7f5c56b424618213","subject":"Limit tox version to \u003c4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44ffbc97b72e60f46a8f471d7f5c56b424618213"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-17 20:14:40.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-01-02 12:13:09.000000000","tz":0},"subject":"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init","message":"Move registration \"ChassisBandwidthConfigEvent\" to OvnSbIdl init\n\nThe \"ChassisBandwidthConfigEvent\" event registration is now done in the\n\"OvnSbIdl\" class initialization. That ensures this event is registered\nin a worker thread.\n\nIf the event is called before the \"OVNMechanismDriver\"\u0027s \"OVNClient\"\nhas been instantiated, the event is skipped. When the OVN placement\nextension is loaded, all chassis configurations are read and loaded,\nthus any previous event is not relevant.\n\nThis patch also adds a \"match_fn\" check to\n\"ChassisBandwidthConfigEvent\". If during an update event, the bandwidth\noptions are not modified, this class does not update the resource\nprovider inventories.\n\nCloses-Bug: #1998108\nRelated-Bug: #1578989\nChange-Id: I74883041c642b9498299ebf2b5bf885776e241e0\n(cherry picked from commit 49eab7d308b598e4c4966110d6aa61cd091cc3f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74b6c1dd0bd9034578d7be81940b36b553ffe627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74b6c1dd0bd9034578d7be81940b36b553ffe627"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
